)]}'
{"id":"openstack%2Fneutron~950673","triplet_id":"openstack%2Fneutron~master~Ic12b6ed3cd84dca16f5081dc646717d0a5be318b","project":"openstack/neutron","branch":"master","topic":"bug/2111475","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-27 05:49:06.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"21113":{"account":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"last_update":"2025-05-27 22:58:30.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-05-27 20:53:16.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ic12b6ed3cd84dca16f5081dc646717d0a5be318b","subject":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts","status":"MERGED","created":"2025-05-22 16:01:54.000000000","updated":"2025-05-28 14:22:10.000000000","submitted":"2025-05-27 22:58:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950673-bug/2111475","meta_rev_id":"e08f3ea53bd2267558cba3231fb4aa8b8ff953b2","_number":950673,"virtual_id_number":950673,"owner":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 22:58:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-05-27 05:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-05-28 14:22:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-05-27 20:53:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-05-28 14:22:10.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-05-27 20:53:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-22 16:03:20.000000000","updated_by":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-05-22 16:18:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-05-22 16:23:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-22 17:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-22 18:54:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-28 14:22:10.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"889b061932a01988799748c27b6c146ff5a7525b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"date":"2025-05-22 16:01:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c9346c16d873fa5cede8f8b7ce2f29e7426140c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"date":"2025-05-22 16:15:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2a727934f58c25d417207db6d85278200364587d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-22 16:18:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"edebaafc3a80b9537a47f478320c9b5b374fdaa3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-22 16:23:53.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/73/950673/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"fac28154635cd229a8215da400282af2e5efe447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"date":"2025-05-22 16:29:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"873011c6326626bc1ba02ea737d687f46dbfee81","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-22 16:45:27.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/73/950673/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"e6e3a9379e1d34e6a9dc1e89a63fa46fd4384000","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-22 17:07:39.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/73/950673/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7f0e9bbd5b3e54af53d3ab2a94de7b5237a71ca6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:41:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a26a52d5d201451aa27ee4f6cd700380\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dcd87d65fc83459f8b5d47607622edcb : SUCCESS in 45m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cac82d5a51f54fb4a0e7480f03232e10 : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b2a5ddb03fc46429597e228cc476b3e : SUCCESS in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1202782ef029ba947d183b4b2a796a9b47d04979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 18:54:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b046e704b224c77a387207fd7b83f0b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd372d4785374eccad66f366812915fc : SUCCESS in 2h 22m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2da29229ff5466ba635b87e6b8da6c5 : SUCCESS in 42m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e671b60b360a40e7a7be933b7117bf5c : SUCCESS in 10m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/655b87463c67473f828ff9bb0d4c7130 : SUCCESS in 21m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5b29214ce7a4864ae42f6ffc48c260a : SUCCESS in 16m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b998421bf9e2426695d5f37eeaea6bb0 : SUCCESS in 23m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7584b4e75d4c97a102ccafec41c2bf : SUCCESS in 12m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d4826485d5cf4ba2b06e89b289db3a22 : FAILURE in 58m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4e74b6f3f5742269af2a8c798d90766 : SUCCESS in 1h 07m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f3331d1f3fda48bf8cdeecd7e924805c : SUCCESS in 51m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3bc68f83df934a048b8156476cbbc52c : SUCCESS in 1h 16m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/772978b9eb9743bc8e42c898138d0f3a : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":3},{"id":"1229943f1262a6b693ebabd069c6618db169871c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"date":"2025-05-23 01:11:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4b1200984c662adeda072e8d1f9999ee3f76ffcb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-23 01:34:02.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/73/950673/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"6c81cdb2871d6e86b2d3b4f8ae3632460d08ae0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 01:43:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9096316455914d8993e7523710abfe7a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0922c09e89d34aa8afe9c25dd3773b73 : SUCCESS in 29m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d6d5b97fb044c768b6c74a204f6d2e6 : SUCCESS in 25m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/73aca0824fa44aa8af3c5becc01c4afd : SUCCESS in 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"68892ff8b2d29255c8851ebc645849ebe91bb994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 03:33:28.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6704520aa7740c1940d0e71ff095331\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5f1c019ba2dc44b6b5f38a2139f169a3 : FAILURE in 2h 11m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbf2eb669038450397bcb783644874da : SUCCESS in 43m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e7805c3fc84aa6991c5f3a77f555ba : FAILURE in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb8330d3568c4656b89e6d4c7282e53c : SUCCESS in 15m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e49af87c99334de5b0694db8deb91552 : SUCCESS in 14m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8122ed27e57b4b6785f32a5b6b18585f : SUCCESS in 27m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309652d99dda473faf7d215cf8a9afb9 : SUCCESS in 14m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5b66ca8b30d04bacbc125bb04a844911 : FAILURE in 1h 05m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/46746a6a44c14f25bc408788e0508578 : SUCCESS in 50m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5d85be6b8b8a4212a2f6b2e72068b949 : SUCCESS in 46m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c27dc20ee062410d81f8744884149a90 : SUCCESS in 1h 10m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4ea0342b79f74cfa9c0eb0b8a13e0fa0 : SUCCESS in 1h 18m 52s","accounts_in_message":[],"_revision_number":4},{"id":"992ec31d16c5b7b5c8c4aaff02aaedef2f99e0dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"date":"2025-05-23 04:07:55.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7e7a411a6dc2264d1c32f9f89fd5afb28f08edc6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-23 04:29:56.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/73/950673/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"921a37bf1f2716e93b77ad66d5f9a56feb939d36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 04:31:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ce82a24706942b68d300a16b1bfe9e0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9cc3d4696eeb4b9ea8a63bb828ac3e84 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e98803b55fc4619aa56520d65edd271 : SUCCESS in 20m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2125f4a70b1244f0a2e33233a65c244c : SUCCESS in 20m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7790f425eaad436df0ea3b4719af22b9041c2ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 05:45:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c686a4f32047079478fd78b6e5b821\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4bf44550663a407cb2f4c12c6a176688 : SUCCESS in 1h 36m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6125d285ca81468dbe2ba82a6b798c74 : SUCCESS in 47m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4b15e297904834b0459bcc94430418 : SUCCESS in 10m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7528a32ccf2f4174a145e50994212c08 : SUCCESS in 19m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ff9e3b9c01742c68debd3656f8d8b36 : SUCCESS in 13m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/def3f00b57d94ff2be60cf44ff9682b8 : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da925fd5f8904fc68229ba53cc51686f : SUCCESS in 9m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f3e3591ceca74fe4af54715fd9d2a5d1 : SUCCESS in 1h 24m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/81271583e5d240918c55ee7d620fa18e : SUCCESS in 1h 04m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b1a19f758b89481cb5d1064ba29df0e5 : SUCCESS in 46m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3add3d51f0014442aa5d6ecd8264d373 : SUCCESS in 1h 17m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c8a5d0532dc84aa78b6b4eec98493fd3 : SUCCESS in 1h 22m 27s","accounts_in_message":[],"_revision_number":5},{"id":"ab8f037d85f32bed3172e0d795cb4278ca21e7ad","author":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"date":"2025-05-23 10:22:02.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f964e0630670329682f70d3be059b01935b79b71","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-26 12:43:32.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6d8be9eeba4a3afdee4ad08a80d52f5ca0f6eeb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"date":"2025-05-26 13:40:34.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"7ede11b03d9618b4bc95352fa1c35221c727aa01","author":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"date":"2025-05-26 13:57:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"afac3cae5764f3d27cb6e335a7ff75a269f7a240","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 14:07:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14cc88dfa3b64b098ee04947f52968dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b10024c2c0ae467287ccfee7d26452c3 : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/913974650a4a4dd1b6c31c1ca7f39b89 : SUCCESS in 25m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5cf5e0cf2a4c4fd7ada1dfe96b34f7ab : SUCCESS in 22m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e0e6550ebbf9906f41bad6826a4b0291c8666fa0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-05-26 14:14:57.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2025/Arista_Tempest_Testrefs/changes/73/950673/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"e729257e415027c95b1ed797353cc3c8c64013f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 15:48:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97130b258b654b709447c4c4249f8d89\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9d6e21d1dc4f44cd9bd87356ff0b824f : SUCCESS in 2h 05m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6af1375834c4b558ed14a517946fa9d : SUCCESS in 36m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a5daeca33549849e052b5d7c6fbf9d : SUCCESS in 11m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/471f50b116dc46fca7f1398c923f097f : SUCCESS in 18m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9d9b3dc9666470ca8a0e1094e44782b : SUCCESS in 20m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cad8ff9a67f4f3d936efd65d9fc0b09 : SUCCESS in 26m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65051a85552d4cae923bb5c774ea7639 : SUCCESS in 11m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ad9d92ac912b4548be23c20aa3357287 : SUCCESS in 1h 22m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b191899b107d48ac8cddc07e05875dd0 : SUCCESS in 1h 13m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6fce2aad35a6441d9a9721bb2cad2fb4 : SUCCESS in 1h 04m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/84ab43c99d9d4106873adfc9a4c1a14d : SUCCESS in 1h 11m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1a91b83649b1455180994f1ec6eac099 : SUCCESS in 1h 35m 07s","accounts_in_message":[],"_revision_number":6},{"id":"c981289c6e670a59bb04c24facf0c8cf2e255000","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-27 05:49:06.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5bf51ba83d76eba5c6764cf314406d669f914e41","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-27 20:53:16.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"7b387d633f20f3b4df61f414b2cb78799b42e9ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 20:53:42.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3f5c0d81c1ea3d5974f661707f347e2ebb8b3861","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 22:58:30.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfa4057ea7e1438986ec89066fb0d299\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a33a9ebba55942aab3a6b814c751cd16 : SUCCESS in 2h 03m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd86cafc10645d3bc2e3f8440ddf28c : SUCCESS in 10m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c92515614f1b4be49af2cdcf7c808489 : SUCCESS in 16m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d37e23ceb174b6a828d93f5864b7897 : SUCCESS in 24m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cad64497d5949a8b41d1eb318d5ccb9 : SUCCESS in 9m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/feb93d748710404ebac3921072033ec5 : SUCCESS in 1h 15m 36s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bee635dfdb5948bdbab9bc7256490100 : SUCCESS in 55m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d16c6318182d451988c57d643b43cef4 : SUCCESS in 1h 02m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d687eecb70984ec7a21981709c943ace : SUCCESS in 1h 27m 49s","accounts_in_message":[],"_revision_number":6},{"id":"fc2477c1f02dcc512469abb1738c0f1397e2d55c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 22:58:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6edc686827c86cf602808e27460f62f831bb35d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 22:59:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e283fa47da834d4f8b6c9dfb312bbdf5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f08eec93c40441097729d08e96d63e1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":6},{"id":"e08f3ea53bd2267558cba3231fb4aa8b8ff953b2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-28 14:22:10.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c6f5c05778076c653e47c6cb7e726bff7016b860","revisions":{"1b98809dba08c0d54fee4474c60fdd085e73fb29":{"kind":"REWORK","_number":1,"created":"2025-05-22 16:01:54.000000000","uploader":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"ref":"refs/changes/73/950673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/950673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/950673/1"}}},"commit":{"parents":[{"commit":"5fbb3c1a3ffd895acb4869e54643c63981e38f34","subject":"Merge \"Fix some pylint warnings in the test tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fbb3c1a3ffd895acb4869e54643c63981e38f34"}]}],"author":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-22 15:56:51.000000000","tz":480},"committer":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-22 15:56:51.000000000","tz":480},"subject":"If the metadata agent is not present in the local AgentCache and the report time (agent_down_time) is too long, the Neutron API will delay the Chassis_Private update. As a result, the agent will not update the corresponding record properly. Consequently, the newly added metadata agent will not appear in the output of the openstack network agent list command.","message":"If the metadata agent is not present in the local AgentCache and the report\ntime (agent_down_time) is too long, the Neutron API will delay the\nChassis_Private update. As a result, the agent will not update the\ncorresponding record properly. Consequently, the newly added metadata agent\nwill not appear in the output of the openstack network agent list command.\n\nThis patch aims to update OVN_AGENT_METADATA_SB_CFG_KEY to nb_cfg as soon as\nthe metadata agent starts.\n\nChange-Id: Ic12b6ed3cd84dca16f5081dc646717d0a5be318b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b98809dba08c0d54fee4474c60fdd085e73fb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b98809dba08c0d54fee4474c60fdd085e73fb29"}]},"branch":"refs/heads/master"},"0e67207da2ced72830a6db13ccd1a2d7fd6ff91a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-22 16:15:36.000000000","uploader":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"ref":"refs/changes/73/950673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/950673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/950673/2"}}},"commit":{"parents":[{"commit":"5fbb3c1a3ffd895acb4869e54643c63981e38f34","subject":"Merge \"Fix some pylint warnings in the test tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fbb3c1a3ffd895acb4869e54643c63981e38f34"}]}],"author":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-22 15:56:51.000000000","tz":480},"committer":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-22 16:14:39.000000000","tz":480},"subject":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts","message":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts\n\nIf the metadata agent is not present in the local AgentCache and the report\ntime (agent_down_time) is too long, the Neutron API will delay the\nChassis_Private update. As a result, the agent will not update the\ncorresponding record properly. Consequently, the newly added metadata agent\nwill not appear in the output of the openstack network agent list command.\n\nThis patch aims to update OVN_AGENT_METADATA_SB_CFG_KEY to nb_cfg as soon as\nthe metadata agent starts.\n\nChange-Id: Ic12b6ed3cd84dca16f5081dc646717d0a5be318b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e67207da2ced72830a6db13ccd1a2d7fd6ff91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e67207da2ced72830a6db13ccd1a2d7fd6ff91a"}]},"branch":"refs/heads/master"},"57a2f03697f1aa658d0ec8f028b91be2095b399d":{"kind":"REWORK","_number":3,"created":"2025-05-22 16:29:33.000000000","uploader":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"ref":"refs/changes/73/950673/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/950673/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/950673/3"}}},"commit":{"parents":[{"commit":"5fbb3c1a3ffd895acb4869e54643c63981e38f34","subject":"Merge \"Fix some pylint warnings in the test tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fbb3c1a3ffd895acb4869e54643c63981e38f34"}]}],"author":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-22 15:56:51.000000000","tz":480},"committer":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-22 16:28:57.000000000","tz":480},"subject":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts","message":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts\n\nIf the metadata agent is not present in the local AgentCache and the\nreport time (agent_down_time) is too long, the Neutron API will delay\nthe Chassis_Private update. As a result, the agent will not update the\ncorresponding record properly. Consequently, the newly added metadata\nagent will not appear in the output of the openstack network agent\nlist command.\n\nThis patch aims to update OVN_AGENT_METADATA_SB_CFG_KEY to nb_cfg as\nsoon as the metadata agent starts.\n\nChange-Id: Ic12b6ed3cd84dca16f5081dc646717d0a5be318b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57a2f03697f1aa658d0ec8f028b91be2095b399d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57a2f03697f1aa658d0ec8f028b91be2095b399d"}]},"branch":"refs/heads/master"},"caf890f485a5248f8ce49f4e10d1e4c4feea9e93":{"kind":"REWORK","_number":4,"created":"2025-05-23 01:11:37.000000000","uploader":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"ref":"refs/changes/73/950673/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/950673/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/950673/4"}}},"commit":{"parents":[{"commit":"5fbb3c1a3ffd895acb4869e54643c63981e38f34","subject":"Merge \"Fix some pylint warnings in the test tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fbb3c1a3ffd895acb4869e54643c63981e38f34"}]}],"author":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-22 15:56:51.000000000","tz":480},"committer":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-23 01:11:17.000000000","tz":480},"subject":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts","message":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts\n\nIf the metadata agent is not present in the local AgentCache and the\nreport time (agent_down_time) is too long, the Neutron API will delay\nthe Chassis_Private update. As a result, the agent will not update the\ncorresponding record properly. Consequently, the newly added metadata\nagent will not appear in the output of the openstack network agent\nlist command.\n\nThis patch aims to update OVN_AGENT_METADATA_SB_CFG_KEY to nb_cfg as\nsoon as the metadata agent starts.\n\nChange-Id: Ic12b6ed3cd84dca16f5081dc646717d0a5be318b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/caf890f485a5248f8ce49f4e10d1e4c4feea9e93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/caf890f485a5248f8ce49f4e10d1e4c4feea9e93"}]},"branch":"refs/heads/master"},"fa34547f7391d31c2c02383154ae593847e8e14a":{"kind":"REWORK","_number":5,"created":"2025-05-23 04:07:55.000000000","uploader":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"ref":"refs/changes/73/950673/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/950673/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/950673/5"}}},"commit":{"parents":[{"commit":"5fbb3c1a3ffd895acb4869e54643c63981e38f34","subject":"Merge \"Fix some pylint warnings in the test tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fbb3c1a3ffd895acb4869e54643c63981e38f34"}]}],"author":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-22 15:56:51.000000000","tz":480},"committer":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-23 04:07:25.000000000","tz":480},"subject":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts","message":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts\n\nIf the metadata agent is not present in the local AgentCache and the\nreport time (agent_down_time) is too long, the Neutron API will delay\nthe Chassis_Private update. As a result, the agent will not update the\ncorresponding record properly. Consequently, the newly added metadata\nagent will not appear in the output of the openstack network agent\nlist command.\n\nThis patch aims to update OVN_AGENT_METADATA_SB_CFG_KEY to nb_cfg as\nsoon as the metadata agent starts.\n\nChange-Id: Ic12b6ed3cd84dca16f5081dc646717d0a5be318b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa34547f7391d31c2c02383154ae593847e8e14a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa34547f7391d31c2c02383154ae593847e8e14a"}]},"branch":"refs/heads/master"},"c6f5c05778076c653e47c6cb7e726bff7016b860":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-05-26 13:40:34.000000000","uploader":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"ref":"refs/changes/73/950673/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/950673/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/950673/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/950673/6"}}},"commit":{"parents":[{"commit":"5fbb3c1a3ffd895acb4869e54643c63981e38f34","subject":"Merge \"Fix some pylint warnings in the test tree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fbb3c1a3ffd895acb4869e54643c63981e38f34"}]}],"author":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-22 15:56:51.000000000","tz":480},"committer":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-26 13:40:15.000000000","tz":480},"subject":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts","message":"Update OVN_AGENT_METADATA_SB_CFG_KEY when OVN metadata agent starts\n\nIf the metadata agent is not present in the local AgentCache and the\nreport time (agent_down_time) is too long, the Neutron API will delay\nthe Chassis_Private update. As a result, the agent will not update the\ncorresponding record properly. Consequently, the newly added metadata\nagent will not appear in the output of the openstack network agent\nlist command.\n\nThis patch aims to update OVN_AGENT_METADATA_SB_CFG_KEY to nb_cfg as\nsoon as the metadata agent starts.\n\nCloses-Bug: #2111475\nChange-Id: Ic12b6ed3cd84dca16f5081dc646717d0a5be318b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6f5c05778076c653e47c6cb7e726bff7016b860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6f5c05778076c653e47c6cb7e726bff7016b860"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
