)]}'
{"id":"openstack%2Fneutron~950856","triplet_id":"openstack%2Fneutron~master~I2c52805a497d2c92168261a8cd4e964d708ca45c","project":"openstack/neutron","branch":"master","topic":"normalize-wsgi","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-27 10:07:03.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-05-27 13:00:36.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-05-27 10:54:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-05-27 10:07:25.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-05-27 13:00:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2c52805a497d2c92168261a8cd4e964d708ca45c","subject":"Move grenade skip level jobs to wsgi","status":"MERGED","created":"2025-05-26 11:33:46.000000000","updated":"2025-05-27 13:02:48.000000000","submitted":"2025-05-27 13:00:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950856-normalize-wsgi","meta_rev_id":"b4665b7c2d3604f1977442523fc2017daac3bb20","_number":950856,"virtual_id_number":950856,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 13:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-05-27 10:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2025-05-27 10:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-05-27 10:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2025-05-27 10:54:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2025-05-27 10:07:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-05-26 14:59:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-26 13:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-26 14:59:16.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-05-26 14:59:16.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-05-26 14:59:16.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-05-26 14:59:16.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"f48bb3396b80b8b8d5d6b7bf7b34e4f35712bbce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 11:33:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bc69812dcb979d21ea416df90f6c6d1ac4a6f7a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 11:34:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"099eb1a915aa0dfae3cfb0f2b7ce2a20e1b930be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 13:27:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d7905f0635a410ab3c6f6fe8fbdbbf3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a5e17242c9a34e6c9326cf6bf6576e08 : FAILURE in 18m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ca251f3b8fd64179bc33657d1ea344c0 : SUCCESS in 1h 52m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/99ef899a66664e2d9a97cf6c8b60bf93 : SUCCESS in 43m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05174a7edf754a7babf4173414adc8b7 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242a8137837d4525b054fc49be4ef0b4 : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6dedbdf5871fa111b72729d1704b66982783ca70","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 14:42:56.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a74426623a3e403495fb80b202bf23d8\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/ab8f5535cfd14b30982411c0cdacf170 : SUCCESS in 3h 02m 19s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/177c4c0c8f074026bb390a53817f5576 : SUCCESS in 1h 45m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/95d4468e621d45ffbcab3065598c5014 : FAILURE in 1h 59m 41s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3740b20e5c4d492194ad24739dfe832f : SUCCESS in 1h 53m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91c017387b994726a6d699999a21bf3e : SUCCESS in 1h 54m 54s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/13e13647793a4640993bb85006493028 : FAILURE in 1h 54m 49s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/4b249f5e13ed4c8ea03acd8ba0f9956f : SUCCESS in 2h 46m 22s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/acdd0e92eeb1480c9bf905eb0b89a94e : SUCCESS in 1h 08m 20s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/416bc1f5d9c6471c89f73e83b606821c : SUCCESS in 1h 06m 36s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/47ec8fb3f938463185658bfb89076b15 : SUCCESS in 50m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/09b7e2a766284902b52198be5baa5ac5 : SUCCESS in 59m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7988b3cbd57846e4865ea8fef8514526 : SUCCESS in 1h 26m 01s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/c8805bab065d4efbb2ffe1afe5eb8145 : SUCCESS in 1h 59m 59s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/c191dc60dc07456abe8f39855c36c6f8 : SUCCESS in 22m 56s\n- neutron-tempest-plugin-ovn-with-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/18541711f96a495986aaf40bbeb59794 : SUCCESS in 2h 18m 55s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/31826f88b5864c02b518a94a5a300c90 : FAILURE in 39m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d8abb98ade2f4faeb6bc9e6763d94bd1 : SUCCESS in 1h 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dff3fa01f47c6f00c40ec9a9cb7603accfc94a8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 14:43:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6410cd344a68e6816dce9ba47529a510aaafa120","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 14:48:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04536869e80490f304def5521c4ac8fc7a0e9646","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-26 14:59:16.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"37021196d691bf41506e0b25aff118e90c340c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 16:31:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bdf9e45ec734e108933b2d72bd02cfd\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/207080be8d8445ad9d14942d70bc65dd : SUCCESS in 1h 42m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0c4eb0851ff947aabf40e12f895ea73e : SUCCESS in 1h 45m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7fc45e82dc8a4d96a71c92941a4b6731 : SUCCESS in 32m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07505feb4adb499282eb04f95a358148 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816d8ff5f2fc44a3b4a930ae8cf2695e : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":2},{"id":"46965e2833a198321e86727d3399f83f40f3a9ed","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 17:39:20.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbc7a4ecc384030bd922622b6d0e322\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/59f2f9c0a3a144cda22f5ea15edcac60 : FAILURE in 23m 54s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/59db53f35bfd4c919bca0a5ede1dcc2c : FAILURE in 21m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/c4e9300358a1403db68907bc4076ec03 : FAILURE in 2h 08m 23s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d857b002a8184b60bcab2ee80e79284a : SUCCESS in 1h 56m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d01e5e34d1e643f4854f0d94d83a7bc5 : SUCCESS in 2h 02m 36s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/4bc1d652d07a489db22270281ba4692f : SUCCESS in 2h 49m 15s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/2103af8293f541ce973ab48ed34f2d85 : SUCCESS in 2h 08m 41s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9e2684e7784d48e48de223474acad83a : SUCCESS in 1h 06m 18s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/827e74997ed14d49a22859e46f5a8fbc : SUCCESS in 1h 16m 30s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/52317c5a7d41468489b68d639929c86b : SUCCESS in 1h 03m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5ffcbc5afb554051b4429f7dc0c17604 : SUCCESS in 1h 05m 46s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0d5e094f60c344e8b9031ef46cc8fbf2 : SUCCESS in 1h 24m 38s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/03068ac41965468fa6db5685e3aececf : SUCCESS in 2h 01m 38s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/aa0b16b6906046aa8f2f9bf2611927f2 : SUCCESS in 25m 25s\n- neutron-tempest-plugin-ovn-with-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/6f8fd8b268c548fea173d9995e54f0ab : FAILURE in 2h 28m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6aca598821e74ae295e9af4f84cb09ad : SUCCESS in 1h 02m 24s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/06f1ef6beca049c896b572caf52261f5 : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6fc1c049b181abdc2ac118a05869035de7ff376a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-27 10:07:03.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"8708d25703b4ad3643cc78b9ad710cc4e8aa7f3e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-27 10:07:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cd0e773eaa0772fef123a61648ee9c5216e8cfd5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-27 10:54:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8885d4058bbe33694519866602f3c57d67c100bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 10:55:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b65cd27b8bacafa13a1aeb6e062861da3456248d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:00:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a7c3d3de7b34962b78b2ba6233891be\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6ee56f932a1740f48f6c3d5ee2983ecb : SUCCESS in 1h 42m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/33f5808c3ca942acae40697586f473fe : SUCCESS in 2h 04m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fa93480f4134ec0b59e911e41e1ea0b : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1ad4789be04e5c9dd5e01d6dc55297 : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":2},{"id":"5cd1e05baddaa4f2dc30b017e11ec50a2fbf9c4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:00:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b4665b7c2d3604f1977442523fc2017daac3bb20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:02:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05848232296e44179c79a2e41dc9d250\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38323113178c4f87a26e23266725c009 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd3957a0f90e9653b87f221e1174c9f471ee23a1","revisions":{"cdb364e4d97be40a3deaeee4da317e8e4de15412":{"kind":"REWORK","_number":1,"created":"2025-05-26 11:33:46.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/56/950856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/950856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/950856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/950856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/950856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/950856/1"}}},"commit":{"parents":[{"commit":"1610455a4b3c0591c9b5f6cf288316aae206d20b","subject":"Merge \"Remove code for suse from post-devstack installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1610455a4b3c0591c9b5f6cf288316aae206d20b"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-26 11:29:38.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-26 11:33:27.000000000","tz":330},"subject":"Move grenade skip level jobs to wsgi","message":"Move grenade skip level jobs to wsgi\n\nWith [1] we can\u0027t deploy neutron with eventlet in\nmaster and wsgi is default mode[2], so let\u0027s switch\ngrenade skip level jobs to deploy with wsgi.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/932203\n[2] https://review.opendev.org/c/openstack/devstack/+/932199\nTopic: eventlet-deprecation\nRelated-Bug: #2069831\n\nChange-Id: I2c52805a497d2c92168261a8cd4e964d708ca45c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cdb364e4d97be40a3deaeee4da317e8e4de15412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cdb364e4d97be40a3deaeee4da317e8e4de15412"}]},"branch":"refs/heads/master"},"cd3957a0f90e9653b87f221e1174c9f471ee23a1":{"kind":"REWORK","_number":2,"created":"2025-05-26 14:43:34.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/56/950856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/950856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/950856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/950856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/950856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/950856/2"}}},"commit":{"parents":[{"commit":"1610455a4b3c0591c9b5f6cf288316aae206d20b","subject":"Merge \"Remove code for suse from post-devstack installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1610455a4b3c0591c9b5f6cf288316aae206d20b"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-26 11:29:38.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-05-26 14:42:27.000000000","tz":330},"subject":"Move grenade skip level jobs to wsgi","message":"Move grenade skip level jobs to wsgi\n\nWith [1] we can\u0027t deploy neutron with eventlet in\nmaster and wsgi is default mode[2], so let\u0027s switch\ngrenade skip level jobs to deploy with wsgi.\nJust dropped the overrides as parent grenade jobs have these\nalready set.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/932203\n[2] https://review.opendev.org/c/openstack/devstack/+/932199\nTopic: eventlet-deprecation\nRelated-Bug: #2069831\n\nChange-Id: I2c52805a497d2c92168261a8cd4e964d708ca45c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd3957a0f90e9653b87f221e1174c9f471ee23a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd3957a0f90e9653b87f221e1174c9f471ee23a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
