)]}'
{"id":"openstack%2Fneutron~951739","triplet_id":"openstack%2Fneutron~stable%2F2024.1~Ib237c7c1e1ba29626a03765a0521180fc2718c32","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2102083","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-05 10:38:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib237c7c1e1ba29626a03765a0521180fc2718c32","subject":"Allow to delete singleton instances","status":"MERGED","created":"2025-06-04 05:50:05.000000000","updated":"2025-06-05 10:41:09.000000000","submitted":"2025-06-05 10:38:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951739-bug/2102083","meta_rev_id":"99256b6cb19a4fca9a640311d33786f305d3803f","_number":951739,"virtual_id_number":951739,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-05 10:38:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-06-05 07:53:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-06-04 14:50:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2025-06-05 07:53:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-04 08:14:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-04 14:50:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-06-05 07:53:06.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"1e3dffc3d3aa43d47a119d6a9c23777a632a7a41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-04 05:50:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2c1a30f563a4134d25d1373278f67894d23de9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 08:14:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c3fbd4d02dc4b7f9d37a7fe15be6c6e\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/97e35ce9855d46c59f8478e10a64fbea : SUCCESS in 1h 44m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/1adbe48733d24eb283c04cdd23eb080a : SUCCESS in 1h 48m 11s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/4a473e6a25724a3a9cc964b5154f0d24 : SUCCESS in 2h 10m 52s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/71e50f49312b4ac3b9b749d424645076 : SUCCESS in 39m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bd6d9f42d8f46ef8dea7d4fa537f083 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b1aaefbfa09412a815d8ba3a6ff1602 : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01842f2b7dd549efa439d7661519e6d9 : SUCCESS in 25m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa94b1f3eba446019673912192b8d360 : SUCCESS in 22m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60968277cd26450a83cfca8706baf424 : SUCCESS in 10m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/27b7d453304147b7a589c86d4a285468 : SUCCESS in 1h 21m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9793cb1b28c44ec199b2e8b516f26179 : SUCCESS in 2h 22m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bbd55b42a87c4b0ebd452ce6b42a438e : SUCCESS in 1h 03m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/74bca713e29a49bb83a4f7ff14f9bb77 : SUCCESS in 1h 07m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/038ef812899a4429a7c4d98f878aefd3 : SUCCESS in 1h 23m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/581f0e49d28849fab92e87e0a2517367 : SUCCESS in 1h 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e390f74c4b3a3dc60a6d04377e2b9ec15f563687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 08:50:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18b1f3c46b4a4cceac2f1d1e849f0a7d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/864e17c67cce428593ee9045208be617 : SUCCESS in 1h 01m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ac9ed3eb25f428e87bb85045648dde7 : TIMED_OUT in 1h 21m 01s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7d9739f6b36a43c28bd49aae9d9b84d9 : SUCCESS in 1h 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98748fc77bf0e13dbd30ce3858981644a6c3ac6b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-04 14:50:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5546aef0bc5041ad844aafdc00b81b6ec657dfc6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-06-05 07:53:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db95696774ffad3e4dd08dfd1642ef794d60aa3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 07:53:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"027d5b14806092e9fdc4456d9917df6e0adfd6d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 10:38:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d89dedb371e4f9e9dd0fa7c66d410a9\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/091aff59e81e4dc9bfe43bb8a84e653f : SUCCESS in 2h 04m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc93afeb3d784217bdd050593d1e0849 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09c4f60f3ce642289d8d2cff74350176 : SUCCESS in 23m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73482e5afc9a407f8a63ed6eb023e284 : SUCCESS in 20m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7546271d9f5b4731b78fe14cb2095eee : SUCCESS in 6m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5784e2d5ff3243c09c3e17698870e33e : SUCCESS in 1h 04m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8af8928ed6842a2a62a054d917784c7 : SUCCESS in 1h 32m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f6107f02f94a4802af244018578eb043 : SUCCESS in 52m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e880e79d29a544f9bc5776602c1072f0 : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":1},{"id":"323a27f9c71c91dd01ba992a3258daaae1e76be3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 10:38:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"99256b6cb19a4fca9a640311d33786f305d3803f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 10:41:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e5465a3e399439fbc2f35510277e244\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea05fb40ddec483ca1aa9d947b7c7ad8 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"790b51985a56374fe6b564eeb40d229805db5b12","revisions":{"790b51985a56374fe6b564eeb40d229805db5b12":{"kind":"REWORK","_number":1,"created":"2025-06-04 05:50:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/951739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/951739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/951739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/951739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/951739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/951739/1"}}},"commit":{"parents":[{"commit":"e3a56a401f8fc94accb543d0f87a5b9fc2654d3d","subject":"Merge \"[OVN] Support address group for ovn driver\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3a56a401f8fc94accb543d0f87a5b9fc2654d3d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-03-13 08:39:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-04 05:49:56.000000000","tz":0},"subject":"Allow to delete singleton instances","message":"Allow to delete singleton instances\n\nNow the decorator ``SingletonDecorator`` forgets the instantiated object\nif all references to this object have been deleted. It is needed to\nkeep at least one reference to the decorated class in order to make\nany new instantiation idempotent.\n\nConflicts:\n  neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n  neutron/tests/unit/common/test_utils.py\n\nCloses-Bug: #2102083\nChange-Id: Ib237c7c1e1ba29626a03765a0521180fc2718c32\n(cherry picked from commit 1e560ecd8a2a620b7b6d677bae962764302136a0)\n(cherry picked from commit a710c78e33536ebc3c62e8493325c910ae305a09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/790b51985a56374fe6b564eeb40d229805db5b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/790b51985a56374fe6b564eeb40d229805db5b12"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
