)]}'
{"id":"openstack%2Fneutron~952301","triplet_id":"openstack%2Fneutron~master~Ibf4a48d19e42432708658af56577237255e9d80d","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-08-18 16:33:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf4a48d19e42432708658af56577237255e9d80d","subject":"Fix a random segment unit test failure","status":"MERGED","created":"2025-06-10 21:42:41.000000000","updated":"2025-08-18 16:35:41.000000000","submitted":"2025-08-18 16:33:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952301","meta_rev_id":"73cfcc44c603ec1493ca6dca73b5a90d1649382f","_number":952301,"virtual_id_number":952301,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-06-10 22:46:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 16:33:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-18 14:12:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2025-08-18 14:12:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-06-18 11:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-18 14:12:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2025-08-18 14:12:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-07-02 19:26:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-10 22:14:42.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-10 22:20:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-18 11:20:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-08-18 14:12:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-08-18 14:12:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"ef1de013f26e466049e9349112dd9b1fa0aedee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-10 21:42:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a6302a7fab67c014a1969e0a04de9390fcf8167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-10 21:53:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cac24637ec5d5c5ac0eb3f793e996745321a6ace","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-10 22:14:42.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/01/952301/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9afa214e776fc4687142cfb8d084914a357d7c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 22:20:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44d4337af5d04ec2aa3acbd2eb5aea24\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/131d26eda7e2417abfa1e0e9a5eedcae : SUCCESS in 27m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4f2a42e764464698b6d74629fc0f69 : SUCCESS in 9m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02756fcfdb644b60944259f23a1b120a : SUCCESS in 18m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/041b43da5ca34a7faff04a0ed6c5be65 : SUCCESS in 19m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bde5fb640554492381878fbc90db68af : SUCCESS in 18m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e78d4009044db2874d799f219dc1d5 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":2},{"id":"890108bad2358474393caa134ad5e2c1c33f8fd5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-10 22:46:58.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/01/952301/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8aa52a083e577c12c47cb909b9031703dcc19f3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 23:39:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffdf894c409d4ab3b883958c04ff0e95\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2dea1715e9f243ea8e024d78aa018bb7 : FAILURE in 44m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/87c4c1c15cf04babbd15cdae7684a2bc : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d1e791a243a24b35b32858777e34ea0c : SUCCESS in 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec769b1f3ec10789f62ebe6c1d36be4dbcd870ae","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-18 11:20:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c9b883fe6907c1c3cf410b504ece77ec9d2cbea","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-02 19:26:43.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"20bc594764598767734255a112822190ec7a598b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-08-18 14:12:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"af45fedd2b4e37f7f077fa64f53313fd3c623937","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-08-18 14:12:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5941e1abc4b42497ed4dab6110c855895df4c3bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 14:12:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6d83ae82880a8f3644c962d2a2d7d66367abff5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:33:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f16438be61f748678a73b506a194819c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/120a108736b24136960135c9de454277 : SUCCESS in 15m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34a2f58436c048479cb4eefc992cf8fa : SUCCESS in 15m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/132dc9d1cbfe4dce9dea5ddd819ac3e2 : SUCCESS in 20m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3b273c18bd43d79c6cf52af0841927 : SUCCESS in 14m 28s","accounts_in_message":[],"_revision_number":2},{"id":"68a3d56bad435039b0e9779aafeab1497d4aae1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:33:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"73cfcc44c603ec1493ca6dca73b5a90d1649382f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:35:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b177f1bbb2df4135a34ba1e396b66258\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098199007e374e27939441b4f9d5ab05 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"614353e8f94aa6bbeac9abe309104cf259c5c522","revisions":{"8ddd77bbb790ed22c9f66609467293c47cbcd138":{"kind":"REWORK","_number":1,"created":"2025-06-10 21:42:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/01/952301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/952301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/952301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/952301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/952301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/952301/1"}}},"commit":{"parents":[{"commit":"01be010d412c7082b64cc9dd450b753b006d9941","subject":"Merge \"Actually require python \u003e\u003d 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01be010d412c7082b64cc9dd450b753b006d9941"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-06-10 21:32:34.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-06-10 21:42:26.000000000","tz":-240},"subject":"Fix a random segment unit test failure","message":"Fix a random segment unit test failure\n\nThe unit test test_delete_network_and_owned_segments()\nwill randomly fail if run last, and print a traceback\nending with this error:\n\n  TypeError: \u0027\u003e\u003d\u0027 not supported between instances of \u0027MagicMock\u0027 and \u0027int\u0027\n\nThe issue is that the network delete event it triggers\nwill create an additional NovaSegmentNotifier object, and\nit will not have mocked either of the placement or nova\nclient return values out, so the comparision is invalid.\n\nChange the test to mock out the _get_clients() class object\nso that it returns the same values the test is using,\navoiding the error.\n\nAlso had to move the override of send_events_interval after\nthe super().setUp() call since it does not exist before that\npoint, otherwise trying to run just that test class alone\nwill trigger this failure for all tests:\n\n  oslo_config.cfg.NoSuchOptError: no such option send_events_interval in group [DEFAULT]\n\nFound that while trying to debug the issue.\n\nPartial-bug: #2038373\nChange-Id: Ibf4a48d19e42432708658af56577237255e9d80d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ddd77bbb790ed22c9f66609467293c47cbcd138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ddd77bbb790ed22c9f66609467293c47cbcd138"}]},"branch":"refs/heads/master"},"614353e8f94aa6bbeac9abe309104cf259c5c522":{"kind":"REWORK","_number":2,"created":"2025-06-10 21:53:05.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/01/952301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/952301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/952301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/952301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/952301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/952301/2"}}},"commit":{"parents":[{"commit":"01be010d412c7082b64cc9dd450b753b006d9941","subject":"Merge \"Actually require python \u003e\u003d 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01be010d412c7082b64cc9dd450b753b006d9941"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-06-10 21:32:34.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-06-10 21:52:55.000000000","tz":-240},"subject":"Fix a random segment unit test failure","message":"Fix a random segment unit test failure\n\nThe unit test test_delete_network_and_owned_segments()\nwill randomly fail if run last, and print a traceback\nending with this error:\n\n  TypeError: \u0027\u003e\u003d\u0027 not supported between instances of \u0027MagicMock\u0027 and \u0027int\u0027\n\nThe issue is that the network delete event it triggers\nwill create an additional NovaSegmentNotifier object, and\nit will not have mocked either of the placement or nova\nclient return values out, so the comparision is invalid.\n\nChange the test to mock out the _get_clients() class object\nso that it returns the same values the test is using,\navoiding the error.\n\nAlso had to move the override of send_events_interval after\nthe super().setUp() call since it does not exist before that\npoint, otherwise trying to run just that test class alone\nwill trigger this failure for all tests:\n\n  oslo_config.cfg.NoSuchOptError: no such option send_events_interval in group [DEFAULT]\n\nFound that while trying to debug the issue.\n\nPartial-bug: #2038373\nChange-Id: Ibf4a48d19e42432708658af56577237255e9d80d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/614353e8f94aa6bbeac9abe309104cf259c5c522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/614353e8f94aa6bbeac9abe309104cf259c5c522"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
