)]}'
{"id":"openstack%2Fneutron~952564","triplet_id":"openstack%2Fneutron~stable%2F2024.2~I3bdd260f0f6b0259ff15cfe16a111bfe93b40749","project":"openstack/neutron","branch":"stable/2024.2","topic":"bug/2112453","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-13 18:43:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3bdd260f0f6b0259ff15cfe16a111bfe93b40749","subject":"Allow empty gateway IP in subnets from subnet pools","status":"MERGED","created":"2025-06-13 10:10:26.000000000","updated":"2025-06-13 18:44:38.000000000","submitted":"2025-06-13 18:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952564-bug/2112453","meta_rev_id":"8473879dbbc2cbfb9827bf744cdd0bd2adcf7071","_number":952564,"virtual_id_number":952564,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-13 18:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2025-06-13 12:14:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-06-13 15:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-06-13 15:12:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-13 10:27:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-06-13 10:27:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-06-13 10:32:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-13 12:14:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-06-13 12:21:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-13 15:12:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"c2b96d2d34c9ef3f3c2b6e0c65e7829f963e63d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-13 10:10:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7a6d3d9dda19ea456bbfa1edc6c0a90013798e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 10:32:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713b358acdfa4307822201ead4ee2eb7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81906191a2344fd0a16c776a98b33dab : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9db9b11e5f4d43d69e2cae50a02abaf5 : SUCCESS in 19m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56ef2810388c76430b9287e2496d6d7a17645d22","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-06-13 12:14:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8abac081db7db9ee682cd92f37fc197ff018cec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 12:21:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0615129bee3e4863b7a97677f001a39f\n\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/3601b5c5c237406d95ca3858f533b588 : SUCCESS in 1h 08m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/0e62351c04e44b6182e324960bdfa98d : SUCCESS in 1h 52m 39s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/9e50fa5413564adab036731251c2b381 : SUCCESS in 2h 09m 30s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/314397a366344cf8a12e812005e17e42 : SUCCESS in 36m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c9a7da26c714f77b904cf22a641bb2a : SUCCESS in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2020b6925e2349308fb860b5d3ad79ec : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a7c01d121c64fef8a53db4b8438d4b8 : SUCCESS in 23m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd455ccbf0e244918c0a358038ea6fb9 : SUCCESS in 17m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23becd104c6649929db35c317a1b2d9d : SUCCESS in 17m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0376fec86404678a115a2d2005d6855 : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a22f9458ce9d4effbabbd455332e8125 : SUCCESS in 1h 35m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/57f6458ec64c48b8ae08b71432b35004 : SUCCESS in 1h 49m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e56957b715a41aeb774a445ff1c74a8 : SUCCESS in 57m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5fdce08e2841412da014bb0d12b2af00 : SUCCESS in 55m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/15bcd5a4023f4f61bdc2d6255827b4a9 : SUCCESS in 55m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a5b51c946b9c4be1ad40fdd6b861f862 : SUCCESS in 59m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b3e35e9b67bc4e65a358ea0d46df9ccd : SUCCESS in 1h 04m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8aaff7d6d7a84640bb5ae8f40bbc2c20 : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a73c050c1a03ab1637c31d1b17800879203fc11f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-13 15:12:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"117f25858a494fc63d25f0b41022fe681fd5a9ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 15:13:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"773fe5c14366234cf968c91aa48a05a6ff6d41f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 18:43:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d2af57cc324db7bf9323a5015c3d69\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/6be7e743c86d4ccd8ae43341148e2389 : SUCCESS in 1h 50m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc85c10a8e954e5eb449226cc029a0be : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2a4908b58664020a0df83f7e2bf7dce : SUCCESS in 25m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3517289ac4c4cc1b31884261feb6e97 : SUCCESS in 22m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97483b834c845f687f1f06787e5dc6f : SUCCESS in 13m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2e0ee751f5b4e61b86cc0bd6d50bf29 : SUCCESS in 1h 26m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d901bd5e3a3a49dd938b89f63fc7fffc : SUCCESS in 1h 33m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/061282b845944816865992b18c5b1993 : SUCCESS in 1h 18m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/20a251d3d6ed40e48f1d27d092efc472 : SUCCESS in 55m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/824174ecad28422498ad700876f8cccd : SUCCESS in 1h 02m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20919a4acbda42ac8f2352407999bfcd : SUCCESS in 1h 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6645981585d018bd948341cc0f9c3d37bab1e3cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 18:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8473879dbbc2cbfb9827bf744cdd0bd2adcf7071","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 18:44:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09d1126f35ef4eb4b1965c87cb39dc56\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8020ff5a4ad9440891681213363462b9 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71cf3f22cb8af43a3cfc32b30260e64e441eedad","revisions":{"71cf3f22cb8af43a3cfc32b30260e64e441eedad":{"kind":"REWORK","_number":1,"created":"2025-06-13 10:10:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/64/952564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/952564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/952564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/952564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/952564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/952564/1"}}},"commit":{"parents":[{"commit":"5b7721882f54475000e2dcf2c5cca435465e3e80","subject":"Merge \"Fix nova notification on port unplug\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b7721882f54475000e2dcf2c5cca435465e3e80"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-11 08:17:22.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-13 10:10:03.000000000","tz":0},"subject":"Allow empty gateway IP in subnets from subnet pools","message":"Allow empty gateway IP in subnets from subnet pools\n\nWhen a subnet is created from a subnet pool, now is possible to undefine\nthe gateway IP. The new subnet created will have this value assigned to\nNone.\n\n  $ openstack subnet create --subnet-pool shared-default-subnetpool-v4 \\\n      --network net14 snet14 --gateway None --format value \\\n      --column gateway_ip\n    None\n\nConflicts:\n  neutron/ipam/requests.py\n\nCloses-Bug: #2112453\nChange-Id: I3bdd260f0f6b0259ff15cfe16a111bfe93b40749\n(cherry picked from commit 5cccd2112fad736967d70768e1d6ac60c7d47415)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71cf3f22cb8af43a3cfc32b30260e64e441eedad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71cf3f22cb8af43a3cfc32b30260e64e441eedad"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
