)]}'
{"id":"openstack%2Fneutron~952663","triplet_id":"openstack%2Fneutron~stable%2F2025.1~Iaa0648979cc09b126af801f1f74a996c9df614fe","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2111660","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-19 09:47:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa0648979cc09b126af801f1f74a996c9df614fe","subject":"Update OVN_AGENT_NEUTRON_SB_CFG_KEY when Neutron OVN Agent starts","status":"MERGED","created":"2025-06-16 10:55:47.000000000","updated":"2025-06-19 09:48:34.000000000","submitted":"2025-06-19 09:47:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952663-bug/2111660","meta_rev_id":"54c807c9a78a46bacf1da3c84c5eba943e1f26e7","_number":952663,"virtual_id_number":952663,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-19 09:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-06-19 08:07:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-06-16 21:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2025-06-19 08:07:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-16 10:55:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"state":"CC"},{"updated":"2025-06-16 11:24:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-16 12:32:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-16 21:26:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-06-19 06:53:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-06-19 08:07:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"9faa814f1cf4ecc93476ce8b46e178bd3610e6c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-16 10:55:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44ef96eb3e1dd10b1351a25d84cc94108ffd3d04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 11:24:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44d663a5e6684880822aef2f34c23cfe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f714890872f4719b0fe1b31ebcbefdb : SUCCESS in 27m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5b3685ac43c34ed29fdf05a1e4c8454b : SUCCESS in 22m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a6736d86994127d434640f7f920f28c6458c2d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 12:32:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffe23cc1851641b09af64383764982f8\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/6a5be33ff18a465abd3438ad75f7af3d : SUCCESS in 1h 36m 24s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/f2423c76f3f34a53b66954a4ad3ad407 : SUCCESS in 1h 08m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d574245d34e44d1a905a1b60fb8b23c6 : FAILURE in 36m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d3b475e3f44dd49644cd1808fa9c4e : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e75ae5b5e0a4160880a351e0ccedbec : SUCCESS in 21m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65fce7d8ddc14dd490938c07f23771e6 : SUCCESS in 13m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc797826ab84bd2a99195b65a042a2f : SUCCESS in 5m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/760c41e8e84345f6b13d1b5d33d741b5 : SUCCESS in 50m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/07b9bafe07d34f4d8d4c524d092586b5 : SUCCESS in 1h 28m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbd7a3103ccc44109bb84f6c41c3a7bd : SUCCESS in 59m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/40fcfed181034adda69035d15c54817b : SUCCESS in 58m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/11f45a0206b64773adffe6e09113feaf : SUCCESS in 1h 14m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f76c907fae5f4e8ab27048f3e2cdf39c : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"78f7f39a1dfb27a6659f52ae0a8c6eba71812837","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-16 15:27:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04cf07cde69358941545f5a48dc4f160a2fc5461","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 16:20:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650ac65884d8438e83c80ff8cdf545fe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60b69b7367774b42a8b09df6cb9a75bf : SUCCESS in 49m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4adf71af3c2451281fceb092eac10e0 : SUCCESS in 45m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34b967647eb0a43493280369333bd0a1df1ec86a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 17:11:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67b50dc481cb42e1a2d585c477f9fc42\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ab14311769e1495fa7a783722266dc58 : SUCCESS in 1h 43m 10s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/91a7f97f2d9a45d68a5a48f44b302bfb : SUCCESS in 1h 11m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53e1eca82ec14e078e87085851ec1fba : SUCCESS in 45m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6374d06235124d3b82e036150f70aef0 : SUCCESS in 12m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/888b7bb222474ca893e4ae160fd5baa9 : SUCCESS in 20m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20e7eca5c3f94f949e6027bb5cd522ec : SUCCESS in 18m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee7c7dd3cc504ae792acda708bbbc08a : SUCCESS in 11m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8d5ee3f2879741118abccca8693d7384 : SUCCESS in 59m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/feeb0c44de1649bf8079e960e400369f : SUCCESS in 1h 11m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd58f065758046349291727cc25ba694 : SUCCESS in 1h 00m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6b6c0af5d2f24057b96e007c55f9f54d : SUCCESS in 53m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/124071f4e88744d398ed7f080c658286 : SUCCESS in 1h 14m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5e5d6df191134ecf89f4d7635c51f119 : SUCCESS in 1h 25m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d1f4f498a20000f8c4e6e6615cb8562eaea78bfd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-16 21:26:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3817000419d3092c008faf0da9497f0b5dded801","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-19 08:07:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4b26196365275d31fb6b543146206ae8fe22e3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 08:07:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93b049d6626fcc6a790292f0002c3dbfe121d477","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 09:47:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c22e8f8abaef4c6b978835975e62760b\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/da4b6e13ebe2483c8b1195f98f5bc8f6 : SUCCESS in 1h 39m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d0ba246b39d4e3f9b2bd60c134f2a8c : SUCCESS in 11m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7daa3ffb9644daaa4d3b01ee452731f : SUCCESS in 19m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/618a7322efd247daac553268fe4cc050 : SUCCESS in 23m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2fe49047cb43338c7c35579508f096 : SUCCESS in 13m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a138e2cf55b34d4cb8ac3754793e1561 : SUCCESS in 1h 19m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9b2221b4a44480992bd141a7f96d66d : SUCCESS in 54m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/40ef320cc3e74113bf8ced6c36a75113 : SUCCESS in 1h 07m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/265db7af8b564d5c90a8728f939c6e27 : SUCCESS in 1h 24m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bfab32e20a2d9a9f3475e67f811f58dd00bc6029","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 09:47:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"54c807c9a78a46bacf1da3c84c5eba943e1f26e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 09:48:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92e23bbc7d1a4352967c87a564f4e70f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caef3258c284445789f134b0437b1295 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5a77aca0409c33f3e8fe5d854fbfa5c66762926","revisions":{"b5a77aca0409c33f3e8fe5d854fbfa5c66762926":{"kind":"REWORK","_number":1,"created":"2025-06-16 10:55:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/63/952663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/63/952663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/952663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/952663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/63/952663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/63/952663/1"}}},"commit":{"parents":[{"commit":"9854c4c213bd07e08794efa368dfa6ab69957e3d","subject":"Merge \"[FT] Improve ``test_assert_pings_during_br_phys_setup_not_lost*`` tests\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9854c4c213bd07e08794efa368dfa6ab69957e3d"}]}],"author":{"name":"minsu","email":"minsu@cisco.com","date":"2025-05-25 13:14:40.000000000","tz":480},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-06-16 10:55:32.000000000","tz":0},"subject":"Update OVN_AGENT_NEUTRON_SB_CFG_KEY when Neutron OVN Agent starts","message":"Update OVN_AGENT_NEUTRON_SB_CFG_KEY when Neutron OVN Agent starts\n\nIf report time (agent_down_time) is too long, the Neutron API will\ndelay the Chassis_Private update for OVN_AGENT_NEUTRON_SB_CFG_KEY.\n\nThis patch aims to update OVN_AGENT_NEUTRON_SB_CFG_KEY to nb_cfg\nas soon as the neutron ovn agent starts.\n\nCloses-Bug: #2111660\nChange-Id: Iaa0648979cc09b126af801f1f74a996c9df614fe\n(cherry picked from commit 3bf7f8ec80072c70396a60dc78b065c3c47eaf2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5a77aca0409c33f3e8fe5d854fbfa5c66762926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5a77aca0409c33f3e8fe5d854fbfa5c66762926"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
