)]}'
{"id":"openstack%2Fneutron~952767","triplet_id":"openstack%2Fneutron~stable%2F2024.1~Iec481f5c008be8cac621ff44afa8eb3921f52e34","project":"openstack/neutron","branch":"stable/2024.1","hashtags":[],"change_id":"Iec481f5c008be8cac621ff44afa8eb3921f52e34","subject":"Reduce metadata port lookup log message to warning","status":"ABANDONED","created":"2025-06-17 13:50:49.000000000","updated":"2025-06-17 14:46:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":951870,"cherry_pick_of_patch_set":1,"meta_rev_id":"ec07724e4f1d8fd07a55312944b9d56f7b5de9bc","_number":952767,"virtual_id_number":952767,"owner":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-17 13:50:49.000000000","updated_by":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-06-17 13:59:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"35b1f003a1a377dc7206dade87c2c32497927d7e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"date":"2025-06-17 13:50:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* neutron/agent/metadata/agent.py\n* neutron/agent/ovn/metadata/server_socket.py","accounts_in_message":[],"_revision_number":1},{"id":"6e8026dfe3ef26e942c8ed66232a3d621de73b38","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"date":"2025-06-17 13:51:37.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"56c27fc2842f33741dc4fe668c9508e00d50802a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 13:59:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b316b18c535448c87a13b467ec450bc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3e194f4606b44f5a1e69c5254faa107 : FAILURE in 6m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e78c089f6e684e15a5d2bc39ad9bc025 : FAILURE in 5m 36s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/206e10c6aacd4dcb9d09628dc2177955 : FAILURE in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a11915ce0d4f28c6b8e324f17d8eab298e214ab8","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"date":"2025-06-17 14:01:28.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"a8e66d39fba6b49625ab43836bee8c622f47f268","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"date":"2025-06-17 14:03:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"97beffd7d61cf94619f47435a320e79cf746b69d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"date":"2025-06-17 14:04:37.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3b40e67bf6e9b51c1e20b82b3c29f4bfc928e1c6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"date":"2025-06-17 14:04:43.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"1428ec8a2cbec7bb564f8724f9ca663fe4468ffb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 14:34:46.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e52b71216294fd8b128b0e34c00232d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b57e21c39c2a4c35a6085e47efa6887a : FAILURE in 29m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56079b8255f740719ecb3e740e89833c : FAILURE in 27m 00s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6db8cf04b1514629a1de0ea7409960a0 : FAILURE in 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ec07724e4f1d8fd07a55312944b9d56f7b5de9bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"date":"2025-06-17 14:46:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"84b4f0da82b310442a9b51f4c591f48a57b4b1e9","revisions":{"9682a992a5b813c04decf430f7c4e548fe577e05":{"kind":"REWORK","_number":1,"created":"2025-06-17 13:50:49.000000000","uploader":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"ref":"refs/changes/67/952767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/952767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/952767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/952767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/952767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/952767/1"}}},"commit":{"parents":[{"commit":"3ea28968c8005105e0d1d0be3632f5d1323ac4c4","subject":"Merge \"Allow empty gateway IP in subnets from subnet pools\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ea28968c8005105e0d1d0be3632f5d1323ac4c4"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-06-05 15:15:06.000000000","tz":-240},"committer":{"name":"phillip toohill","email":"dead2hill@gmail.com","date":"2025-06-17 13:50:49.000000000","tz":0},"subject":"Reduce metadata port lookup log message to warning","message":"Reduce metadata port lookup log message to warning\n\nWhen a neutron port is not found for a metadata request,\nwe currently log at an error level. Since there is\nnothing the admin can do about it, short of finding the\nclient making the call, reduce the level to warning.\n\nChange-Id: Iec481f5c008be8cac621ff44afa8eb3921f52e34\nPartial-bug: #2112130\n(cherry picked from commit 81e1976ec625c74a5846854a3ed4c1ded662f5df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9682a992a5b813c04decf430f7c4e548fe577e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9682a992a5b813c04decf430f7c4e548fe577e05"}]},"branch":"refs/heads/stable/2024.1"},"49ea17d6099c4dec3c0c45b2c1b2603fe89a506a":{"kind":"REWORK","_number":2,"created":"2025-06-17 14:03:14.000000000","uploader":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"ref":"refs/changes/67/952767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/952767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/952767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/952767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/952767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/952767/2"}}},"commit":{"parents":[{"commit":"3ea28968c8005105e0d1d0be3632f5d1323ac4c4","subject":"Merge \"Allow empty gateway IP in subnets from subnet pools\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ea28968c8005105e0d1d0be3632f5d1323ac4c4"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-06-05 15:15:06.000000000","tz":-240},"committer":{"name":"phillip toohill","email":"dead2hill@gmail.com","date":"2025-06-17 14:03:14.000000000","tz":0},"subject":"Reduce metadata port lookup log message to warning","message":"Reduce metadata port lookup log message to warning\n\nWhen a neutron port is not found for a metadata request,\nwe currently log at an error level. Since there is\nnothing the admin can do about it, short of finding the\nclient making the call, reduce the level to warning.\n\nChange-Id: Iec481f5c008be8cac621ff44afa8eb3921f52e34\nPartial-bug: #2112130\n(cherry picked from commit 81e1976ec625c74a5846854a3ed4c1ded662f5df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49ea17d6099c4dec3c0c45b2c1b2603fe89a506a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/49ea17d6099c4dec3c0c45b2c1b2603fe89a506a"}]},"branch":"refs/heads/stable/2024.1"},"84b4f0da82b310442a9b51f4c591f48a57b4b1e9":{"kind":"REWORK","_number":3,"created":"2025-06-17 14:04:37.000000000","uploader":{"_account_id":10284,"name":"phillip toohill","email":"dead2hill@gmail.com","username":"ptoohill"},"ref":"refs/changes/67/952767/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/952767/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/952767/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/952767/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/952767/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/952767/3"}}},"commit":{"parents":[{"commit":"3ea28968c8005105e0d1d0be3632f5d1323ac4c4","subject":"Merge \"Allow empty gateway IP in subnets from subnet pools\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ea28968c8005105e0d1d0be3632f5d1323ac4c4"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-06-05 15:15:06.000000000","tz":-240},"committer":{"name":"phillip toohill","email":"dead2hill@gmail.com","date":"2025-06-17 14:04:30.000000000","tz":0},"subject":"Reduce metadata port lookup log message to warning","message":"Reduce metadata port lookup log message to warning\n\nWhen a neutron port is not found for a metadata request,\nwe currently log at an error level. Since there is\nnothing the admin can do about it, short of finding the\nclient making the call, reduce the level to warning.\n\nChange-Id: Iec481f5c008be8cac621ff44afa8eb3921f52e34\nPartial-bug: #2112130\n(cherry picked from commit 81e1976ec625c74a5846854a3ed4c1ded662f5df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84b4f0da82b310442a9b51f4c591f48a57b4b1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84b4f0da82b310442a9b51f4c591f48a57b4b1e9"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
