)]}'
{"id":"openstack%2Fneutron~952791","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I1a2119221be57afaa94a94ecb1448196802f041b","project":"openstack/neutron","branch":"stable/2025.1","topic":"bandit_exclude_b311-stable/2025.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-06-18 19:19:23.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I1a2119221be57afaa94a94ecb1448196802f041b","subject":"Bandit: add B311 to skiplist","status":"MERGED","created":"2025-06-17 21:05:37.000000000","updated":"2025-06-18 23:46:32.000000000","submitted":"2025-06-18 23:45:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952791-bandit_exclude_b311-stable/2025.1","cherry_pick_of_change":952768,"cherry_pick_of_patch_set":1,"meta_rev_id":"7970ef9a279508a94e089862c4a790b442f0c384","_number":952791,"virtual_id_number":952791,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-18 23:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-06-18 06:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-06-18 05:39:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-06-18 06:15:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-17 21:05:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-06-17 21:29:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-17 23:23:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-18 05:39:12.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-06-18 06:15:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"36486442416f8dc63cbf574337bc97784d11e815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-17 21:05:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aef4115b2685d182d29ec86eded9ff5d02ef8bb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 21:29:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05f2f9919ff443b3a4c10dad2e0f4ad2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9cb7bdbb6a234204a9191743b2bbac27 : POST_FAILURE in 21m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd7104070ba9495fa467007b9bd7fc3b : POST_FAILURE in 16m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3f6c82935b73db3c3a560563421e54430561afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 23:23:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f67403739644abc8a9a6d7f0598c578\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/846475699a1542bea9bb6e13a2a1e3f7 : POST_FAILURE in 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f981a124f1fe4364933559e598a769d1 : POST_FAILURE in 11m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68c71b71d5664789a68be61b8d8215bf : POST_FAILURE in 23m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7195e2e5da54a92b1c58f89ad1ada8c : POST_FAILURE in 18m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e418ecbcb24920b9cddb3d056418b0 : SUCCESS in 23m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/834723d367a347378d18e684cf674e3d : POST_FAILURE in 13m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/225c9dbedb2449f9b74eff07f0ca80a0 : SUCCESS in 1h 25m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/74bd7f8d47a54ff4a58b71c0d6efdebc : SUCCESS in 2h 00m 46s","accounts_in_message":[],"_revision_number":1},{"id":"deb96733d3f1ceb05498a8f1174b1426aafc8f7a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-18 02:20:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40f0a01e5b7d59eebbc3119a464e2c1d2078c7e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 03:56:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f596683af93e40ac9fabf3c39015edc6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6659d51b9e8249b6bf7315108f343d31 : SUCCESS in 24m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b2fa1b9b1db9418d944f4072eb30ea16 : SUCCESS in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efc9e8162693ccdb82984b058b6e86d558f5dc08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 04:11:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/195b138ba88e49a29509b989d169177b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a75f2ac88a8841e1b8b2d6f3be459455 : SUCCESS in 40m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2752cae3c0a4ad3a0dd23a2f2810c93 : SUCCESS in 12m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd261f13f1b3490abdc3ddeafc3c21d9 : SUCCESS in 23m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8059bc637c174f86910364bf15032eff : SUCCESS in 18m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa46b98215a94ba799518aedcf44cae0 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2bb5b0869ca47d8ab773b4a509b586e : SUCCESS in 17m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d1886488544145bc82aab591ecb11b2a : SUCCESS in 1h 19m 04s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/db568dbb9b5042f2ad8b88bf5e8f9574 : SUCCESS in 1h 46m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b94b05308d0beaf53e9f224fa9f698e10f914cd1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-06-18 05:39:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"15418bd556cff56b08ec2675d2817d09905693ac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-18 06:15:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbe8e1b5c405b42f8208cf2191204aa04e906856","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 06:15:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eee82043474ce4ab5c625b15d2734c864a4148af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 08:09:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d89daf4055cb418b9edc4dbf31415c3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45e61b8d862b423e9e7322709d912e2b : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acb840b35f804a02a9cb7b354c01c95b : SUCCESS in 20m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38ee9dd1bcf44db88486a6599c0017c3 : SUCCESS in 22m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21c2e867b6b4557bddbefa09b67caf5 : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99d747de9efb4f1f9afa41c9daaa52a4 : SUCCESS in 10m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b3d118cb870c4cb59ac23464384cbf8e : SUCCESS in 1h 31m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f54160dd818049ff9f6ccae86a37d362 : FAILURE in 1h 45m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f4bf6371a63480c8fe28accf5c53dbec26cd3e4c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-06-18 08:54:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30f78e54e03cd184027a1bcb9fdc6e45ee0086ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 09:17:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cbc569c593044539f848fc247ebf909\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42dfe237075d4fd4b00f6129ac563e26 : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f72f852caefb49348390c558180a3a31 : SUCCESS in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a340f6dc0f2ae6dd6695e309b94cbdf5b452dbd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 11:30:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d2bfe6df8684fca955bcd8744ad6b47\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed833b5831274e9db42c82a9da8a9d48 : SUCCESS in 37m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a96c5824ec284446a16834f840e5f38f : SUCCESS in 11m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b095bd5e914d4a90b7ee8ae3fa55860b : SUCCESS in 19m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6190b96c886c41c2992046175536bd30 : SUCCESS in 19m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81faf72559ed4ff3b2995c2132763518 : SUCCESS in 14m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c46cc7ae92045a8b2ce8b4b7c015873 : SUCCESS in 15m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c377532427cc43a59622657e229df9e1 : SUCCESS in 1h 31m 11s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5446a09e19d142fdbcdfb344073ed104 : SUCCESS in 2h 34m 13s","accounts_in_message":[],"_revision_number":1},{"id":"03ee1cdc46c0acba94284107c09c3ee86f8be31f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 11:31:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1789bdeeff3bd05cee81c56bad65d493abc4c2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 15:03:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8899e74d707441e9a53e36c054efd120\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52d7c51caa22411a9c73179a32d64fb2 : SUCCESS in 10m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bdccfcec9c34aaebaede6535bbdf11e : SUCCESS in 17m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11e05f791f8d4fd2af93a0200b302439 : SUCCESS in 16m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d26ed32f6b4ab09ce07c3132d09664 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7e919fd319547afa191ca4c4a60bf2c : SUCCESS in 17m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a858fb193a21437987d557b6de06dbdc : FAILURE in 1h 07m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/750673f1adf84145b8a457155a1362dc : FAILURE in 1h 52m 07s","accounts_in_message":[],"_revision_number":1},{"id":"efe58f5a173739c1b591828910523e39ea6e1e12","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-18 19:19:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efefbaaf1b083b1299c0c8e8d68d93845c38c430","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 20:27:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3d54d0a3254bc3b57ef45916f92a46\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a6872fa9e18403993c3d35f0d3ec07f : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/61dd5e773b214d819fbcc34765407c9b : SUCCESS in 1h 06m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eff7d7219c7529a1c450d022f7ec4e95e4178398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 21:32:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4566d2ef723a405387bad15405fda630\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/386c05cdebe44d9fa7e9037730aed144 : SUCCESS in 52m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba65ee7b2ade4599af21eef0077a92ed : SUCCESS in 11m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6b77920fa5f460c8b9784b830f76934 : SUCCESS in 21m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63f9b00e500f4caca754c2e24c4c4da6 : SUCCESS in 18m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20588ec455c44e94917b64ba6a8dfb2d : SUCCESS in 14m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1819fd24b93414791ca8fe7761a4f60 : SUCCESS in 11m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ab20b6fb73864ddcb6870623bcf59f63 : SUCCESS in 1h 22m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/7121eb984cd34fb89fc4f29c1d2cb74a : SUCCESS in 2h 08m 42s","accounts_in_message":[],"_revision_number":1},{"id":"33258cf155be45f3b1c88f62a5298645cbfd84fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 21:33:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77c10f5ba22ea9f7fa59b5d334e0d48674d26968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 23:45:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/267ebe70b0264de28bc585efed3f66e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf5b40c92964542acaaab178e2d4afe : SUCCESS in 11m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/323753b1eed9473989adc3fc4783191d : SUCCESS in 24m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/afe18689d5a643cfa2fc400678df6452 : SUCCESS in 23m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8223d83de445434da3c71a35f85cdcd7 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9789b9ca454d4861833e068d1910a055 : SUCCESS in 19m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6d84d44a5b5e4d6a9095f35c59c3075e : SUCCESS in 1h 22m 52s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/aadd7691b6d6470d8f3e43541ae6b3ed : SUCCESS in 2h 04m 21s","accounts_in_message":[],"_revision_number":1},{"id":"42bb5e6d1c88bbd3c86fb32ae91749facb1ea63d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 23:45:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7970ef9a279508a94e089862c4a790b442f0c384","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 23:46:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de0e6571472e418cb4b9a760f9914050\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24fde63ccaf64dd7b76dfcf44c7563f4 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4bed517117e4abdb3cf6d015601921a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc74c31900700e2fe78ca47712462ec3fd7e6835","revisions":{"dc74c31900700e2fe78ca47712462ec3fd7e6835":{"kind":"REWORK","_number":1,"created":"2025-06-17 21:05:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/91/952791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/952791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/952791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/952791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/952791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/952791/1"}}},"commit":{"parents":[{"commit":"9854c4c213bd07e08794efa368dfa6ab69957e3d","subject":"Merge \"[FT] Improve ``test_assert_pings_during_br_phys_setup_not_lost*`` tests\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9854c4c213bd07e08794efa368dfa6ab69957e3d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-17 14:02:32.000000000","tz":120},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-06-17 21:05:37.000000000","tz":0},"subject":"Bandit: add B311 to skiplist","message":"Bandit: add B311 to skiplist\n\nBandit 1.8.4 is more strict for random functions, see: [1],\nto avoid pep8 failures due to the few random calls in Neutron,\nskip B311\n\n[1]: https://github.com/PyCQA/bandit/releases/tag/1.8.4\n\nChange-Id: I1a2119221be57afaa94a94ecb1448196802f041b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc74c31900700e2fe78ca47712462ec3fd7e6835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc74c31900700e2fe78ca47712462ec3fd7e6835"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
