)]}'
{"id":"openstack%2Fneutron~952977","triplet_id":"openstack%2Fneutron~master~Idd9f4ade3911b21691f220c5eafa09463689b8b3","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-06-24 13:01:07.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-27 10:16:59.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-06-27 10:16:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd9f4ade3911b21691f220c5eafa09463689b8b3","subject":"Drop unused method","status":"MERGED","created":"2025-06-20 12:48:28.000000000","updated":"2025-06-27 10:18:49.000000000","submitted":"2025-06-27 10:16:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952977","meta_rev_id":"b04876427c7b43f00499ac73fbc43dd5362e80c3","_number":952977,"virtual_id_number":952977,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-06-24 09:08:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-27 10:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-06-24 13:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-26 08:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-06-27 08:23:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-20 13:17:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-20 13:21:04.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-20 13:31:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-06-20 15:59:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-22 18:47:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-06-26 08:47:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"0199a72c006aaababb689873cb92d59a562fc937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-20 12:48:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59a43c4d713cc565300d88ce1a6edf2c386ac6f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 13:17:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feb1fbe34bea4b2ab2689488d78d7351\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a35f0798357845e983e1752dce7fe32c : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec273eb3436f49d282e2f32b6bd7a7ff : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8322ed22c43f4380ada6e369698fc400 : SUCCESS in 27m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a37f91c7ff8fad65b0c5f173232007b0b5a2de41","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-20 13:21:04.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/77/952977/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9c5afcaf5e502ce2a8e1bf6de5630cc3324f86ff","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-20 13:31:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63912e9721f4056c7e2072d1d48154de234d404c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-20 13:33:16.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cdd9402b2d68318085966e3e782af5bec53e92d1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-20 13:33:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"11356b0b24cc100e806462c45d04913d1ae7fa38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 14:05:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9934b187f2704aa4b1981dd34c85c5c6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da5bcbb1d4024717a7f5403da1bc73aa : SUCCESS in 24m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/98fb7fd1ac3d4388bd053d399d6802bc : SUCCESS in 30m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e23379a050cc405081ceb4b05c5502ee : SUCCESS in 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ceac71c6debabbd2b172b0d5dd5dc20bfe2347f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-20 14:06:31.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/77/952977/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6a21e4df180133517fe5e51127fc5239cbbf62a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 15:59:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/854836e1c314462882b88af9d3e42b57\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0c393f67286f4e61997a1f6e60c8d1b0 : SUCCESS in 1h 53m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f8c3ce36222f405eb9e08bb9c08d8a49 : SUCCESS in 2h 24m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/725c6b376aa54da38e1afd3b43bf322c : SUCCESS in 45m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4652f271386241b9962d11d41d8ad857 : SUCCESS in 43m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497cedd2bc58434ab8f24a98badb9914 : SUCCESS in 12m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/835df633dcd14c51a2ecb5c1fb1c45ee : SUCCESS in 22m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5eef5b5c5667411e81107b1c58f9c86a : SUCCESS in 16m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27f0373a955c40508583e9bdac986a17 : SUCCESS in 23m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2825d1d8acc94066b383a1ab30b83d80 : SUCCESS in 13m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/635f500566564a6ba0c2755ea4d59289 : SUCCESS in 2h 13m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2606c5d9dae74b5ebcf80d0bf5db25b8 : SUCCESS in 1h 39m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/20978b0c98ca4a0691e301d3db894897 : SUCCESS in 57m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/dab93d6c0f244e139402e0ea80d55d09 : SUCCESS in 1h 01m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dd6d27ba922f43d5a84d81751adf5b63 : SUCCESS in 1h 52m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/697e1705f82d4bc28f646985b94cdb70 : SUCCESS in 1h 19m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2bcb0c75a5bf4366bca25b7573d6d6e6 : SUCCESS in 1h 01m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d094f6bcff9d4027a845f569e73a4b25 : SUCCESS in 1h 00m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/741284b1d6ec453499e0832fa3941479 : SUCCESS in 1h 15m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f98efab45cd543afbf62fd8e75f7804c : SUCCESS in 1h 37m 07s","accounts_in_message":[],"_revision_number":2},{"id":"09739848eb6e0d7c23ca79d527badec00da0e008","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-22 18:47:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8023d86616c59967b1b4b191c737542ab6aab76d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-23 10:22:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf6e9aa7d5dca046e7d1524cefcbd8d5856ad131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 10:22:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8454149576c4fccd2b992c920a25dbf25bb4a444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 12:24:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f666e06683ec4b22ac71a0105a8d3a43\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/90f3ce988c394886ad29c638b0eb7cda : SUCCESS in 1h 30m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d0aca115a34d4ce4b0ddf4be07061a46 : SUCCESS in 2h 00m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ca6e5669447473b8d451d3a84d98d03 : SUCCESS in 11m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0481ea599fb747f58aa5c26a1e2b1bf2 : SUCCESS in 19m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c7685159c394bf29d1dda426bb7eb21 : SUCCESS in 20m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b183530d9824650b3a22ae143c68814 : SUCCESS in 13m 13s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1613ef0d6a414ba1960f6193c46fc036 : FAILURE in 1h 56m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3fdca69c663a447fa75d6c62da239bfc : SUCCESS in 1h 15m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3ce95239355b442b8aa9741386a714ec : SUCCESS in 1h 37m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b093d195dc5947138cf41992e7218383 : SUCCESS in 1h 06m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c04467f6adea4ec68b4eb4c80fa53d91 : SUCCESS in 51m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/72db0132b76741d4a526f498dce79241 : SUCCESS in 49m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/18b11eeda38043438eb2b88aadfbe19d : SUCCESS in 1h 10m 52s","accounts_in_message":[],"_revision_number":2},{"id":"90ff7190dcbdba714ee57ac335a6fd87c875906d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-23 13:13:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f15d29ad755c7db1b0e1cdba8aa390b82129202","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-23 13:47:23.000000000","message":"Patch Set 2:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/77/952977/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"24f02f66562bcba4b11502775dd595c8b7418238","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 14:15:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b1e839edb4541168188de8f7237f592\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c03adbe3a9b04ff792cc652b5e863398 : SUCCESS in 36m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/893d78fea8f040509ba17ff137d76ed0 : SUCCESS in 39m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9099bf3abaed44dcb9b78f4669283517 : SUCCESS in 49m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4eda621e04953b6398b2f0f1fcfe1b8e3cc7d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 15:06:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35aefdc1a9e74210990b7397e16ee58d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/95d28f67cba341829f0a87ae0f4c1290 : SUCCESS in 1h 20m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/745162f970794ff185a5fa911310ed57 : SUCCESS in 1h 50m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5ac8a443703f4dd4840503a946bebb7b : SUCCESS in 48m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4c2c69479724f179036663b8f60e35f : SUCCESS in 35m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed32ec8f4344c21b474df135c42fe1d : SUCCESS in 12m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a79fadfa689747729cb61ab8bcd7c1c2 : SUCCESS in 20m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0fafcf38ec85419fa0b239ba03823fdd : SUCCESS in 21m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/580c8668111c4113bdfa91fb0f933c69 : SUCCESS in 32m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b43e82c974e74f56bad5f9623ed3ed23 : SUCCESS in 11m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5d3896b477904fbfbf8cd0045fb847ae : SUCCESS in 1h 26m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/a5a09a5bfb0c41f587ef526b3a426603 : SUCCESS in 1h 13m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1774030f8f884f2e8b7e119188d74f50 : SUCCESS in 57m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f0517f690274133b40b4850f684be67 : SUCCESS in 1h 06m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1c14a92d55874610aa588d51f8e170ad : SUCCESS in 1h 32m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e1a61d9a04a540c498823ab00c453e65 : SUCCESS in 1h 31m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/788f317096c44d16a4f9f47ca71c10a3 : SUCCESS in 57m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b1ba78c8160f4629aeb67f74b3b37bb3 : SUCCESS in 59m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0c9f15ad0ce847d88badfd8696f56b7e : FAILURE in 1h 26m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5f8387a45e1e4572969f54cbb059b512 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":2},{"id":"2159aa431242dc88d27aa74781e65758f0a737d5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-23 18:02:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3be71b7e040822422e170b5f1fd69e69aeec2930","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-23 18:35:41.000000000","message":"Patch Set 2:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/77/952977/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8b254435eca42908cf50305c66289f11be875f13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 18:46:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f208261d6b1b4b3387aeb4d213ffe4a3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d683074c3f5e40429705c85f3747b6e1 : SUCCESS in 35m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53f8c7e629644bceaccffd23cb6daac1 : SUCCESS in 34m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6c9c9a3b09c4f4aa59a3221bc971f4f : SUCCESS in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"165cf1be14b599cc656c5ce0ace3ab4009697b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 20:51:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b374e42adbdf41b1bf6cdc7fa156ac9f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/60503ec6c76d4fb0b7b57013593afccb : SUCCESS in 1h 36m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/23113e6ed5774712910b037e1e9ae8ad : SUCCESS in 2h 47m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/322226717d104e7cbd28ed8d40129caa : SUCCESS in 31m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b91067963f5a4c23bed94efdbae58092 : SUCCESS in 33m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b9c4d92320c4c249d9f099e6af48f8b : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bf7f0a6f1f949168bd440beb42cef19 : SUCCESS in 17m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ea387f4a386421cb48f69d1067da45b : SUCCESS in 17m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b0fc17633f348feaccdca6fcce5197f : SUCCESS in 15m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/194011d9e71e462b9071f0a7a2d2476a : SUCCESS in 9m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cb8d52b90bcc450e863143264e2d52d5 : SUCCESS in 1h 22m 29s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/5e339df4a1104f4082ce38155be66482 : SUCCESS in 51m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f733f3c27234eaa80110514e3040bb7 : SUCCESS in 1h 01m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2286944e6364edea3762ebba3e09fd5 : SUCCESS in 35m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c291d437a5f74a5bbef6e2a158db5dea : FAILURE in 9m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a3ec93372f994c588ec05b34d773effd : SUCCESS in 1h 37m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f8add9f91201419c9178f7f467f5bd84 : SUCCESS in 1h 00m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/29e1bdb9df4e4a9ba583d8f5a6b3759b : SUCCESS in 1h 04m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/99806bfac8eb47ce9d2b069a062bc98d : SUCCESS in 1h 18m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4c079c07aa9c4e47806a3e7847aa4246 : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":2},{"id":"8c875df74f454e00e1251e54cba57374979b03fe","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-06-23 21:48:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9c406a3e713e88ec3afb3a71515f373b2b8994f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-23 22:21:56.000000000","message":"Patch Set 2:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/77/952977/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"57d37aa52b1fcb8265a82cc77b7eb55d1059e190","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 22:32:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca213c277c7e4a87aa135fe88f51e53a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3762bb21c8294b36aada471a97d9fdcf : SUCCESS in 35m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4ef7b802945240448c2a72a98290e357 : SUCCESS in 33m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05e8427cf2374dc2a91792d2d2ea9590 : SUCCESS in 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"943e2fe9aa8918773d638d3ec4d022bffaab79a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 23:48:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/931e646c1e094c0ebc980a82664e0e27\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e61515f442b4c9f96fd55187637f7f6 : SUCCESS in 1h 41m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d0b87b51ed0c416183eacd078823a62e : SUCCESS in 1h 20m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec75b53759c3472a99e4283791a839c9 : SUCCESS in 34m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d96552ccb044dba9c6c49cf42ef7203 : FAILURE in 31m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/773c524da834445595e9cb90c309e720 : SUCCESS in 10m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce7dd25c40694f39adecb65681c540b5 : SUCCESS in 14m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a6c7c11ca8243b4a24786159e786c7f : SUCCESS in 16m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84106ad053e34d09ab3b8bec51aebf6a : SUCCESS in 29m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b50030501f614fd8974792ec1456d416 : SUCCESS in 6m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d65b8cc9635040c2bf5a33a8fee7750d : SUCCESS in 1h 42m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b426aeb9fb534f68a371ff24a292d25e : FAILURE in 1h 27m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fac6cbe2e47b4b20a61b49446fbb28d6 : SUCCESS in 52m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf5d1910b794499596ee7a1c042df8cd : SUCCESS in 1h 06m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f30cbce9225846dd9070cc77e4af2c4b : SUCCESS in 1h 47m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/40d8219d1f074b8ba21f9ed73ec9f3d6 : SUCCESS in 1h 29m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/129e58582cd04e7380680c5eeb347bf1 : SUCCESS in 1h 08m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/31c4ef181cce49ccacc98083682acd85 : SUCCESS in 56m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd57e37592ba472d8c0d996bd92d56d8 : SUCCESS in 55m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9eeb149cc55945f284acbe0d9deca327 : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":2},{"id":"65e231f15a509460a664e998de88f00852a4e0db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-24 08:34:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"77d37bc472428323cdaa35aad6024e183567c0a5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-24 09:08:19.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/77/952977/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f38303353ca6848751c891ecf02d627d001d8def","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 09:24:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3a3021d9cd243a1aeecd429d85da2c4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc2cee1745db4f0bb91a054fc25bd708 : SUCCESS in 41m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4033819c633d4a009bcd70b4d8ca9897 : SUCCESS in 37m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fb288792714c4a16ac1a2482b6e2f3c2 : SUCCESS in 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"304f284736592d7bcbf4450d48ce359dff7ec4eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 10:47:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0848460437b4d9e9a3d64f8f9ed0fc4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f8d4b62fc914acf9a2a70352518b29c : SUCCESS in 1h 50m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3abbc9a6eea0485396ace4cb850ce801 : SUCCESS in 2h 10m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0e487da40f0d40e39f29758a77458db6 : SUCCESS in 45m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de6140bb4d52454885e64f6b222d5cdf : SUCCESS in 37m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8dbee2229db43a59e9a67bcb14429dd : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5417a990204843ffb63ffb48bcc9cd41 : SUCCESS in 14m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/800aabe76b8d485cbdd8d97442a59527 : SUCCESS in 14m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77aa647291604283b177682da2fe6405 : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad388c26c984932a6a878c274eaf05b : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1f725765e3947c79564348f1e6f6bf9 : SUCCESS in 12m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c453ab462f1849a6a61fa59f72b68cfc : SUCCESS in 1h 37m 08s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e5d6b5a6e3564a2699bdfaddfc99310c : SUCCESS in 1h 36m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0f229a6cd48429c9ccddcfc4bf5aa89 : SUCCESS in 48m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1381f4b5f1504d8fa1d3a85001d37bf3 : SUCCESS in 57m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/42205b06d7854748ae996d5e4ff89eac : FAILURE in 53m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2ad59225c5c343d5bfbc82e5d0048f47 : SUCCESS in 1h 23m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ad553906c634494fa077ec354b5dce99 : SUCCESS in 45m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d37841e9f26f453f9dee7251366e27fc : SUCCESS in 1h 01m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd0126de624a4bd5a448f073b4eaa0e7 : SUCCESS in 1h 12m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/62f6488538554712bfec2b966b209037 : SUCCESS in 1h 17m 44s","accounts_in_message":[],"_revision_number":3},{"id":"23b9fff86b47f216b52057558670ed95f14d208c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-24 11:24:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"effe9e73b18fc908082cda557d8271abfbad9298","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-24 11:57:32.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/77/952977/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"53f60ae8ae6a4067aaced0862a9abf39ce979550","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 12:03:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/302c097bfa0c41fab3566bc5f37f9239\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/675c73ac5592472cb40cb559420a16e6 : SUCCESS in 37m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d2a62f8f967441fa45fd4fb03653456 : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3bb282c6a9564a3abaa3da5ac316a3cc : SUCCESS in 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c732ea30ddaf05f9f2e2036be54976d3b19c996f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-24 13:01:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"87fb0c82f0c9bb7affa7045cd6c117e95d70dd64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 14:07:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9fa2920500c4928a56e69dec493c752\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/07604da8e8f84d5da31ba237b1f367fa : SUCCESS in 1h 51m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b0b419b7ece4b519cbb17554bf754de : SUCCESS in 2h 41m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9853125a44fa4588a7fba022e1dc02be : SUCCESS in 45m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f66c7500e5b46ef9115f8e615714d0c : SUCCESS in 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/008a344d8f874fdea4195a5d6ed6dca1 : SUCCESS in 9m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/227f5c589c8f48a8b08caecd70b7c08f : SUCCESS in 20m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8c1e76a70904ba98b4dfd50a34d0304 : SUCCESS in 23m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/979f3b6309244bf19d2db3643f4306c6 : SUCCESS in 24m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/952e542a089d4b689052d7c45b5176d0 : SUCCESS in 14m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faaa5fbf60b84e3399cf085698ff16d6 : SUCCESS in 22m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/098ad63a06fa40a588ae64e9b68cc9a5 : FAILURE in 1h 16m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/3899c5e473ce49a6a0fa55e491448dac : SUCCESS in 1h 23m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6596de71c4dd42308347f42910ccec19 : SUCCESS in 49m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/05013b3abbc947abb236fd34478be0c1 : SUCCESS in 1h 09m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/55fe076f97ad4669a63cd3abd7eef08d : SUCCESS in 1h 35m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a1135ec5a1bf4c879aa627396a9c6ab0 : SUCCESS in 1h 35m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b2303298edb54f4aacea369a70cbd9e7 : SUCCESS in 46m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3e004f16bc3844d895a9957870ae2e91 : SUCCESS in 59m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/99ec1670b8f44927b24cb31d97f881fb : SUCCESS in 33m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e9dd5e132e34ed1b56a08a4dc30ee8a : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7d1351b092f52a83d6c27c0993e7cd0b473696bf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-25 07:37:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96af6ac406aa1331a45c51676983138cdc7b9232","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-25 08:09:55.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/77/952977/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"42d52b9df0dee0e716b1c274e79f842556a5117c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 08:59:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/592526c7aa234a329cd2e35f805771dc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b50b8b0015eb47b991c1239660c350e2 : SUCCESS in 43m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/59df07c792b1425e9fcd6848face16a5 : SUCCESS in 37m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/685673f23b234a74ba60fc43f4fd52b0 : SUCCESS in 47m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78da983fb77ab42d2f246e4191788b1cb8d81d79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 10:01:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/700ae1c8b1b54347a4ab82176f56afe0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/655274b94d6341f3b4ffb1d391268fb1 : SUCCESS in 1h 40m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7c2f6de58e474c77befa656320a7220a : SUCCESS in 1h 57m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d172490adaa345c189cdaf84b95a1907 : SUCCESS in 46m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/853d549e8ed8400eb94834757bc430b8 : FAILURE in 45m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38c70ea3efe249a7aac35c7f1011f0ac : SUCCESS in 11m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/805b1e2d0cf842df848a107ca3b077ce : SUCCESS in 21m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d608bc26e5b847879675368419789fc9 : SUCCESS in 19m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2be5c5ab9b1f4274a8e0d37bfa6dd61c : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f902921a55e44e109304b81f7ef8edaa : SUCCESS in 13m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b03ac7e6d9e14846b61d27f353a0ea80 : SUCCESS in 19m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ba0ea9458f454423b8c2ddc50d9d99ba : SUCCESS in 1h 26m 32s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/26d137237cb040148275db6fa8e6b840 : SUCCESS in 56m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/30863291de5c4b7084191d26bafcaab5 : FAILURE in 13m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/14a2209aa23e4c3d9495805e026cfaac : FAILURE in 20m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6b2466460fdc43f7997b3c971990e135 : SUCCESS in 1h 47m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/df56eeb5edec4c43bbd53b88464d36fb : SUCCESS in 1h 24m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a15ea3f4da414007a1bb5a65aaa79d80 : SUCCESS in 1h 06m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/179eecf4ca064553ac7fcf26428d14cc : SUCCESS in 59m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8d23f46d81ef4922aac8e60b5750db91 : SUCCESS in 1h 20m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/10a7ed294d3446d19b4939817aad8161 : SUCCESS in 1h 13m 53s","accounts_in_message":[],"_revision_number":3},{"id":"3c0ea324e6cb745cc18cab86fb99665267e15040","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-26 08:47:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"09406d3277db38fa28043601df3579332a473596","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-26 08:50:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0392c242b4517d5a09261db734400a65ec477704","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-26 10:22:13.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/77/952977/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"52ef2bb0444c89472128f8ab095dd065710b9e2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 10:55:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38089fc7e6f94dc6a6428665a695a038\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/68e7eb0355354416baafcc4ab5391e6c : SUCCESS in 1h 22m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3a87a66e3cc04657b98a9bf64d18e411 : SUCCESS in 1h 49m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0eef786946d54303bc215bfb9c559af7 : SUCCESS in 32m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62d4deaba83245de8a2a55b7cd21b90d : SUCCESS in 41m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae3783f70ab44dc919bcec9968afbf2 : SUCCESS in 9m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94f2fab138344fe793d2d5a132dfac2f : SUCCESS in 20m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77b630f6594f4b0995fc36ee027f0ce8 : SUCCESS in 19m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d397bdff9ec46c9826bea2c58846f90 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f057226c1e034b81946ea80344d930d4 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da94fb55487b40d7bada13ed2dd48d09 : SUCCESS in 16m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d5ef1fd5691444c38b827816027cf673 : SUCCESS in 1h 50m 33s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9a001b813c2344c195885affffa5195f : SUCCESS in 1h 28m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/67af9e7aca984202bd96dc118d0ba4b7 : SUCCESS in 53m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f463430afc744b56bba9229f1368f2d3 : SUCCESS in 1h 07m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ea5bc9f1c2f84e6abb82a71c1542a4a2 : SUCCESS in 1h 53m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fa814ae1b71847789129a8877d78f8d4 : SUCCESS in 1h 30m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/edb87d5f251d4d76bc6a8da685727874 : SUCCESS in 52m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5bbe3d0e55ca4da9987c07869c1d1144 : SUCCESS in 52m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ee140d0a03a04ce0848dc4f21cd82096 : SUCCESS in 1h 14m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/09a77294f7ed4f80831ed0f114941d4c : SUCCESS in 1h 12m 33s","accounts_in_message":[],"_revision_number":3},{"id":"b9cdc18a260097bbcb91b61b6ab46753ef49ad68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 11:03:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/033d5bf247b2424a943a365bb19398de\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc869c4509964c41be470f251cd10c14 : SUCCESS in 41m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80bbade2c49e4e4584d7c9d7e1dc2760 : SUCCESS in 40m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f364f39609a4401c9e7e17a37f9fe48f : SUCCESS in 50m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ed5fa63048c497efe1b6857bfa81ac6a32bd1bd1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-26 11:50:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9539f452597d7900d5d42d723896c8ac7f42184a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-27 08:23:56.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7446ebed66429a78389e39c2fc432194cc17df7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 08:24:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6614677eae79ed5847fc99166e671ab16d71c304","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:16:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ea215dda77a4ac7a411e5e67a0c0cf3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/036ea8c6958e49709bd3264becbe0749 : SUCCESS in 1h 27m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/01902bbe41584b28af56a5368c53f38d : SUCCESS in 1h 49m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee904726cb742daa55f2fb0cd440f5a : SUCCESS in 11m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2b5796f24ad49418a5927c30918efa7 : SUCCESS in 22m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4e83fa316264e949752031cdd7d3a89 : SUCCESS in 17m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142e495c62cd4e42b12dabbc0b6cfb38 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ac6bcffc5a146d8bc40b1275e04a44e : SUCCESS in 18m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ffb70a64ac4542c7adaf5327cca45ad2 : SUCCESS in 1h 25m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2c36e7eedbd746a88a983b261aa8779a : SUCCESS in 1h 22m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3e188af0bfd94500936d18e1451b61b1 : SUCCESS in 1h 48m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c553d50c2e2444a8e3808ddc651410c : SUCCESS in 1h 13m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f5fa9d18d584023a166bb0486a72c64 : SUCCESS in 1h 00m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5d4182372b8d4629bb66ee6d7211bce6 : SUCCESS in 1h 04m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/073bd3a0a17c4f1c85f98ebe92c74776 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4f2813a5ddab2dd71c9d416e78f4aa08afb76ae8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:16:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b04876427c7b43f00499ac73fbc43dd5362e80c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 10:18:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cd67b380538497bb8b9a7bfaaa69152\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b368d383dc4125943cd1701ff43aad : SUCCESS in 1m 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/342d663795fa416d81086d3eed85a5be : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fbdd642df6d31a8d3762df0b3c67fe8ce8d1a23c","revisions":{"ffc60b8a44ac4fecb84f249a7b35ad44241b7550":{"kind":"REWORK","_number":1,"created":"2025-06-20 12:48:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/952977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/952977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/952977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/952977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/952977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/952977/1"}}},"commit":{"parents":[{"commit":"39b183645e071f9c2926f0bd870723fb6e300cb3","subject":"Merge \"[eventlet-removal] Remove \"logger\" mechanism from config examples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39b183645e071f9c2926f0bd870723fb6e300cb3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-20 12:47:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-20 12:47:40.000000000","tz":540},"subject":"Drop unused method","message":"Drop unused method\n\nAs described in the TODO comment, this method is not at all used.\n\nChange-Id: Idd9f4ade3911b21691f220c5eafa09463689b8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffc60b8a44ac4fecb84f249a7b35ad44241b7550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffc60b8a44ac4fecb84f249a7b35ad44241b7550"}]},"branch":"refs/heads/master"},"27536b3af0bbb0fb62b9cb3603aced6dbb59c353":{"kind":"REWORK","_number":2,"created":"2025-06-20 13:33:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/952977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/952977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/952977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/952977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/952977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/952977/2"}}},"commit":{"parents":[{"commit":"39b183645e071f9c2926f0bd870723fb6e300cb3","subject":"Merge \"[eventlet-removal] Remove \"logger\" mechanism from config examples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39b183645e071f9c2926f0bd870723fb6e300cb3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-20 12:47:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-20 13:33:16.000000000","tz":0},"subject":"Drop unused method","message":"Drop unused method\n\nAs described in the TODO comment, this method is not at all used.\n\nChange-Id: Idd9f4ade3911b21691f220c5eafa09463689b8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27536b3af0bbb0fb62b9cb3603aced6dbb59c353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27536b3af0bbb0fb62b9cb3603aced6dbb59c353"}]},"branch":"refs/heads/master"},"fbdd642df6d31a8d3762df0b3c67fe8ce8d1a23c":{"kind":"REWORK","_number":3,"created":"2025-06-24 08:34:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/952977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/952977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/952977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/952977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/952977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/952977/3"}}},"commit":{"parents":[{"commit":"b8e21fd54e18efa9134aa24dbf40f57786993d1d","subject":"Merge \"[UT] FIx no_dhcp_release6 test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8e21fd54e18efa9134aa24dbf40f57786993d1d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-20 12:47:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-24 08:34:34.000000000","tz":540},"subject":"Drop unused method","message":"Drop unused method\n\nAs described in the TODO comment, this method is not at all used.\n\nAlso loose the minimum boundary of coverage percentage, because\nthe limit matches the exact percentage of current tests and can be\neasily broken.\n\nChange-Id: Idd9f4ade3911b21691f220c5eafa09463689b8b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbdd642df6d31a8d3762df0b3c67fe8ce8d1a23c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbdd642df6d31a8d3762df0b3c67fe8ce8d1a23c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
