)]}'
{"id":"openstack%2Fneutron~952988","triplet_id":"openstack%2Fneutron~master~I64ad033e600c2c87af4716736c10e11c143afea2","project":"openstack/neutron","branch":"master","topic":"bug/2084446","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-08-29 15:27:50.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-08-29 17:32:31.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-08-21 20:59:49.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I64ad033e600c2c87af4716736c10e11c143afea2","subject":"Delete tunnel endpoints if endpoint agent is deleted","status":"MERGED","created":"2025-06-20 13:39:26.000000000","updated":"2025-08-29 17:36:24.000000000","submitted":"2025-08-29 17:32:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952988-bug/2084446","meta_rev_id":"51915411b73b47299d5b5013c405754b39f655e6","_number":952988,"virtual_id_number":952988,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-08-21 14:07:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 17:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-08-29 15:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-21 20:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-08-29 15:27:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-20 14:27:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-20 14:39:03.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-20 16:08:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-22 18:48:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-07-07 20:00:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-28 11:50:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"0c83aba8b99a66652223c8d973ac2e2fb41e16be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-20 13:39:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af887aabcc22824b2cbc0952defa233b7b95010b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 14:27:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d779f805514fd38e0f7f52d8a6feb1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a654778a04f4318845861cb96a7ef88 : SUCCESS in 39m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c54e636b7da470eba6be23d65d2901b : SUCCESS in 37m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4b472004baa24fdc99d9b552b36c6ba0 : SUCCESS in 45m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f437b5ad4918e7ce1c6d0a81713ff5e062d7c62e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-20 14:39:03.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/88/952988/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b9dbd73d07bc4774206eeef3a591dd0ddcba3722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 16:08:35.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/426b451822984a4fa3190b47f8605af0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a7244664975944cca6d5f42fed2bd0f4 : SUCCESS in 1h 44m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ac1e52add4b3407293597d1087b70172 : SUCCESS in 1h 55m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ed3642485844c509699a7547e650967 : SUCCESS in 43m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/694e3e083b0b48c189c0c4e0cfbbfc0a : SUCCESS in 35m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff618de5210a4670ae986298f37eed3b : FAILURE in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc44036afd4f40e3a96f7d3484aabf9f : SUCCESS in 21m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41fc2f33135d48e99b5362dd4f4f11da : SUCCESS in 20m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8166761ab3674940839ec7344b80a99c : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b317646e10472b8b57eea84adc711a : SUCCESS in 12m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/59cf7cfeecd34eb990ff92875b80b6d2 : SUCCESS in 1h 35m 25s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f7e6698cb8dd44498fb697223882004c : SUCCESS in 2h 26m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/71fe3827f9e64f6c9f13fcec8982e139 : SUCCESS in 55m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9484f5c5936545838b56a669f690bf89 : SUCCESS in 1h 12m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1ef46bb51c3848bc811cac0aa46f4cda : SUCCESS in 1h 59m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aa4e5cb7f8634083beb85984d18e8cfb : SUCCESS in 1h 40m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e68a4ed844bd454788d0b664a523a562 : SUCCESS in 1h 11m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1f0fe5d0ac7d4d258d8dd6b60b5f5753 : FAILURE in 1h 04m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/19e002bb72d549769928d4b1770999aa : SUCCESS in 1h 13m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d19c7d27b874cd9ab7cbf38f5f0e5b1 : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5abae4660af4cd6cc00dbe8aa0beb5faf3fcf919","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-23 10:00:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c5d3db8fd42a6f0919bca1aa6a5245adf302e112","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 11:07:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f7cee37f2da4a8783767a6dff7f5b5b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/917006e30b0e402f98140fcbf63fc038 : FAILURE in 40m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/673463577f9448b29bff76f615cd1f0e : FAILURE in 23m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30537f541fb24ac696ba0bcd3a5bae49 : FAILURE in 48m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3471b80bc7358d46c886ca94465bb1bdfcd585b5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-23 11:49:30.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/88/952988/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"f8bee8caa5acac44ec0858878ec1f26fd6b24095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 12:50:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99b86161d1834f5a844ec3bc7c252ba1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/07318efff60b45cf8206c18e6e1a3c22 : SUCCESS in 1h 53m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de3b372124574406a7b8b211caea4b26 : SUCCESS in 2h 48m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e593a679838449789d86ff6bf4c7b1cc : SUCCESS in 41m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d452483e5f02461fad2bf7e76f02de35 : FAILURE in 40m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07644597bcbc4c709983047832b65d01 : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a5561f453234d8897f3ea5ce491ff57 : FAILURE in 16m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba23e64187d64fb1995ec57f397c122a : FAILURE in 16m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33d45ce71d4c4fd2bf2e4a2252465598 : FAILURE in 22m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5247c66b603b49a1a007d50603823d55 : SUCCESS in 10m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fc22fe39674a4f6e98596e8046c35d64 : SUCCESS in 1h 38m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/395392a5382e42179f1600fa1f20481e : SUCCESS in 1h 23m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c4029b92f5145e0a4b392c886e4493a : SUCCESS in 50m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5cf52aad9fb84024b56688d7d6a154cd : SUCCESS in 1h 11m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/615afd42cec74df1bbe2dffc6e9c1dd7 : SUCCESS in 1h 37m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/950e0ca2b150486586e72da9f0d9eac0 : SUCCESS in 1h 32m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f2a76b9097674ba4a0184d6274d7dacc : SUCCESS in 52m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/eb2196ee84d049398de686d8378ef605 : SUCCESS in 53m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0817a3985ee4744b14965f3895b96c6 : SUCCESS in 1h 07m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e2a6b4eb29fe4e8daeb55f31aeb2155f : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":2},{"id":"67ce9afc85397e014709e53da6afed54c4bbdad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-24 15:09:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dfc455f8b3915a037a92f48613f6b8e54d446ccb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-24 15:43:55.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/88/952988/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"de15da63cf7556075e93564976bf317677c143b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 15:59:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/562b1514efe846a4a7e70c735dd207ed\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e35f0ea9043417ea9af7c239fb4d007 : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42aec0b25d4a4b79a1e5004dfbca0ff9 : SUCCESS in 32m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7d1070fe4f6141ddae125c133520273b : SUCCESS in 36m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f626a8a5b2618fee91a24f3ae18601aaeafad9f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 18:20:48.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7451b38dcff2447980966de3fa28602f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8d3cf88624504a41b2f2035a830d8b64 : SUCCESS in 1h 02m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9fc1dba152904ce6820bdd5cffa4c308 : SUCCESS in 2h 28m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/740e50bb96d2484793b23871ee917cf2 : FAILURE in 11m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/517410b2861246d491912ae8e3b8ea05 : SUCCESS in 37m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfd8fbe0be2041e0a13818b939df0d5d : FAILURE in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b05d5490e0a441b7a2e50b23d3988285 : SUCCESS in 20m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0375bdfdb43a486682e4b7f40068d21d : SUCCESS in 17m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c5caaf257ce4816b0565540348da7ac : SUCCESS in 30m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48456c6ce45f42889ab789cef81f80c9 : SUCCESS in 12m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/99ba35e6c09a47abaac85ab9b40197cf : SUCCESS in 1h 42m 35s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c11a9a4c5bc442ea83ff26cbd0c7e78b : SUCCESS in 1h 26m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f69423700ec64df3a294138bd12eb33b : SUCCESS in 50m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a445cdf2c424e6cb5912fa6cecca28b : SUCCESS in 56m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4eeb9818c30c4d67b2560dcc92c611f8 : TIMED_OUT in 3h 03m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/661992a57a124f7e8553f53b9fcf0438 : SUCCESS in 1h 33m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d14e8860bbd54e1e948408f0365c4fc8 : SUCCESS in 51m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fd00ffab29264720be9fde5dcbc63afa : SUCCESS in 57m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d613aac31af04534a540e5ae6cd4e151 : SUCCESS in 1h 16m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/407a1859fe784a508628c7a9184365c6 : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":3},{"id":"84c1bc5dce936d867fab0feb04d5a3a7c79e7e4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-25 08:07:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f0a3533297dea4632ef43a3b9a531ae5d22efc1a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-25 08:42:06.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/88/952988/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"d04d7d2d21a1c61674b53d1bda8867010c8e487a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 10:36:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ec4af20a7494a2aaebe8bff065d1f68\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/59da3413ebb54be1be4ef1714a8eae52 : FAILURE in 10m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b6105f68d0414a4caade192fe7a73e90 : SUCCESS in 2h 23m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89a490cf20f446d594abb7c76308370c : SUCCESS in 43m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3734f3769b8b4455899d17abf1a1988c : SUCCESS in 31m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876aa0531a2a4821bd0d1d04b6c0e39b : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17b76df1168a45ae8d549de39a01925b : SUCCESS in 23m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ae077775cac4cbcb960c1e2f91fbd7d : SUCCESS in 20m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64d91b1791524df8b7c3e2600397fd5f : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8a9065192c4abc891b9682b42e1cf8 : SUCCESS in 11m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/28b7c71573794a2a988f8eeff092758c : FAILURE in 1h 14m 16s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ba4a3c73a9b14d10b54e0d8cba6e5efa : SUCCESS in 1h 59m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/85e735059e4a4fb7bb4ca397174b642d : SUCCESS in 53m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/edd68f57559e469dae7a78aec01d74ca : SUCCESS in 35m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e9d0ddc2082a4707bddfdca825a22b09 : SUCCESS in 1h 38m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/21c078d5982b4dcf851b6c804f5ea332 : SUCCESS in 1h 31m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a21915339626425fbbd51149b0b336f3 : SUCCESS in 56m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/148832e3e1a047c5b31829fc55c63dc7 : SUCCESS in 1h 02m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0ff0931c9ea41e4a3ad0bb0ce239b8c : FAILURE in 11m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/696a4173de374562b10eaf27bc5d59f4 : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":4},{"id":"ff7a58bac6be7d43ed9cc482b2dd78b222d9bf82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 10:59:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec18fc53e56f47debd016c7b9e3acab0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2ddb4c4643a34585bc1c1c377dfe9c05 : SUCCESS in 42m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99430951d9fd4a989d69d0243d21d410 : SUCCESS in 40m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/356dd4ac7ea3459baaea108a487c6faf : SUCCESS in 50m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3f2defccdcf2bf99a822f4cd9d1af0383347fce1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-26 07:05:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"efd269a91d0ab43aa4e098165d32426e6b54eba1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 08:02:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0699d03f3d0498596dd47ce36a5e1c1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/657159103b66486395f3f74e13dd5b70 : SUCCESS in 50m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f49e1441ebd460fb1c2cd4a061c57a6 : SUCCESS in 47m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f7784c3e95646739397e03c810a0ea1 : SUCCESS in 54m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d74916032939834784f4e2ab126795ecb484ff11","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-26 08:11:08.000000000","message":"Patch Set 4:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/88/952988/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"0c6962edb398327d2b904602b426e5b5afd7f783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 09:27:43.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b00f0be866a14bfab9c3755f670322ca\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7d402971828348b7b04dd50c6c07fab3 : SUCCESS in 2h 06m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e1ab6b2a4c684f8383f94a885708a721 : SUCCESS in 2h 17m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e253156b441b432a89e5086de6532c6a : SUCCESS in 47m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc23f8d9c2b847eca6398de295643da6 : SUCCESS in 33m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7384ac0d63c64fbf97215cbcc580d926 : SUCCESS in 12m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/727a821d610a4ef7a3200e7a49ef9965 : SUCCESS in 20m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d72f54221b904a61885cf8a0e298305b : SUCCESS in 18m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/896d28fc00484268991e45b85d5f4fdd : SUCCESS in 23m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9ca4f08610411bad126283ab60d309 : SUCCESS in 11m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bcf14e163cf346e1a4c0a3d73f7da60b : SUCCESS in 2h 04m 53s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/935e273c28b24b3ebedb62fa1b308e86 : FAILURE in 1h 10m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/32d31133571d4011ab8b5add09339353 : SUCCESS in 51m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8d241fd0d4f4b0bafae2c535e268405 : SUCCESS in 1h 09m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3e53946d38c944bdbc7d874f9bfaa0d6 : SUCCESS in 1h 55m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a626163eabd9440ea1cc75f32a1818e5 : SUCCESS in 1h 24m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/524d7cfdcb7c4445aea28bb22e12ef68 : SUCCESS in 54m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/27f1102f4267475a855219657bfaeb87 : SUCCESS in 1h 02m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f7569da2c0d4c779643f57d2b28fb2a : SUCCESS in 1h 16m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fa55c688d84b4040ae7d208864040b05 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":4},{"id":"3673239d3a82a7d76662ce73408761b2238daac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-02 07:04:38.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"39d107626c9baf7e004ee4c8ae70cad4978ebf44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 07:40:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e4be3cea22c4208b47d1d36dd18f3b6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/923e78d74633464cb666e36184142ba3 : SUCCESS in 34m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/17ab1848a55a464f8eeb9f52898b9f4f : SUCCESS in 32m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0e7e6bc481ad417aa4b2c68dff538d21 : SUCCESS in 27m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9de1024732e9ccd2c6d9f1de7d42232817e49530","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-02 08:48:10.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/88/952988/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"00d0e153c6e657381712e03dd45e0ef66dd96699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 09:17:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35c741b649f94143b0a447b623477899\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3975f3e6cc90432e899ae60da2b7b217 : FAILURE in 19m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/58006d01cb884ff480993042636b6857 : SUCCESS in 2h 03m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/640c637d53b441fdab5b1f9e18fb4887 : SUCCESS in 48m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f828e2ce0bbe4b8c994d126d9cab909e : SUCCESS in 44m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a6bad060ff48f4b78e4b91c1363072 : SUCCESS in 11m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c70f68bf3a94dcab15c2197f2043da5 : SUCCESS in 23m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/384ad2b405194d7ebe47c84227282519 : SUCCESS in 20m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35bee89db89848988cfbbac4a1096d08 : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e770d39e374190889b4a6702c3a030 : SUCCESS in 12m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9656768a105443e0b33a3fa9a03e3ef4 : TIMED_OUT in 2h 12m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/88211e6904de4d178a69cf68a3229161 : SUCCESS in 49m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/107943adefc64d40bcd26edce3c3a28a : SUCCESS in 1h 04m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b5ad45dc94c844d48c4f9e0bb4863a5c : SUCCESS in 1h 44m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d1cdf2aa76364c96bfb0b8368677d474 : SUCCESS in 1h 24m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/af64c64e0c4547c9ac7c175ecef79913 : SUCCESS in 1h 05m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f06ce53eb6e840fd9d2127ef7d31f5d8 : SUCCESS in 52m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e19d80c4798b486abd255b089fcbbdb0 : SUCCESS in 1h 17m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ba6648cb36004baf818d081f8e9fed9f : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":5},{"id":"b7aa546480e5524a17b66e70d1418c6b4c00911b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-02 14:35:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"283262d1bbc9df2c3c159f0096a4d563a0ac6764","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-02 15:44:24.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/88/952988/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"27028701de6fbe89fbb3be93a9cede400841e343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:38:42.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44af3c2e6d5d4d08a77b10942299b5e0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c3fcb051691049599b61dd48718afc41 : SUCCESS in 1h 15m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1388b4e787a942aab1cd3d9fa08da716 : SUCCESS in 1h 46m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a24e56f789034815aabde6704fff6475 : SUCCESS in 24m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/122ee9c4f4134c0283461b0f636d6233 : TIMED_OUT in 1h 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/850f9c4ec32245aca6a3152c11ab3b5d : SUCCESS in 11m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/748ce17d33ff4564a06dac3ed3a60f0f : SUCCESS in 17m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/084e61a65ca04b54b3f9b9e423532142 : SUCCESS in 16m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b02fc8b7ee5048b5955ebd0c85ad47d1 : SUCCESS in 18m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69829988d584ec9a5654380ae373d0e : SUCCESS in 10m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/89942c79f26a4775a8f5d801f6ac1d28 : SUCCESS in 1h 25m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9ea6975518541a495b7b7eaa2a89be7 : FAILURE in 53m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6c3f89fda494942bfd5be2139fcaa54 : SUCCESS in 1h 04m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/45fd3ceae4fc40489df9735befeb6a25 : SUCCESS in 1h 45m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0396c6f411a84ea280d2f70c87cbb1c0 : SUCCESS in 1h 37m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a8be4c8dd23d4672adb3ff3a4d8278cf : SUCCESS in 1h 08m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dbe8c3d404c1479aad5fadd79b95ebdc : SUCCESS in 46m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9bf30494485a4da38fe9329187c1a78d : SUCCESS in 1h 04m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/355117c5f4e74411b3e5fd153bd33e66 : SUCCESS in 1h 59m 10s","accounts_in_message":[],"_revision_number":5},{"id":"b6c1a53d0fd22817e98669ed404d4d61df17ea53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:57:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79b7f615b8724fd3a39f7ea39d915285\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/09f5e860d1ff4d52a8dfc37c45a81f9b : SUCCESS in 46m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b59b71953604e3b8d457c2038a657b2 : SUCCESS in 38m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ac9bcad59b1c4d4fa67d1c1fc66ac9be : SUCCESS in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4abeeae4fe259936a6fff1ba71afeab4477026b6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-04 07:43:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0af9a6cfc27b33848b7ca8250bb123e94c01b315","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 09:15:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3065a6af5af14aa2bac3848aaf0910d3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7f0107a29c5465e9082da975e8aea6e : SUCCESS in 36m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e48fd9b91b9b4cecab21f89ed24c8f06 : SUCCESS in 21m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/737260e4ca024b82a363b098eb1c9d2a : SUCCESS in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"624f85decce6709bea6342ba553e1eef7de94c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 10:45:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/955c75a5c59e41a183572507715830ed\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8e38b566a2e548118483c7e376e0e878 : SUCCESS in 1h 55m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9d52769b01694ac290052920620aac59 : SUCCESS in 2h 54m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9adc14d899154639a7eaf18e34d296c2 : SUCCESS in 47m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5ab44f13670459ca5ec62bc67c563a3 : SUCCESS in 33m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d349f20f81a44acbf66e3cc9a6d29e6 : SUCCESS in 9m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c01be2482e041e7abf1c87ad02741d5 : SUCCESS in 23m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f03c0c27997547bbbf577c66dce9e708 : SUCCESS in 19m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a14e45323166488fa6a7696655e6acaa : SUCCESS in 26m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba72d10f8da54c318eee2cfcaf14117b : SUCCESS in 13m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/28eadd4d937948c8b2d0c074c0f4a51a : SUCCESS in 1h 36m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a3ec96dc3b747759831504cb22b4379 : SUCCESS in 57m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff5b9db10ee04e57aab21621b05ad3bf : SUCCESS in 1h 03m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fed01a1c052b414e96a42f7c9357d851 : SUCCESS in 1h 42m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b5a7899c1233495582963c83b3861485 : SUCCESS in 1h 24m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/61d5223169904b6da0db26d91e11a5e7 : SUCCESS in 1h 17m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7289ee6108e840b38442fb110bde1fa0 : SUCCESS in 53m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bebd30f311e1495494212ab814295ec5 : SUCCESS in 1h 12m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/00ff26b78885468ba64a78fbcf8f3361 : SUCCESS in 1h 17m 57s","accounts_in_message":[],"_revision_number":5},{"id":"3835884545161e5384d1b079b66f796e11618289","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-07 20:00:24.000000000","message":"Patch Set 5: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"df4fb78abaa5a8da5d13a91cd446e8a38f740316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-21 13:59:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"29ae76589223ed5a08fa923f73509e5a8daff2b4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-21 14:01:23.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4d0c1433037541acaffbe3f9d97d97681cc4e63b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 14:37:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/281ab2f6c1b3490586e584f2bb95be39\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e70fd2982f944119115fee5dd9a3bba : SUCCESS in 35m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/827f20bcf65441f183617830f8953960 : SUCCESS in 35m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/edb6148029604ef6929b72fc1bf579f1 : SUCCESS in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e458528c9e322084ac777cddeaf19d0e81bd662a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 16:04:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e62d41d590e249a482a6fbb715671b74\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a6308cc33edb4993adfdd0092a732c6b : SUCCESS in 1h 26m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/79668eb52d8f4375be5eab92be8e2333 : SUCCESS in 2h 03m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/71d466e83f584b62be7c7e9ecd124c12 : SUCCESS in 34m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f7b004542eb40f0b880d4852b9897f5 : SUCCESS in 34m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23e7f53c884e4d578be1486bd9719ebf : SUCCESS in 14m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1b9e7707b5f4f78b6fe84044692f67c : SUCCESS in 18m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a377e3cb0324e35b14e3915f589e597 : SUCCESS in 16m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba0490cc64a1483f8f93eab826d6513b : SUCCESS in 16m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd9750f69da4e10b7dca334603645ea : SUCCESS in 12m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2e3baece7e3f41cda263a97afa5f64df : SUCCESS in 1h 25m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/11878c6104334188b0dd9b17afb1f9ef : SUCCESS in 59m 06s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/83a6ee8a47fe470da5e0838a41b0b025 : SUCCESS in 1h 10m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32891d59c1b84f189bfcda3719efc8fd : SUCCESS in 1h 39m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/55b1bbea5e6d493aa7195018a3d40f59 : SUCCESS in 1h 25m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2cfbb4c638d74e44832ce504a7b45c22 : SUCCESS in 48m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9f86c5d13dee4db8abbbb287712cfc48 : SUCCESS in 1h 00m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fbdcbd3303674cf39bbc3d3bc7db68b2 : SUCCESS in 1h 01m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6fac319532c243a7bfd230179fc1661b : SUCCESS in 1h 17m 12s","accounts_in_message":[],"_revision_number":6},{"id":"343368f24d4d07d59750ab16ca1d266729d77cdd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-26 02:17:05.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/88/952988/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"f0b1d7db1076e765768af3021219fb92c626d377","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-28 11:50:47.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"74910651ceeaf05b9b9b166f851644bbfd2b2491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-29 11:29:12.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1497139b40e985e1b08f2b6c84fc3805c208579f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-29 11:30:22.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4ca1bcff228d30ad5a5282e6b10cb34e6bc096fd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-29 12:12:27.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/88/952988/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"f1f856dd0ea7f5d4eb81147647f214daea92a56f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 12:21:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bbcd71807684b5da09348f68e69e6de\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2726e5f7db924a6a87b11264f601be8f : SUCCESS in 36m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1624f99e995a4c3d9b16cf2f9ec1b52c : SUCCESS in 41m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ee7e0c78a66f44c2bd5b6d60e78f0401 : SUCCESS in 49m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"19e51b45aeedf4344c3f3de0fd5b495fbee30b73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 13:39:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c276dffee4348289ad368ee14470bb5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/62b2bf633a784d83b40af2f45a9219d9 : SUCCESS in 2h 08m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/89ae46096ead4613a39a6489d7fc9f05 : SUCCESS in 1h 56m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d40f250a26c148eea03d9fd30f36cefc : SUCCESS in 34m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2ffa13408b94dfe92ad4f11fca807c8 : SUCCESS in 37m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ada8532f4544eafa8068818838f0ae9 : SUCCESS in 14m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eea3122d4b8e4c35a30c980f6006c753 : SUCCESS in 23m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aacbe3b8b34f4001a5f60930acb5ab61 : SUCCESS in 19m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e542d3407014f8194ef74e32bddfe01 : SUCCESS in 20m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1dc1e48c599477e9c08becacfa10fc8 : SUCCESS in 11m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/db9a00c44da843c095fe3d227af9c248 : SUCCESS in 1h 36m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/44251e94b1604b438fe44184d4c34eaf : SUCCESS in 1h 02m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/151c884ca5ec41659bb53462b7e232e1 : FAILURE in 49m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/96a560a091c04ec187e1091ed6752de6 : SUCCESS in 1h 38m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b0af7f344ea7428eb23f9af25c5a7ee7 : SUCCESS in 1h 42m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/320972ca4730481fb96c5e747979355f : SUCCESS in 1h 00m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/24a720cd9d094ac0b47fbed277eb0f19 : SUCCESS in 1h 05m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6bfc9c57de7744debfb9399b95a02f23 : SUCCESS in 1h 16m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ee5ccb2a2ec1414e996d09d443b87afc : SUCCESS in 1h 23m 42s","accounts_in_message":[],"_revision_number":7},{"id":"44f79a4059fbe34a31fd8f2da1e2a92758167fca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-29 14:21:58.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"698f92615ac604e50c37c8991aa124a93831792f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 14:58:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd9c1503da4d42b990e8fce0d69725fb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c635f438e774105bed80d8af88bf9f0 : SUCCESS in 27m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/122f3712229046679ce9212a7f456588 : SUCCESS in 33m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f5925f9b9a9d45e2bf707989f819fa37 : SUCCESS in 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"aa7b8aa8420366edba3df605cbc2de75142d5983","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-29 15:04:35.000000000","message":"Patch Set 7:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/88/952988/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"6c54971c97d6ed9737a12f3f767c834fe5c6676c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 16:34:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe3f246bcb6c4e958584aaab64315ea8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/99a810bc6aef4615a376fe8f3cc20c2f : SUCCESS in 1h 31m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/47c3493a400846218c6ff4941f310270 : SUCCESS in 2h 12m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4e19cad61a7e40ef9d482f71145dead1 : SUCCESS in 32m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f36b6fb7b1a145b29824e8b2be284856 : SUCCESS in 29m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89cbbc8a61e5495682d2f75efc0eb4d2 : SUCCESS in 13m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/451a45d0a4f64e809774cb79a842a64f : SUCCESS in 24m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76a98631d64f42f1bdbb8836dace7cd5 : SUCCESS in 19m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e89083bd77ad4d6990c605a2b4dcfa92 : SUCCESS in 23m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d1f011e79f41fe82174f0ba7bc977a : SUCCESS in 9m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5f172967a26f47d0b877f405d189775b : SUCCESS in 1h 38m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d66b2486752d44c690629be248c73b39 : SUCCESS in 1h 01m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/29ef94105b2e499bba1952b609adcf47 : SUCCESS in 57m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8f5117ac4f6748d9b31ca37c10e95486 : SUCCESS in 1h 43m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/78d3d64c668e493c8f600d51571fc64c : SUCCESS in 1h 25m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bb971333a5e342d5b4a6f992c1848229 : SUCCESS in 57m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e262f4bf7b3b4231af25cf4e138d49c7 : SUCCESS in 1h 38m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c840a73c782a4ec49ee9630797264e7d : SUCCESS in 59m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/10fd23de1f4b4a3e841a0c4aaeb00143 : SUCCESS in 1h 16m 39s","accounts_in_message":[],"_revision_number":7},{"id":"01d22e6bd6a02ab549b38565a84a864495c1ddb9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-29 18:51:38.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"31465ae195e5ec76187c39aa2ad8c6a6e6501294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-08-21 13:20:44.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"a7c94c51b3d465d4d06146ba7ec671d8a533e458","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-08-21 13:21:42.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"cbbe50bc2b5cbb1795963de6a12ec013e99b780d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-08-21 14:07:01.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2025/Arista_Tempest_Testrefs/changes/88/952988/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"d211ecd3468ae7f65cdda156b113fd548fc8a600","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 15:13:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2e89259f71740e98c15c2a0e9ffc4ac\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f2ed5341cda14db1899135361244289c : SUCCESS in 48m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad158a8cb79b4846a673eb94d6b05a96 : SUCCESS in 40m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f0c3cf2b6c6b4a13b182504b532bb16f : SUCCESS in 56m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"daf029080f2fb2416d75259dd94e694b5cbc22a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 15:41:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cef0eb8220b469d8f576acce45c925e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/85ec23a756ce4cf3bb51e1fc7093cb8d : SUCCESS in 1h 37m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5bf7f82fa8134fb1869bf268867f7c7e : SUCCESS in 2h 19m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e3bd75df5235425586002fcf661374c0 : SUCCESS in 31m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1a9afd0a25d421d984b261bf0ea9df0 : SUCCESS in 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e7a9557d8d24a2eb040e6a150bc3647 : SUCCESS in 7m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebba4518ef8649ba9658564802e880e4 : SUCCESS in 19m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31d4150b10e8464f99516f0d32611194 : SUCCESS in 18m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b2cb222538f4a54a16f4c0a351c4de6 : SUCCESS in 27m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ae2b5dad05454e8532fa0bc7327163 : SUCCESS in 17m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/776617bb729342c3bdff648ec33192e8 : SUCCESS in 1h 33m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c820cc1e2b34dcd9a7cd4f3f84c5cd3 : SUCCESS in 30m 37s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a850ee8d5044f2ca630ff3f6c8510e4 : SUCCESS in 1h 03m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7e28b5956d814dc4b504c345f80ea8b9 : SUCCESS in 1h 58m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/13ac025071b14aecb6a7a20092743367 : SUCCESS in 1h 46m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c0507205076f46718b7149d5fad0deec : SUCCESS in 56m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b37c2f861c9b45f8b62570d6ae874903 : SUCCESS in 26m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/be9fc179e1714ee7aeaad62ccb1389ee : SUCCESS in 1h 17m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0a413a7c6b184d4c84be8f199d66d24c : SUCCESS in 1h 31m 08s","accounts_in_message":[],"_revision_number":8},{"id":"9781677ce35641e74ec1a7a6c4edc32f6c6b4e04","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-08-21 20:59:49.000000000","message":"Patch Set 8: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"745c6577d2150b6ef8f944948a7524000311c6ee","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-08-29 15:27:50.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"cb1039272a9d6058517b78fd94b01129362984cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 15:28:48.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a25676cad094a6913c3b54cb2bcffe2ef8b43cac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 17:32:30.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f59cbeadc19b41319812d283c6ecf233\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b0e71c765f1542cbbd45779c702a4291 : SUCCESS in 1h 34m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e79980d8df28490885ba4495207b5bba : SUCCESS in 2h 03m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95b7d5d34ab74b3d9f2fbe54291f5a52 : SUCCESS in 15m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbbfe70d88324a9b9deace4f413ec348 : SUCCESS in 18m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98ec84481b12464788abdf96ed23dd44 : SUCCESS in 14m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0833aa8476af4d6dafee6f8011ffda35 : SUCCESS in 14m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/045eaa5441ff4d648b7ab353ce9d71a8 : SUCCESS in 1h 18m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bc369963e62648fa8eeb6ebff902546b : SUCCESS in 1h 52m 54s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bbccfa2dfb445fc95116997ee782889 : SUCCESS in 1h 07m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/df14f4d706b342bb9c062a95a8131418 : SUCCESS in 28m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/76f5382e39384c359f1029b629e60c39 : SUCCESS in 1h 02m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d1b5ce293f444830b754601029b941b6 : SUCCESS in 1h 22m 02s","accounts_in_message":[],"_revision_number":8},{"id":"94536ea4d924bc5db7a6e9f08dc550992961ac11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 17:32:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"51915411b73b47299d5b5013c405754b39f655e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 17:36:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a58770cb8554378b202228b548f37e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c705e8a5fed483db2a0f076ca547dd5 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9001ddcb1c74b018fc3d47ddb04a76edf38168e1","revisions":{"f7e9021407be2022a23b6902810bbe176edb6543":{"kind":"REWORK","_number":1,"created":"2025-06-20 13:39:26.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/952988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/952988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/952988/1"}}},"commit":{"parents":[{"commit":"f56494bace6e710aded9e309268185c77166adc4","subject":"Merge \"[FT] Match the correct event in ``test_subport_delete``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f56494bace6e710aded9e309268185c77166adc4"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-06 13:18:55.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-20 13:38:28.000000000","tz":120},"subject":"WIP: Delete tunnel endpoints if endpoint agent is deleted","message":"WIP: Delete tunnel endpoints if endpoint agent is deleted\n\nCloses-Bug: #2084446\nChange-Id: I64ad033e600c2c87af4716736c10e11c143afea2\nSigned-off-by: elajkat \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7e9021407be2022a23b6902810bbe176edb6543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7e9021407be2022a23b6902810bbe176edb6543"}]},"branch":"refs/heads/master"},"125c7df4df3ee60cbc575be41af4569eb202c9d4":{"kind":"REWORK","_number":2,"created":"2025-06-23 10:00:26.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/952988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/952988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/952988/2"}}},"commit":{"parents":[{"commit":"b8e21fd54e18efa9134aa24dbf40f57786993d1d","subject":"Merge \"[UT] FIx no_dhcp_release6 test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8e21fd54e18efa9134aa24dbf40f57786993d1d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-06 13:18:55.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-23 10:00:14.000000000","tz":120},"subject":"Delete tunnel endpoints if endpoint agent is deleted","message":"Delete tunnel endpoints if endpoint agent is deleted\n\nCloses-Bug: #2084446\nChange-Id: I64ad033e600c2c87af4716736c10e11c143afea2\nSigned-off-by: elajkat \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/125c7df4df3ee60cbc575be41af4569eb202c9d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/125c7df4df3ee60cbc575be41af4569eb202c9d4"}]},"branch":"refs/heads/master"},"2e97ba6c4dfe29a13b643f39efdbf0ea92c22381":{"kind":"REWORK","_number":3,"created":"2025-06-24 15:09:35.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/952988/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/952988/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/952988/3"}}},"commit":{"parents":[{"commit":"b8e21fd54e18efa9134aa24dbf40f57786993d1d","subject":"Merge \"[UT] FIx no_dhcp_release6 test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8e21fd54e18efa9134aa24dbf40f57786993d1d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-06 13:18:55.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-24 15:09:22.000000000","tz":120},"subject":"Delete tunnel endpoints if endpoint agent is deleted","message":"Delete tunnel endpoints if endpoint agent is deleted\n\nCloses-Bug: #2084446\nChange-Id: I64ad033e600c2c87af4716736c10e11c143afea2\nSigned-off-by: elajkat \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e97ba6c4dfe29a13b643f39efdbf0ea92c22381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e97ba6c4dfe29a13b643f39efdbf0ea92c22381"}]},"branch":"refs/heads/master"},"e1da5bef5546d154cbd25291d6416baeb1d8ebf3":{"kind":"REWORK","_number":4,"created":"2025-06-25 08:07:33.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/952988/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/952988/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/952988/4"}}},"commit":{"parents":[{"commit":"b8e21fd54e18efa9134aa24dbf40f57786993d1d","subject":"Merge \"[UT] FIx no_dhcp_release6 test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8e21fd54e18efa9134aa24dbf40f57786993d1d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-06 13:18:55.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-25 08:07:20.000000000","tz":120},"subject":"Delete tunnel endpoints if endpoint agent is deleted","message":"Delete tunnel endpoints if endpoint agent is deleted\n\nCloses-Bug: #2084446\nChange-Id: I64ad033e600c2c87af4716736c10e11c143afea2\nSigned-off-by: elajkat \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1da5bef5546d154cbd25291d6416baeb1d8ebf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1da5bef5546d154cbd25291d6416baeb1d8ebf3"}]},"branch":"refs/heads/master"},"f11e8e32aa484bd93a82fbf167478716bdcd3add":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-07-02 07:04:38.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/952988/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/952988/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/952988/5"}}},"commit":{"parents":[{"commit":"1971c4a261ca54f077e2dec8b59774602bbbe3f5","subject":"Merge \"[eventlet-removal] Remove the eventlet WSGI Neutron server code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1971c4a261ca54f077e2dec8b59774602bbbe3f5"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-06 13:18:55.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-07-02 07:04:38.000000000","tz":0},"subject":"Delete tunnel endpoints if endpoint agent is deleted","message":"Delete tunnel endpoints if endpoint agent is deleted\n\nCloses-Bug: #2084446\nChange-Id: I64ad033e600c2c87af4716736c10e11c143afea2\nSigned-off-by: elajkat \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f11e8e32aa484bd93a82fbf167478716bdcd3add"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f11e8e32aa484bd93a82fbf167478716bdcd3add"}]},"branch":"refs/heads/master","description":"Rebase"},"618c06b02f2a62819ae76871b9ad920c145f62ed":{"kind":"REWORK","_number":6,"created":"2025-07-21 13:59:27.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/952988/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/952988/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/952988/6"}}},"commit":{"parents":[{"commit":"29a6bc1c2d3b0986d77c105f12f458842e98785e","subject":"Merge \"[doc] Include the ML2/OVN minimum bandwidth support added in Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29a6bc1c2d3b0986d77c105f12f458842e98785e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-06 13:18:55.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-07-21 13:59:10.000000000","tz":120},"subject":"Delete tunnel endpoints if endpoint agent is deleted","message":"Delete tunnel endpoints if endpoint agent is deleted\n\nCloses-Bug: #2084446\nChange-Id: I64ad033e600c2c87af4716736c10e11c143afea2\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/618c06b02f2a62819ae76871b9ad920c145f62ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/618c06b02f2a62819ae76871b9ad920c145f62ed"}]},"branch":"refs/heads/master"},"22f22c0b32687fdba64730ff37b6c77e72a49a35":{"kind":"REWORK","_number":7,"created":"2025-07-29 11:29:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/952988/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/952988/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/952988/7"}}},"commit":{"parents":[{"commit":"758dbfa135e4bb1c2bee0823f2203a66adcdd5a2","subject":"Merge \"Use ALLOWED_ADDRESS_PAIR resources constant from neutron_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/758dbfa135e4bb1c2bee0823f2203a66adcdd5a2"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-06 13:18:55.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-07-29 11:29:00.000000000","tz":120},"subject":"Delete tunnel endpoints if endpoint agent is deleted","message":"Delete tunnel endpoints if endpoint agent is deleted\n\nCloses-Bug: #2084446\nChange-Id: I64ad033e600c2c87af4716736c10e11c143afea2\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22f22c0b32687fdba64730ff37b6c77e72a49a35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22f22c0b32687fdba64730ff37b6c77e72a49a35"}]},"branch":"refs/heads/master"},"9001ddcb1c74b018fc3d47ddb04a76edf38168e1":{"kind":"REWORK","_number":8,"created":"2025-08-21 13:20:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/88/952988/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/952988/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/952988/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/952988/8"}}},"commit":{"parents":[{"commit":"758dbfa135e4bb1c2bee0823f2203a66adcdd5a2","subject":"Merge \"Use ALLOWED_ADDRESS_PAIR resources constant from neutron_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/758dbfa135e4bb1c2bee0823f2203a66adcdd5a2"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-06-06 13:18:55.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-08-21 13:20:29.000000000","tz":120},"subject":"Delete tunnel endpoints if endpoint agent is deleted","message":"Delete tunnel endpoints if endpoint agent is deleted\n\nCloses-Bug: #2084446\nChange-Id: I64ad033e600c2c87af4716736c10e11c143afea2\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9001ddcb1c74b018fc3d47ddb04a76edf38168e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9001ddcb1c74b018fc3d47ddb04a76edf38168e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
