)]}'
{"id":"openstack%2Fneutron~953095","triplet_id":"openstack%2Fneutron~master~I82e96063bff82815ec0a529e6139912e2f8e8c84","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-21 08:34:51.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-07-21 08:34:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I82e96063bff82815ec0a529e6139912e2f8e8c84","subject":"Use NetworkMTUSubnetConflict from neutron-lib","status":"MERGED","created":"2025-06-23 13:16:19.000000000","updated":"2025-07-24 23:10:37.000000000","submitted":"2025-07-21 08:34:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953095","meta_rev_id":"5bd9e71b5432e5f60b5bdb9ebd6f46f21d6a5106","_number":953095,"virtual_id_number":953095,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-07-24 23:10:37.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-21 08:34:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-21 06:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-07-16 15:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-21 06:31:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-23 14:20:19.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-23 14:34:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-23 15:40:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-23 18:03:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-06-26 08:39:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-07-16 15:52:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-21 06:31:05.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"dd9719a0539a6f4963e2afbf242ebdc82438d9bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-23 13:16:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"637aa10ddc0ee5b3c38a74d02719a00b7eb0e43c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-06-23 14:20:19.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2025/Arista_Tempest_Testrefs/changes/95/953095/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"530d85788dd3aabe1d5b166568ec2132195ebbee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 14:34:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b4b8b49dd97459f9002e2a2a1a7be56\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5c9d648b9ce4b14ae3571360072e275 : SUCCESS in 21m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c16922aa94c4cc5b5048bd271f7fefb : SUCCESS in 38m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5c36aef865ee4bea982be10bddaf0b20 : SUCCESS in 42m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa78e5df75b1b5dedecfeb77b247b6287a386abb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 15:40:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bcb91c681a443db8ce1c46861e628b1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/35c164651ada4e8eb163a836274588d7 : SUCCESS in 1h 53m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b65c6422ac7242aaa9df6623feb2ed4d : SUCCESS in 2h 01m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e18b12f841f94200af355a474a33e86a : SUCCESS in 47m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8e0a60bdbbb4507903f4d5745c04a7d : SUCCESS in 41m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd299e68a6614cb7b901817b18e7e9b1 : SUCCESS in 12m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d71eff26514b44ed9f11042397895477 : SUCCESS in 24m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ae83ff30df34e6a8ce127c471d11b1c : SUCCESS in 20m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fe0c13c915c48f6b7cc8c571873e8b8 : SUCCESS in 27m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20618cccca534bd983bdf80801e1d138 : SUCCESS in 11m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/aa7b64709f364148bbd0d76027062dc1 : SUCCESS in 1h 44m 50s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/160e6f13ae474acca0052dca825110d8 : SUCCESS in 2h 06m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/807e8b0c7bbc4237af045158708a4819 : SUCCESS in 1h 06m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2276b644c4ce44a8acd36a6fa35aa80a : SUCCESS in 58m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/80de05d8119f4c41b5e1a4c2ea3c220a : SUCCESS in 1h 38m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b6c59e0ff10f4095920591e9449e18c8 : SUCCESS in 1h 28m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f1a7464eefbd4d74a54736579d5aa8cf : SUCCESS in 1h 07m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dc639a3a96614cb2b712bf142de51e7d : SUCCESS in 55m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/92d0c66deea74dcc9b3099ff96a7f5aa : SUCCESS in 1h 17m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/79e65bd4b3674c41aa12d9fd16fa3ead : SUCCESS in 1h 18m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3574d7166afce2da83971930e6a7c24bc637b394","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-26 08:39:12.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18b243f9149291b20bc26b412feb5ae6342cf9a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-11 13:51:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9c0c35e8514a8385c65b197f791b7a88785b43a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-11 13:55:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bde95719a1e6bee92c4b800a92f0996f003fc9c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 16:13:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1638c7f1c3d648ecb4e1c9666cd0d74b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/19725f0958b64ec0a0098c7d770a4817 : SUCCESS in 1h 23m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/84140edc3f034223953eee1fdc639655 : SUCCESS in 2h 04m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ea814b15b904f0bb229ad315dae5c0c : SUCCESS in 38m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69f95b79a736459cbde72674f4b1e931 : SUCCESS in 40m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed772e066a54b88a7333caa5929466c : SUCCESS in 12m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dfee8f775ac430cb33ca9f03349a8a4 : SUCCESS in 22m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f956bbdef5794e418aa5fefd9559967d : SUCCESS in 18m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/446f6b3db1c24fb9b9623aee8636a964 : SUCCESS in 24m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8efa4822de04897be870cdc37e7e7b5 : SUCCESS in 14m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/74e0f161dbcb4ed090a09a406aebf18e : SUCCESS in 3m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bfccd74b1e184452a232b60cc0fe7b99 : SUCCESS in 1h 44m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e9ef35cee524fecaec47871b6d4ba86 : SUCCESS in 58m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c45288d1c1c47a6b1d3efe418a19726 : SUCCESS in 1h 04m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e95f84b48c9d4ad8a8ac6fbbc65e9b78 : FAILURE in 1h 43m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9061e78ad0844ce2a12f34f0697926ac : SUCCESS in 1h 25m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a37c5e9abb8542339e82d9390dbdb872 : SUCCESS in 53m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/802adbbc3cce4dd6956656d311c2fee0 : SUCCESS in 53m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3ca2a1bd037b4ca28b6804268327cd5f : SUCCESS in 1h 07m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4bf618ab7d1f4ed18655ef5ece438e3e : SUCCESS in 1h 23m 23s","accounts_in_message":[],"_revision_number":3},{"id":"2e440ec2a9d179243acc41e01dd40a9cb5bd85a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 22:12:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/192faa2f608d4398978b2a78c9ff640e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7eab8bd9ffd4a37a4f40c8ffe65d8c7 : SUCCESS in 43m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94bde166f7e44cab8cbdc5acee511054 : SUCCESS in 34m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d39a8c0cb83a4ba1be76820b8233f737 : SUCCESS in 46m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e44dbdde70630358f92f570d37660277daa6fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-16 13:29:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ccd16eec3ae2caa4de2ccce61c32748fb1927dc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 14:39:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcfebbc6dd054b25a0bfc3f68fd7e897\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0adf630d7d0449584cfa0d6b3e5b88a : SUCCESS in 43m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/923f9065c2cf41f1b40d2b7cf8c1380b : SUCCESS in 27m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eab8c83c620c4c7eb3e5584c3b2cf9f6 : SUCCESS in 30m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6a16ee2241d4f9cfe06a4edd885f92f4c03dbf75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 15:36:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/998f12c589e9481f87ae9b3e83b6da79\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e280191d00284f60ba1edfcb012c7304 : SUCCESS in 1h 18m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3d2277d95554479180cf4196c2d30e6a : SUCCESS in 2h 06m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a2c308731c284d02b507483a03e52f06 : SUCCESS in 36m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cf0fa27519946f8925e5b232dde5c9f : SUCCESS in 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0884fed4cf434755822c90b906763246 : SUCCESS in 12m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e424af0140641f49e78d8b1a55bbb2f : SUCCESS in 19m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5dde24ccbe604725a15a26346af4390d : SUCCESS in 17m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ac69bfcf4ff453da3a2566c1eafb701 : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e1efd2d4e3411a9af3f0d299c28449 : SUCCESS in 13m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1403d5fddaf74a08b22845eb6d4945a8 : SUCCESS in 2m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/99fd4f05b2e34f82ad409d19fea735b9 : SUCCESS in 1h 27m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d24373d56014f008d8a9c6ca5e94aef : SUCCESS in 54m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/541fd53447034084872c5079ee86cc3f : SUCCESS in 1h 04m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/253ea82941334bff8314ef2cfc1a0711 : SUCCESS in 1h 39m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fdf1bf1e6eb84b0dade093310b92ffb0 : SUCCESS in 1h 23m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1d85936bd2df460685dfa1d2590a7703 : SUCCESS in 1h 02m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3647a22c94ae47f2a63c3c0f6b4570a6 : SUCCESS in 1h 01m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f2bba8f53fdd4e719fe711ca2f5bb842 : SUCCESS in 1h 13m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d2d5c3a7ecf74ee2a056ab910df951cf : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":4},{"id":"2711f1187d66d80290647d37a378dc814352ecb8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-16 15:52:18.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2bc0d61023a4597e6fa24b22f3c586b4b60ba374","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-21 06:31:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"231e1df7913ed5b6dd5fa6f94e6d570f76076572","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 06:31:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"82489ecaf0de7e65d90b6e6deadf6419d00634d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 08:34:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47fde6939d494854946b1199d330a41e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d4f6b4cb682c4ff6a6bf0285bc9a7c94 : SUCCESS in 1h 24m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/842dd3be60e04a80b9a60abc002efdf9 : SUCCESS in 2h 01m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9beab1c0a03c4632a6c54ba15d5b4dd1 : SUCCESS in 11m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d68220cf189d4fad855f90a4b00498cf : SUCCESS in 19m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3630cf443b4646269742c34b5b89ba8f : SUCCESS in 15m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1236109d45f46ae85cbcc05be592beb : SUCCESS in 9m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/afb25c8831fa4eb8b26bbcbcc13677ed : SUCCESS in 2m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/669e1751c1964b9580681b4296e366f8 : SUCCESS in 1h 20m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6e605d94adc649898e99440cedcdc596 : SUCCESS in 1h 04m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f50fa845ede44e079b2e9a910bd320de : SUCCESS in 39m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7f8fe544efb418eb4c291ca6862b57d : SUCCESS in 58m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/858d68fc703c43eca741130cd0fd8318 : SUCCESS in 50m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d2a9c598a9724e0ca41f2cb6d3e296cb : SUCCESS in 1h 30m 18s","accounts_in_message":[],"_revision_number":4},{"id":"3a424e90e051a0925f1415b9764bd61da00f72c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 08:34:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"62997618b49d883d95e976f24576848500e1a7ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 08:36:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4eff983c5dc412d9a030ce3bc6182ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b0d8a72c6a24772952c0c31d93314d4 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":4},{"id":"97860b0b542ca01ef8a6310a277e5a9ca1b350bb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 12:36:49.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/95/953095/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"dfe7e2f50207a6f19f77261db905165c2a07078e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 13:18:27.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/95/953095/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"5bd9e71b5432e5f60b5bdb9ebd6f46f21d6a5106","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 23:10:37.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/95/953095/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2983babbac9241715c9e3bf2937e0031291e9626","revisions":{"b2a7c22c205a8669dac5a3f0474c341a6cbba48a":{"kind":"REWORK","_number":1,"created":"2025-06-23 13:16:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/953095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/953095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/953095/1"}}},"commit":{"parents":[{"commit":"39b183645e071f9c2926f0bd870723fb6e300cb3","subject":"Merge \"[eventlet-removal] Remove \"logger\" mechanism from config examples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39b183645e071f9c2926f0bd870723fb6e300cb3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-23 13:15:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-23 13:16:07.000000000","tz":540},"subject":"Use NetworkMTUSubnetConflict from neutron-lib","message":"Use NetworkMTUSubnetConflict from neutron-lib\n\nIt was added to neutron-lib.\n\nTODO: Bump min version of neutron-lib once new release of neutron-lib\nis created.\n\nDepends-on: https://review.opendev.org/952978\nChange-Id: I82e96063bff82815ec0a529e6139912e2f8e8c84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2a7c22c205a8669dac5a3f0474c341a6cbba48a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2a7c22c205a8669dac5a3f0474c341a6cbba48a"}]},"branch":"refs/heads/master"},"31f1a71cb8ebf94cee13bb89521a0ee3a8dfa1d1":{"kind":"REWORK","_number":2,"created":"2025-07-11 13:51:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/953095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/953095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/953095/2"}}},"commit":{"parents":[{"commit":"47b54e94ae56c0ab1bcbbd8c51d35046b4b2fc37","subject":"Merge \"Fix mocks in the test_l3 unit tests module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47b54e94ae56c0ab1bcbbd8c51d35046b4b2fc37"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 13:51:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 13:51:33.000000000","tz":540},"subject":"Use NetworkMTUSubnetConflict from neutron-lib","message":"Use NetworkMTUSubnetConflict from neutron-lib\n\nIt was added to neutron-lib.\n\nTODO: Bump min version of neutron-lib once new release of neutron-lib\nis created.\n\nDepends-on: https://review.opendev.org/952978\nChange-Id: I82e96063bff82815ec0a529e6139912e2f8e8c84\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31f1a71cb8ebf94cee13bb89521a0ee3a8dfa1d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/31f1a71cb8ebf94cee13bb89521a0ee3a8dfa1d1"}]},"branch":"refs/heads/master"},"c3dc8d66efce882865e05a94fab9915b6f58bd4e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-07-11 13:55:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/953095/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/953095/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/953095/3"}}},"commit":{"parents":[{"commit":"47b54e94ae56c0ab1bcbbd8c51d35046b4b2fc37","subject":"Merge \"Fix mocks in the test_l3 unit tests module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47b54e94ae56c0ab1bcbbd8c51d35046b4b2fc37"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 13:51:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 13:55:13.000000000","tz":0},"subject":"Use NetworkMTUSubnetConflict from neutron-lib","message":"Use NetworkMTUSubnetConflict from neutron-lib\n\nIt was added to neutron-lib.\n\nDepends-on: https://review.opendev.org/952978\nChange-Id: I82e96063bff82815ec0a529e6139912e2f8e8c84\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3dc8d66efce882865e05a94fab9915b6f58bd4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3dc8d66efce882865e05a94fab9915b6f58bd4e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2983babbac9241715c9e3bf2937e0031291e9626":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-16 13:29:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/953095/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/953095/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/953095/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/953095/4"}}},"commit":{"parents":[{"commit":"0b23123a09f0d93fa5bc7c7a5b0c730db53dd9a5","subject":"Merge \"trivial: Autofix low-hanging linter failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b23123a09f0d93fa5bc7c7a5b0c730db53dd9a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-11 13:51:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-16 13:29:02.000000000","tz":0},"subject":"Use NetworkMTUSubnetConflict from neutron-lib","message":"Use NetworkMTUSubnetConflict from neutron-lib\n\nIt was added to neutron-lib.\n\nDepends-on: https://review.opendev.org/952978\nChange-Id: I82e96063bff82815ec0a529e6139912e2f8e8c84\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2983babbac9241715c9e3bf2937e0031291e9626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2983babbac9241715c9e3bf2937e0031291e9626"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
