)]}'
{"id":"openstack%2Fneutron~953437","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I4d2c3db355e29fffcce0f50cd12bb1e31d1be43a","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2095152","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2025-06-30 16:30:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d2c3db355e29fffcce0f50cd12bb1e31d1be43a","subject":"Limit trunk ACTIVE state hack to OVN","status":"MERGED","created":"2025-06-26 12:31:12.000000000","updated":"2025-06-30 16:32:33.000000000","submitted":"2025-06-30 16:30:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953437-bug/2095152","meta_rev_id":"1393eeff18c3f17a915003a07d13b35bd0bd5e3d","_number":953437,"virtual_id_number":953437,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 16:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-06-30 14:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-06-27 07:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2025-06-30 14:15:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-26 14:08:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-26 14:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 07:39:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-06-30 14:15:02.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"e1d1efcebb3017f13abf49f91b75b1c0189df1a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2025-06-26 12:31:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ea9813a85d3492d159a822e87ae3ed8eb1683ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 14:08:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafadfc9f2594c7cae2bfd31e345b1ca\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/488b464dab6a469fb9b9c0e2feedf8ea : SUCCESS in 37m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7749476bc6a14be590a48fcedc1a193c : SUCCESS in 46m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95b0f829e545419ffdb71d8ebe43a609f27829ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 14:27:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/228af33f759e4a5dbcff76a7af87f31f\n\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/99744cd796074177b833f3e8e1f9859d : SUCCESS in 1h 11m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/bb6cd700b4374b95b6c393979fd36adb : SUCCESS in 1h 54m 32s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/e7c0a64016b2474da5c4953da8c9ee0f : SUCCESS in 1h 46m 32s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/336cb6698983416ab188fb9538e1b287 : SUCCESS in 1h 07m 39s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/5206f4323da24b0f85ac094fa436b2a1 : SUCCESS in 55m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fcb6953b8794a208819f7cb3d994fc6 : SUCCESS in 27m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cece4138b739437fb75abd8d0fe48a3a : SUCCESS in 15m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48f2d21b41e4405ab030cb2fafa10be7 : SUCCESS in 31m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73dae442215b461ebe9f176108b706c3 : SUCCESS in 20m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf94f185b164bc585d29c552498434f : SUCCESS in 12m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3aeae68530804c5b86c9e0f0487fe298 : SUCCESS in 54m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/dbf195a96ded445a8196a91e38dd3815 : SUCCESS in 1h 03m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc1a4a48ced34a98a65a8d4c01afc990 : SUCCESS in 1h 07m 09s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/177663600b1141c8922c5a0c0923b017 : SUCCESS in 1h 43m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/575bd17cb0c64862b8445e9858643726 : SUCCESS in 50m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/21a35468e8e44f08b0eb609624bcc1f3 : SUCCESS in 1h 04m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/921ee9b70ca94f109e2cd9e49575d5b1 : SUCCESS in 1h 42m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9ab512ac53f44f2395eebb41a0be9725 : SUCCESS in 1h 28m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dd30595d241d4c73bfa91168070cb292 : SUCCESS in 1h 03m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/de63d46098924b6ba7ba48f4b67a3ea3 : SUCCESS in 58m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d6967d8f5e974260a6680c5aefd229b2 : SUCCESS in 1h 17m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d9dfb8cbb21419395cdc263d6db9d7a : SUCCESS in 1h 15m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7c77311cff51be29915b0d5ddd77588d1c13542f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-27 07:39:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80080997ce48b6d3c32fe3e41a419360368c5418","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-06-30 14:15:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c81c96440c7c9ea8338c7fe1bfcce8ad3537e566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 14:15:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61ebdfafdb43bc50a274b9c054075a324a9a646f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 16:30:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec3a51bfcbb74716a5acdf76f292d813\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/c2cfbb09d32e4d7ab6ee297567353d0c : SUCCESS in 2h 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb9e3f3e37f4b9383d5262e419339ee : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/780d3aac2ede4272a5723d5da0495fe7 : SUCCESS in 21m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f47626fe62244dea3ea19a7a69c14d5 : SUCCESS in 13m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73144fcc65a749a485c3d07cde5139aa : SUCCESS in 12m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3f80438b331a494485ee9a87a8453553 : SUCCESS in 1h 12m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/62b914dda70344429ce5456423334428 : SUCCESS in 1h 54m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8cdfeba58aca4482b1b7eeca47e46017 : SUCCESS in 1h 41m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b79357fa9158464ba81a5edbafdb8333 : SUCCESS in 1h 04m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/27b352051548428e8eeb24ca7326a228 : SUCCESS in 58m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/04652218be984d998e61d676d0f1b479 : SUCCESS in 52m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb5916a0cef74a9aaf489b9f9668cba2 : SUCCESS in 1h 09m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9892d1b87346db62a2f5dcca475326b073ea2250","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 16:30:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1393eeff18c3f17a915003a07d13b35bd0bd5e3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 16:32:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef4a9ecd3ec947079d35c3bf48e32b68\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a695e080a74c8abd312a964afeb2f7 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2224c9ea1e5a35acaad753a7f677d6788a612ec8","revisions":{"2224c9ea1e5a35acaad753a7f677d6788a612ec8":{"kind":"REWORK","_number":1,"created":"2025-06-26 12:31:12.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/37/953437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/953437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/953437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/953437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/953437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/953437/1"}}},"commit":{"parents":[{"commit":"d11f0be262a1814dd9f3a962da33ab8a195c7bde","subject":"Merge \"Fix bad i18n and fstrings interaction with flake8\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d11f0be262a1814dd9f3a962da33ab8a195c7bde"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2025-05-07 14:29:58.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2025-06-26 12:05:20.000000000","tz":120},"subject":"Limit trunk ACTIVE state hack to OVN","message":"Limit trunk ACTIVE state hack to OVN\n\nIn https://review.opendev.org/c/openstack/neutron/+/853779 we started\nmoving a trunk to ACTIVE when its parent port went to ACTIVE. The\nintention was to not leave the trunk in DOWN after a live migration as\nreported in #1988549. However this had side effects. Earlier we moved a\ntrunk to ACTIVE when all of its ports were processed. That means we\nunintentionally changed the meaning of the trunk ACTIVE status. This\naffected all backends and not just live migrate but create too.\n\nThis change moves the logic of propagating the trunk parent\u0027s ACTIVE to\nthe trunk itself to the OVN trunk driver, so we limit the undesired\neffects to ml2/ovn. By that we restore the original meaning of trunk\nACTIVE for all non-OVN backends. Ideally we would want to limit the\neffect to live migrate (so we don\u0027t affect create) but I did not find a\nway to do that.\n\nChange-Id: I4d2c3db355e29fffcce0f50cd12bb1e31d1be43a\nCloses-Bug: #2095152\nRelated-Bug: #1988549\nRelated-Change: https://review.opendev.org/c/openstack/os-vif/+/949736\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n(cherry picked from commit e69505d293ef12999f948531acbde75e16a65cd4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2224c9ea1e5a35acaad753a7f677d6788a612ec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2224c9ea1e5a35acaad753a7f677d6788a612ec8"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
