)]}'
{"id":"openstack%2Fneutron~953439","triplet_id":"openstack%2Fneutron~stable%2F2024.1~I4d2c3db355e29fffcce0f50cd12bb1e31d1be43a","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2095152","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2025-06-30 16:30:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d2c3db355e29fffcce0f50cd12bb1e31d1be43a","subject":"Limit trunk ACTIVE state hack to OVN","status":"MERGED","created":"2025-06-26 12:33:10.000000000","updated":"2025-06-30 16:32:35.000000000","submitted":"2025-06-30 16:30:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953439-bug/2095152","meta_rev_id":"651f6bb9b3ac30d6f07b2fbf2df670162422a38d","_number":953439,"virtual_id_number":953439,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 16:30:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-06-30 14:15:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-06-27 07:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2025-06-30 14:15:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-26 14:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 07:39:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-06-30 14:15:24.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"f9a0790b101f7039d6e1f10bfd3ea59f14216f79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2025-06-26 12:33:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2479ff51524f11b5c791dd83520913bcb3a3678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 14:10:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fcd37b394914dc5ae743d1213d3e6a9\n\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/743a7a5c30724d18bb5aaaadef9e5582 : SUCCESS in 1h 09m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/15ab6e8ce75e48caa5b6bda3f421a4a7 : SUCCESS in 1h 30m 28s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/3c50c59766bd46a09fa8a57ab3d039f6 : SUCCESS in 1h 20m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1513244736b845598029ccd62f024893 : SUCCESS in 32m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec8212148817448e9c1526a90d0d28a2 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5910c9348da8438097973e690808b123 : SUCCESS in 24m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81565d8f51a14c74a792fa713e1e3f40 : SUCCESS in 21m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b1f8d5311749f19b5634976c9986b0 : SUCCESS in 11m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb8071215ae042a997f0c2d39609b4ff : SUCCESS in 1h 30m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/75de4f5628014173a31da4cebe6b59b4 : SUCCESS in 1h 11m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1f6cfcd4227d49d289e64562c3821104 : SUCCESS in 38m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/65123b0ebd8949c8b49e88d2cf6bd67c : SUCCESS in 42m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/30b8d763f6fb4fbb83b174bffaecfc2d : SUCCESS in 1h 28m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1c83dace20ef49e6b30ceca62b35dcad : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":1},{"id":"587aedc3e5d6eacd49f6db6181a953f31d3a2452","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 14:41:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58a943dcf3a946d985d58d7c0b27ab19\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/166082eea0fb47ec86d511971b86172a : SUCCESS in 47m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6778ebdd80014bc4bffdb494cbad22d3 : SUCCESS in 1h 01m 54s\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae7b1bf3a27142529e2724d2a54d9738 : SUCCESS in 52m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"820b0340b393ded1a41a85100d84c7288d840e1d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-27 07:39:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3938da30344d05f391641313044aa94c7643b9e6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-06-30 14:15:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a33a047b7b7285563d4776119063563e5a2e51ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 14:15:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6343328afc488127c34515920d032cc4b4247861","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 16:30:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6548c5cb61346e28b2aa394bf8ada03\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/bf138991cf8c43eda6b3705d2ca5c585 : SUCCESS in 1h 42m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebf3e811aac842d29cbd1e90091acf2f : SUCCESS in 11m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d66437a16484192bae4200fb50ca0fa : SUCCESS in 27m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e81e956dd20249c98377243a461e1e3c : SUCCESS in 25m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6613ddc875cc4cfd87336d8f4d52747c : SUCCESS in 11m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/59a566c1606d4100bc9317ba74c8f38b : SUCCESS in 1h 34m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fc0fb4bb2424a558672e8ad5b2f8ec5 : SUCCESS in 1h 40m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ea1785644a784c77a8e5340c96c4aaf7 : SUCCESS in 48m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0f84afaf3470406eb9c1703ae7aa6150 : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":1},{"id":"eeb8040bd6594049a3d9fa7fc3115933b612aaf6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 16:30:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"651f6bb9b3ac30d6f07b2fbf2df670162422a38d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 16:32:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fa3631da78e469586ff30d6fa960ff0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78a02cda76c4a1881168617b4e0e8c5 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24c492cf394ed23e431c72ac0c87f8a0246ee007","revisions":{"24c492cf394ed23e431c72ac0c87f8a0246ee007":{"kind":"REWORK","_number":1,"created":"2025-06-26 12:33:10.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/39/953439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/953439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/953439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/953439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/953439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/953439/1"}}},"commit":{"parents":[{"commit":"95b1b1d03c68779d1b2f419a2b5725275ace8a1d","subject":"Merge \"Consider logging options when using OVNdbsync\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95b1b1d03c68779d1b2f419a2b5725275ace8a1d"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2025-05-07 14:29:58.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2025-06-26 12:13:13.000000000","tz":120},"subject":"Limit trunk ACTIVE state hack to OVN","message":"Limit trunk ACTIVE state hack to OVN\n\nIn https://review.opendev.org/c/openstack/neutron/+/853779 we started\nmoving a trunk to ACTIVE when its parent port went to ACTIVE. The\nintention was to not leave the trunk in DOWN after a live migration as\nreported in #1988549. However this had side effects. Earlier we moved a\ntrunk to ACTIVE when all of its ports were processed. That means we\nunintentionally changed the meaning of the trunk ACTIVE status. This\naffected all backends and not just live migrate but create too.\n\nThis change moves the logic of propagating the trunk parent\u0027s ACTIVE to\nthe trunk itself to the OVN trunk driver, so we limit the undesired\neffects to ml2/ovn. By that we restore the original meaning of trunk\nACTIVE for all non-OVN backends. Ideally we would want to limit the\neffect to live migrate (so we don\u0027t affect create) but I did not find a\nway to do that.\n\nChange-Id: I4d2c3db355e29fffcce0f50cd12bb1e31d1be43a\nCloses-Bug: #2095152\nRelated-Bug: #1988549\nRelated-Change: https://review.opendev.org/c/openstack/os-vif/+/949736\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n(cherry picked from commit e69505d293ef12999f948531acbde75e16a65cd4)\n(cherry picked from commit 2224c9ea1e5a35acaad753a7f677d6788a612ec8)\n(cherry picked from commit fd9d90706da7965dabfc0521bdc15b468f6af899)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c492cf394ed23e431c72ac0c87f8a0246ee007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c492cf394ed23e431c72ac0c87f8a0246ee007"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
