)]}'
{"id":"openstack%2Fneutron~953447","triplet_id":"openstack%2Fneutron~unmaintained%2F2023.1~I4d2c3db355e29fffcce0f50cd12bb1e31d1be43a","project":"openstack/neutron","branch":"unmaintained/2023.1","topic":"bug/2095152","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2025-08-27 19:14:19.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2025-08-27 15:34:31.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"I4d2c3db355e29fffcce0f50cd12bb1e31d1be43a","subject":"Limit trunk ACTIVE state hack to OVN","status":"MERGED","created":"2025-06-26 13:31:48.000000000","updated":"2025-08-27 19:18:33.000000000","submitted":"2025-08-27 19:14:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953447-bug/2095152","meta_rev_id":"df301d9ad848e4a40f5d4df3dfe616e1d0d7d040","_number":953447,"virtual_id_number":953447,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-27 19:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-07-23 12:51:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2025-08-27 15:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2025-06-26 15:12:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-08-27 15:34:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-26 15:12:36.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-06-26 15:18:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 07:41:57.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-06-30 14:11:51.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-07-21 14:08:23.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-07-23 12:51:58.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-08-27 15:34:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"d15ea1ac5871ff742cf4d890dc4ed71a8988ad2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2025-06-26 13:31:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12a09272fd674d83b13e835d09da77c10f814c43","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-26 15:12:36.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/1919671a73e74846a9a3b8bc457e5bef\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/60316f6a6b2b483f87185b2167f88f76 : SUCCESS in 1h 39m 31s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/298fc30fa75c4426962b30be151426f5 : FAILURE in 24m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfe89326ab9da81685867b3b2a089d085a2b15ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 15:18:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67e72c5437724996b82c3ad5f398043b\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/59f6d5e2410c40f59f1c4773d9fe58ab : SUCCESS in 1h 13m 26s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/1ec4c19e8b224524b1ab392bf74b2c86 : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/82c90a3fee3a42ed86079fc682afcdb8 : SUCCESS in 1h 40m 56s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/3834169d713045da924936c21e181d4e : FAILURE in 1h 41m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5a3f7f8e194456f84735b3961a79c79 : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d3ec3d359143c0b9f44ce2490d93c8 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0b579d5a8df4f5f924d796d7b72f846 : SUCCESS in 21m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2855e8d0317f4a01a5fd6c947e26cbab : SUCCESS in 16m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bba2414260e4fbfb9e722a7e631ca71 : SUCCESS in 10m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1941f4e03e3a4bd5aa878553db073a38 : SUCCESS in 1h 01m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f6e49c8b19b54232be9eadbaf5cac284 : SUCCESS in 1h 01m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/467b972b3615470c96661f4a15b6d243 : SUCCESS in 55m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4fc53ed630b64bc9a1c5c11b828c68d7 : SUCCESS in 55m 11s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c5a3d08474b04cc384bd802a8c6ad727 : SUCCESS in 1h 05m 36s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/73d39922292c46f9b6201a1abf388d65 : SUCCESS in 50m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e9fb84f6bdfa4c3e833a4a83f34b7c47 : SUCCESS in 56m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f77bfc89465843ef8b7013b7165fb0b0 : FAILURE in 6m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7bda80ab19444ea1bfa216bda749540a : SUCCESS in 1h 15m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/354e29c5b0c54610a2f41a95a9e0aaa6 : SUCCESS in 1h 13m 30s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/81f8c5cf7a8d4d75a67ea29d3b37a9ea : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1c99b6f11e12d679f79e0201c76f846d6ccc9658","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 17:39:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8a47f4a5ec7414db6721ad9aa57777f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66a10a1524464b9a936e6b389a164f24 : SUCCESS in 49m 22s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8577689f3fea4ef9a9c66e82249ac80c : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e380f7846fa6404d248941b865b2023259bdb781","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-27 07:41:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"141770b0231986c69bf8b3676695c87049d7e27c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 08:52:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1d909a503d947b28def1f621c97a59d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/313b747e48a246f6be99e8abf61bea4e : SUCCESS in 48m 07s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f900169837d447c872850f5b0fbff11 : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":1},{"id":"092ea38ebbeb6b2da6c566d1cde29b5ecd9a5159","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-06-27 09:15:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/a00443c6fa9c4202b86bd9892fddb36e\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/bf6311b3f577467c824c51305d5c65f6 : SUCCESS in 1h 32m 05s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9d1a81ebf7c040179acea9a12a23b80d : FAILURE in 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9dea1ed7bce9980414eeff6dd9d624bf7144b93a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 09:39:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddf4bdb9e5024767be9de8f504ceecea\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/ce720a18999e4e268692a15dc79a1444 : SUCCESS in 1h 43m 42s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/a7ac546101eb44ff8b51ba54d4b66d6c : SUCCESS in 1h 27m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/865e5b545d1d402d92d6bcb621dffbcc : SUCCESS in 1h 51m 49s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/ec80c7839af24811b8bfcd07f231eafd : FAILURE in 1h 50m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bc83d875507484d98fbdffec69a41fa : SUCCESS in 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c3de67ff5f41098d5ca7e696508cca : SUCCESS in 12m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1afb59c689f547839d2642f408067d97 : SUCCESS in 18m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f2080a79ba24536ad6045c77e24adfa : SUCCESS in 22m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03eec546e1184e0d991903acd93f5dfb : SUCCESS in 9m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a943276f75eb4ba8976322001e431886 : SUCCESS in 42m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5daade79cb3f497a946f63f35f701a76 : SUCCESS in 1h 14m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/59fb5a6132cb41aea3b2a1d347c0f6ef : SUCCESS in 46m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c8573df370744dfd9a8a81d8a1ec9254 : SUCCESS in 23m 53s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/10d53de19dc4443fb95aaf24f4e0aa32 : SUCCESS in 1h 04m 06s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4927d44f92924ab2837caff9a0234a33 : SUCCESS in 53m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/05c447f4a25a4eaf912f26ffe588acce : SUCCESS in 53m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/27ef8b44ba474df88623219ce6f6281a : FAILURE in 9m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1bf6840636f84ddf877ff0e0b1dc4a84 : SUCCESS in 1h 21m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0245d6f1fbcc432481b185a640d1d762 : SUCCESS in 38m 46s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/7013e3cf02da422e91019745c24e56fd : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5ea65f1fe7d0562f195f51519bc3f0b4f1313c7d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-06-30 14:11:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bc6b9fe8a3200ceed08a812d3be7d1ab99b5f01","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2025-07-03 08:44:19.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ae7fbadaac1c6c079b7281595d5584eb67d9eb2a","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2025-07-17 10:59:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e19233f8895dba244b2fe82a1123f0c6d06f9b34","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-21 14:08:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bc8e8412f674854ec145862b94c789fe4f790ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 14:49:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df31a8be5cf14222b37cf94b90d4a31d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7ad40ecb6da4847bf313841e0987a66 : SUCCESS in 25m 09s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13e10bc9c865423182cb4b3fd28eb786 : SUCCESS in 39m 19s","accounts_in_message":[],"_revision_number":1},{"id":"cdfc43e8c7705d3e71de4eb539b0f34aa2ae5da1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-07-21 15:46:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/dad0b08115814c90ad95a7411a1fda4d\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b20be7cd4ce14bbb9c14d720b7f1a706 : SUCCESS in 1h 36m 41s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/44bdc18c1c4341aab17182d25371c721 : FAILURE in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"338b96cf6380805599f3eab4dbb156089b594983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 16:59:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b84c8ec000046fb8fe1c5f35d3409e5\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/f46406215a104650b921557115d77bbf : SUCCESS in 1h 30m 05s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/d812145efd264ba283e0ada20e989c57 : TIMED_OUT in 2h 49m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/c43aae61fd954602a18b773ecf7d383d : SUCCESS in 1h 30m 09s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/13f31163a17148e5a87dbef1900390d6 : SUCCESS in 1h 19m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef48abcb7acf4ca8b34be8fc82dbc839 : SUCCESS in 23m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec73313efa1c430abd2c10c7eaaf69c3 : SUCCESS in 13m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e58056811d6452c992cfbfeb0262e8f : SUCCESS in 15m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5534630d14044d27bcb42e02e6daa94e : SUCCESS in 16m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84139a6bb1a94cc7b51ce87f0261a229 : SUCCESS in 5m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d26ac11b39e544a59613f8a44b296248 : SUCCESS in 45m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ececbc243f449ceb73432c859a26661 : SUCCESS in 1h 26m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3fdaf305fe64495bbf72aa22acfc8d7 : SUCCESS in 51m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bab929b5af914276a8662d8069ec9dd4 : SUCCESS in 56m 23s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0380f750f27c4322bb49991589d23634 : SUCCESS in 1h 07m 03s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0f9ba10a98bd4167ab865654edb47cc2 : SUCCESS in 55m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/29c2fe607f68498a8b245115fc2b377f : SUCCESS in 50m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/25e62782e5df4de29f812cb72e047f1c : SUCCESS in 1h 01m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c09137f339eb4f048dafe238cddfb466 : SUCCESS in 1h 11m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b72c70a5e07f419188e08046c4edbb64 : SUCCESS in 1h 01m 00s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/644596d9c32c42149410dd7c49869834 : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0193b0360e9ba3f6d85194ee053382e272dfb7fd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-21 17:54:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb92ca68a7f9820f9e305c8e3d261718f088c4ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 19:14:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fe36b58298e423bbbe2555f7e96646c\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/d840f01b521a4798bd4031cf090a72a9 : SUCCESS in 1h 16m 49s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/b09727c632a349de9a237bef7d4ac2bd : SUCCESS in 59m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/eecc5e7846de43e4b9b4034428ab3796 : SUCCESS in 1h 09m 38s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/bc4f52abe3544b5cae418f8d872594f0 : SUCCESS in 1h 12m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c0c67e1882f4924ada4218eb250eb08 : SUCCESS in 22m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41bfb705e0e3403292c7177d67a0ed5d : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cfab9493bc2486ab2fea4a9e38b3a50 : SUCCESS in 16m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/244a34e76db04f17a91613ceae2ce92b : SUCCESS in 15m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ccf72089cbd4f66a240d400b3cda51a : SUCCESS in 11m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d08e33c9f2b147cbaf120497fd5a4e50 : SUCCESS in 47m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9811d6f190204a6fa60df5f39116a2a6 : SUCCESS in 1h 18m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2505f86b2877438da54d595d20cf30d7 : SUCCESS in 49m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3a894d3c8fe24001ab1b88acf38e327a : SUCCESS in 59m 40s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/163dd08679d9417ab7760b7d9030b23e : SUCCESS in 1h 07m 22s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3b6ae0f554b34814ab659bc111da2589 : SUCCESS in 54m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cef6f775ee1a45208cc0553445d61f31 : SUCCESS in 55m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/69ebc47f64774eb59efc4d36d174a527 : SUCCESS in 1h 01m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/16b5483773fa47bcbb27b3ed0de06b91 : SUCCESS in 33m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bac2993327d1470495f8ff117cb1b6d2 : SUCCESS in 34m 38s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/c41477fc5b654d048b64be09d608b271 : SUCCESS in 28m 54s","accounts_in_message":[],"_revision_number":1},{"id":"186db5b32a64ee57eebccc5da6ce9c1ccacc8931","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-07-21 19:25:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/2119c4cf52ea4be78ef8a966b8e984e2\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/00a5d391b27b42cb9861b7b6ef76bd74 : SUCCESS in 1h 30m 04s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/6415d68813624044b6cbafd98c620af5 : FAILURE in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93b9662e139ecd58b7feb0dbda1a1f2c7a41d783","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-23 12:51:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54793d46a1b71ac98946c2bfe14fbcaeb5f6ad04","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-08-27 15:34:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1df3955645d8fee53c0b9ab86e414fd94d4744f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 15:35:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3b87db67b21c0e8b75fa47c122279caa2ae425c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 19:14:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41917ed7e024425ab5f351ccf1c6e877\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/8ee726cdba2b4e4692e83811031ecfea : SUCCESS in 1h 31m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc34820f4b04d6aaec944719d7ad5f9 : SUCCESS in 12m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ccd2308f297439399b883a311d4f467 : SUCCESS in 25m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/850c71d50ed54c7fbc1603ef53d7a563 : SUCCESS in 21m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb2bb7b84834ca3877e6380e88d76a6 : SUCCESS in 11m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/57e233592a1043e482b917b500be7cfc : SUCCESS in 43m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/825266d66585415bbfa3e5b4ab0d6985 : SUCCESS in 1h 22m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa2f6fa9604e42aeb9f2587b741cd831 : SUCCESS in 51m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8cf637de1c6d4bd1aac6fcba14733c1f : SUCCESS in 55m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4e699128564b41578498083635b5c2b5 : SUCCESS in 52m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/140901578d0346cc91b303f2952431d4 : SUCCESS in 1h 00m 20s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/0c57535ddbca4a6bbdb8748ee9dc12a2 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ccbafbf36877d6cb19a3709932c6af9c58c171ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 19:14:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"df301d9ad848e4a40f5d4df3dfe616e1d0d7d040","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 19:18:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f871bf546fa244e7981f51a8cc1b46d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/617f71ee1385475aabcd8ac42edc0140 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52aa69d16c92afab0cb45df95f217a6ce69716b2","revisions":{"52aa69d16c92afab0cb45df95f217a6ce69716b2":{"kind":"REWORK","_number":1,"created":"2025-06-26 13:31:48.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/47/953447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/953447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/953447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/953447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/953447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/953447/1"}}},"commit":{"parents":[{"commit":"59c9ad47d876bb589d116c496d6a893c502d0e7e","subject":"[stable-only] Remove \"devstack-tobiko-neutron\" job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59c9ad47d876bb589d116c496d6a893c502d0e7e"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2025-05-07 14:29:58.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2025-06-26 12:22:04.000000000","tz":120},"subject":"Limit trunk ACTIVE state hack to OVN","message":"Limit trunk ACTIVE state hack to OVN\n\nIn https://review.opendev.org/c/openstack/neutron/+/853779 we started\nmoving a trunk to ACTIVE when its parent port went to ACTIVE. The\nintention was to not leave the trunk in DOWN after a live migration as\nreported in #1988549. However this had side effects. Earlier we moved a\ntrunk to ACTIVE when all of its ports were processed. That means we\nunintentionally changed the meaning of the trunk ACTIVE status. This\naffected all backends and not just live migrate but create too.\n\nThis change moves the logic of propagating the trunk parent\u0027s ACTIVE to\nthe trunk itself to the OVN trunk driver, so we limit the undesired\neffects to ml2/ovn. By that we restore the original meaning of trunk\nACTIVE for all non-OVN backends. Ideally we would want to limit the\neffect to live migrate (so we don\u0027t affect create) but I did not find a\nway to do that.\n\nChange-Id: I4d2c3db355e29fffcce0f50cd12bb1e31d1be43a\nCloses-Bug: #2095152\nRelated-Bug: #1988549\nRelated-Change: https://review.opendev.org/c/openstack/os-vif/+/949736\nSigned-off-by: Bence Romsics \u003cbence.romsics@gmail.com\u003e\n(cherry picked from commit e69505d293ef12999f948531acbde75e16a65cd4)\n(cherry picked from commit 2224c9ea1e5a35acaad753a7f677d6788a612ec8)\n(cherry picked from commit fd9d90706da7965dabfc0521bdc15b468f6af899)\n(cherry picked from commit 24c492cf394ed23e431c72ac0c87f8a0246ee007)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52aa69d16c92afab0cb45df95f217a6ce69716b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52aa69d16c92afab0cb45df95f217a6ce69716b2"}]},"branch":"refs/heads/unmaintained/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
