)]}'
{"id":"openstack%2Fneutron~954292","triplet_id":"openstack%2Fneutron~master~I69d49cd3737b3bee8c8b4221149085375e9ac1e2","project":"openstack/neutron","branch":"master","topic":"bug/2116087","attention_set":{},"removed_from_attention_set":{"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2025-07-14 16:38:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69d49cd3737b3bee8c8b4221149085375e9ac1e2","subject":"[OVN] Fix remote-managed binding profile validation","status":"MERGED","created":"2025-07-07 19:20:22.000000000","updated":"2025-07-14 16:41:26.000000000","submitted":"2025-07-14 16:38:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954292-bug/2116087","meta_rev_id":"a68b736645b372e7d118ee5efa3c66854bab7d02","_number":954292,"virtual_id_number":954292,"owner":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-14 16:38:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2025-07-14 11:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2025-07-14 14:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2025-07-14 14:29:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-07 20:56:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-07 22:48:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-07-14 11:50:49.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-07-14 14:29:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"d2242e17e3bb281907a829769099b6ea5a3dfc89","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2025-07-07 19:20:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8f664a48211ca27110bb1a5f3def6d537d89289","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2025-07-07 20:04:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"e0e413a5c53a764a726accba08384a2ead4a09c5","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2025-07-07 20:05:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"067e2350eb71d39af46f85740247001cf421c68e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 20:56:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd22864d46344562a05e89ec952b68c7\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3d661a9c5cc546a7a12f620b6d5c1f94 : SUCCESS in 1h 29m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2eab31e7bf2e41428fb37b4599c754fc : SUCCESS in 34m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aafdd44118e44ed9bf422d1d9df455dc : SUCCESS in 30m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51edbae55cd1497aa62fd71260ee568f : SUCCESS in 11m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94f022c5204b4721a2cc2bf905a6fb67 : SUCCESS in 15m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96e1ea37cb324848b7485408bfff033d : SUCCESS in 17m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abe2dd4a69c544d79a2aa8adb7dcea1d : SUCCESS in 18m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3628a661c7084ed5a545a390ef60dd86 : SUCCESS in 9m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/611dc6f1fb394d82b8810cc9cb23a54b : SUCCESS in 1h 34m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f919d00fdd6842808d56f8a8891b40f5 : SUCCESS in 1h 07m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2d6fd77e87864520b152fc2850f14b97 : SUCCESS in 55m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5546b266d8b74852b984b04294dcf26e : SUCCESS in 53m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/09b760c5be394d0ea80c8ca1a1c163b1 : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3315facbe3f9e377a3be543ac05502adf461a745","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 20:59:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f678bdb39d54b10b10992cceb5f01a2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54a1261a854a4351bdc52473917466dc : SUCCESS in 1h 03m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3212f0f8379f46b8bfe9c7552a0e272b : SUCCESS in 22m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e17b8f6e906446aea820e3b15fba5fdc : SUCCESS in 1h 04m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"680830a647c72f04796811d239123859e3177490","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-14 11:50:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ebb3b0039b327ea229eff7ae2f6773ed6fb87bad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-14 14:29:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7bfd6a0058a028165e95923ae466e87f1d32d87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 14:29:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2efe21dcc8b25c7cf202170c93eb074483fc20ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 16:38:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c301071f65f14d929464e31740fad7cb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/22b54b01e59c424b8c44b234b5f0a1a7 : SUCCESS in 2h 07m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b7633f21494b9ba4a5a3eb2ecb5eab : SUCCESS in 12m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/832e38d134784b38afd18e26e7ac429f : SUCCESS in 23m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50a81f4ca4ef493a8c7bd32beb71b3d0 : SUCCESS in 26m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033d9b1b9de64a1db2b38b4f20f3e813 : SUCCESS in 12m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/34b6d5c4e3424b7d9359f76c695a5a4a : SUCCESS in 1h 21m 59s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e136e0d7e974615b47d18a0d1d023a3 : SUCCESS in 1h 13m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/820ae1acc9694606bc8bff295f17fbcc : SUCCESS in 54m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/caf60b7285514cc78a3be0c8cd0f0d34 : SUCCESS in 1h 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ffb66ab584bce57f6756e390197993bc2b6f5c0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 16:38:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a68b736645b372e7d118ee5efa3c66854bab7d02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 16:41:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6327303805cb43939b2354ad32d383e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d15e955c424ef6ab3240ffbd4f5c51 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e01e6a7130a570092e6b7324b7f9d051a34735d","revisions":{"1e01e6a7130a570092e6b7324b7f9d051a34735d":{"kind":"REWORK","_number":1,"created":"2025-07-07 19:20:22.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/92/954292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/954292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/954292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/954292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/954292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/954292/1"}}},"commit":{"parents":[{"commit":"71962255dee06d75258984b39d3a5c6c6293534d","subject":"Merge \"[OVN] Set OVN bridge in Chassis_Private in OVN agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71962255dee06d75258984b39d3a5c6c6293534d"}]}],"author":{"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","date":"2025-07-07 18:35:20.000000000","tz":0},"committer":{"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","date":"2025-07-07 19:20:09.000000000","tz":0},"subject":"[OVN] Fix remote-managed binding profile validation","message":"[OVN] Fix remote-managed binding profile validation\n\nWhen binding profile parameter validation was extended in\nchange I8493cba28e92b9b36bdb952c9737c0fea6fb7b75, a subtle bug\nwas introduced.\n\nNamely, if a binding profile specification specified ``None`` for\nexpected capability, the code would literally look for an instance\nof ``None`` in the list of capabilities, as can be read from the\ndocumentation [0] this was not the intended behavior.\n\nUpdate unit tests that asserted the incorrect behavior.\n\nAdd new unit test specifically for the case of a remote-managed\nvnic that also has extra capabilities.\n\n0: https://opendev.org/openstack/neutron/src/commit/71962255dee06d75258984b39d3a5c6c6293534d/neutron/common/ovn/constants.py#L340-L353\nCloses-Bug: #2116087\nChange-Id: I69d49cd3737b3bee8c8b4221149085375e9ac1e2\nSigned-off-by: Frode Nordahl \u003cfnordahl@ubuntu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e01e6a7130a570092e6b7324b7f9d051a34735d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e01e6a7130a570092e6b7324b7f9d051a34735d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
