)]}'
{"id":"openstack%2Fneutron~954294","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I7bf4dc396a044d0d67d2d2f6e68140c063f3e3d8","project":"openstack/neutron","branch":"stable/2025.1","topic":"bug/2115952","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-07-15 17:49:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7bf4dc396a044d0d67d2d2f6e68140c063f3e3d8","subject":"Use the OVN policer for max-bw only QoS policies","status":"MERGED","created":"2025-07-07 19:53:47.000000000","updated":"2025-07-15 17:50:48.000000000","submitted":"2025-07-15 17:49:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954294-bug/2115952","meta_rev_id":"b922ca35e27f34b67953819dd47b6fce8c669e3b","_number":954294,"virtual_id_number":954294,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-15 17:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-07-15 15:51:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-07-08 15:27:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2025-07-15 15:51:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-07 21:00:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-07 21:59:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-08 09:13:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-07-08 09:13:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-07-08 09:13:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-07-08 15:27:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-15 15:51:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"790de6c584f1c57172c6be8a653b04ed8136cad4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-07 19:53:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3935177b6fbd1fadaacdb3e80f5109c5409860e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 21:00:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccf3242fda8b467daa317ed6957d55f6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/792ad052773c4762b8339c09decc5e19 : NODE_FAILURE Node(set) request 300-0027361700 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ca67ec98ca9a4c1f95e9f97f49b9ffee : SUCCESS in 1h 00m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57458c3c6e56434a6b27e1d3c4921f35ea949f8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 21:59:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11ed3267cca8411391944339ef46446a\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/12ddc8c7baff4d1487f29871316c4654 : SUCCESS in 2h 04m 26s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/c674319ce5f1487186f9797f1440be9f : SUCCESS in 1h 19m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86737b44ca3f4857b5e1327a9e981753 : SUCCESS in 33m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf5b0992f9ae41639e82ce2ea3fc0d6d : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00646533a48d469ead1ac4b0d349ad58 : SUCCESS in 14m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64904963b5ef473fb41a27094343a71d : SUCCESS in 15m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ad59ee965545f4a929b06d59d103f9 : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3464bbef540042e791a31fe0e3661c7e : SUCCESS in 14m 35s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bedfbee419a447948dd86792c1adf63f : SUCCESS in 49m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a834c2569a22482fa84890ac5c65378d : SUCCESS in 1h 26m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad4c147e5fa640838378fb02dfd42934 : SUCCESS in 1h 10m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1464603c2a8c49a19a21fdd31d980a25 : SUCCESS in 48m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4765605ad24f495a9555eff618053c84 : SUCCESS in 33m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/55c7ff7f8d224607ad74dc830457fbb3 : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6d01c696b0320ca8ed822e4e97a77fbdd343092e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-08 15:27:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"56db90231c1b62df59a20d464115b8d9129daa8a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-07-15 15:51:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebe7b7d3484852c2d5140d59c1cb7797c8ab80e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 15:51:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ec64bea5ae85509b446dc722be4d1cba4d4971b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 17:49:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e562e82bd0ad4a6c84bfc932ae95cc4c\n\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/741c1832787d41de8199f427e5293a1f : SUCCESS in 1h 57m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e37cc27d7ff4cc380f3c909db319072 : SUCCESS in 12m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd526aac1f4446ad827332aca6eceb06 : SUCCESS in 14m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2ed26b524c14854a19af19d3b2a0e89 : SUCCESS in 13m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147d4d9dd52b468781610342c47bd6a5 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3e6ac8be5044bd58983297c772ae593 : SUCCESS in 15m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e97350fd320458bb25d0e03459bf323 : SUCCESS in 1h 01m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/98ede0cc914347198fb866cbc147e65f : SUCCESS in 1h 09m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d7919d1e378043fa88ed94b19bde16d2 : SUCCESS in 45m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b5128620e1484483bf0c9627d9f6412c : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":1},{"id":"683fb687bbc9d3c2c64ee57963accbf3b81d325a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 17:49:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b922ca35e27f34b67953819dd47b6fce8c669e3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 17:50:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da636532a91f46e48150923ffdfcc981\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217f447241144f39a8cfc4c8d2c860c4 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/038b5ece39504534abd3e5d56bd736ad : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63ba2c92943f2a06236e7359dbc0d80a94fbc446","revisions":{"63ba2c92943f2a06236e7359dbc0d80a94fbc446":{"kind":"REWORK","_number":1,"created":"2025-07-07 19:53:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/94/954294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/94/954294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/954294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/954294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/954294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/94/954294/1"}}},"commit":{"parents":[{"commit":"62606a017eb97df849287d8b4f72427afff3e873","subject":"Merge \"[doc] Change OVN agent \"extension\" section to \"agent\"\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62606a017eb97df849287d8b4f72427afff3e873"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-07-04 14:17:18.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-07-07 19:53:34.000000000","tz":0},"subject":"Use the OVN policer for max-bw only QoS policies","message":"Use the OVN policer for max-bw only QoS policies\n\nAs reported in the related bug, when [1] was implemented, that\nintroduced a regression when a port located in a physical network,\nwith maximum bandwidth *only* QoS rules and egress direction, was\nbound to a compute node with a bond interface on the physical network\nbridge. The TC commands have no effect on the interfaces of the bond.\n\nThis patch restores the backend enforcement for this scenario (physical\nnetworks, egress direction, maximum bandwidth rule only) to the OVN\npolicer, using the OVN ``QoS`` registers, instead of relying on the\nphysical interface bandwidth enforcement, using TC commands.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/934418\n\nCloses-Bug: #2115952\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I7bf4dc396a044d0d67d2d2f6e68140c063f3e3d8\n(cherry picked from commit 4ebd69f2c5456d8f3a357eefac1e07c46c54fea3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63ba2c92943f2a06236e7359dbc0d80a94fbc446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63ba2c92943f2a06236e7359dbc0d80a94fbc446"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
