)]}'
{"id":"openstack%2Fneutron~954615","triplet_id":"openstack%2Fneutron~unmaintained%2F2023.1~Ie038a896b1d3d779a04a1262b29791f996f40b15","project":"openstack/neutron","branch":"unmaintained/2023.1","topic":"ovs-rally-job-fix-osprofiler","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2025-07-21 13:28:40.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-07-21 13:28:40.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-07-21 12:14:54.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-07-21 13:28:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie038a896b1d3d779a04a1262b29791f996f40b15","subject":"[stable-only] Pin osprofiler to 4.3.0","status":"MERGED","created":"2025-07-10 14:21:30.000000000","updated":"2025-07-21 13:32:40.000000000","submitted":"2025-07-21 13:28:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954615-ovs-rally-job-fix-osprofiler","meta_rev_id":"e384d747eb5244c956aa4e3cde33a26a3a8c3f87","_number":954615,"virtual_id_number":954615,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-21 13:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-07-21 06:41:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2025-07-21 11:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-07-21 12:14:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-07-21 06:41:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-10 15:50:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 17:50:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-14 08:31:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2025-07-21 06:41:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-07-21 11:15:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"151a576696051592301d6f3239141051f6c1451b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-10 14:21:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63a76395a56e48b3dba668f7b78680c5952f8035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:50:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b4d4d9cab414e7cb0340defbb73338c\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/bf785289f4a3450a813a48536e812aed : SUCCESS in 1h 11m 30s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/30201712762f40999282c1d7c2e5cd4a : SUCCESS in 56m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/7d6ae1e5ec7d44dba344bc387708bf5a : SUCCESS in 1h 12m 30s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/f7676dbe1d13406ab28518ab20e0205f : FAILURE in 1h 27m 36s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/2dad37d090f24e668f154c932635d9f0 : SUCCESS in 27m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a397f5fc1b147cdbc5f0549fddd0030 : SUCCESS in 14m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa38146ab3d4263b8e057686dda708b : SUCCESS in 10m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d01b35ae0baf4a7fb08aa0ba281bc74b : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ce8e5d2905dd29755a10969691e569cea47e03cd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-10 17:50:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fe7808fe9875939776fcca6a507f5b2816f240c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 19:23:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e8e00263337487dac8bf5c2d9312f06\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/1faca0ac1e744653a3c76d46812be470 : SUCCESS in 1h 21m 55s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/9e5f9632737c4733815c1edbf7e978ba : SUCCESS in 58m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/bea4df3ceff54da3a16f6cff2c76cefa : SUCCESS in 1h 05m 06s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/2b3dbd4a3bc8443188d15ef06b7446c3 : FAILURE in 1h 29m 02s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/e7005b0069314c97bdaad49414a033bb : SUCCESS in 27m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b04d1c5e09b64f238be8346acf1d405f : SUCCESS in 13m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8930344c4c41259bedd31bf6b253e0 : SUCCESS in 11m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/422a6c32838a43e781058f5eb0eedd54 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d5fdea733725daa7f3b12f343b4bc1cb0debcb89","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-14 21:45:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"478252a4b30b4d33952aed4067d932c0910bac3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 23:04:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7b853fb4ee8416583ebcbf1bda1dde9\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/5b45e638ddf74718bcc5113fd955fbdb : SUCCESS in 1h 17m 52s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/fac554081c104396b6c04194e9e3788f : SUCCESS in 1h 03m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/f09b3aef1378427aa9afedb3f2585912 : SUCCESS in 1h 02m 11s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/970d3e76ad1d499db413f9b3b5e9fc37 : FAILURE in 1h 08m 20s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/53885b7023dc4f239051292d6e14ac8e : POST_FAILURE in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e9fb5e9c2c431d9ec0276e7a5a520e : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd44d9c6d59f498ab73ad8a6a6b91344 : SUCCESS in 11m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5a3346a7369c42df903d8aac15ac091b : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c9a5b13eea799e5500949d27b7154dbcbf6d9a13","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-18 02:00:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"345f65e0abd60b3785ffe12828d5469dc0297787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 04:00:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5ea8e8eeb214c82b762ff81bbb20c7a\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/6813f11fa201403caf1b982bb0852ddf : SUCCESS in 1h 37m 41s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/7605039d20604c6e845eeebd89135b30 : SUCCESS in 1h 19m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/451545afe4a64eaf96d11aa6585e036c : SUCCESS in 1h 41m 02s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/c90a9bbd84044cb8b2eca046d2be1d1c : FAILURE in 1h 58m 56s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/c482513ce8d34a90929e6d5e71f783cd : SUCCESS in 43m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b218f638a20443e79ec4ac774026a5cb : SUCCESS in 14m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6c3f4416664f5ba4138d58b0f07b2b : SUCCESS in 9m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/acc9e75a1d934fcf9b7a56517c56b134 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2730c5f03368196b4c17204d2920b2cd6ee11af3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-21 06:25:16.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7e92d8055060518bd7d3e1818598bc7b99d3f7f0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-21 06:41:06.000000000","message":"Patch Set 2: Code-Review+1 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"5ba8db739bc69ca45740aa0ea63f02b4a5c98bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 07:48:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b40943637ef4074929b061dd3ce7248\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/478514a62d2b4ceaa9997eec93ca6080 : SUCCESS in 1h 17m 09s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/72ad49cc9ef0438c8eadf8772825f9f1 : SUCCESS in 1h 03m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/137aa8992fe642ed9fa96342982c596c : SUCCESS in 50m 08s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/0e5368f8c4a34c41ad8be7067a6647a0 : SUCCESS in 1h 12m 23s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/c934f809decd47989dcaf4643d2c1a65 : SUCCESS in 29m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b89dd48d8a14fd2a735b6524078ab24 : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d11b40ce53e497c9d1a5e17e89a4c88 : SUCCESS in 9m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b81441c0fa284c3ca0f3c1192f878a86 : SUCCESS in 1h 17m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4a09ca9bdf9c1f60343954317b9e2efe90383829","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-21 11:15:01.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4849da93ec191609f091e1966f6badfacd6b6b58","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-21 12:14:54.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7bd52c2b2d0c78d802179e418b3722684694d7f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 12:15:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"194507816746db5dffe64de53724835afe44555b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 13:28:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a9daedaee244ecc8d9111be9dbaa0a4\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/a2fa422bf3f24f2fa75e88c66056cefd : SUCCESS in 1h 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/234aad208323499c91b9d288aa615c6c : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca22b1f7bccd4a8391fc3621e2b828af : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d849f42b4297ffaa04387505ff39ef48fd493738","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 13:28:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e384d747eb5244c956aa4e3cde33a26a3a8c3f87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 13:32:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a662977dc54d6ebb3a1c9a09382d1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95293979ad24db29ba1eeef549e7be2 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"838bc618b1a67b5ec8d9b0e78e283e330ad67717","revisions":{"20398696c0709c6b313462e8674aaa7728ea2e23":{"kind":"REWORK","_number":1,"created":"2025-07-10 14:21:30.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/15/954615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/954615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/954615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/954615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/954615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/954615/1"}}},"commit":{"parents":[{"commit":"59c9ad47d876bb589d116c496d6a893c502d0e7e","subject":"[stable-only] Remove \"devstack-tobiko-neutron\" job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59c9ad47d876bb589d116c496d6a893c502d0e7e"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-07-10 14:17:07.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-07-10 14:20:57.000000000","tz":120},"subject":"[stable-only] Pin osprofiler to 4.3.0","message":"[stable-only] Pin osprofiler to 4.3.0\n\nneutron-ovs-rally-task job started to fail now that pyproject.toml was\nadded in osprofiler for package building.\nThis patch pins osprofiler to previous release that does not contain\nyet the new build method.\n\nChange-Id: Ie038a896b1d3d779a04a1262b29791f996f40b15\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20398696c0709c6b313462e8674aaa7728ea2e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20398696c0709c6b313462e8674aaa7728ea2e23"}]},"branch":"refs/heads/unmaintained/2023.1"},"838bc618b1a67b5ec8d9b0e78e283e330ad67717":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-07-21 06:25:16.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/15/954615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/954615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/954615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/954615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/954615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/954615/2"}}},"commit":{"parents":[{"commit":"59c9ad47d876bb589d116c496d6a893c502d0e7e","subject":"[stable-only] Remove \"devstack-tobiko-neutron\" job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59c9ad47d876bb589d116c496d6a893c502d0e7e"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-07-10 14:17:07.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2025-07-21 06:25:16.000000000","tz":0},"subject":"[stable-only] Pin osprofiler to 4.3.0","message":"[stable-only] Pin osprofiler to 4.3.0\n\nneutron-ovs-rally-task job started to fail now that pyproject.toml was\nadded in osprofiler for package building.\nThis patch pins osprofiler to previous release that does not contain\nyet the new build method.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/955450\nChange-Id: Ie038a896b1d3d779a04a1262b29791f996f40b15\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/838bc618b1a67b5ec8d9b0e78e283e330ad67717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/838bc618b1a67b5ec8d9b0e78e283e330ad67717"}]},"branch":"refs/heads/unmaintained/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
