)]}'
{"id":"openstack%2Fneutron~954789","triplet_id":"openstack%2Fneutron~unmaintained%2Fyoga~Ieb1b116c4d0866bd8208f9b8b440c1e274c82b1c","project":"openstack/neutron","branch":"unmaintained/yoga","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-07-12 10:04:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieb1b116c4d0866bd8208f9b8b440c1e274c82b1c","subject":"Remove non voting arm64 jobs from stable branches","status":"MERGED","created":"2025-07-11 19:23:04.000000000","updated":"2025-07-12 10:07:17.000000000","submitted":"2025-07-12 10:04:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954789","meta_rev_id":"f6de85ce9ac762f5079cd6f7fbf37dc1c499ff90","_number":954789,"virtual_id_number":954789,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-12 10:04:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-07-12 08:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-07-12 08:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 19:23:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-11 20:35:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-12 08:56:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a1f5ab233d5418137e2fb70b7cff74d2dc8f5f81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-07-11 19:23:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d63c4ae28cb1942169587b9ef2e359a5979fda67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 19:23:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f14dd463f92c43caaca7703c9ed051cc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f05fde6cb4b4dd8a5a968c3e99dd0b3 : NODE_FAILURE Node(set) request 300-0027362238 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/475014c694954ef2b12054a5ed40c5af : NODE_FAILURE Node(set) request 300-0027362239 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bc56731893d9d09259afb0f05fd1783b0d4a375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 20:35:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f626ec1539d4cd48103a9b6b01a5596\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/a70eea74b09b453a867bb66e405212e9 : SUCCESS in 47m 50s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/883b3938642c40a6913fa386ec32f15f : SUCCESS in 1h 06m 08s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/56a888e0d9924e6ab1e1f361570411d1 : SUCCESS in 46m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/a9242083553b4edfb249fc97e1201373 : SUCCESS in 1h 07m 42s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/8fdeb1f817a6463b8589f48dee0d3def : SUCCESS in 1h 10m 23s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/1fdd87662d2547408c532e5d0e81d7a8 : SUCCESS in 25m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5908e99399b48d5a6a404eb5d525ad9 : SUCCESS in 38m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9366fcb7449b4992bb2ec78b70e8d9fb : SUCCESS in 12m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d508548d4b144f4f827e02dbfb7b6dba : SUCCESS in 32m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d38f7769d25c4e828c07088b4fb031cd : SUCCESS in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7355bd97054fd9907cbd08e8fdd330 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":1},{"id":"29264e6186979e9893f9145684c6973c2f86fde1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-12 08:56:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ebb83efb848cd357b1e01a745f515d4e3c1fb24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 08:57:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f53b81f71b19b87b89d2bf0c68c5217a02491b60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 10:04:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b1c4f0adcb4c1c8783aa1f9949fe9c\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/1758e40bba1e4698a71dd2cbe96858e0 : SUCCESS in 52m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921ca63c1c4a4a2095c3f7c41e71edc7 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0437e032e8f4d12a97dd786a57895a0 : SUCCESS in 20m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b44e73499a34c2cb2a7a48c1863d7e3 : SUCCESS in 17m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae7dad1751fc42c29f719764572ea6b2 : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b736c1112e238ec0d0f918f54921edf0cfb390bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 10:04:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f6de85ce9ac762f5079cd6f7fbf37dc1c499ff90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 10:07:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f3c3b8f1ff249b0b0f03a6b8f723953\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c96923d5064fb585bf297feeb63bd2 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"106598d0b168e7a24b62205d995132de513b2730","revisions":{"106598d0b168e7a24b62205d995132de513b2730":{"kind":"REWORK","_number":1,"created":"2025-07-11 19:23:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/89/954789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/954789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/954789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/954789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/954789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/954789/1"}}},"commit":{"parents":[{"commit":"18f8a72f3bf1fd3f64d2ebea05dd725e059641d1","subject":"Don\u0027t print traceback if standard attr is missing on update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18f8a72f3bf1fd3f64d2ebea05dd725e059641d1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-07-11 19:01:53.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-07-11 19:22:52.000000000","tz":-420},"subject":"Remove non voting arm64 jobs from stable branches","message":"Remove non voting arm64 jobs from stable branches\n\narm64 jobs are failing 100% and those are not fixes on\nmaster. It does not make sense to carry forward these broken\ntesting to stable. until those are fixed on master, we should\nnot run them on stable.\n\ngeneric template openstack-python3-jobs-arm64 will take\ncare of running arm64 jobs on stable branches once jobs\nare passing.\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/954781\n\nChange-Id: Ieb1b116c4d0866bd8208f9b8b440c1e274c82b1c\nSigned-off-by: Ghanshyam Mann \u003cgmaan@ghanshyammann.com\u003e\n(cherry picked from commit d7e79d13371a49889243a6b6a217846ec3e36ca5)\n(cherry picked from commit eea4c1855e5adade2de3212f88363017ba55157d)\n(cherry picked from commit a35f555d60a31c6c3e334b5eb37c7653fa33dad5)\n(cherry picked from commit 9d51d8ff538f5f21b375e99e6d688f94f75a31db)\n(cherry picked from commit 7227063997e4827a0cea5d59a466c55c53e914d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/106598d0b168e7a24b62205d995132de513b2730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/106598d0b168e7a24b62205d995132de513b2730"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
