)]}'
{"id":"openstack%2Fneutron~954791","triplet_id":"openstack%2Fneutron~unmaintained%2Fwallaby~Ieb1b116c4d0866bd8208f9b8b440c1e274c82b1c","project":"openstack/neutron","branch":"unmaintained/wallaby","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-07-12 10:05:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieb1b116c4d0866bd8208f9b8b440c1e274c82b1c","subject":"Remove non voting arm64 jobs from stable branches","status":"MERGED","created":"2025-07-11 19:24:54.000000000","updated":"2025-07-12 10:07:29.000000000","submitted":"2025-07-12 10:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954791","meta_rev_id":"b53d77e6a55851b4bbbdcd669df4dcf05aa7123e","_number":954791,"virtual_id_number":954791,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-12 10:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-07-12 08:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-07-12 08:57:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 19:25:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-11 20:16:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-12 08:57:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1659ae04971cb669295eeeb23e4491e6555bf9c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-07-11 19:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e8dc2b2806494d7e11d32c17054175444453da5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 19:25:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdb79d8c5f8f41d48c9260c7bcc730f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d36df531f7440e8933e84fea45005c5 : NODE_FAILURE Node(set) request 300-0027362242 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40c9bc82b8a54452a561e7c5abb3e911 : NODE_FAILURE Node(set) request 300-0027362243 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4d2b54c115363c7b5e064c378fc14ac85f1d488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 20:16:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5265bd53eee34c1f9a483aba0d1cf584\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/c0b090bb46734b71a6d2c3eb5fbd8d2e : SUCCESS in 47m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cd7750aec4f4be89a79121d8484200a : SUCCESS in 39m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/773fa8f8bb804da49579ae5272cac637 : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/947e1770b14e4c75b4905efcf776299b : SUCCESS in 28m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c8a4d7aa2fa48bf8268ab005a062bcc : SUCCESS in 25m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05e6f225409b4d6bb3f1a58d368302da : SUCCESS in 32m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32d71dc6d9a4111a9239f7ed166a7fa : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c6790e82fd36ad5dc2f8348cfc9236fe84b8b1df","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-12 08:57:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8a8762da39a461ecf228b7f48f49e1ecf11ddc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 08:58:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3bbeaa52c54a1f216a4a40c064721fc7d80f9ce0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 10:05:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a0d127cb14e4599b94e73627fa260ff\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/9b1b1e5ac03a4e7496635757bc62f55f : SUCCESS in 26m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bda54fb47cb4c028ccdc18704e384b9 : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ea3fc5d527c4bd2affb2115dcd5ad9d : SUCCESS in 30m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b36da0fadde44909ba5735cbbc91d2ea : SUCCESS in 20m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/404193c1f19845248a55c036b4eb278a : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2636cf3108926ea20500d69feac434234ef325e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 10:05:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b53d77e6a55851b4bbbdcd669df4dcf05aa7123e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 10:07:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f996f7bf727473b8c62d38841b52160\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deaa0a2a63234d65bb3678cb91526fd1 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea43dfe3910575431954454cc9d2b425963362cb","revisions":{"ea43dfe3910575431954454cc9d2b425963362cb":{"kind":"REWORK","_number":1,"created":"2025-07-11 19:24:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/91/954791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/954791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/954791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/954791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/954791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/954791/1"}}},"commit":{"parents":[{"commit":"9aac096007cc01c625bef5820a7ca893969ce832","subject":"[unmaintained only] Drop sdk job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9aac096007cc01c625bef5820a7ca893969ce832"}]}],"author":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-07-11 19:01:53.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-07-11 19:24:43.000000000","tz":-420},"subject":"Remove non voting arm64 jobs from stable branches","message":"Remove non voting arm64 jobs from stable branches\n\narm64 jobs are failing 100% and those are not fixes on\nmaster. It does not make sense to carry forward these broken\ntesting to stable. until those are fixed on master, we should\nnot run them on stable.\n\ngeneric template openstack-python3-jobs-arm64 will take\ncare of running arm64 jobs on stable branches once jobs\nare passing.\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/954781\n\nChange-Id: Ieb1b116c4d0866bd8208f9b8b440c1e274c82b1c\nSigned-off-by: Ghanshyam Mann \u003cgmaan@ghanshyammann.com\u003e\n(cherry picked from commit d7e79d13371a49889243a6b6a217846ec3e36ca5)\n(cherry picked from commit eea4c1855e5adade2de3212f88363017ba55157d)\n(cherry picked from commit a35f555d60a31c6c3e334b5eb37c7653fa33dad5)\n(cherry picked from commit 9d51d8ff538f5f21b375e99e6d688f94f75a31db)\n(cherry picked from commit 7227063997e4827a0cea5d59a466c55c53e914d2)\n(cherry picked from commit 106598d0b168e7a24b62205d995132de513b2730)\n(cherry picked from commit 8fbbfc7728cb3d238bd094a6d971f6b88388721b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea43dfe3910575431954454cc9d2b425963362cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea43dfe3910575431954454cc9d2b425963362cb"}]},"branch":"refs/heads/unmaintained/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
