)]}'
{"id":"openstack%2Fneutron~954908","triplet_id":"openstack%2Fneutron~master~Ia7ffb977401aa5c9cfe14df61d496fd0815e4c29","project":"openstack/neutron","branch":"master","topic":"pre-commit","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-07-14 13:40:21.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}},"hashtags":[],"change_id":"Ia7ffb977401aa5c9cfe14df61d496fd0815e4c29","subject":"Revert \"tox: Fix indentation\"","status":"ABANDONED","created":"2025-07-14 09:18:03.000000000","updated":"2025-07-14 14:42:05.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9b18f622f3c7aeb2356f2c0e190ee71ac1b625b3","_number":954908,"virtual_id_number":954908,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-14 10:37:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-07-14 09:40:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2025-07-14 13:40:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-14 09:40:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-07-14 09:55:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-14 10:37:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9886a1656460be1352ae6f3fe5ab941a13ae621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-14 09:18:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84fe128f0e37d635211ff4e600c1180a4abbc8dc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-14 09:40:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d675000c5cf6fe551c46bda505c73d527d46cabd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 09:55:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a45daf24f414f6eb86568173ce0ce39\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1d756505047f44c386fe9c12e77c0bda : SUCCESS in 35m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a4f10a66acb942df85499a0e906911df : SUCCESS in 24m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c9250de8ce9f416c9d21b8d374791e81 : SUCCESS in 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6917fbd955816734018af187614b2beb8d5df380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 10:37:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1038bb3bb2de4d17916611af96e52127\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a78d2d5e01d4d4195fde452953663b0 : SUCCESS in 30m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18bfa27b51e04bc08521a6438ff51c0a : SUCCESS in 14m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb6c90b2637445ea95ff67bb6f06b171 : SUCCESS in 14m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04f5ffc612604cf0b3bebeab6c44ad10 : SUCCESS in 28m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5504a6384c914a12952ccbf2829bc156 : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b4a5bf04c34f6a99979d0412626258 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f464a0d59c2341748ce0fdbecd38dccd : SUCCESS in 19m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a4edbfd3596d479180c8be8c21eb9ac7 : SUCCESS in 1h 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3842bd0618ae2cd0602e0526d07ea3b51a5ddec8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-14 13:40:21.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b18f622f3c7aeb2356f2c0e190ee71ac1b625b3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-14 14:42:05.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/neutron/+/954894","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6144b597348b6c59eae7eccddbc35727a2e6e72","revisions":{"e6144b597348b6c59eae7eccddbc35727a2e6e72":{"kind":"REWORK","_number":1,"created":"2025-07-14 09:18:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/08/954908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/954908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/954908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/954908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/954908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/954908/1"}}},"commit":{"parents":[{"commit":"47b54e94ae56c0ab1bcbbd8c51d35046b4b2fc37","subject":"Merge \"Fix mocks in the test_l3 unit tests module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47b54e94ae56c0ab1bcbbd8c51d35046b4b2fc37"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-14 09:16:41.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-07-14 09:17:32.000000000","tz":330},"subject":"Revert \"tox: Fix indentation\"","message":"Revert \"tox: Fix indentation\"\n\nThis reverts commit d72adaa8d64291e9f7ca05de4c681c71b894af93.\nThe original commit did much more than fixing indentation\nand broke jobs which rely on TOX_PYTHON to be overridden.\n\nSigned-off-by: yatinkarel \u003cykarel@redhat.com\u003e\nChange-Id: Ia7ffb977401aa5c9cfe14df61d496fd0815e4c29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6144b597348b6c59eae7eccddbc35727a2e6e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6144b597348b6c59eae7eccddbc35727a2e6e72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
