)]}'
{"id":"openstack%2Fneutron~955065","triplet_id":"openstack%2Fneutron~master~I86fdf306775002922ae9832b2f8c48a90af3908b","project":"openstack/neutron","branch":"master","topic":"use_neutron-lib","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-07-21 09:23:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I86fdf306775002922ae9832b2f8c48a90af3908b","subject":"Use placement constant ``ALLOCATION``","status":"MERGED","created":"2025-07-15 13:45:36.000000000","updated":"2025-07-24 16:08:17.000000000","submitted":"2025-07-21 09:23:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955065-use_neutron-lib","meta_rev_id":"c3f7b3425ed56fc8cdd42d9a0a58fbd6113f72fd","_number":955065,"virtual_id_number":955065,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2025-07-24 16:08:17.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-21 09:23:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2025-07-21 06:34:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-15 16:56:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2025-07-21 06:34:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-15 14:47:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-15 16:05:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-15 16:56:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-16 10:07:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-07-21 06:34:07.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-07-24 16:08:17.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db8a9886d36ba14ce2a7aecd243045c76307048d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-15 13:45:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de2b7bac6f96586c4bdf2605a3a983c1ab7252df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 14:47:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7993b8bd01ad446187ed86e3cce75162\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6c1ddda05a624855aa423ee1522e855f : SUCCESS in 27m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/abd99a63e7b34ccabf23a0111477e1b5 : SUCCESS in 17m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f13872c8cfc04260a35f19ae35fe829a : SUCCESS in 30m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02d776f37bee92c73ceaffd2840d2ef60a3ca4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 16:05:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e7bfdc87dbc419585a40a4833ec7796\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f97196dda12c449fbca1576c65ad8803 : SUCCESS in 2h 18m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3820a2f87cc9403c9f09ac384ce70423 : SUCCESS in 31m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a642fa6d602d432ca9dd9ad76e4226f6 : SUCCESS in 12m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78fda4e1dd3a48ca996ed96657735b2e : SUCCESS in 19m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f026ac99cdc44f4aee56d8159083886 : SUCCESS in 19m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45734620ab5a4d38b9fab3f58b191a93 : SUCCESS in 16m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf8b15ee7b14601aed28ac9ebc3492b : SUCCESS in 12m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ce51031a59f6460f823eb8c50a143c9a : SUCCESS in 1h 32m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe4326983efe42448ff078006d7b0fe2 : SUCCESS in 1h 07m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/27da7ba670fb4e738f4abb42f66f86aa : SUCCESS in 1h 01m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/436a532377d2424592525d8454b5408a : SUCCESS in 36m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7e633700522544babc4510ab86dc4ec8 : SUCCESS in 1h 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d30ffc4ea7898d1d91d9e46f9661fdfc16da92e9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-15 16:56:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8ae0710109826d80712cdc6cb3e8a54d0145a35","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-07-21 06:34:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f5e954cb6dbe1e84a3105e1f602b60ac0f7563f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 06:34:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5ddb34b6685bcd538379623f5ae63405694460a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 09:23:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70518394fafe480a86ab28950c68c949\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/96fb95e7812543b1b67bc08aae6a3c98 : SUCCESS in 2h 26m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/212eeeade86a423f84a6b8061ee06654 : SUCCESS in 10m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/978cea4c6f334a5fbfe6a1de12627cc7 : SUCCESS in 20m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9ee015711ff486784c81b0dcfffa71e : SUCCESS in 15m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f173357c87614a5788270bf045807140 : SUCCESS in 11m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ac086db1e8454c90ad54f62d6479fb93 : SUCCESS in 1h 43m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/59e4ec86048b4a81b41702baf6d72995 : SUCCESS in 54m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cfb2a0e04eb74fd98287c9d4bc05e906 : SUCCESS in 55m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/129327a9649248f9a956f3e310074709 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0e3e3a92b6487025e354ce67c5c8e424c020c052","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 09:23:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87858b52f0b3d626020d6e86566e2e925f98db92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 09:27:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e26549db3a69471ebbf54bac3deb0322\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac79907e67804998816e4309d9b8b95b : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1},{"id":"c3f7b3425ed56fc8cdd42d9a0a58fbd6113f72fd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 16:08:17.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/65/955065/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0295a1918e00eb8dac81fdf24ce39c0bda6c95d7","revisions":{"0295a1918e00eb8dac81fdf24ce39c0bda6c95d7":{"kind":"REWORK","_number":1,"created":"2025-07-15 13:45:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/955065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/955065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/955065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/955065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/955065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/955065/1"}}},"commit":{"parents":[{"commit":"a8b8a646482fa807776c4ab2b4ccb06067e384c1","subject":"Merge \"Add a temporary grenade job using the OVN Metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8b8a646482fa807776c4ab2b4ccb06067e384c1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-07-15 10:44:08.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-07-15 13:45:20.000000000","tz":0},"subject":"Use placement constant ``ALLOCATION``","message":"Use placement constant ``ALLOCATION``\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I86fdf306775002922ae9832b2f8c48a90af3908b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0295a1918e00eb8dac81fdf24ce39c0bda6c95d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0295a1918e00eb8dac81fdf24ce39c0bda6c95d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
