)]}'
{"id":"openstack%2Fneutron~955067","triplet_id":"openstack%2Fneutron~master~I90b27357324023d1edbd97c75718db7a53aa6711","project":"openstack/neutron","branch":"master","topic":"refactor","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-07-16 13:16:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I90b27357324023d1edbd97c75718db7a53aa6711","subject":"refactor: Remove OvnIdlDistributedLock._last_touch attribute","status":"ABANDONED","created":"2025-07-15 13:48:29.000000000","updated":"2025-07-16 13:16:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"28e779b201a517630d80431130e5649cffd46676","_number":955067,"virtual_id_number":955067,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-15 15:58:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2025-07-15 16:56:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-15 13:48:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2025-07-15 15:24:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-15 15:58:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-15 16:56:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8abe1ddd678b90a509542838942dfd682802f79e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-15 13:48:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dc12735e5379e105d137349c1b818a5f3cf2a84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 15:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ed2e436d20546adb41c59a7ab7d5cea\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68de5abaf86f435d9b0e4065944b7dab : SUCCESS in 36m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/76d199d050484dceaa4d5c896f9a94a5 : SUCCESS in 19m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6d550572af654aa8b3fd4c61163221ec : SUCCESS in 31m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2f93976283297fe02eec344f3ce3d102bcf634b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 15:58:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6764400a8abb49c9a1387d49be8a7914\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6b500bb166aa418e9cc68071774abf78 : SUCCESS in 2h 09m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de009287125d44a098e2c7a858b4cb64 : SUCCESS in 39m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad1ddfe7ec74b83934690d4a495caf3 : SUCCESS in 11m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46a43b00d46f48f599a1fd2690a50e08 : SUCCESS in 20m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2988345f5d64ca18b929014106d530b : SUCCESS in 18m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dc35ce2072446e6a65ebdb2665e699f : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2024ddbda764f43ba31aaa81345573e : SUCCESS in 12m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3df85c48b2854d67890e65a6a161b7ae : SUCCESS in 1h 34m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/719b239f94c14f9481e0aa58647fdcf3 : SUCCESS in 1h 00m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e4b2c1ca475b4bfb943b25361dcfeb05 : SUCCESS in 57m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bad7b8fb9403445485a3b27ae15ba179 : SUCCESS in 1h 05m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a009c2b717824bf79be8c0638b069c69 : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":1},{"id":"54eb0403c746e82a034db00ade9c4636698bd4ed","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-15 16:56:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28e779b201a517630d80431130e5649cffd46676","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-16 13:16:36.000000000","message":"Abandoned\n\nSorry, I don\u0027t know how I pushed this patch. The correct one is https://review.opendev.org/c/openstack/neutron/+/939359","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"09dba442171766dfd64fa01bbe3a34fe4e43d640","revisions":{"09dba442171766dfd64fa01bbe3a34fe4e43d640":{"kind":"REWORK","_number":1,"created":"2025-07-15 13:48:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/67/955067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/955067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/955067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/955067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/955067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/955067/1"}}},"commit":{"parents":[{"commit":"a8b8a646482fa807776c4ab2b4ccb06067e384c1","subject":"Merge \"Add a temporary grenade job using the OVN Metadata agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8b8a646482fa807776c4ab2b4ccb06067e384c1"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2025-01-15 15:28:30.000000000","tz":-300},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-07-15 13:48:10.000000000","tz":0},"subject":"refactor: Remove OvnIdlDistributedLock._last_touch attribute","message":"refactor: Remove OvnIdlDistributedLock._last_touch attribute\n\nIt\u0027s overwritten on every notify call.\n\nIhar Hrachyshka \u003cihar.hrachyshka@gmail.com\u003e\nChange-Id: I4524034c51940532e6156dfdf2320813e5e91b5d\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I90b27357324023d1edbd97c75718db7a53aa6711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09dba442171766dfd64fa01bbe3a34fe4e43d640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09dba442171766dfd64fa01bbe3a34fe4e43d640"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
