)]}'
{"id":"openstack%2Fneutron~957579","triplet_id":"openstack%2Fneutron~master~I5f7900acf94beb1bc5dc6e28fdb0da2c7a036904","project":"openstack/neutron","branch":"master","topic":"bp/policy-service-role-default","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-08-26 15:18:57.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I5f7900acf94beb1bc5dc6e28fdb0da2c7a036904","subject":"DNM: Testign Nova service role","status":"ABANDONED","created":"2025-08-16 04:19:20.000000000","updated":"2025-08-28 16:21:12.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44a9eee3df3cfd10afa16f2a76f490b42fec93cf","_number":957579,"virtual_id_number":957579,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-26 17:13:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-16 07:29:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00463de0c0ff68f01b28e46d5a28031f3a8dbfb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-16 04:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1e3738bb9e8aad1a926b0732e30d8a1397e73dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-16 07:29:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f323ee7ba5c4c0ca4e56a2f158e1e72\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/442b4332960e409b89d33cd6939711c9 : TIMED_OUT in 2h 49m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c251b32341154e5f93c51b31496e470f : TIMED_OUT in 3h 04m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/20ce02c404064798b7b5e78bee2093ff : FAILURE in 36m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8941e0c87adc4790bbd7e12804f993a0 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f4bc0821de34397abca087988c78c47 : SUCCESS in 5m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/7d1067c51fd9442bbd580d19c6503e29 : SUCCESS in 1h 19m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ae2109315834729bb4123457c1eb51e : FAILURE in 49m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/37423834660749b0af6366489b8f7029 : FAILURE in 1h 05m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/260ebd71a21f430fa075b1eda888f286 : FAILURE in 2h 53m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6c7cd6d45fa94c27b4870e7cf4680cb2 : FAILURE in 3h 00m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/52fdfe7fae4c4ad18c5d8f071d0df3d4 : TIMED_OUT in 3h 02m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/64bb2c5397e44cefbb1a0391667cc3fa : FAILURE in 9m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/07d9d365622f4364ae67336c0602e450 : SUCCESS in 44m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/13598134f44b41af8ee0df3a20be8528 : TIMED_OUT in 2h 04m 03s","accounts_in_message":[],"_revision_number":1},{"id":"06a7c6ca26153a33380a14d6a2533aa4e6de5d6b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-16 19:03:15.000000000","message":"Topic set to bp/policy-service-role-default","accounts_in_message":[],"_revision_number":1},{"id":"be8e0a35590b5f6384cae801096395fa3056a0cd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-22 17:08:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9f87bfbba0eb2d0a5f81f3a48e07bfd4d7b630a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:55:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/492d421b337b4516a62080cc219a13ae\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4165830d2b01457982f419727d3772d0 : SUCCESS in 1h 21m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6c4d1daa5564a08865d77ea12add36f : SUCCESS in 1h 45m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0c4d29dcb9e54926b8e211f5ea05794b : SUCCESS in 31m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5967c77398c44c88883db22a362ad74 : SUCCESS in 15m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902dd170fdf74894a8f7e7bba7185f03 : SUCCESS in 12m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fb065b05b9fd41e3aa2d60f041f04d8c : FAILURE in 18m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8a9318698f94f98a41018fe0e6ce12b : FAILURE in 1h 06m 58s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/033559d273fd43e9ba9bc87ff578e1fa : FAILURE in 1h 02m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/717fab89babd44e69d0f65e17d1ab092 : FAILURE in 1h 17m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/401a2fa403734500beffba86e001205f : FAILURE in 1h 14m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8868df24c8a948b49e591111ad4e1e29 : FAILURE in 57m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/31b33f9df0104d53a71fe27ae299ad5b : SUCCESS in 59m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/688237ae030d41d99a754cb19160ef59 : SUCCESS in 31m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1bb0832ee4fd4edc8137009b06cd4fcf : FAILURE in 45m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f1365c927ddc43607e922ef6ec417e96d3741af6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-26 02:47:53.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"479b25c79d7a52bc88316ee84fbf8a60403d4dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 03:12:07.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/892639 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"7104955a32f4559485436118ad0101bcb16daff1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-26 15:18:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06796343a5612d11f7a33693ae0e08de9fe05094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 17:13:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f446a761a6964eabb1e4f0e60e6257fd\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7f7c47f519a84b2d81ef2549d768ca69 : SUCCESS in 1h 11m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/662b2c5081994d4bb107d8ed25f0203d : SUCCESS in 1h 42m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1bd6d099e8f54b148828c56c55209355 : SUCCESS in 32m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37afef363a3646fd8105926755cc4e35 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96f3eb057788498aa46e8283926bf181 : SUCCESS in 10m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/74b1c37e2c8f4b4dabfc808d3fad5ad1 : SUCCESS in 1h 39m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/af1bb5a308d94b02a9877a77b1ed3918 : SUCCESS in 48m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/68d1a790905249bfa81a83040df9675b : SUCCESS in 1h 12m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/21170b53af93492f87b055d66c2a0ef5 : SUCCESS in 1h 27m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6ea9e3f14dbe46ef8b1094cfbb7a11e6 : SUCCESS in 1h 23m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4a001b2646c64c049cdccf644643e1ff : SUCCESS in 1h 11m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0601676e8c8d485984e51d0e1c001c68 : SUCCESS in 58m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/014401b109e3481f85926a37d622ea71 : SUCCESS in 1h 22m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9f16682cfdbe40a9ba4f2431a3128a17 : SUCCESS in 1h 45m 05s","accounts_in_message":[],"_revision_number":2},{"id":"44a9eee3df3cfd10afa16f2a76f490b42fec93cf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-28 16:21:12.000000000","message":"Abandoned\n\nall good here in testing","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e51176cd8f819a758d127eb697d885aaffe81115","revisions":{"29274dd4428cd1ab01ba9014963d3f17a43ceb74":{"kind":"REWORK","_number":1,"created":"2025-08-16 04:19:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/79/957579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/957579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/957579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/957579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/957579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/957579/1"}}},"commit":{"parents":[{"commit":"758dbfa135e4bb1c2bee0823f2203a66adcdd5a2","subject":"Merge \"Use ALLOWED_ADDRESS_PAIR resources constant from neutron_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/758dbfa135e4bb1c2bee0823f2203a66adcdd5a2"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-16 03:57:16.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-16 04:19:18.000000000","tz":0},"subject":"DNM: Testign Nova service role","message":"DNM: Testign Nova service role\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/957578\n\nChange-Id: I5f7900acf94beb1bc5dc6e28fdb0da2c7a036904\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29274dd4428cd1ab01ba9014963d3f17a43ceb74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29274dd4428cd1ab01ba9014963d3f17a43ceb74"}]},"branch":"refs/heads/master"},"e51176cd8f819a758d127eb697d885aaffe81115":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-26 02:47:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/79/957579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/957579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/957579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/957579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/957579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/957579/2"}}},"commit":{"parents":[{"commit":"758dbfa135e4bb1c2bee0823f2203a66adcdd5a2","subject":"Merge \"Use ALLOWED_ADDRESS_PAIR resources constant from neutron_lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/758dbfa135e4bb1c2bee0823f2203a66adcdd5a2"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-08-16 03:57:16.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-08-26 02:47:53.000000000","tz":0},"subject":"DNM: Testign Nova service role","message":"DNM: Testign Nova service role\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/892639\n\nChange-Id: I5f7900acf94beb1bc5dc6e28fdb0da2c7a036904\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e51176cd8f819a758d127eb697d885aaffe81115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e51176cd8f819a758d127eb697d885aaffe81115"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
