)]}'
{"id":"openstack%2Fneutron~959345","triplet_id":"openstack%2Fneutron~stable%2F2024.1~I69d49cd3737b3bee8c8b4221149085375e9ac1e2","project":"openstack/neutron","branch":"stable/2024.1","topic":"bug/2116087-stable/2024.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-09-04 10:38:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69d49cd3737b3bee8c8b4221149085375e9ac1e2","subject":"[OVN] Fix remote-managed binding profile validation","status":"MERGED","created":"2025-09-03 16:02:40.000000000","updated":"2025-09-04 10:42:25.000000000","submitted":"2025-09-04 10:38:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959345-bug/2116087-stable/2024.1","cherry_pick_of_change":954292,"cherry_pick_of_patch_set":1,"meta_rev_id":"c1592f863f9317728b4e7b47df14d9fa64e36d59","_number":959345,"virtual_id_number":959345,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 10:38:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-09-04 06:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-09-04 06:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-09-04 06:57:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 16:02:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"CC"},{"updated":"2025-09-03 17:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-04 06:30:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-09-04 06:57:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"6f46b1904ae3b9e0fc2713591273b382b7442e0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-03 16:02:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e4f2d38300a151c5ea72e01a3a04689a61943771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 17:45:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/484d3193bdae4e96ad1c9e5b60a47075\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/a76e2226e28545a3893c71904824a16c : SUCCESS in 1h 34m 24s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/5180350a79754da8b0050cd70e226589 : SUCCESS in 31m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46c5516664be444d8fbcc13f93ee6adc : SUCCESS in 38m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637e80185a89431589419d1386514355 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4026f77f1ab84a7a8523118b32f380c2 : SUCCESS in 26m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49a26e6d56594f0f8060c07310a0b847 : SUCCESS in 23m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b7f9659f554440845c3b0c3c4128f0 : SUCCESS in 11m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e169f51a0234931ab4beb2fb435c829 : SUCCESS in 1h 33m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7002cd3497d34952bc162e12f11f7f35 : SUCCESS in 1h 37m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dca9b3b6924444d3961d32499bd75e16 : SUCCESS in 54m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a1cc15c750cc4177933399ad390abf5a : SUCCESS in 35m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1678021e0eb2463abc6b82522534a2a0 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9cf69d1db185ae9f8026590203802c038d86450c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 20:13:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb9792062dd94dcf9ced94088b5a09c4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79d2c39494404b4eb6c1da62321a857e : TIMED_OUT in 1h 01m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e0a91d554bd49d1bb6c5eac2690be24 : TIMED_OUT in 1h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19d3f80da2fc73186cabb07c661e093f45b79fe2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-04 06:30:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2750237db2768eb0f6de5161578aa08e086bde15","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-04 06:57:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5554ea60ec2423372c26a149cfea93b32e289f27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 06:57:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c11cc385b49c28279c866c8732ebda12671a44d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 10:38:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a300ac769046e6a0a6ecbe753b3839\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/929fb15318fe4614b995211be1e9a604 : SUCCESS in 1h 32m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/627ad1170f084aa7a224cbc5494c97cc : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c42410479923450799176ea24d54e587 : SUCCESS in 18m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72a640cfd8694e82968b0857e7b23cad : SUCCESS in 25m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24673d4c4f0488a80b3e90c03c498bb : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b98652590f74b5fa7eb1c19c3f72dbc : SUCCESS in 1h 25m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b92000292184709a9d2a47ce60c5387 : SUCCESS in 1h 35m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/52c9b36b0ad94694a28779ad2011144a : SUCCESS in 37m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7babd329c61e4bf99e9028875e6f2872 : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e999decde539402357e15c5c90e05d66c252e622","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 10:38:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1592f863f9317728b4e7b47df14d9fa64e36d59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 10:42:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/467ca5a971a84e5a92a165c3e3ad3587\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faaf3461ae8f4072b4afc9f5a40bf401 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5a4e771d587c28816c3ffbf17c675d173a7376f","revisions":{"a5a4e771d587c28816c3ffbf17c675d173a7376f":{"kind":"REWORK","_number":1,"created":"2025-09-03 16:02:40.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/45/959345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/959345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/959345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/959345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/959345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/959345/1"}}},"commit":{"parents":[{"commit":"f94d3effc5a1a841e96789f2db48418cc429d359","subject":"[OVN]: fix ovn-db-sync to include gateway chassis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f94d3effc5a1a841e96789f2db48418cc429d359"}]}],"author":{"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","date":"2025-07-07 18:35:20.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-09-03 16:02:40.000000000","tz":0},"subject":"[OVN] Fix remote-managed binding profile validation","message":"[OVN] Fix remote-managed binding profile validation\n\nWhen binding profile parameter validation was extended in\nchange I8493cba28e92b9b36bdb952c9737c0fea6fb7b75, a subtle bug\nwas introduced.\n\nNamely, if a binding profile specification specified ``None`` for\nexpected capability, the code would literally look for an instance\nof ``None`` in the list of capabilities, as can be read from the\ndocumentation [0] this was not the intended behavior.\n\nUpdate unit tests that asserted the incorrect behavior.\n\nAdd new unit test specifically for the case of a remote-managed\nvnic that also has extra capabilities.\n\n0: https://opendev.org/openstack/neutron/src/commit/71962255dee06d75258984b39d3a5c6c6293534d/neutron/common/ovn/constants.py#L340-L353\nCloses-Bug: #2116087\nChange-Id: I69d49cd3737b3bee8c8b4221149085375e9ac1e2\nSigned-off-by: Frode Nordahl \u003cfnordahl@ubuntu.com\u003e\n(cherry picked from commit 1e01e6a7130a570092e6b7324b7f9d051a34735d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5a4e771d587c28816c3ffbf17c675d173a7376f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5a4e771d587c28816c3ffbf17c675d173a7376f"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
