)]}'
{"id":"openstack%2Fneutron~960271","triplet_id":"openstack%2Fneutron~master~I5fa8e2011203e75e8754a1b9035ee8c6f5af2919","project":"openstack/neutron","branch":"master","topic":"bug/2122429","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-09-10 10:03:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5fa8e2011203e75e8754a1b9035ee8c6f5af2919","subject":"[logapi] Wrap inside a DB context the method ``_check_port_bound_sg``","status":"MERGED","created":"2025-09-09 14:36:29.000000000","updated":"2025-09-10 10:09:22.000000000","submitted":"2025-09-10 10:03:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960271-bug/2122429","meta_rev_id":"58b3c7d9b872ffc91099356109f9be4c1e5fae21","_number":960271,"virtual_id_number":960271,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-09-09 15:22:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-10 10:03:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-10 07:48:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-09-09 16:28:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-10 07:48:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-09 14:38:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-09 15:22:20.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-09 15:49:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-09 16:28:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-09-09 16:30:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-10 07:48:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"7236d4451cf75114331c9dfa116082ced0618135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-09 14:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96d3ecc2368380bb3a4183c9dfed1e98fa3a016b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-09 15:22:20.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/71/960271/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"2794fa9adb40cf4324a99598027bf96dba9b8617","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 15:49:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1b9a927ebf449a3be6ee0b20dcc11b9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e13aebf7d0594b22bc1f137d44c293b4 : SUCCESS in 29m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c836173e56c4064a3dd455973b9a013 : SUCCESS in 43m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bdfc6036bee64eaebc4306695f94c4f4 : SUCCESS in 53m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"067bc9daa6a965598141d5bc83acb4fe1d80c83a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-09 16:28:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"847ae90eeecca112b4d1c612e8d184610068e9f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 16:30:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9eaa1a837d449949560642067cb2185\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6faf27395d9b47eb93c8ecb583848bd4 : SUCCESS in 1h 35m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/25eae55d90a34910ad88770ca5121416 : SUCCESS in 1h 53m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ad21d2329d645dda03f71dbd6e72eda : SUCCESS in 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea6a71146d94eb082dbd097a14bfe30 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6eb0f8b36ff41fd96792de0662dc1d7 : SUCCESS in 27m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec362d2fa9de49c2813c78e9464ba347 : SUCCESS in 24m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfc02306747f4baf97923d9273334690 : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd699fadcc784fb0b5648214f0dd2388 : SUCCESS in 5m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/23107f093aab4437ab9bebf02e4582ad : SUCCESS in 1h 43m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7afd9dec515f413b9d8e89295c0e621a : SUCCESS in 42m 09s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/44214eb6493c4421bfc0a144af02ab6f : SUCCESS in 1h 20m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/df3c25929ff74ddaba6316b531e214da : SUCCESS in 1h 46m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/27e94ebd25fb4112882a3179422094f1 : SUCCESS in 1h 36m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9f7261f0897948b694efbc41d1921c27 : SUCCESS in 1h 22m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/376b7aa55b064957aec791ef0f97917a : SUCCESS in 1h 01m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c6186c36235b4a74a9f92c79a308c8f3 : SUCCESS in 1h 14m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f76d43c6eea444dbb0bb076d4df4321 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":1},{"id":"10766f49e49da7021588036bb71f9a8182b990e3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-10 07:48:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4eafa63b55d8b9258ec8115f67b7eca61467411a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 07:49:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5143493583b3bd43e6d03b0a1ab5e4ab0587efa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:03:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d11de47dfd484e6cb96988b2b1e6be9b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/66e7e851d79f40cdb8738e7ffb6f585a : SUCCESS in 1h 18m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0bd9ac00e8c94463ad2126cbf80fbd21 : SUCCESS in 2h 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb9af7988eef4d69ab9fb6033de19d76 : SUCCESS in 13m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3620cd2391d044e3b8cb5dac18e37944 : SUCCESS in 23m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4af236036fec485ab09c00847e33f993 : SUCCESS in 18m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f23815cce7824e7eb419596c44cdc634 : SUCCESS in 12m 57s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c844751bfe8e49f3bb7bb60fe128bace : SUCCESS in 1h 25m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ee2393aa961640459218ebbdf0917295 : SUCCESS in 1h 44m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/18fccba49dcb48f9ad62afa128ae7e11 : SUCCESS in 1h 08m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/73c16cba7f7b4e91a117dd1ba114c8ce : SUCCESS in 58m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7ae5e19da2fe4c77a4e8dc83c739a024 : SUCCESS in 1h 10m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1c0e2564e35b4560b893b72ebc825e7c : SUCCESS in 1h 17m 32s","accounts_in_message":[],"_revision_number":1},{"id":"65e81ddd59a32bd426365e96ae97448ee3875b2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:03:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"58b3c7d9b872ffc91099356109f9be4c1e5fae21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:09:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e95268fb29f24f6086b0ea3049618b02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef0c8484b8644dfa2cc28029466572c : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3fb3f8e213e4c092a05e426d9f711f76eaeb4c0","revisions":{"d3fb3f8e213e4c092a05e426d9f711f76eaeb4c0":{"kind":"REWORK","_number":1,"created":"2025-09-09 14:36:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/71/960271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/960271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/960271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/960271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/960271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/960271/1"}}},"commit":{"parents":[{"commit":"cd3bf2f88c77f069c8652568a62749c296d97321","subject":"Merge \"pep8: Return ability to run pylint only on given patches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd3bf2f88c77f069c8652568a62749c296d97321"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-09 14:34:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-09 14:36:06.000000000","tz":0},"subject":"[logapi] Wrap inside a DB context the method ``_check_port_bound_sg``","message":"[logapi] Wrap inside a DB context the method ``_check_port_bound_sg``\n\nCloses-Bug: #2122429\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I5fa8e2011203e75e8754a1b9035ee8c6f5af2919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3fb3f8e213e4c092a05e426d9f711f76eaeb4c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d3fb3f8e213e4c092a05e426d9f711f76eaeb4c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
