)]}'
{"id":"openstack%2Fneutron~960368","triplet_id":"openstack%2Fneutron~unmaintained%2Fyoga~I4defe14c768fd18fa1e65568e2579a22cf1d6a88","project":"openstack/neutron","branch":"unmaintained/yoga","topic":"neutron-py36-gate-fix","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-09-11 08:49:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4defe14c768fd18fa1e65568e2579a22cf1d6a88","subject":"[CI][um-only] Force use of ansible 9 for py36 job","status":"MERGED","created":"2025-09-10 12:01:40.000000000","updated":"2025-09-11 08:51:10.000000000","submitted":"2025-09-11 08:49:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960368-neutron-py36-gate-fix","meta_rev_id":"7a9100333031805bbd39b6ae15bbc301686bd9f9","_number":960368,"virtual_id_number":960368,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-11 08:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-09-11 07:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-09-10 16:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2025-09-11 07:50:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-10 13:10:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-10 14:28:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-10 16:33:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-09-11 07:50:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"07f7f5ea86dc51549b32d97c8785a68587db505b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-10 12:01:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a404c0d67f18374604e0ff6dbe34e41faa65abe1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-10 12:04:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c8eebe1ff89615b9a8a827b1eb3ba8c382b1626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 13:10:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3302e122d624c718ebc912adab4375d\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/40ea427a81a64171bf2e26147c80b335 : SUCCESS in 57m 47s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/d72f185db4174ff0809240254e8ea01e : SUCCESS in 1h 04m 22s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/4fcdbf8b9eff48bcae2cbe88d008646e : SUCCESS in 47m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/8e7cb91501fc4b099e070e185e91e4c4 : SUCCESS in 1h 01m 39s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/2928e151f3de4a2990f9fbc25539f076 : SUCCESS in 1h 06m 39s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/a2bd8df4d5fc4fa98fb27895f0d2b988 : SUCCESS in 26m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0168da66ec54cadbeb2d557ee353855 : SUCCESS in 39m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f04e2fbed8954b27a0ee9ca9072e6a90 : SUCCESS in 13m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/329d11a7dca94c48a989bc363b0b732e : SUCCESS in 30m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4bd07281cd843ef92e428e1d06f74f0 : FAILURE in 25m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877ccc9f0de24c7683ffd6357b3dc14c : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2e857478283f661ac73c24faf8fc2dd8ca68739c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-10 15:11:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"314b7991fbad02c967466e73fa55e54ca41c8415","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 15:12:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/151f4dd39a2545759cdcbb55caa66048\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a0059b83bd147a79b740bf969fabbb9 : SUCCESS in 49m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da82f1e666864a938d8f451bb8a31689 : SUCCESS in 42m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f75d54ec37b7b4bd657770ad87d5c500f2d39fc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 16:21:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51fe814d99064ace8efef2ed8f3508f7\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/b49ab297f13344c8b84d8e721bdb59d2 : SUCCESS in 57m 04s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/b52e280d8c354bdd993e5aab5f562bba : SUCCESS in 1h 04m 44s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/340772055dfa4f53aeb1f720eae9a62e : SUCCESS in 47m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/a74253ad0db045a78753621f4369f2a9 : SUCCESS in 1h 01m 47s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/487a7461a23540fc802b41f0781d3f1e : SUCCESS in 1h 07m 00s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/74ece69492914ef6b9a631a5e6e00cce : SUCCESS in 26m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8639ebb65f90400e8641060395ec35e6 : SUCCESS in 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cbf42b70f6a41fda1f428392d4ee512 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3ac489e46d74f24a6650c230dc54cd0 : SUCCESS in 27m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ec7c1fb8e604003bfe96d21210ed17d : SUCCESS in 23m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/262d5bd5908048738b65e40d0719eb6a : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"338d27765e5f2befb0ed1a8735910aa6500aa656","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-10 16:33:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da67429d2a141d11ed496edead006a2a78ad8ba4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-11 07:50:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5d0306ccc9bef7ea9d3a9a3a6f92d45e77452df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 07:50:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"912ff536e955142a080b2e4d8277009821951695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 08:49:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9d5afdc1ef645e7bf12f5b9df8d8c1e\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/fe6ee6c7731d4970875e1af418d1f1ca : SUCCESS in 52m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3185212078aa48a5b5ac14ae37367f4e : SUCCESS in 14m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c739a53233540918b6baf9b20baacc6 : SUCCESS in 17m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67d6fdedbac34bf39629710c0494a4ca : SUCCESS in 25m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95c9045ad6b465dbf29ecaa1aa35f8e : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":1},{"id":"fee5308c68fe39efe55055aa6f9141af16438460","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 08:49:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a9100333031805bbd39b6ae15bbc301686bd9f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 08:51:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e72858270da482fb706860f3faa16ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49fee6ffe08c4b0b8f70a2fa143e9bb4 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f74cd382002e1432577a10d93914fa6beaf007db","revisions":{"f74cd382002e1432577a10d93914fa6beaf007db":{"kind":"REWORK","_number":1,"created":"2025-09-10 12:01:40.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/68/960368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/960368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/960368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/960368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/960368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/960368/1"}}},"commit":{"parents":[{"commit":"69780ba39ea33a0c9e5ba1f4b2c26aa12aa11900","subject":"[unmaintained-only] Do not use flow_group_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69780ba39ea33a0c9e5ba1f4b2c26aa12aa11900"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-10 11:58:49.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-09-10 12:01:05.000000000","tz":120},"subject":"[CI][um-only] Force use of ansible 9 for py36 job","message":"[CI][um-only] Force use of ansible 9 for py36 job\n\nThis branch still has py36 as supported runtime (with ubuntu bionic\nnodeset). But that requires ansible version 9, which is not supported\nanymore in Zuul, but can be forced to use it until that is completely\ndropped from Zuul.\n\nChange-Id: I4defe14c768fd18fa1e65568e2579a22cf1d6a88\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f74cd382002e1432577a10d93914fa6beaf007db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f74cd382002e1432577a10d93914fa6beaf007db"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
