)]}'
{"id":"openstack%2Fneutron~960384","triplet_id":"openstack%2Fneutron~master~I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e","project":"openstack/neutron","branch":"master","topic":"bug/2122483","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-09-15 15:17:14.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-09-12 18:58:31.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e","subject":"[OVN][FT] Do not lock the DB in the ``TestOvn*Sync*`` tests","status":"MERGED","created":"2025-09-10 14:05:54.000000000","updated":"2025-09-15 15:20:08.000000000","submitted":"2025-09-15 15:17:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960384-bug/2122483","meta_rev_id":"24884b885f218faaf581f897135176bb7fcb419b","_number":960384,"virtual_id_number":960384,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-09-15 12:33:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-15 15:17:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-15 10:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-09-15 09:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-15 10:38:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-10 14:15:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-10 15:11:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-10 15:12:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-11 15:44:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-09-15 10:38:04.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"58822fc705b72e633eec5e388b35f643ae3e150e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-10 14:05:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbababf1962b259ff0814b05c71cd6a3942563c9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-10 14:26:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"995feee3655a1d8254111bfcb21fe8281e1ac62b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-10 15:11:40.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/84/960384/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"431bb99ba232c0cdbe799e13b08eb62e9d45d266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 15:12:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1622883ddfca456ea345e4e6d59ba7c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944b721c3a5e4b878914beb3ccf63e7e : SUCCESS in 20m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4ae73a559547fc9d604a0125b1e68c : SUCCESS in 10m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/07bbf99e90f7443a95dcffc665fdc585 : FAILURE in 1h 04m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a95a7f5745473e67f19160e1f709bde5f26c4d55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-11 08:46:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d94fab07e6bbcc1f1ac8a685486b99c893b2fd3d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-11 08:46:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c497223b80a40f6e1d6489254ac20968fbe4507c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-11 08:54:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e61c8c8bcec4e363d4c7b1f1c322673818f53e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 09:15:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6339a57f5370477a9e966130d079502d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0648988e3d1b45739d51522d959d3ec5 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da688cee328d427ab1acedd233c26c90 : SUCCESS in 10m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d012bfd5ff84408ebcb6fdbac6b00ce1 : FAILURE in 18m 58s","accounts_in_message":[],"_revision_number":3},{"id":"fde5141857654f48ab67d5626f2eb3e1cd6622d8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-11 09:32:26.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/84/960384/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c986a5c3e34beb876c27516c949c75f2091dd7a7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-11 11:29:50.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/84/960384/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"673b1aea85ba1b67fed76693c8d9a5aa3cc3a307","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-11 12:25:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c2afd68e9ccf696719585f16d1fb8ef7640d0b89","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-11 13:29:35.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/84/960384/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f7b0f13d13eb0b093f72495b0231fe4730fb66b1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-11 14:02:07.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e95452e6db0d43612e0eb88983770b9661279fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 14:53:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9c81167216413db486365e74280ba6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/353ebd52cc8e4d22aa04f45913f2a576 : SUCCESS in 12m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45bac28ecd214d28a421d13e51a0b803 : SUCCESS in 8m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/83d62f0743164c8b93bcf7e3dacd175c : SUCCESS in 1h 23m 16s","accounts_in_message":[],"_revision_number":4},{"id":"82c4c57d9c11bcebc053cdbad216ef3b347a6a74","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-11 14:56:34.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1207ed96a1ea03c5daec4af553f7e8d6ee153717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-11 14:56:43.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c0033afdb9bd8b2df6567c9a38b42ce0841f6247","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-11 15:44:41.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f708f37766f9f4f5a5c8e9e353b95c1594e0f595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 16:22:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac9254a15884afc9cd8085c772ff4e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ece13854a404b51a94b778b507263e0 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af720ce186414656adb766aa6ee4aef1 : SUCCESS in 6m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e639f5d0e5594717bcea02ce073309fe : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":5},{"id":"c69ef4a27e8b0b8e29fa9dc1601bf33bd130673d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-11 18:15:57.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/84/960384/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"789d0416bab2a26587bffd97554f99edc9b51e5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-12 07:45:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a666822b88484b3f724cd428cba7787cffb18659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-12 07:46:05.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"bcf519a64bde97f45fcf5e36b0e41f635bc0e1c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-12 08:30:55.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/84/960384/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"d80e5c696d5f50e390ca54b168653832337fb7ae","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-12 09:17:35.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/84/960384/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"bd0f3479aba60ef0bbc767b4424736bd29a575e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 09:35:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6895f415bea440cbb73ceb14f3a266e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0da3e86f49a47c9ba76f78740b4109a : SUCCESS in 13m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffac2143de0b433cbd5f5baa7d6f2225 : SUCCESS in 9m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b2ec311badf143faab68fafe266d79be : SUCCESS in 1h 44m 42s","accounts_in_message":[],"_revision_number":7},{"id":"c5de73ba0a5ffd80a8c7b7e4391c5bb141837982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-12 15:38:16.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"363f02a50c172467c477dcf9df28202d9147e7e1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-12 16:25:32.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/84/960384/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"da2916748a83a9ad143627df938f25c231e4082c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 17:25:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ceaf4af4634852863c1b4eb1dcecca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6c4fa283d594415b578b6ee22abb742 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/036f94202dda416dacd3291794374a42 : SUCCESS in 11m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d2565256d5ce435fa9f36d0993a2c36b : SUCCESS in 1h 45m 13s","accounts_in_message":[],"_revision_number":8},{"id":"fd3dd2e6486e2d494b056d8cafee07e048828fd6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-12 18:58:31.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"73865475021592da34c0cf3e8c34637a3da30259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-15 09:44:25.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"675cb066a4cbbbeffe68c20cae4a25ce4f1099f5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-09-15 10:38:04.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"28f4dde7863ced54cc2e5fb5c0786b9a2fa99646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 11:53:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b124e681a2a84c50a3954d9dcb57ce94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c437e2d324417ba6e4740ad53c2a83 : SUCCESS in 14m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b813c9efc6a94042b16d3f64eb7ac8d9 : SUCCESS in 13m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e4ef638fdb0a464aaf26feb62e36956b : SUCCESS in 1h 57m 53s","accounts_in_message":[],"_revision_number":9},{"id":"2791bd9e3946404209e66b9e0b77b98300d6df42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 11:54:25.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0f016ee66a8a5090de51b04a84628c46da51e892","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-15 12:33:07.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/84/960384/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"5b08b47182eaf3f0dd40c680b87aee78f6e36504","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 15:17:14.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ab2eda04db84061b388b84fadfad857\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b3da4a1206442de88351627353883cf : SUCCESS in 15m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4650f52488034a98a08191fafa9d97a9 : SUCCESS in 11m 23s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e52d799df5fe4f0abe94e25984c4a80a : SUCCESS in 1h 39m 05s","accounts_in_message":[],"_revision_number":9},{"id":"e2bf91d1bcd5f73dab5a27cf371e167d375c13d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 15:17:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"24884b885f218faaf581f897135176bb7fcb419b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 15:20:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5876b5566064f05ba7d45aa6450cfab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8eb92d790eb4d8286827450e0e5098b : SUCCESS in 50s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"9581ed379dea1b4850ec9bb55a3cd5451265a599","revisions":{"9fc8cadeda756bebb141e64278a1ef6d426e9ace":{"kind":"REWORK","_number":1,"created":"2025-09-10 14:05:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/960384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/960384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/960384/1"}}},"commit":{"parents":[{"commit":"e988197b22aa0a6be32b48c776370a292c7b512a","subject":"[OVN][FT] Wait for \"Chassis_Private\" update event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e988197b22aa0a6be32b48c776370a292c7b512a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"subject":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests","message":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests\n\nCloses-Bug: #2122483\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9fc8cadeda756bebb141e64278a1ef6d426e9ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9fc8cadeda756bebb141e64278a1ef6d426e9ace"}]},"branch":"refs/heads/master"},"30363b86998c63fd713c811905a1891da807b046":{"kind":"REWORK","_number":2,"created":"2025-09-11 08:46:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/960384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/960384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/960384/2"}}},"commit":{"parents":[{"commit":"e988197b22aa0a6be32b48c776370a292c7b512a","subject":"[OVN][FT] Wait for \"Chassis_Private\" update event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e988197b22aa0a6be32b48c776370a292c7b512a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-11 08:39:50.000000000","tz":0},"subject":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests","message":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests\n\nCloses-Bug: #2122483\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30363b86998c63fd713c811905a1891da807b046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30363b86998c63fd713c811905a1891da807b046"}]},"branch":"refs/heads/master"},"4e1c2d283d6750927ff4aaff05f279bfcf25e92c":{"kind":"REWORK","_number":3,"created":"2025-09-11 08:54:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/960384/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/960384/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/960384/3"}}},"commit":{"parents":[{"commit":"e988197b22aa0a6be32b48c776370a292c7b512a","subject":"[OVN][FT] Wait for \"Chassis_Private\" update event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e988197b22aa0a6be32b48c776370a292c7b512a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-09-11 08:54:51.000000000","tz":0},"subject":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests","message":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests\n\nCloses-Bug: #2122483\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e1c2d283d6750927ff4aaff05f279bfcf25e92c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4e1c2d283d6750927ff4aaff05f279bfcf25e92c"}]},"branch":"refs/heads/master"},"efa69a9661ba5760a8520837888918d122aff662":{"kind":"REWORK","_number":4,"created":"2025-09-11 12:25:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/960384/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/960384/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/960384/4"}}},"commit":{"parents":[{"commit":"e988197b22aa0a6be32b48c776370a292c7b512a","subject":"[OVN][FT] Wait for \"Chassis_Private\" update event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e988197b22aa0a6be32b48c776370a292c7b512a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-11 11:17:32.000000000","tz":0},"subject":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests","message":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests\n\nCloses-Bug: #2122483\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efa69a9661ba5760a8520837888918d122aff662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efa69a9661ba5760a8520837888918d122aff662"}]},"branch":"refs/heads/master"},"7b3dedd00d4de72d07720dcc43cd5ba83d24f0a3":{"kind":"REWORK","_number":5,"created":"2025-09-11 14:56:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/960384/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/960384/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/960384/5"}}},"commit":{"parents":[{"commit":"e988197b22aa0a6be32b48c776370a292c7b512a","subject":"[OVN][FT] Wait for \"Chassis_Private\" update event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e988197b22aa0a6be32b48c776370a292c7b512a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-09-11 14:56:43.000000000","tz":0},"subject":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests","message":"[OVN][FT] Set an IDL lock in the ``TestOvnSbSync*`` tests\n\nCloses-Bug: #2122483\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b3dedd00d4de72d07720dcc43cd5ba83d24f0a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b3dedd00d4de72d07720dcc43cd5ba83d24f0a3"}]},"branch":"refs/heads/master"},"bb78929983cddcbb9738e51588193c02397ecfd1":{"kind":"REWORK","_number":6,"created":"2025-09-12 07:45:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/960384/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/960384/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/960384/6"}}},"commit":{"parents":[{"commit":"e988197b22aa0a6be32b48c776370a292c7b512a","subject":"[OVN][FT] Wait for \"Chassis_Private\" update event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e988197b22aa0a6be32b48c776370a292c7b512a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-12 07:41:03.000000000","tz":0},"subject":"[OVN][FT] Do not lock the DB ``TestOvn*Sync*`` tests","message":"[OVN][FT] Do not lock the DB ``TestOvn*Sync*`` tests\n\nThese tests call the ``has_lock`` method several times, when the\n``DBInconsistenciesPeriodics`` class is instantiated and then restarted.\nBecause the tests are running isolated and no other process has access\nto the OVN databases, this patch mocks the ``set_lock`` method,\npreventing the IDL to gain this lock.\n\nCloses-Bug: #2122483\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb78929983cddcbb9738e51588193c02397ecfd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb78929983cddcbb9738e51588193c02397ecfd1"}]},"branch":"refs/heads/master"},"900a6c2d7d50d9a92c5db498dd0fd1bafe328189":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-09-12 07:46:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/960384/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/960384/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/960384/7"}}},"commit":{"parents":[{"commit":"e988197b22aa0a6be32b48c776370a292c7b512a","subject":"[OVN][FT] Wait for \"Chassis_Private\" update event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e988197b22aa0a6be32b48c776370a292c7b512a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-12 07:45:52.000000000","tz":0},"subject":"[OVN][FT] Do not lock the DB in the ``TestOvn*Sync*`` tests","message":"[OVN][FT] Do not lock the DB in the ``TestOvn*Sync*`` tests\n\nThese tests call the ``has_lock`` method several times, when the\n``DBInconsistenciesPeriodics`` class is instantiated and then restarted.\nBecause the tests are running isolated and no other process has access\nto the OVN databases, this patch mocks the ``set_lock`` method,\npreventing the IDL to gain this lock.\n\nCloses-Bug: #2122483\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/900a6c2d7d50d9a92c5db498dd0fd1bafe328189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/900a6c2d7d50d9a92c5db498dd0fd1bafe328189"}]},"branch":"refs/heads/master"},"725dfc89610511c616117e007dd55f2090ec68bd":{"kind":"REWORK","_number":8,"created":"2025-09-12 15:38:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/960384/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/960384/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/960384/8"}}},"commit":{"parents":[{"commit":"e988197b22aa0a6be32b48c776370a292c7b512a","subject":"[OVN][FT] Wait for \"Chassis_Private\" update event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e988197b22aa0a6be32b48c776370a292c7b512a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-12 15:32:35.000000000","tz":0},"subject":"[OVN][FT] Do not lock the DB in the ``TestOvn*Sync*`` tests","message":"[OVN][FT] Do not lock the DB in the ``TestOvn*Sync*`` tests\n\nThese tests call the ``has_lock`` method several times, when the\n``DBInconsistenciesPeriodics`` class is instantiated and then restarted.\nBecause the tests are running isolated and no other process has access\nto the OVN databases, this patch mocks the ``set_lock`` method,\npreventing the IDL to gain this lock.\n\nCloses-Bug: #2122483\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/725dfc89610511c616117e007dd55f2090ec68bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/725dfc89610511c616117e007dd55f2090ec68bd"}]},"branch":"refs/heads/master"},"9581ed379dea1b4850ec9bb55a3cd5451265a599":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-09-15 09:44:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/960384/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/960384/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/960384/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/960384/9"}}},"commit":{"parents":[{"commit":"0c59b19c03ede0662080ca61a23c5c59945cf4a7","subject":"[OVN][FT] Wait for \"Chassis_Private\" update event","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c59b19c03ede0662080ca61a23c5c59945cf4a7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-10 13:52:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-09-15 09:44:26.000000000","tz":0},"subject":"[OVN][FT] Do not lock the DB in the ``TestOvn*Sync*`` tests","message":"[OVN][FT] Do not lock the DB in the ``TestOvn*Sync*`` tests\n\nThese tests call the ``has_lock`` method several times, when the\n``DBInconsistenciesPeriodics`` class is instantiated and then restarted.\nBecause the tests are running isolated and no other process has access\nto the OVN databases, this patch mocks the ``set_lock`` method,\npreventing the IDL to gain this lock.\n\nCloses-Bug: #2122483\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3b5b9e72b2aaaa65eb7917f8077b72820ab6816e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9581ed379dea1b4850ec9bb55a3cd5451265a599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9581ed379dea1b4850ec9bb55a3cd5451265a599"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
