)]}'
{"id":"openstack%2Fneutron~960749","triplet_id":"openstack%2Fneutron~stable%2F2025.2~I8c1158bd6517f822bc19ae11d137c10cf7ae7481","project":"openstack/neutron","branch":"stable/2025.2","topic":"create-2025.2","hashtags":[],"change_id":"I8c1158bd6517f822bc19ae11d137c10cf7ae7481","subject":"Update .gitreview for stable/2025.2","status":"MERGED","created":"2025-09-12 09:04:11.000000000","updated":"2025-09-12 13:58:41.000000000","submitted":"2025-09-12 13:56:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960749-create-2025.2","meta_rev_id":"946a2a87ed0bcf088559b68dd4890f1df31b1d3a","_number":960749,"virtual_id_number":960749,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-12 13:56:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2025-09-12 13:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2025-09-12 13:21:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-12 09:44:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-12 13:21:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"6d343f6e4df8a6e79fb35dea12f7e79678f2f7d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-09-12 09:04:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e6fb94c51f038b8bea9ec799d20ac306582cb09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 09:44:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae1e5fc40a6b42b0a5f51c7f3b594e46\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/661531a56ac94eb28e58f8158ddbf79b : SUCCESS in 36m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb7edaa750f847b2b55d21b3b41d81a0 : SUCCESS in 7m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09a2dec0512d498eb0cbcfa5c9a8c5b8 : SUCCESS in 15m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fed8dcb395c74e77829ff8f8e703d0a7 : SUCCESS in 16m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cdc3706618e410e8177500a7e1cff88 : SUCCESS in 21m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6ec0f7fbc64aad82212db10e87f314 : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"dc5c8828cce78a6b1f4437763b638118ba5b2ea9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 10:49:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98c841d1bc864790a2fad880969797e6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/884ace400ff24380a36c589116162ad1 : SUCCESS in 51m 35s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/573802387bb74e51809ff0c748fae2ff : SUCCESS in 50m 11s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4944acf1d132491a8934eb547d05a2a6 : SUCCESS in 29m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1a509d409fd79b11fe4560fa2f280fb14de00db8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-12 13:21:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca0b7a0f5609366769ff9c95e862d54cf25b1e71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 13:21:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33dc60efcb09e221445c49351973c877bde533d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 13:56:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/536959fcf7e841f99daf794c045cfc27\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91a964bd8374678a3c83316cd135ff4 : SUCCESS in 21m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98390067f602453c8776baf7ed1bef09 : SUCCESS in 24m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/659eefcdbdf6400bbfc75ae672ba4d4a : SUCCESS in 20m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4306fc1a59344c96bb4ee45fa0aef30b : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"81b855c6d1dd10feac0c64494cc47a9cb9ebc9f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 13:56:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"946a2a87ed0bcf088559b68dd4890f1df31b1d3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 13:58:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ece14ed0f7ce4767baa3708e40cd8352\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5781e4672ac4b1886c72501a6cf8146 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d16d63bb63c8053ec6850e9a1ad6a49df0a55931","revisions":{"d16d63bb63c8053ec6850e9a1ad6a49df0a55931":{"kind":"REWORK","_number":1,"created":"2025-09-12 09:04:11.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/49/960749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/960749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/960749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/960749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/960749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/960749/1"}}},"commit":{"parents":[{"commit":"46f4ef447537d80bff2bb64da6a61fff40e8fcc3","subject":"Merge \"Remove stale DVR FIP check introduced in Pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46f4ef447537d80bff2bb64da6a61fff40e8fcc3"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-12 09:04:09.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-12 09:04:09.000000000","tz":0},"subject":"Update .gitreview for stable/2025.2","message":"Update .gitreview for stable/2025.2\n\nChange-Id: I8c1158bd6517f822bc19ae11d137c10cf7ae7481\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d16d63bb63c8053ec6850e9a1ad6a49df0a55931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d16d63bb63c8053ec6850e9a1ad6a49df0a55931"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
