)]}'
{"id":"openstack%2Fneutron~960892","triplet_id":"openstack%2Fneutron~master~I044940c8a45df8b1f373de9ca6bbc747a05a5d41","project":"openstack/neutron","branch":"master","topic":"bug/2122706","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-05-19 17:52:51.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I044940c8a45df8b1f373de9ca6bbc747a05a5d41","subject":"functional: Spawn Open_vSwitch ovsdb-server for OVN agent","status":"ABANDONED","created":"2025-09-12 21:51:54.000000000","updated":"2026-07-30 18:01:14.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"da7189d73ab5acc4d3a4f19f36f688a7b7c11e0a","_number":960892,"virtual_id_number":960892,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-09-13 04:09:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-12 23:47:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-09-15 18:10:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-12 22:37:58.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-12 23:47:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-15 13:37:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-16 13:15:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"e3ac92faf108daa8c6360912ac096c119e6121b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-09-12 21:51:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"934bbdd56c9e52aeab34fc20aeb7e38aaefb33ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-09-12 21:56:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d63484240715001af1c62f47e135728566a2d9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-09-12 22:17:49.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4f5a8f4f15a9ce6f8c9ef4a5c0bd283844390d68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-09-12 22:20:16.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"494c629d98c95ade8adb7b76c7a10564675ce831","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-12 22:37:58.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/92/960892/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b3b5f185b6c9a6bb7a1b0d40567ee0e7bc87c04b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-12 23:23:51.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/92/960892/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"048b18084c70150e486e3886cd8a3eebbb471f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 23:47:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c920710399a4a1a8d3a411796be2b86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42584a92072943808c6d3074445cdccc : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4757bb17a2644bee8a6002e5c0734aa7 : SUCCESS in 12m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cceb2f65de0849388bea9ac9296e1a1e : SUCCESS in 1h 25m 01s","accounts_in_message":[],"_revision_number":4},{"id":"13747ea2636b43eb55e4f9449298980cd11034b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-13 00:10:24.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/92/960892/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"96f674a422eafc9237bcd663779d632c5f4baa09","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-13 04:09:28.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/92/960892/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"b534752e184b3d7ca1bc3de0eaea47e3a0723179","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-09-15 18:10:09.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6f5a824341d2fdf91b83031f00a04b25bfc3a9b2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-16 13:15:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"98b876eaececb0c06eaab7b7e3a234a5471e43d4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-09-16 21:10:23.000000000","message":"Topic bug/2111276 removed","accounts_in_message":[],"_revision_number":4},{"id":"21b3bdf2622dbc9793d334ea1edeb4d0bfdd27ca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-09-16 21:10:28.000000000","message":"Topic set to bug/2122706","accounts_in_message":[],"_revision_number":4},{"id":"eb8205c035455df633911b36b2ff3e38095064ab","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-05-19 17:52:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"da7189d73ab5acc4d3a4f19f36f688a7b7c11e0a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-07-30 18:01:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"685cc1947c78a09b0cd0d267707fd0c7397b9aae","revisions":{"007c0227a6154daf77af417beb1bc9513b56385f":{"kind":"REWORK","_number":1,"created":"2025-09-12 21:51:54.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/92/960892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/960892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/960892/1"}}},"commit":{"parents":[{"commit":"46f4ef447537d80bff2bb64da6a61fff40e8fcc3","subject":"Merge \"Remove stale DVR FIP check introduced in Pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46f4ef447537d80bff2bb64da6a61fff40e8fcc3"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-09-12 21:50:14.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-09-12 21:51:40.000000000","tz":0},"subject":"functional: Spawn Open_vSwitch ovsdb-server for OVN agent","message":"functional: Spawn Open_vSwitch ovsdb-server for OVN agent\n\nThe patch spawns an ovsdb-server process with Open_vSwitch schema.\nThis is needed in order to isolate OVN agent functional tests as the\nagent configuration might be monitoring Open_vSwitch table just as\novn-controller does.\n\nResolves-Bug: #2122706\n\nChange-Id: I044940c8a45df8b1f373de9ca6bbc747a05a5d41\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/007c0227a6154daf77af417beb1bc9513b56385f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/007c0227a6154daf77af417beb1bc9513b56385f"}]},"branch":"refs/heads/master"},"ccbc6f039ca4f4722a6a13361d39fe2780c50228":{"kind":"REWORK","_number":2,"created":"2025-09-12 21:56:50.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/92/960892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/960892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/960892/2"}}},"commit":{"parents":[{"commit":"46f4ef447537d80bff2bb64da6a61fff40e8fcc3","subject":"Merge \"Remove stale DVR FIP check introduced in Pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46f4ef447537d80bff2bb64da6a61fff40e8fcc3"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-09-12 21:50:14.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-09-12 21:52:36.000000000","tz":0},"subject":"functional: Spawn Open_vSwitch ovsdb-server for OVN agent","message":"functional: Spawn Open_vSwitch ovsdb-server for OVN agent\n\nThe patch spawns an ovsdb-server process with Open_vSwitch schema.\nThis is needed in order to isolate OVN agent functional tests as the\nagent configuration might be monitoring Open_vSwitch table just as\novn-controller does.\n\nResolves-Bug: #2122706\n\nChange-Id: I044940c8a45df8b1f373de9ca6bbc747a05a5d41\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccbc6f039ca4f4722a6a13361d39fe2780c50228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccbc6f039ca4f4722a6a13361d39fe2780c50228"}]},"branch":"refs/heads/master"},"d7cb42890ea6950ff82e12cdc09a3b2f80f8d87c":{"kind":"NO_CHANGE","_number":3,"created":"2025-09-12 22:17:49.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/92/960892/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/960892/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/960892/3"}}},"commit":{"parents":[{"commit":"46f4ef447537d80bff2bb64da6a61fff40e8fcc3","subject":"Merge \"Remove stale DVR FIP check introduced in Pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46f4ef447537d80bff2bb64da6a61fff40e8fcc3"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-09-12 21:50:14.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-09-12 22:00:24.000000000","tz":0},"subject":"functional: Spawn Open_vSwitch ovsdb-server for OVN agent","message":"functional: Spawn Open_vSwitch ovsdb-server for OVN agent\n\nThe patch spawns an ovsdb-server process with Open_vSwitch schema.\nThis is needed in order to isolate OVN agent functional tests as the\nagent configuration might be monitoring Open_vSwitch table just as\novn-controller does.\n\nResolves-Bug: #2122706\n\nChange-Id: I044940c8a45df8b1f373de9ca6bbc747a05a5d41\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7cb42890ea6950ff82e12cdc09a3b2f80f8d87c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7cb42890ea6950ff82e12cdc09a3b2f80f8d87c"}]},"branch":"refs/heads/master"},"685cc1947c78a09b0cd0d267707fd0c7397b9aae":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-09-12 22:20:16.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/92/960892/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/960892/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/960892/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/960892/4"}}},"commit":{"parents":[{"commit":"46f4ef447537d80bff2bb64da6a61fff40e8fcc3","subject":"Merge \"Remove stale DVR FIP check introduced in Pike\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46f4ef447537d80bff2bb64da6a61fff40e8fcc3"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-09-12 21:50:14.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-09-12 22:20:16.000000000","tz":0},"subject":"functional: Spawn Open_vSwitch ovsdb-server for OVN agent","message":"functional: Spawn Open_vSwitch ovsdb-server for OVN agent\n\nThe patch spawns an ovsdb-server process with Open_vSwitch schema.\nThis is needed in order to isolate OVN agent functional tests as the\nagent configuration might be monitoring Open_vSwitch table just as\novn-controller does.\n\nCloses-Bug: #2122706\n\nChange-Id: I044940c8a45df8b1f373de9ca6bbc747a05a5d41\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/685cc1947c78a09b0cd0d267707fd0c7397b9aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/685cc1947c78a09b0cd0d267707fd0c7397b9aae"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
