)]}'
{"id":"openstack%2Fneutron~961810","triplet_id":"openstack%2Fneutron~master~I926db2b9e53a7b127d85cf8b6337e1632aebaf57","project":"openstack/neutron","branch":"master","topic":"ovn_supports_external_ports","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-02 15:30:39.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-09-29 08:21:38.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I926db2b9e53a7b127d85cf8b6337e1632aebaf57","subject":"[OVN] Remove the check for external ports support","status":"MERGED","created":"2025-09-19 14:52:58.000000000","updated":"2025-10-02 15:31:38.000000000","submitted":"2025-10-02 15:30:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961810-ovn_supports_external_ports","meta_rev_id":"7dee150c1957b072273db6ce8504d164b0b0cb2f","_number":961810,"virtual_id_number":961810,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-09-24 11:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-02 15:30:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-29 08:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-10-02 13:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-10-02 13:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-19 15:39:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-19 16:15:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-19 17:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 07:49:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-10-02 13:30:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"ed098f2ab2ca03c039242765747cdc0f58a16ad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-19 14:52:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d8f1a6a4802857120e65a042a02451f7645d193","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-19 15:39:40.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/10/961810/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"de516868e9a196bcff06da370869dfdf54eadb79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 17:22:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50dc73d7829043678238edf63da10e31\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/33ce3ea450cf46b988327a2c52deab26 : SUCCESS in 2h 11m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9780b40882714b08b79cc9ecfc85fd73 : FAILURE in 36m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d189066325604055813d76c2ed6cc8af : SUCCESS in 7m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/042195f7983444dbbb45e6057f13a960 : FAILURE in 16m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d9bc36476004951ba68c45265a18c84 : FAILURE in 16m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7815b108e3242be87f00abc22d8d926 : FAILURE in 28m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3589b1be713b447f87fc9df27a69c9af : SUCCESS in 11m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c5cc1967db61437bb33c9eb090850a97 : SUCCESS in 1h 38m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8bdf2937d274986b586493dd872cab6 : SUCCESS in 53m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/949922a133b24cb78b0f189e417d1e5b : SUCCESS in 54m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bfb17fa9b04944d29b5484d4e024164f : SUCCESS in 1h 16m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9eb1eb96ae7245e8a563754a9fdfb053 : SUCCESS in 1h 18m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2adbb23653cda08abfc22f174db0f3f293f9f47e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 18:48:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38be950054e841a991ab900e0839cb9a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf7359315e74469daafd3358f730458b : FAILURE in 40m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4afb244e56c9484bb5505cbb6365599c : FAILURE in 45m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb64d72c152b4559a088736098a64a1b : FAILURE in 1h 07m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af7ebe4902f9ced8319e2820e29e8422a1ebc8d3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-23 07:49:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1edd6bc086f3c2907b4624ef47283b9467229b0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-24 10:10:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"570f9d9fe5a2ce40effd3a7d347f4476327a4f5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-24 10:32:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a102af993123eab91097fb67c837c7dc17dbca74","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-24 11:18:07.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/10/961810/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6ea149173e5a141cf577a01f355f5bba1bf43a49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 11:41:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43cf2a2a565548e49eae424344f2dbd7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2298681497de4af899efd49d847b0800 : SUCCESS in 41m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f244aff441324628a901a0a4b8ce318e : SUCCESS in 28m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f83357c8b0dc4ad795bbf3b539f25e81 : SUCCESS in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e02236d0eaabd6b0993f36892bf2d97fecc63d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 12:42:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31adce721dcb40f0a1ad3ae5df6e8f03\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/da32a53623a840bfbb9c26c67ac5fc5d : SUCCESS in 2h 07m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07d919b76dd64f2bbcc5e2aaa26bdaf9 : SUCCESS in 26m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1770915b8d894f73a8f844db18b7504e : SUCCESS in 9m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c64fd6ae99c34dc3a4ab2b4ba9b80d32 : SUCCESS in 22m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc74ff4419da4d2e99d8ca0089a36097 : SUCCESS in 20m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b28a395ab6b4ac78f934f2d959b1d22 : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50268e4c5a7744d284e1eb8b2cc3366c : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4c9b7ec284d4875ad9eb9e363593bf9 : SUCCESS in 16m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a51f919149b7470eae7655a55aa7bbb2 : SUCCESS in 1h 31m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d143c122759422ca4fe4b4f4e03b993 : SUCCESS in 1h 04m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4b6153ef4cd94b6d9447eb62943bf380 : SUCCESS in 51m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0193c2ee1fb54fc9834eb4a49997517e : SUCCESS in 1h 14m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c2245074d2d4bb1974c4cc2df66f5bd : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7f72ff1d404c7315af5157bda121c8a2745ba4c3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-29 08:21:38.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0f59f3ff77f0e3f646d641800408c1b7633308c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-02 13:30:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c7aa11361f6804b28c11081cd16c177c7316e9d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 13:31:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b76112740c28d8d83b773fd68d77f19f73ce9ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 15:30:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/832cd06389544da492e4070732f8a03d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5833f99406a3476a915ad8d5bccaff04 : SUCCESS in 1h 59m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d95370effe64c128c552cec2d81d1cd : SUCCESS in 10m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/735f9e0f1df3433c8d2c1dc1cc7e846e : SUCCESS in 21m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ecd662d5b2a4520b12dc1b9c33a40ef : SUCCESS in 19m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf2bbd71bb4467c98e4e0069d320010 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14acc6681a444f84bfa18e8e4ff5fc87 : SUCCESS in 16m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0ab42193ae7d46e0bf8634b2133154b0 : SUCCESS in 1h 33m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c7837d858d94907889378837962bd2a : SUCCESS in 1h 03m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/83a5da81444f41869f628b20c75f0360 : SUCCESS in 45m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6b8d1c9a617d4170804683fa2a360bab : SUCCESS in 1h 36m 42s","accounts_in_message":[],"_revision_number":2},{"id":"39dd30050b12698c07a66df3d310ae1023005adf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 15:30:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7dee150c1957b072273db6ce8504d164b0b0cb2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 15:31:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/881d5066e61546a58090955b3d0f7df1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc00fe6988764cd595cbc12d12d9eec0 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54b24d9ca5a449a4a7b33df5d54f97b1 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a26933cc3dcbb2a981c0ada63f9b5a8600839bc","revisions":{"5b410d6afd2267729a3e5c338bbc9aa626863a84":{"kind":"REWORK","_number":1,"created":"2025-09-19 14:52:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/961810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/961810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/961810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/961810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/961810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/961810/1"}}},"commit":{"parents":[{"commit":"5b91772ac022ea3eb301d8147ece8ed6d29d5fc2","subject":"Merge \"Add unique constraint for network segment ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b91772ac022ea3eb301d8147ece8ed6d29d5fc2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-19 14:48:29.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-19 14:52:42.000000000","tz":0},"subject":"[OVN] Remove the check for external ports support","message":"[OVN] Remove the check for external ports support\n\nThe support for ``HA_Chassis_Group`` in the ``Logical_Switch_Ports`` was\nprovided in OVN v20.03.0 [1] and thus the support for external ports\n(baremetal nodes, SR-IOV).\n\n[1]https://github.com/ovn-org/ovn/commit/b31c76000bef314b68e776d318d1ce4cf152450b\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I926db2b9e53a7b127d85cf8b6337e1632aebaf57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b410d6afd2267729a3e5c338bbc9aa626863a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b410d6afd2267729a3e5c338bbc9aa626863a84"}]},"branch":"refs/heads/master"},"6a26933cc3dcbb2a981c0ada63f9b5a8600839bc":{"kind":"REWORK","_number":2,"created":"2025-09-24 10:32:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/10/961810/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/961810/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/961810/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/961810/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/961810/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/961810/2"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-19 14:48:29.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 10:31:32.000000000","tz":0},"subject":"[OVN] Remove the check for external ports support","message":"[OVN] Remove the check for external ports support\n\nThe support for ``HA_Chassis_Group`` in the ``Logical_Switch_Ports`` was\nprovided in OVN v20.03.0 [1] and thus the support for external ports\n(baremetal nodes, SR-IOV).\n\n[1]https://github.com/ovn-org/ovn/commit/b31c76000bef314b68e776d318d1ce4cf152450b\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I926db2b9e53a7b127d85cf8b6337e1632aebaf57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a26933cc3dcbb2a981c0ada63f9b5a8600839bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a26933cc3dcbb2a981c0ada63f9b5a8600839bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
