)]}'
{"id":"openstack%2Fneutron~962155","triplet_id":"openstack%2Fneutron~master~I125d87ffbb33541ff3a4bb2a90915d20d0a171ad","project":"openstack/neutron","branch":"master","topic":"bug/2125553","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-11-21 10:43:38.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-11-21 15:11:58.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2025-11-19 14:11:59.000000000","reason":"\u003cGERRIT_ACCOUNT_34271\u003e replied on the change","reason_account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}}},"hashtags":[],"change_id":"I125d87ffbb33541ff3a4bb2a90915d20d0a171ad","subject":"[OVN] The external networks GW chassis must the same as the GW LRP","status":"MERGED","created":"2025-09-24 07:14:29.000000000","updated":"2025-11-21 15:13:12.000000000","submitted":"2025-11-21 15:11:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962155-bug/2125553","meta_rev_id":"8bfc995a2f228b7c25757a67b7992a150c0443cf","_number":962155,"virtual_id_number":962155,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-21 15:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-20 14:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-11-20 14:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-20 14:21:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-11-20 14:21:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-24 07:56:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-24 08:01:38.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-24 09:24:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-25 00:28:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-29 11:16:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-23 14:49:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-11-17 14:14:23.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"CC"},{"updated":"2025-11-18 07:43:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-11-18 20:45:21.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2025-11-19 13:28:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"4831724e39aaff7fe786c6503c76a26f88749de3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-24 07:14:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24eb6b8ea316a32274c3031dc4656ec90ba4414a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-24 07:14:39.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"52a186f4d2993c8709f37f592c5d19ff0c960f8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 07:56:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a4c28fcf7874d43aecd07d052c84b47\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4922920f3d8428f905326e30a36f511 : SUCCESS in 39m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a677027b7b0f4d43b886fe42aece68dd : SUCCESS in 19m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3459964feb0b4a8897b876aab4e8592f : SUCCESS in 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46a4d27dfa12a7062d591558dadcf46e564d3e37","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-24 08:01:38.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/55/962155/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ec7f16031c477cd72dee75a40a01446e7ef7b9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 09:24:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c965f964c6b9412bb49428b6eedff69a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ded484fcf93b43ffae8c30f18d1ed045 : SUCCESS in 2h 09m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65a758906ad144baa945bf9a9d5c9d42 : SUCCESS in 31m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55540c08bd3846e99eb7af7fffc3be59 : FAILURE in 8m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86eb2c03c3f348bdaf24c91ec00ae899 : SUCCESS in 19m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66e49972be114623aa2b3398b3c1dc57 : SUCCESS in 18m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f046c170eab4f9581d955993f61fda8 : SUCCESS in 23m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53599ed5e5d843ab9b13d340bad6cf56 : SUCCESS in 6m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5695ba2ea812424fae0e904275d7b4cd : SUCCESS in 1h 33m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f17004e6ac124aca84e227d5d2547ac9 : SUCCESS in 35m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/74d6bcfea335429a915d2935b4fd09d5 : SUCCESS in 1h 00m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4ffc56d807ef452cb495868e123c0974 : SUCCESS in 1h 19m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/743ea802091d4b768af95ef0544e0739 : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":1},{"id":"30595906341a1abd39b632a7eddaac8eeeaf9135","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-25 10:17:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"d5f0ad816effb9f8f9551371159b257c7f079b35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-25 10:18:25.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"05b5c9075bd0ebea7f5863cf1b614abe662f3bb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 11:59:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/421a70b32810409e876f4ad0ff92b617\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45ec70e5505a441aaf5b2dddd13754bc : SUCCESS in 1h 03m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cbef359cd050443993782efe987b9552 : SUCCESS in 32m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e80106795ed9442c9d77bf418c81c791 : SUCCESS in 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d600568c9fc7cc8288acdda90c89f3d43300748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 12:19:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5f194655ca143b2aef7cd8747cb3c2f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4018b32008cb45c78ca9bc3506798e66 : SUCCESS in 1h 59m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/360ff9454529480c9959f3ed5738630e : SUCCESS in 45m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2e503faa654ee891383f687fe0ca3c : FAILURE in 10m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c11331536ce7462583cf3ac838fbde6f : SUCCESS in 22m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6dca2b7169e94c92b640dd9476f3da80 : SUCCESS in 19m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7ec346141bb4a699c21879db844f3b8 : SUCCESS in 29m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e674cbbaf8b476499ee5ac9166eb36d : SUCCESS in 12m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6179060c178945c4be287f090b87b159 : SUCCESS in 1h 24m 18s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/199103a4464f4828ac97c89b874c7b60 : SUCCESS in 49m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cf7338113bfd44e5b46e0e849cd413a1 : SUCCESS in 1h 09m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/99c043837b0c4574aba791a398d06a5b : SUCCESS in 48m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e0e33386a8b24f9c8315948ce5d3d201 : SUCCESS in 1h 15m 08s","accounts_in_message":[],"_revision_number":2},{"id":"ba7e9ef69ce7ad2d8390b5f5e6fc684cebe29cbc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-25 14:08:57.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/55/962155/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5e7fe64a7ba161d9f36ce8f87b3d51d62c9aa65a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-26 20:27:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f81501be8e47fb38b00eb09e95b8aa3f0f4e31cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-26 21:18:19.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"70a9f41c3d3ed79490b768ad2dd9946aa8c9242c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-26 22:20:14.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/55/962155/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"e34b0e53d3b332c554b0530b5191a41c1862776b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 22:54:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/560bb4f408134eee8d6c3060c28e471d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68c6fc44194d4c358e57a4f99214af5a : FAILURE in 33m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/33295e5621ef41c8802cd79adb7cedf7 : FAILURE in 49m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eeb580c13fc1491a8dcf182522d3c684 : FAILURE in 57m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"733fb7cc3ff4ba389505f3c22829fc02b897b0e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 22:57:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40bf7c790cd0410dbf9287d9ed70aab5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6efe4a7f65634716a4cb48912ea402f6 : SUCCESS in 1h 38m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3deac35525854a01bcca8e3bf22212ce : FAILURE in 25m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a3281dc63d14002a86f168e5aed2e0d : FAILURE in 10m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bbfc041d6a48403a82c734fa03ab2c38 : FAILURE in 24m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab16bb590ba749ce9e67830745a8eefe : FAILURE in 18m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/245c8510571b44cda5f60c4455823784 : FAILURE in 21m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04590e91e46f4595869788faa7690aaf : SUCCESS in 12m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5a0e3807425a4979ba98b764927f6457 : SUCCESS in 34m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5d0aa1676c0a408591b1f878d8dd4322 : FAILURE in 30m 31s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/43a99ce4ea5c46e081b57032696577e1 : SUCCESS in 34m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f1700e1b6081431db1eef8c65e03f9e2 : SUCCESS in 53m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/acb7d16dc3a14dc7a4a4300690d19887 : SUCCESS in 1h 13m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/50153283e4bb4ef791c83de94a812a5b : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":4},{"id":"443895df2da025992cb9716aaf90860181f401bf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-27 03:40:42.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/55/962155/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"42c6c4330bb6d5fcfe380381af410f72c507458c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-30 14:12:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a413a41b4177bc26f0c41e807009c7c6348cc86a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 15:08:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/994756f589c84db6879047e7e887fc33\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58b9ca0fe6d841d991db5978c9d6b0eb : FAILURE in 20m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c77e6d41fad4a918dbb788c10a6689f : FAILURE in 40m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6d9ba2bf370846538e2b15c560ef7e9f : FAILURE in 53m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"38a1d52bc42ae55878cfcc9b123cf4f9c924ac43","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-30 15:27:51.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/55/962155/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"13792f573b6c299a7f3ed06bc5de94c289d05d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 16:06:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa9275318cc2441383cdb4620a321eb1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5fb09089e6444ffd9affbe08d1848573 : SUCCESS in 1h 30m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d323c283682540efae7f00517fc0ba22 : FAILURE in 45m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df1de597fc724b738baae73432827114 : FAILURE in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e60dfff03f8457ba7f2181f92d88557 : FAILURE in 27m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4dff9ec608f42ca8d470f597f0f3979 : FAILURE in 13m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84ff2eb014214effa5ecc497833b8cc0 : FAILURE in 32m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1390561423497cbd297ed0b795c1d9 : SUCCESS in 6m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/55695d3d92424337b63fea2459236474 : SUCCESS in 58m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2c60e973f2ba423b8193681eaee35c7b : SUCCESS in 1h 51m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9857610d057b456b82b0f1bd08382265 : SUCCESS in 1h 08m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1e1f45dbed914ee8bbb43aab05d0b9f9 : SUCCESS in 55m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/725f5b9bae2f49fea010c53493623e7b : SUCCESS in 1h 20m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f7921b70d82469ea9cd1cd9f6e3f565 : SUCCESS in 1h 19m 51s\n\nWarning:\n  Comments left for invalid file neutron/common/ovn/utils.py","accounts_in_message":[],"_revision_number":5},{"id":"f416eb4a1cb69d36724c0df34209013b3522cb86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-01 09:18:21.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8aca25e407d9f3fb86d904af15be2f08e6025230","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 10:01:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/382aab581da748e8a025ce73b0497810\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd84037200564fcbbe89c7caa55bfc65 : FAILURE in 8m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fbed9779e6d943b1bf9d18e51b8e887e : FAILURE in 7m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/60951477054b4d51a8cb333ac36f585c : FAILURE in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"65d28db6242d76ad2fd3e31c77d74b27a99a61ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 10:23:57.000000000","message":"Patch Set 6: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c15629abe3bc4bd4aba6916ff4a6b25c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2329833b86f4b77a7756773cb2731f3 : FAILURE in 14m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1068b44fbbf849babafe2d08b3270e7c : FAILURE in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99bd61467e564e9c85524e04cc289bc9 : FAILURE in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf3f1876a4584f7384e233a550fd1ae4 : FAILURE in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94237fd6b2014c7080cfaf88972426f0 : FAILURE in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2066d2e4534e4852837051ebfc9daf9d : FAILURE in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb62ce2f5d19493aaabc428d56a939d0 : SUCCESS in 12m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ff79ec3d4cdc4954a6564cc5d35b7232 : FAILURE in 52m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cda6b6d101b448529a149331218e81cd : FAILURE in 16m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2121399595c4ff79be1cade419d45ab : FAILURE in 49m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b292e26a87894054a66e1c0701923003 : FAILURE in 21m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/020a289a66ae419f8c98e4d046d883ef : FAILURE in 12m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c3ce6e497037474c9ee23ebf82e8cfcf : FAILURE in 13m 13s","accounts_in_message":[],"_revision_number":6},{"id":"363d34f4245020cdbd3eebc343a74b991595f89f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-01 10:51:36.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"9ef13414ecc3e2b9d71b77ed96e6d2e53904aa75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-01 14:57:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"955e6783960862710214f91a7d88e3f9b0030eca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-01 15:43:56.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"d8f6a74f71785b7601306a00365455bd0525d62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 16:47:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08290280b50c4e34991bd97d13a74dc6\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a3b0939db9cd4c97a5adfc29fdf8a5cf : SUCCESS in 1h 48m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfe24b6c55b54b1ba0732fde83097831 : FAILURE in 40m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7247cf30b0645dc8afe72a042ae1459 : FAILURE in 7m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e77b93d926c472eb4a45f3e2078a2cf : FAILURE in 22m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62f381671c0b4656ba95f2603ed7fc0a : FAILURE in 16m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b3fa62e560545ba8322fbc9139370c0 : FAILURE in 30m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8916b789c12f4ba1a9252c27384bc848 : SUCCESS in 10m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7fffcf2754d645aa9719e9d60ffe172d : SUCCESS in 1h 11m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f00138ab336d4475822d368887526a3b : FAILURE in 19m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0de8fa74f02145129dfca2748bac0d74 : SUCCESS in 1h 05m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e7cb495b5eb3403d819c55d9339ed74b : SUCCESS in 1h 00m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/479989a674874028b6f475fee6a15efa : SUCCESS in 31m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/74ef2c01c4f9444b8e9fce052b993345 : SUCCESS in 1h 15m 33s","accounts_in_message":[],"_revision_number":7},{"id":"abe4938f65bee9be4bacc34e69b192f049cd092c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 17:44:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06117bbd8220425bac32d57c8a449dd6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf77f0edd49f483fa9bb262d8d9c3ea5 : FAILURE in 22m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e90baa103539469486ba566b2331d03f : FAILURE in 43m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/db3ffede3ca342c6afd1961b698f1ee1 : FAILURE in 50m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6dac270aee6a02c3cffd56f7af72c1bc9b99ff69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-02 16:02:03.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"fbf23cddde037ee2ae68a7e2c9a4a7a0985a1ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 18:23:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cde04863b5f4acca99a4f1edad10921\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e0b808807c4e4f5aa851113398cbf3fe : SUCCESS in 2h 16m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d1263c64b254d5d8def18ebeaa60234 : FAILURE in 43m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b2fdfd86aa545fd8227e6645fe9b515 : FAILURE in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba87b81012bd4c46aeee3c4a649f3f60 : FAILURE in 22m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aff2019c58bb4006a1f3f4b16e4c624c : FAILURE in 22m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48315ecb72fd44f1b9fce529d773c9ae : FAILURE in 16m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/059e0ba63d2a456e90611e34817573b8 : SUCCESS in 12m 41s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6593cd10f2964ebca42131d111e999c8 : SUCCESS in 42m 39s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1bb60dba4b8b48d3ba6f1e5f0d611f97 : FAILURE in 27m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce566bc84c214558bbc4da805b080fac : SUCCESS in 34m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cb6de7554c5e4cb892df564d57f99e29 : SUCCESS in 58m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4a40e88bf0e34e60b0c8e1f363717e97 : SUCCESS in 58m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/01aa76e222ad4378abcefc2f2799d3ab : SUCCESS in 1h 16m 13s","accounts_in_message":[],"_revision_number":8},{"id":"71d8194e04151b48ac9407860f969c5e2b646547","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-02 20:47:20.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"93f8246b02526c80ab1b2c0890f85a8ca7e3f478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 21:01:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ff5505f443b4e40afa31acbe7e0cb91\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/61963de1ce524ceb88f6db98f9a4c91b : FAILURE in 24m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3fa5d37d6a740dbb238acb3a46065a8 : FAILURE in 39m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/37ddf8db1ee440fc9532b54836f03a97 : FAILURE in 49m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7a174b5db2fe557a18d0bc632fb82d615f9e4341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-07 06:58:47.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"a0cd9d4621c33c0f33cd5b7b4bf176a8e87e4fbc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-07 07:44:37.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"eb96b803358c40ae15a442e8cf8dbcc16822489e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 07:50:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c01cd1beec45f4bfacfdd34baad8e4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/576e39884c294348baff9a6a5fbbce63 : FAILURE in 41m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2cd0890efe244abb893a4d3c6a3c4573 : FAILURE in 19m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7f4511db6d4446509d76ccd116f02ffa : FAILURE in 49m 52s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"32ecfb085a2708eede8532a3e6b2d49c4bda47a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 08:54:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8825f2a95ce4a699b62423e7018cd41\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/39b34d4da4ab4974b6c74567d0d95d4a : SUCCESS in 1h 55m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd8ff23052884621b45cfff0320a96d2 : FAILURE in 43m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e5bb97d1264b07986a86ec4d91744f : FAILURE in 7m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a78d08180a6246229c581a5288d07862 : FAILURE in 21m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb2585e1ffdf4c5d906090fb712aaf39 : FAILURE in 21m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d4498e23fd943ea834e7c6af68810bd : FAILURE in 19m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49cd90d49e9a4cd19ab2d0a4732ba729 : SUCCESS in 6m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/36d43f67114341a19aa172957a4b7430 : SUCCESS in 47m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/82322c1406a54c5ba7b7f8febfbdcac6 : FAILURE in 31m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e06f5cb76c24968b11d080257ca64e9 : SUCCESS in 1h 05m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c36d45c962944d10b6bc23eace54f8ff : SUCCESS in 57m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1bbb616044d243d493bfd6ada31a9187 : SUCCESS in 1h 16m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c24d8f274dd649869ab5e3c758ccb5e8 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":9},{"id":"2f01d3c4007b64dccfc347366562f6fb9c814a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-14 15:03:38.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"5a6280e3409122dafcf3f3c6fb1539414a1241b0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-14 16:17:13.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"bcd6113e5c7cea87dd7eb961b2026e064c4f4a52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 16:26:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06e286ed1aca4f7d9ad9d846c205d0e8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f912cd2d2f0f459c908946e3741b4be6 : FAILURE in 43m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e21666aa6c5d4d0d8b6445f269e68504 : FAILURE in 38m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2264288f92744940a7e91f6608941eed : FAILURE in 49m 36s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"37ac5a18b18174d7b506bbde0aee89e48ad00866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 17:05:57.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ff2996760e949d3815123e8a1aa60c9\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3254011825a842079172abedfbd5e41d : FAILURE in 1h 50m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e0effe11cc764387b7c137e0d0710635 : SUCCESS in 33m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f40657d2a5145c395e77d8904e4300d : FAILURE in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/411a7b9f9c9e4c8ca59797e9175f4cd8 : FAILURE in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f44a2532f35a460fb5eb0630691ed235 : FAILURE in 23m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1419149f10a47a5919aac222fa32ad5 : FAILURE in 16m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/551ef7963cfc44ec928510d470ed846f : FAILURE in 28m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c53eeca641e4def9e2ea39508058675 : SUCCESS in 11m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f01831ed599c43cead2c714eeac77b69 : SUCCESS in 34m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/be6f74b3d5e44fde91c6cbf934b7c868 : SUCCESS in 1h 57m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/728ecf1e2ee14060970430d4b7ca7781 : SUCCESS in 1h 15m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b809e0280266402fbda2d7108f6e800e : SUCCESS in 1h 02m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8d5c4b8dbea048379959b514bed34a5a : SUCCESS in 1h 14m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7156b8b2eb344fba8b6a6f1c0a2e6c4c : SUCCESS in 1h 30m 41s","accounts_in_message":[],"_revision_number":10},{"id":"285d5b87ed2a9b1a4ab20562e23e0efee3376d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-15 09:23:57.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"f603e940ba47e8fc70eea4d0c6173a13db57e10a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-15 10:12:23.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"7e4cd55b14276ebb616136a8a89e9d9a2015ec87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 10:29:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf38e0bc192d4ac2ba2e613e70cd3c74\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6f0653f1b8904ba9b697e83686e061a9 : FAILURE in 1h 03m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7674d7507d6748c1af54981fbd3077bc : FAILURE in 23m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c5c54a56d6745e5b0cca682baf729f7 : FAILURE in 1h 01m 40s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"aad4dd90c060b373961a9924ad24d2f14e6c820d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 11:02:20.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47cf4941662a4c49ab5eb7306d3d395d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95ea419d4ffc4c659c51c253839e114b : FAILURE in 1h 37m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5d25a8f96f645d88cc2e06bbf9ebef1 : SUCCESS in 32m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60552e31797e495ea312b90ab16b6a7a : FAILURE in 38m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35272b15517b46efabfe7c2554df127c : FAILURE in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e22cc856c1a94147b92f80a80e8f35cb : FAILURE in 35m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f1da2ec566942148c8de2570e0e9dba : FAILURE in 24m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af366c056630428087fabf17d79bb459 : FAILURE in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcfbf07968c949afac71e5721c0a4136 : SUCCESS in 12m 18s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d0c2af917aef4539a87f84a99a8cb036 : SUCCESS in 1h 11m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d11433718e6b4c86a5b970d0e8f1e3f7 : FAILURE in 33m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a99dcfac667944ad83d1b54bb70a091f : SUCCESS in 33m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/104f61b034c54281a4a4168cdc8d8aa8 : SUCCESS in 54m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c3559adfcae4478cbad772d140297971 : SUCCESS in 1h 26m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b2e8d043522b4efcbce3045780bbcceb : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":11},{"id":"171e1921c45308c15d054d7293748efa2eedaf47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-16 10:22:34.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"466e38e0a972357fb82071ac89a19ddf78c97b1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 11:26:59.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a03e7ab77041dab9e434da14d65d3f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb479e17372e455d9b30c3e1898b7c81 : FAILURE in 1h 02m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b728c17c57344f21acb2e6a500fd3480 : FAILURE in 52m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/86deec4ea6a54c4382f466beabf9ce9d : FAILURE in 48m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bf297580581979eb4b7fd1b508bd78fdf99fa70b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-16 12:01:52.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"666753e4b1ac1fd2ab27749a6d7efa6b0f4b324f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 12:47:20.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d21a08225744882a826a6954966ad7d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/56d9412e83984f28945153fd7ab1ce6b : SUCCESS in 2h 22m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba791e49c3cb4f28b8040ef7dc37a804 : SUCCESS in 32m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e750fbb8a3d4b0cb243b16e07f3e35a : FAILURE in 56m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76b17ebb873d411fa7671a639858ef5f : FAILURE in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a25f764f5ed4122a6802fd3a069a73f : FAILURE in 30m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d935990cbb1c4ad8a5e82a70777efbf9 : FAILURE in 23m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97de9f056851483bb71877c91b5e1389 : TIMED_OUT in 1h 00m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd225e1b4dac4ab9a830f166fa583328 : SUCCESS in 6m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d47b348ae641438f872707c762c6e531 : SUCCESS in 53m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/55cbf9d84b9a41e6b5274ee5fd57bab6 : FAILURE in 22m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1fff945ad7564418871ec250cf96b6d8 : SUCCESS in 1h 11m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0890e033e5f540c38c6f0b861e712f0b : SUCCESS in 57m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9a7e20ee15894edeb45958ea7b7500de : SUCCESS in 1h 05m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/022a066ec2c24890a6f1137cd2f7c0a2 : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":12},{"id":"96e89de6bdc5ad666e11fe4f165e5c98a3d29576","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-16 15:06:33.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"a547272a8df7c3a47a379406db7bd210457ac615","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-16 15:57:19.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"59ba9bdc9e56879b7df7806d95757eca928e4a5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 16:07:54.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0e0849f089243ccb12dda324efaaa27\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7016990557f6486b944d31e7f6c005bb : FAILURE in 58m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc064e0dc03a46398ea34c6468cede77 : FAILURE in 55m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/89b32ddc034746d19166a283bbe5a802 : FAILURE in 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"eb9386a3ea28e7b4a43a8d1068a450d44f48a387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 17:33:48.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6af3e26d6c7c405680b8bd474c0bbe43\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6edb11c9cc154e75ac0432d25829c5e2 : SUCCESS in 2h 25m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e57107741b37430f868ffbd61aa1541a : SUCCESS in 32m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/946393070ed442bf87ad5928fc1d2a01 : FAILURE in 46m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9dea96bdb884a5f8354a98dfe40e156 : FAILURE in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/feaf17a339ee4b289f8f35ea7766b1bf : FAILURE in 22m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8cd8b831b1d429aa0e20a65a19ade92 : FAILURE in 24m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02c79fb8a2b244d79780fe9a01212501 : FAILURE in 37m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96a7ad84e66642998657cb3e0a89e641 : SUCCESS in 13m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/93f236e77ad7422d89b41d888b509469 : SUCCESS in 35m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f5b9a64b069a409ba7657a810b1ab855 : FAILURE in 25m 04s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/16871440ce054458ad53a5e04b4ca7ec : SUCCESS in 55m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cd9193bd1dcb467786582837c064fb03 : SUCCESS in 1h 02m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c53268ee885417c80c212bbc8a6e940 : SUCCESS in 1h 19m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/09a9c5793a804ce685c53298b8d9fac3 : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":13},{"id":"505afb4add887e12c32ae78c5dccaf273b171d51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-17 07:46:04.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"ffbcc0dc931b17fc670ecbf8fdaa48e64b233f44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-17 07:54:33.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ccd12ebcb1de9c7eddd4a439c8a5cd96df2832e4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-17 08:33:17.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"e86149f07b6a7075d3d5bf22f1e792dbf6a5e722","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-17 09:21:15.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"0d2a8b9e640d41c9f145920f14dda89626ef553e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 09:35:23.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/296960bc3de54422b39ea65b52aee424\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e671ce2d35c94727859919ef095d0a8c : SUCCESS in 29m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a246f00a874240a091c474fa9b8fbeb2 : SUCCESS in 43m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/016fdce43465444a8bd89f40ee8ff798 : TIMED_OUT in 1h 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d21804da70dcd56cc9095f46fd450ce375b8f9e6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-17 10:02:19.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a8e8e500e288b1c44d97d8617b9672e79cf0ab99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 10:02:19.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7921500ad614490a1dbfd5d612cbf62\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a7f44d059194cb095900af1f1559f0f : SUCCESS in 2h 06m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65435a57aa6e4445b4996e7ab34db9ba : SUCCESS in 31m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9978a909be09446fb02683c3d85ebb47 : SUCCESS in 46m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce1ad00758f4ab4b14e7119420fe5e9 : FAILURE in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e2c991322464672ab75814d47669f4e : SUCCESS in 26m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6eb0f4df85d4f118e4849d6c21c76e8 : SUCCESS in 26m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b31b80227ebc4366bd2cd94245755f75 : SUCCESS in 40m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69236540a72d48af94a1a2fc1c7c993c : SUCCESS in 6m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/130a151cb35c47f38aa9e5bc0858eab0 : SUCCESS in 1h 10m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/48ecfc2fa2b541e99f6d0f125d865904 : FAILURE in 26m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad6703fe62fa48b9b29e03ab84bb6650 : SUCCESS in 1h 12m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/613eacef852546e98280167c84be1621 : SUCCESS in 1h 08m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9521ccee5b06423dafd959f5e16ce941 : SUCCESS in 53m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/511bb9a88d7b454282c8f8c884188d6c : SUCCESS in 1h 16m 04s","accounts_in_message":[],"_revision_number":15},{"id":"d04eacea9d7f26d4b5d9314b0b3d712f33f44176","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-20 10:07:05.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c8de7671565e5a8e464bdfe4670cf36de7ff7e8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-20 10:07:35.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"cbf9750bcbf98b0d0540199a00b5fdf3f11f13d0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-20 10:53:30.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"8bbdcbbe92ea18f9b40c4e198c3fd7de7e665c6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 11:09:10.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e35c55b0446e48c7b1b2cb434a50bc4d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/835104948a354e748d84dea229f7b12f : SUCCESS in 54m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/694cffbd946e433097a5d8aabfa531d6 : SUCCESS in 46m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6f6e5c52b2924655959e511060705be0 : SUCCESS in 59m 43s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"48c87f883e0cb18c68280780f6efd6759a172472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 12:00:59.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdb782752f4b4cc486e990725568fc43\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d15d83198d204a3898e15de994ea64ae : SUCCESS in 1h 51m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aff4e7a4af3f4beaa25e8320a422d7bd : SUCCESS in 36m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bcd048e43cc481d8b992d3966df98b0 : TIMED_OUT in 1h 21m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674d9820b29c4a09868de142c563c3dc : FAILURE in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/197fbc38c42f484c95fd7b60deb27524 : SUCCESS in 32m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c91ec4a2292640b38bb13307d9e09add : SUCCESS in 30m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a2d80b92bef4869875a5ba133fe54d8 : SUCCESS in 28m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b4d25848034f079e9e6932e697975e : SUCCESS in 12m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/982445768a7845e09eb57be9536df9d0 : SUCCESS in 1h 11m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/97056ed2cc11405686eff200dc894e93 : FAILURE in 22m 41s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6061a2e4cb2a4c7a9dd9c6b950592ef3 : SUCCESS in 1h 07m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c29d0af30d854d9aaad29a0efe302faf : SUCCESS in 57m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/94478da708984245883c94895a102d35 : SUCCESS in 1h 16m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/03597eb1a67c4039bad64007e67e368d : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":16},{"id":"f35bc4e5c0f319b3a675ee59948078b3527e54d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-20 15:41:08.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"3febe096420ac132927735ce013b7c8e92bd44e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-20 16:00:36.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"60af0af920e1db0438096535bae997c484d4ebdf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-20 16:27:42.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"76aefd1099077cef61e744f2a936dede6f46e609","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-20 17:14:45.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"e2f9a8c924eb03734433c6df5a0134008bb714be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 17:19:37.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99071827ed3d4f10a8c3de847863e1a1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/07abd8ac0c54413393de8472c849f4ec : SUCCESS in 1h 12m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/88e172d8755f43758e54e235d9f1b002 : SUCCESS in 1h 02m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/acc050778231400f899b5d99fcd7d4bf : SUCCESS in 1h 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"fde5fdea170797894f88063ae76370084c8e2594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 17:58:46.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9446ce91eec1460580858bf136350fb5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/35b142058b6942edb53da20f0208f7ed : SUCCESS in 1h 56m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/808c33f5e4074d41a8f02b4b19b06c16 : SUCCESS in 33m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18dedc20793243ce9aabe2988d1f0259 : SUCCESS in 41m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10653ead164543d8adcdf70978966d87 : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1882947d135042c88fa2ca2b06af2a02 : SUCCESS in 33m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc156833c0b147eaa0045ab3782da259 : SUCCESS in 20m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e548201edf4d4f61a5bb439533d76c23 : SUCCESS in 26m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98067098d988426aa24d8221ba453208 : SUCCESS in 6m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ca7a15df470a42d0b9c3252384e1b6bf : SUCCESS in 1h 07m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fdba6fcea2b64909b9b38b3f5b55ffcb : SUCCESS in 1h 38m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0805e0fa1034fb1a8a458f969d52252 : SUCCESS in 1h 10m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/163962c0750b4e5bac8891fe5c458998 : SUCCESS in 58m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a6cfe1359d904a45be3230150d17eade : SUCCESS in 1h 20m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/93a255b622f9474b8aabacd3121905de : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":18},{"id":"f5a1a3e3362c20c1fb0880968688151c33722b2c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-21 20:40:01.000000000","message":"Patch Set 18:\n\n(6 comments)","accounts_in_message":[],"_revision_number":18},{"id":"437f194ecd8ce627d2009e91efe1cbe3d4d7609d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-22 14:55:14.000000000","message":"Patch Set 18:\n\n(6 comments)","accounts_in_message":[],"_revision_number":18},{"id":"761b56e828805c7a9e109317b8e6d9f9e818d344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-22 14:55:40.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"51db8810d897fc47d404e1d014f505b7fcc96080","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-22 15:52:12.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"ce6608e7831b5139c831b07c447ea115050497f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 16:11:56.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2adb02b4f5d466898400e7663e9e8c1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6531ba945d5c40f7a6781bd94b438f32 : SUCCESS in 45m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/446cdbd525a04961b3cf7cb39c92dba2 : SUCCESS in 55m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/621dc547b3ff405695d2a459a0754990 : SUCCESS in 57m 32s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"4f638b4695b7330548914fadb1724741203f5c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 17:22:40.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9722573d1e6c487fbcece526b36f8f37\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b1344eceeb5048e1bf9f0b3b194d7f8b : SUCCESS in 2h 25m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/75486012bb584460a5895c0513dd434c : SUCCESS in 37m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92cca7a4c79f48a686f51e04315d2eff : SUCCESS in 48m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d1f5a234544b6c93d2dbab76488519 : SUCCESS in 9m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2df6d7f35f08461eafacd10da9b52c3d : SUCCESS in 26m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0fb14ec0c37946408542c32e507fa1c0 : SUCCESS in 19m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d9c73bafa424361ba831e07aa5fa9bd : SUCCESS in 41m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f520b7a8f746278c4492eb5d5980a5 : SUCCESS in 12m 24s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cf4f4a5750f040a18076787fd7e550e1 : SUCCESS in 1h 14m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/00c02a567e754d6f9485fc643bd58c56 : SUCCESS in 1h 53m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2e1ec207c3b422ab01301959bef7cde : SUCCESS in 56m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3b43b622f54c453db71c49a7155a75f6 : SUCCESS in 1h 09m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6d148f99b8cb4c4f9dd58cd27a8e48e2 : SUCCESS in 1h 10m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77d9ccd4bbbd45e4b3c3d882381dc0be : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":19},{"id":"c6cf718490e672bfe66ccf51dc369abf54e0af17","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-22 18:21:20.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"a642c6f02139972634af54db3c9038d36f2593b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-23 14:25:25.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e78c15dc56e4b7b33020dd0020ca53e6bedd5a9d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-23 14:49:46.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1236b82f28f081314bd343836066dc3487b0ee51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-30 07:18:32.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"2072cd06056a083081873f35589b1ef5bd00c300","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 08:41:38.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f12462fd3c4dd79f023df20bc13220\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c3549172bddb4395a3dcb2d9bc21c44f : SUCCESS in 53m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7916c0cb771643d8a5e08d737bca9f7b : TIMED_OUT in 1h 21m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b06acf43cb142dbbddeceef45df12e9 : TIMED_OUT in 1h 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4b67ccc1632d63ccf6bc87e9113911c6f11ed74f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 09:17:49.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23fc4bf104ae4bb6a6843cf141ee9459\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/51206cad04524c66824359dbb8f764f4 : SUCCESS in 1h 56m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ae0bb4fa3b84779a159ac0f9eed201c : SUCCESS in 30m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38312f79e39f48eabbc8beb0c6334f46 : SUCCESS in 43m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f622161b9a7a42dd8e0a6337ff767f1b : SUCCESS in 14m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9030ec90b3754731b52421bd09d2730c : SUCCESS in 23m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa221ac8cdc441c9ab376db18aac5646 : SUCCESS in 18m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ae4129e9fa646caa0240f4b95907a59 : SUCCESS in 26m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f62c8907c344dea7de0191f4546b80 : SUCCESS in 10m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a14778410e334722adc45545c7583c46 : SUCCESS in 1h 14m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3ded1fe5b8474a439250b732db6540fe : SUCCESS in 1h 31m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5be5e6c021ea4eca82ea2553ba679b78 : SUCCESS in 1h 14m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6af3f6dfcdad464795d3a642a068652f : SUCCESS in 56m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5481721087443cc9de20bda08f9cbea : SUCCESS in 1h 24m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0b92713f84df4f0cbf2c6bb61b0411ef : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":20},{"id":"69e3125d81045907fbcaa64495083d5031361de8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-30 15:47:48.000000000","message":"Patch Set 20:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/55/962155/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"feee4fa6583d15063570b54c99528ada2b5450e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-05 12:37:33.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"9fc84ef89296b2adb49edcfda05636bb476c5544","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 14:03:01.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f47cb5bad54c7cb56addf9c284253a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a2ecaf316bd4afba095cff40000cf02 : SUCCESS in 42m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45981d4f68b648eebea2309914ab7202 : SUCCESS in 45m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/463335138b8343da94fbdd622b686379 : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"756a9ded0d73c3c86428d3a7c72f071ab75a81dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 14:40:40.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e38353cfc4c7428795454a191f75c5dc\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6fc5b52c93854ba0a21d6ee5ce68763b : SUCCESS in 2h 01m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fadd523633b046578ddab681acee3bdf : SUCCESS in 31m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f572156175d4a8aac9bbd4d4b635e19 : SUCCESS in 1h 04m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b24515955c8a4671a71b96159d36561f : SUCCESS in 12m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c6af719e5b84a40a5b7a186aa23be1c : SUCCESS in 35m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c20d13b23e674eeea3735dbddd19264c : SUCCESS in 23m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bda7d5ccf3fd4bdf980cb8f012a07d77 : SUCCESS in 45m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c50e4a3754f645a7843e93a36b1e8579 : SUCCESS in 11m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/69a68b7b97314c068850a510036ef0e5 : SUCCESS in 1h 01m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/eae0a2049b124e15beb4d3a548dd448f : FAILURE in 1h 45m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8374cd53f33b4a598f9e30a2d0f5821b : SUCCESS in 1h 05m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8f81efed0fb64e448d3c14afe53d5a0c : SUCCESS in 1h 06m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/02153cd0ee3f46fd8f344d5f71385dd0 : SUCCESS in 43m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0e994076201b4ca2affdbbdf6e1036f5 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":21},{"id":"3bdd362d00dbccb2119ce2a66f60eec602c53344","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-05 14:52:22.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"cbdef332c424280dbfb02c30b6d170bf1b997366","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 16:49:23.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9153e0ad3cdb48489ac8d08152de3eff\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04f67a9ea3b34ce883e3babc1c9c98bf : SUCCESS in 25m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be87909beef34273b6abfcf3840dcc12 : SUCCESS in 31m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/12f352314c5f43dbb591dcb366ee2a30 : SUCCESS in 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"cf8be72167d5407c3a09d20c596716cf6231426b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 17:27:34.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc56149a424746138ee0a15dea5228f1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/56d0d352ce78409793909198e295f363 : FAILURE in 2h 33m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/05938bec58d74d12aa0f910c654e87ad : SUCCESS in 24m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7833eda60f274b4b867ecde46163504a : SUCCESS in 1h 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e9a76fc0a6c4629ae14cad9de0ea191 : SUCCESS in 12m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb08a8c3329344998cd8f0097e3bd9e9 : SUCCESS in 35m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e984671077a4d9ca991bcd940036edd : SUCCESS in 24m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db90c799984e46ec930ccbd98c4bee83 : SUCCESS in 28m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399eded29dbb4e73877daeab9904236a : SUCCESS in 12m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5ed699876db645db975fc17a41cbb205 : SUCCESS in 1h 13m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e6d53481a88248aeb98648113f137f58 : SUCCESS in 1h 23m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/080257d68e614cefa28198528fb3be72 : SUCCESS in 1h 12m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/33ae57f2799147078b4c51fd02247a35 : SUCCESS in 1h 14m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65d9d99b99f24e0fb173e452c242459e : SUCCESS in 1h 13m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8856f7041975451e95e62515915ffd50 : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":21},{"id":"cad960a11279c2acbaddfa6cc254dd628d3db1a4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-11 10:04:14.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d893e69fb3fb0a781876c3a7c1c74a594684d1ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 11:09:42.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43a50c9b3a6c449c9b09facc9f1984ef\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52c94c268bd54e96a4975f47b9f8c024 : SUCCESS in 1h 03m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/92f9814530244d87bf6b74473749ab5f : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c36c8bbeacad4efbb7a2cfa458ff80e6 : SUCCESS in 1h 01m 49s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"57376d50bd3a45e098c58756cf17e4f58aa69da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 11:58:47.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66509357095f4944a30c8e4f85e35b94\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/261b6de20598481b80da8f5325950f4b : SUCCESS in 1h 53m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e7c9a0710d248c795fe0c39412d6382 : SUCCESS in 31m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d0d23743b364b32b75513fdbc712700 : SUCCESS in 1h 02m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac8cb501e98944479aae7bd0c9e34795 : SUCCESS in 13m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22e954277e1147e4bd174e772e5eac2a : SUCCESS in 35m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9718a253813c4217a15e780bb9252982 : SUCCESS in 35m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd2f59de7e5f43628bc7dc10ce602723 : SUCCESS in 8m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/657f06cd1e234e8ca9ac0f3db6e606c5 : SUCCESS in 1h 06m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3cadb53a8b214508b902494830d9fb8f : SUCCESS in 1h 41m 00s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/5389db2d1e784cc39dc3da780682071e : SUCCESS in 1h 08m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/22cfab1bb17b441fbbac6f16f09b2cdd : SUCCESS in 1h 08m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1b27aec38ee745e8a2112c2877b15098 : SUCCESS in 1h 33m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a67fd1ea77004c5592fbd7782d92ce96 : SUCCESS in 1h 24m 03s","accounts_in_message":[],"_revision_number":21},{"id":"4bd345f2bc44657ca937958a6e744a6320ca44c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-13 15:52:34.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"a38fdaf79acb93e54d26be46896610d79160b49c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 17:35:27.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d745cbf0a0b47a18d13fb9158cf51e6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4e4beddf51c4c128ea408322bd53158 : SUCCESS in 1h 10m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49902e05dd124d0eab8e4f3d8aa49d0f : SUCCESS in 44m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/23b2f6d90b914effbcbf61ec6cf974c2 : SUCCESS in 58m 32s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"b444d671f09c0c655010da4179a25e608ef88a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 18:07:33.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d27b649b215464499b02a4ac4f5ab33\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d76d4c291e6849fcac2397272e2552fb : SUCCESS in 2h 13m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9909a7844827455a83b61e25e8f1e9be : SUCCESS in 34m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/564139ea18de42be810111bb283eed19 : SUCCESS in 55m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934cbcf4f9d149f599e6a90155d3ea29 : SUCCESS in 7m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/999a93db5cf14481b0cbe8f82bafeea0 : SUCCESS in 33m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0d04e9c1fe64e02b4bdb1fb062d5208 : SUCCESS in 35m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03a5634d1ff4d56933da12fa0ac45df : SUCCESS in 12m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ea2f31a954084893aa21fda7d023f075 : SUCCESS in 1h 03m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2e27dd283ac04f27b6aa756c8673d714 : FAILURE in 42m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/506d33906e0742ec8b1ef6556ef13caa : SUCCESS in 1h 07m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/28ef4feee6384ff7b4e2575f0d722c0e : SUCCESS in 1h 02m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc125da28f034e98ab70f4ef9fca07ff : SUCCESS in 48m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/622837acbb0d459a97d66350983c9ae7 : SUCCESS in 1h 08m 52s","accounts_in_message":[],"_revision_number":22},{"id":"0b65463a62fc466434e851f6b3c1bfe98251b45b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-17 07:01:49.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"68836b0269c29aa33ee2d1f7979184ec92778fdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 07:44:52.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5823967f7f7346e595baf8d8341525f8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e66960246dfe4e25981f49a1e8406c3e : SUCCESS in 40m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6dede64e908c4e42a625782afd8cc0b4 : SUCCESS in 39m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8554967a4cef46208848af1d2b440b78 : SUCCESS in 30m 04s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"6512afada36949da0b7d5bca369ef7278234bff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 08:55:12.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e652d309fa34b24af03cc8ec0234048\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ea3b6d7643564115b83fbefb15126299 : SUCCESS in 1h 52m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c556630ddb174e09a1afafed7272f3fd : SUCCESS in 30m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8baecaae280f4058aa3f5a0801c088d4 : SUCCESS in 57m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/984e610969424e93bb6494de2e087baf : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fafefb64af544ec6a7142f3cfad7519b : SUCCESS in 23m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01443aea3ac84353bfbd8510b589135f : SUCCESS in 24m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dac72c527524bc2a1b37b39b62df1f5 : SUCCESS in 6m 00s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/46214e6552dd4dc58f0bd2434ee1da9d : SUCCESS in 38m 00s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/95168068e5b44423a38816d9d387d7ea : FAILURE in 43m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a3222fb9a144d9fb63df3dccc14e916 : SUCCESS in 1h 04m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ab3a583ef8ed424caed5dddc869d18dc : SUCCESS in 51m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ff9516a417ec4624b017c16b460f67ca : SUCCESS in 1h 32m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/333ef6ef55774e77b3d599070ffaff27 : SUCCESS in 1h 23m 02s","accounts_in_message":[],"_revision_number":22},{"id":"5d492179a6f7bc819479a9cdefa70fcbdcb0df0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-17 12:25:10.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"3fbc3ffd1b9cc519ce04c8dc51072c73fc456484","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 13:38:39.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c743a82cb2b54b3baf8b649e765248ee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/977077d6afd841b4a7e86351827ffa17 : SUCCESS in 49m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ca627f0d910a4d928afc903898ce82e4 : SUCCESS in 53m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3be1ab8870fd42b1aa0ef1a4b1f1d77e : SUCCESS in 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"374e73e0f3d8d837523975f5de90830f87957bd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 14:22:56.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/776d19a93aa0463cb7912bd56d5f04a0\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eef585be5f6f45e6af92b42792d65019 : SUCCESS in 1h 55m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f40f7190f8f4c5b9296d13a69ee793f : SUCCESS in 32m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e128bb03c9749fa8ac0f29099809bd9 : SUCCESS in 58m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a240c002fda542a292fb4a4240a4c235 : FAILURE in 9m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bcca2611e98479f9fa24fa5c57d5152 : SUCCESS in 26m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51ddf59e7f934938b1e26d295a4567ba : SUCCESS in 35m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb6557f5e869470caae2e5d21a65fc7f : SUCCESS in 7m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1160931086044330bc56ea4bb69cd06c : SUCCESS in 50m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/58d31fbecd04417c924207b68d6c60b5 : FAILURE in 18m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/47cd12d2e7874029afe29e59b86f97a4 : SUCCESS in 39m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3fb82f8f209a432cb57d269c9a44ad17 : SUCCESS in 1h 10m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b142f8b3bc454e0fa644743e61209109 : SUCCESS in 1h 27m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/789587cdf34e4a2baff4119e89a20dc6 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":23},{"id":"af1d8bcbf2b40c7f8f0c20cc63f9999b73e00cf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-17 14:28:07.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"2cbd194ec261016d8f083482d964aa8d2f683687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 15:39:43.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee1164f6468f49fa98f265b2e8f8bdbc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dfba03077da74b6f9d1be58d9dc6729d : SUCCESS in 1h 09m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8aae7de6eb2c4ac8b83e93c665a50720 : SUCCESS in 53m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d266e430603747c6a86c815f315bb1c3 : SUCCESS in 1h 09m 37s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"3280582570b5bdde8349f9a36fe1f8e88f3f6d4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 16:29:08.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3453f3898db043c0b484f4f191554ae9\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e689a7bef7c14e05b09f75934ed40663 : SUCCESS in 1h 41m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f140374c4086498495892ce518202821 : SUCCESS in 33m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4abe0e6e3d4b46c6bafc585b7641c8ce : SUCCESS in 44m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a8a9ef2c034f098268293b97562b9c : SUCCESS in 12m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/797b55d4328a4d9cbe8080d6ba25fcf5 : SUCCESS in 33m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/024205cd3d944b638262d605d927ab86 : SUCCESS in 21m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9de41414e44786a82fd068fba58005 : SUCCESS in 11m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9aa06b86efa04bd4aa61e2e4bbc18f64 : SUCCESS in 1h 12m 40s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8bc6d2f80d354be6b73724c5cf417b52 : SUCCESS in 1h 55m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d7134294b544c7a8a4f74aad51a2501 : SUCCESS in 1h 16m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9c554fe5401745e6b5d5481c0b48c926 : SUCCESS in 1h 04m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/adc861df9fbc40f1ac9667fe96e2ea0b : SUCCESS in 1h 14m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/60538f018b484015a4dd2b64fbeba5b2 : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":24},{"id":"9e7f1e818f9d47a20cd3d211e22294304083684b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-18 07:42:38.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"da8f8247606ba80a0e38079b3c3bde158041c4a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 08:56:45.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/141b2ab3d05d44ea8f2c7b88917ca3f7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/052687ff08954586b2bda90309f1228b : SUCCESS in 30m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b00ba25fb82c42418c97aeb4075138b8 : SUCCESS in 22m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5c1dbd1bb76240368b7dac20ac34756c : SUCCESS in 1h 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"a863923c1e6f7ff0e5c2f4c1c07efee70f6eb3ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 09:47:10.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab140cfac0284a919ff91481df6af961\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f9e81d72a28a496cb1d6a9286e351675 : SUCCESS in 2h 03m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf1648c568ea47ffa19396f818c78b7e : SUCCESS in 30m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/373d6f22085d426b842a57a9d9dd9fce : SUCCESS in 58m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1efa24d64fc45bbb9d53724915af921 : SUCCESS in 12m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ce8e08257c0437fb2367ccf9ebaa1fa : SUCCESS in 27m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d6655484de14218814c7f4fba36fe05 : SUCCESS in 37m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bcee296c9df40aeb9a4c6d498a41fda : SUCCESS in 12m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ec1db906e23b4da7b6b89cc628564e85 : SUCCESS in 43m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2742e0370e17444a9534c3223c5ee327 : SUCCESS in 1h 50m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4152f6e439342fb86f5b156a59d4353 : SUCCESS in 1h 20m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/63c318e8e0c4447d898b3e0c41f278c1 : SUCCESS in 1h 18m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/595aed7f32f2473795342c3e1c2ccb85 : SUCCESS in 1h 24m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/22cf7a3f08534c8b87e1755d9228dbfa : SUCCESS in 1h 15m 40s","accounts_in_message":[],"_revision_number":25},{"id":"c676f3e551a97ba33b6d81ef6112d3c75a1b7eb5","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2025-11-18 20:45:21.000000000","message":"Patch Set 25: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"c09fbc0ba81b87f9099bffa9eba6f2da56a698cb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-18 23:38:31.000000000","message":"Patch Set 25: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"2e36184b9aa2f0709780b5dc0a21aada9b6a8288","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-19 08:01:32.000000000","message":"Patch Set 25:\n\n(4 comments)","accounts_in_message":[],"_revision_number":25},{"id":"16623acd301f315ab0fff288298fbad2d2f90559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-19 08:02:04.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"d63f9ca9167ded232ac2bcbf45b8cb055bc39b25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 09:23:44.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/122a6101ed984d4ca16a27c9c85664e1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b533821ce4824d3dad75653b8f67f76f : SUCCESS in 25m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/55395236af7f4373858aa95c903d98de : SUCCESS in 43m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3fe3941662ef4c0fbe579aeaa8d086e4 : SUCCESS in 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"c61bbfa43730b141c1f24bbb58ccdb39a4b92b7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 10:22:05.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b89ee0168ee948dc825f627cdafb1958\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1b4d27c252ef4a4ba5775d4ef3b657b6 : SUCCESS in 2h 19m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d8d031faed349869f8ed3ed0fcceaca : SUCCESS in 30m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/209c07d5958446a1baf35b002f05722b : SUCCESS in 38m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25dc04361f424d0cad938ede6d6ef3f9 : SUCCESS in 12m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96ce00dd6f61402dac963c8a6af08f1d : SUCCESS in 26m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d22964ea9b3e4cb5855eef69ea4cc49e : SUCCESS in 35m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3881f27caca490092f2bd68bd428119 : SUCCESS in 7m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8d8ab1a7e318452b9748bf4113fe39f7 : SUCCESS in 47m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/741a725f17314cd4b326174dc2eb1e49 : SUCCESS in 1h 04m 26s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d603c6d3662c484dadbb58bd6f436f71 : SUCCESS in 43m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/553bf25e87894ab7826f025b25a3153f : SUCCESS in 1h 00m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6e1370b953a64fafbd5c4affb977ee42 : SUCCESS in 1h 48m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f6c9f6af6fa749c0aa6f8c4fa49252ba : SUCCESS in 1h 25m 36s","accounts_in_message":[],"_revision_number":26},{"id":"bca3336c30f9df55ffb90004de7a28c8367ae4a6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-11-19 13:28:51.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"77bc51b0999c209ca245bc3b61ffe72d8f559862","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2025-11-19 14:11:59.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"61846c81fe781f98af1bae44ecc75a762e9ce1d6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-19 22:29:58.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":26},{"id":"71d8de7bc3bc1534ad34dcddad8af045f827a1ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 22:30:19.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"5add2359db620d993d3facbac04ada60ebda9a6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 23:20:09.000000000","message":"Patch Set 26: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/966789 is needed.","accounts_in_message":[],"_revision_number":26},{"id":"620ca90336463132486a854e3a38bc71a9219577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-20 14:21:31.000000000","message":"Patch Set 27: Patch Set 26 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":27},{"id":"54611818e8966ac023511fcfeb7b443040984ee5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-20 14:21:39.000000000","message":"Patch Set 27: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"d59640e435fb1a83546d43817a91f320d2ad055c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 14:40:03.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd52a3db700d481ab1b565d2adfc1ffd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db5eccb568c348c180c17cbe8dcac8ba : NODE_FAILURE Node(set) request 914d53ae9d3f4afc8e12fbee514d7adf failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c35950c4f8e46d3b8a78629acd37fdd : NODE_FAILURE Node(set) request 0c5155f18d334f78a6b3e981841ad645 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5eb9efd324a74c1a83cca3c8a97d395d : NODE_FAILURE Node(set) request 7b5de4f2ddad4bf28b7ed90ddc5393e5 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"1703bd77f695ce6bb8120b78d42d905ff2716bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 16:31:33.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ff3dec1c09a4b9f8d8dc48ddf6199c8\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3874a3704d7f4566973125bc625a8215 : SUCCESS in 2h 07m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68984d34f78244a3ae70cd8452417f3a : SUCCESS in 30m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d581d8731ccd4c9aa6bdb65aed7d5383 : SUCCESS in 57m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd57a75a76d94b038b7b0e9e3ee0068a : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1740b6287b2042908305fd9b96a1f853 : SUCCESS in 37m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab0fe155f14d4953b782e0a9944b34ce : SUCCESS in 26m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a18287fdd84017b75f56968bb4aa8e : SUCCESS in 7m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2b0a3f843e9f4ae3ba243800e9a70804 : SUCCESS in 1h 13m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4fba3db9a6b54e3da427ad3291e2cdd0 : SUCCESS in 1h 54m 02s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/39786f93ab734559a1a99bb7aa59fbdc : SUCCESS in 1h 13m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c3742282072d42ebb66a47c9dfbfcd0d : SUCCESS in 1h 17m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dcae3ff07dbb48e284823f1d54624a9a : SUCCESS in 1h 43m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0250cdab10ea4360bcf4cc4b3eeb04eb : SUCCESS in 1h 22m 51s","accounts_in_message":[],"_revision_number":27},{"id":"b1ac99a3f57186f13a143bfdc6f23a1349c5744f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 16:31:58.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"868bc17fb154fd89a7c192c1c1d84cf30cff7b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 18:38:52.000000000","message":"Patch Set 27: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa960425d0aa4ec0a1bdd23360810691\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e7e9e2afe82b43bebc6b4d46fe1b9b6b : SUCCESS in 2h 05m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0e798408fe440a8ba11b748acfa597 : SUCCESS in 9m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6d7369b9525476a8485d5223e055b24 : SUCCESS in 25m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d61b9703126149e2912b94eca18b6301 : SUCCESS in 23m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b2618fe7f54939a14c52cb2faf66e4 : SUCCESS in 6m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d7fe64e3abb04fbcbfe8ee4455724fb4 : FAILURE in 40m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/97cc8562b0f84377973d0eb87207f72c : SUCCESS in 1h 08m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cb57f827b9984664b8bb8bf18caabd2a : SUCCESS in 50m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a41e35eab9174746a396295fc57ff062 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":27},{"id":"de9707be7296016f173a71d9d7c3d37f719c7902","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-21 07:32:15.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"13ba4adbfe7d21c1d89330868d3dc5100838c396","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 08:22:52.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ff1eff753344863a06ddfd3bc925da8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/124b722e9e3e4a62b0c69045e0d3869b : SUCCESS in 36m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/71eb288ecca54e3ba5f95cffe823a0d9 : SUCCESS in 23m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05b44bef06aa4575834cb1e4d4e0218b : SUCCESS in 48m 32s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"fadbb33cb1132878a62a299d9917fe6e560fdfc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 10:00:13.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4d8792f56f7442dacb28cfb07fcceaf\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d7b69418845646dea49ba2c776ca4b7e : SUCCESS in 2h 10m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a2ed1d22f6884ed2acfd9fd9469eec20 : SUCCESS in 31m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d03c40138ee945129742a64bc4255840 : SUCCESS in 1h 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99543739b694e558aa936116aacc866 : SUCCESS in 13m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad990cd4f55b4bfcaa1cbd34a20e3dbf : SUCCESS in 38m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0bbff0f7c3fe459b9cb85c771e04cb82 : SUCCESS in 26m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0285441ea46d4d729205bb59972b4797 : SUCCESS in 9m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cfcd7932f47841958d15fc80ce14bd59 : SUCCESS in 53m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/53ba41a8b02b44f3962064b3771be6fd : TIMED_OUT in 2h 14m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee86aaa878714ea2a6ffcb5463c00c4c : SUCCESS in 1h 08m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b082a55446ac4c1095b4e1bffdcea5a1 : SUCCESS in 1h 21m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b3c6dceea2694321a92425d7d4efecd6 : SUCCESS in 1h 33m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/03e88dc588b74981ac6e37dd96d0e2bf : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":27},{"id":"e732fea9657e4d52e488043e307c11cb72484188","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-21 10:43:38.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"b7e61404d183e7111b56b6e1e40a5f1943282a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 12:54:21.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ecef042b8344f73abb7897ee1069b06\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/624360c1b66547c8aa24cef32ff7a78d : SUCCESS in 2h 06m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9a0ea4aab055413bbe1573dc10b854e3 : SUCCESS in 29m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbe7f9a354b44db58f7e3995d3fec8bc : SUCCESS in 48m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fd28e0948cd4974ad9475e259ab5ba5 : SUCCESS in 9m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/013a736d110b45af9a370e70553dbac5 : SUCCESS in 34m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0711bfeab331464f92d1fdf1a41fb55b : SUCCESS in 30m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0802410443c54fbc87b38ffcfa58d902 : SUCCESS in 12m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/39e45749a03b49028b9cb7ec6a392c5e : SUCCESS in 1h 11m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d4d613f11cc34576819d222398b7062a : SUCCESS in 1h 48m 57s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/567aa9d6e4c041289b1c22f4b20792ed : SUCCESS in 49m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0a0d1e8ddc0b487aa31fb8dfcb96b610 : SUCCESS in 1h 04m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b217568512240f38d22ede98aa02b59 : SUCCESS in 1h 19m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1756a19d11b74966b4a25699bf3ef0e0 : SUCCESS in 1h 19m 15s","accounts_in_message":[],"_revision_number":27},{"id":"15b4c28d4238f3c265a7912eb5c9b3f2f7e47fe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 12:54:44.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"128fc1a8dc1dc588bf3e1cc5f617686f77e0da8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 14:45:49.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87c23a6df2b7443d824abc74ace671ca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7542a50f3f5d4def9aaf26089a9cd690 : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/77f4699a6fa04b4c87a79b574dda74dd : SUCCESS in 57m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/361a374d734e43a9bcf05df2aa28671b : SUCCESS in 1h 03m 14s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"035c69b3d36c3501e677307632f8ff56a624e517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 15:11:58.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46c0ec782fb143728835dfc8827b5e49\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6569cc3112b14e3a8f831fed53274660 : SUCCESS in 2h 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a04f5835aa42cbbc043030c48a293e : SUCCESS in 9m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edc3dea091c4438cb022eabc9b5c40aa : SUCCESS in 38m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d87db59cfaf43a6b4a6612fcb316923 : SUCCESS in 26m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/471bbe2d31794344aea5093ee45c9a17 : SUCCESS in 12m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd9c65ae75ce499eb45fe43986bfa244 : SUCCESS in 1h 46m 40s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a391f93bd1c04d6aa857b72daf7c2482 : SUCCESS in 35m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6d1f2971ebb64e549d84de2b8e9c30e2 : SUCCESS in 1h 06m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/baf7e6c735bd40e0b892e2dc4573a2b7 : SUCCESS in 1h 26m 59s","accounts_in_message":[],"_revision_number":27},{"id":"f6de9a9fd7fd2c86595fc0ce4b7d764747cf7437","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 15:11:58.000000000","message":"Change has been successfully merged\n\n26 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":27},{"id":"8bfc995a2f228b7c25757a67b7992a150c0443cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 15:13:12.000000000","message":"Patch Set 27:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faed48cda12c4d2da27f40c7bdf5b757\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad08bfbfa8cf40c7b970d7679da06ae1 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"e94e7447aa621792b7c31d71d1c581d473128ec4","revisions":{"e22d5e72cb2d80db193101fb100cf54df8beb511":{"kind":"REWORK","_number":1,"created":"2025-09-24 07:14:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/1"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e22d5e72cb2d80db193101fb100cf54df8beb511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e22d5e72cb2d80db193101fb100cf54df8beb511"}]},"branch":"refs/heads/master"},"59ae4af6019a91c2912f3aaac2e5420bac553e9d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-25 10:18:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/2"}}},"commit":{"parents":[{"commit":"686076fba452b5bcdbde5a66aca4e5c5470d43c7","subject":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/686076fba452b5bcdbde5a66aca4e5c5470d43c7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 15:13:11.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59ae4af6019a91c2912f3aaac2e5420bac553e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59ae4af6019a91c2912f3aaac2e5420bac553e9d"}]},"branch":"refs/heads/master"},"1aa92c2c4f4884a9de40293f4ffe145118efb5f9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-26 20:27:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/3"}}},"commit":{"parents":[{"commit":"45e1d5e6f2bdc9feff61510422de40a0107de8c5","subject":"WIP \u003d\u003d [OVN] The external ports HA_Chassis_Group belongs to the network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/45e1d5e6f2bdc9feff61510422de40a0107de8c5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-26 20:27:38.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1aa92c2c4f4884a9de40293f4ffe145118efb5f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1aa92c2c4f4884a9de40293f4ffe145118efb5f9"}]},"branch":"refs/heads/master"},"bdb0d211f235b49246bde11c58b9c2443c3d2b53":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-09-26 21:18:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/4"}}},"commit":{"parents":[{"commit":"1c49818f611f79f9fe8bbb256533cf9861f38c6b","subject":"WIP \u003d\u003d [OVN] The external ports HA_Chassis_Group belongs to the network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c49818f611f79f9fe8bbb256533cf9861f38c6b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-26 21:18:10.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdb0d211f235b49246bde11c58b9c2443c3d2b53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdb0d211f235b49246bde11c58b9c2443c3d2b53"}]},"branch":"refs/heads/master"},"2944b4b3128a7e797ddc0ab195cf6c75604b0fd0":{"kind":"REWORK","_number":5,"created":"2025-09-30 14:12:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/5"}}},"commit":{"parents":[{"commit":"dd6e94527feeab41cb86b2eed17272e52a649d2c","subject":"[OVN] Add method ``sync_unify_ha_chassis_group_network``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd6e94527feeab41cb86b2eed17272e52a649d2c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-30 14:11:38.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2944b4b3128a7e797ddc0ab195cf6c75604b0fd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2944b4b3128a7e797ddc0ab195cf6c75604b0fd0"}]},"branch":"refs/heads/master"},"c04a9542d3cf00b75348fcc65d9f006b1dba7f71":{"kind":"REWORK","_number":6,"created":"2025-10-01 09:18:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/6"}}},"commit":{"parents":[{"commit":"3641ae6f6a1c815373f1c2b9c5e1aa3be043e298","subject":"[OVN] Add method ``sync_unify_ha_chassis_group_network``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3641ae6f6a1c815373f1c2b9c5e1aa3be043e298"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-01 09:18:06.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c04a9542d3cf00b75348fcc65d9f006b1dba7f71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c04a9542d3cf00b75348fcc65d9f006b1dba7f71"}]},"branch":"refs/heads/master"},"2b4ee9673716d5709a68e3ab1a3dd9c72dd4510a":{"kind":"REWORK","_number":7,"created":"2025-10-01 14:57:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/7"}}},"commit":{"parents":[{"commit":"5f9805cd62b464c1b466fec8d2d8c791425601de","subject":"[OVN] ``DelLSwitchPortCommand`` also deletes external ports HCG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f9805cd62b464c1b466fec8d2d8c791425601de"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-01 14:57:18.000000000","tz":0},"subject":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b4ee9673716d5709a68e3ab1a3dd9c72dd4510a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b4ee9673716d5709a68e3ab1a3dd9c72dd4510a"}]},"branch":"refs/heads/master"},"b7c2238d941e47c269e28b37b947a025a2bff1bd":{"kind":"REWORK","_number":8,"created":"2025-10-02 16:02:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/8"}}},"commit":{"parents":[{"commit":"2e4330f06985203089d7fff03a6ddee45fccf317","subject":"[OVN] ``DelLSwitchPortCommand`` also deletes external ports HCG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e4330f06985203089d7fff03a6ddee45fccf317"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-02 16:01:36.000000000","tz":0},"subject":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7c2238d941e47c269e28b37b947a025a2bff1bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7c2238d941e47c269e28b37b947a025a2bff1bd"}]},"branch":"refs/heads/master"},"f6d70b7bb45702f562b2c6de302ea8f5ad6e4300":{"kind":"REWORK","_number":9,"created":"2025-10-07 06:58:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/9"}}},"commit":{"parents":[{"commit":"d7e80423fd300f4cd407a86221932146d74d3a04","subject":"[OVN] ``HAChassisGroupRouterEvent`` is for router HCG only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d7e80423fd300f4cd407a86221932146d74d3a04"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-06 12:50:51.000000000","tz":0},"subject":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6d70b7bb45702f562b2c6de302ea8f5ad6e4300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6d70b7bb45702f562b2c6de302ea8f5ad6e4300"}]},"branch":"refs/heads/master"},"40d8b8a374ae9ecd4881ba21094feef37a94d869":{"kind":"REWORK","_number":10,"created":"2025-10-14 15:03:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/10"}}},"commit":{"parents":[{"commit":"b15869fbe2458ae8387a95ae9bd2af9f6cfe0d61","subject":"Merge \"[S-RBAC] Fix policies for l3_conntrack_helpers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b15869fbe2458ae8387a95ae9bd2af9f6cfe0d61"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-14 15:03:22.000000000","tz":0},"subject":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40d8b8a374ae9ecd4881ba21094feef37a94d869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40d8b8a374ae9ecd4881ba21094feef37a94d869"}]},"branch":"refs/heads/master"},"f053fa46dd031b5d0e5caa618f8b9c5e6fb420a0":{"kind":"REWORK","_number":11,"created":"2025-10-15 09:23:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/11"}}},"commit":{"parents":[{"commit":"4ea604e8875b971b5024953794fcbc1bae031d2a","subject":"Merge \"[OVN] Implement a local Logical_Switch deletion command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ea604e8875b971b5024953794fcbc1bae031d2a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-15 09:23:45.000000000","tz":0},"subject":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f053fa46dd031b5d0e5caa618f8b9c5e6fb420a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f053fa46dd031b5d0e5caa618f8b9c5e6fb420a0"}]},"branch":"refs/heads/master"},"0c351e77fc0ae57bf330aa8614b4f337b3364cdd":{"kind":"REWORK","_number":12,"created":"2025-10-16 10:22:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/12"}}},"commit":{"parents":[{"commit":"ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-16 10:22:19.000000000","tz":0},"subject":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP","message":"WIP \u003d\u003d\u003d\u003d [OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nCloses-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c351e77fc0ae57bf330aa8614b4f337b3364cdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c351e77fc0ae57bf330aa8614b4f337b3364cdd"}]},"branch":"refs/heads/master"},"1a5753f76a15dcd2023334f00d4ba091c1a5c130":{"kind":"REWORK","_number":13,"created":"2025-10-16 15:06:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/13"}}},"commit":{"parents":[{"commit":"ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-16 15:06:17.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a5753f76a15dcd2023334f00d4ba091c1a5c130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a5753f76a15dcd2023334f00d4ba091c1a5c130"}]},"branch":"refs/heads/master"},"a35514c8bdfc5efb74c7f23e30bb47425e525730":{"kind":"REWORK","_number":14,"created":"2025-10-17 07:46:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/14"}}},"commit":{"parents":[{"commit":"ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-17 07:45:47.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a35514c8bdfc5efb74c7f23e30bb47425e525730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a35514c8bdfc5efb74c7f23e30bb47425e525730"}]},"branch":"refs/heads/master"},"4ba4c16730c61e6d50b23f6422b6352b36951b90":{"kind":"REWORK","_number":15,"created":"2025-10-17 07:54:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/15"}}},"commit":{"parents":[{"commit":"ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-17 07:54:25.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ba4c16730c61e6d50b23f6422b6352b36951b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ba4c16730c61e6d50b23f6422b6352b36951b90"}]},"branch":"refs/heads/master"},"4872d145a32e5a1ac1fc2ade01a2b9009bdd96b1":{"kind":"REWORK","_number":16,"created":"2025-10-20 10:07:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/16"}}},"commit":{"parents":[{"commit":"ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-20 10:07:13.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4872d145a32e5a1ac1fc2ade01a2b9009bdd96b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4872d145a32e5a1ac1fc2ade01a2b9009bdd96b1"}]},"branch":"refs/heads/master"},"1c338197c2fb18be1a455543769b60eeb0b6ec9a":{"kind":"REWORK","_number":17,"created":"2025-10-20 15:41:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/17"}}},"commit":{"parents":[{"commit":"ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-20 15:40:55.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c338197c2fb18be1a455543769b60eeb0b6ec9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c338197c2fb18be1a455543769b60eeb0b6ec9a"}]},"branch":"refs/heads/master"},"9b1e73995f53c3b5c104e621a195b0d6d0870c64":{"kind":"REWORK","_number":18,"created":"2025-10-20 16:00:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/18"}}},"commit":{"parents":[{"commit":"ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-20 15:59:39.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b1e73995f53c3b5c104e621a195b0d6d0870c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b1e73995f53c3b5c104e621a195b0d6d0870c64"}]},"branch":"refs/heads/master"},"5c27bdf6a7bba39f538332b57b06c3059c5e08af":{"kind":"REWORK","_number":19,"created":"2025-10-22 14:55:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/19"}}},"commit":{"parents":[{"commit":"ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea5b2da7fcd367c5ffbf995d7f89982b8f76ba9a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-22 14:55:24.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c27bdf6a7bba39f538332b57b06c3059c5e08af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c27bdf6a7bba39f538332b57b06c3059c5e08af"}]},"branch":"refs/heads/master"},"8c6555a726b510962a8c92a39022c93340eecb09":{"kind":"REWORK","_number":20,"created":"2025-10-30 07:18:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/20"}}},"commit":{"parents":[{"commit":"1bcf4a1fb5e544047d347e7d1912007ef670e150","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bcf4a1fb5e544047d347e7d1912007ef670e150"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-30 07:18:06.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c6555a726b510962a8c92a39022c93340eecb09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c6555a726b510962a8c92a39022c93340eecb09"}]},"branch":"refs/heads/master"},"46c07147b9c8545ccf60ea35ac96fac355cb1a43":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2025-11-05 12:37:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/21"}}},"commit":{"parents":[{"commit":"2ed024bb6feaeb8e2ec95efee2cbea1cb2c44da5","subject":"[OVN] Add a post fork initialization method in OVN L3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ed024bb6feaeb8e2ec95efee2cbea1cb2c44da5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-11-05 12:37:33.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46c07147b9c8545ccf60ea35ac96fac355cb1a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46c07147b9c8545ccf60ea35ac96fac355cb1a43"}]},"branch":"refs/heads/master","description":"Rebase"},"5cb800cff9cddea87dc4d79c1632aee253b98bff":{"kind":"REWORK","_number":22,"created":"2025-11-13 15:52:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/22"}}},"commit":{"parents":[{"commit":"d738cff717e58c5a79e80e795c32f2bf0f1bb01e","subject":"Merge \"Remove the ovn_emit_need_to_frag config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d738cff717e58c5a79e80e795c32f2bf0f1bb01e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-13 15:52:23.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cb800cff9cddea87dc4d79c1632aee253b98bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cb800cff9cddea87dc4d79c1632aee253b98bff"}]},"branch":"refs/heads/master"},"05fa795cdf18b9b218d27fd8a81a68fee7a4c485":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2025-11-17 12:25:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/23"}}},"commit":{"parents":[{"commit":"7ab3189576d6cb29846b304b31f1ea90c04a51dc","subject":"[OVN] Wait for the \"Chassis\" creation in the HCG tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ab3189576d6cb29846b304b31f1ea90c04a51dc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-11-17 12:25:10.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05fa795cdf18b9b218d27fd8a81a68fee7a4c485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05fa795cdf18b9b218d27fd8a81a68fee7a4c485"}]},"branch":"refs/heads/master","description":"Rebase"},"bd0c9bbe320788d199d412a1b5824e3456dd713a":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2025-11-17 14:28:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/24"}}},"commit":{"parents":[{"commit":"0f6d12ebaf74db45ae20e82c7597d9b9df9eb815","subject":"[OVN] Wait for the \"Chassis\" creation in the HCG tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f6d12ebaf74db45ae20e82c7597d9b9df9eb815"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-11-17 14:28:07.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd0c9bbe320788d199d412a1b5824e3456dd713a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd0c9bbe320788d199d412a1b5824e3456dd713a"}]},"branch":"refs/heads/master","description":"Rebase"},"bcfcbb0798eadb23d09be62a9f1b02138387258c":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2025-11-18 07:42:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/25"}}},"commit":{"parents":[{"commit":"46fb69ff967e97526c659e3c3eb7a03b482a4c98","subject":"[OVN] Wait for the \"Chassis\" creation in the HCG tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46fb69ff967e97526c659e3c3eb7a03b482a4c98"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-11-18 07:42:38.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcfcbb0798eadb23d09be62a9f1b02138387258c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcfcbb0798eadb23d09be62a9f1b02138387258c"}]},"branch":"refs/heads/master","description":"Rebase"},"fe88393b3daee05edbd26c7662546d0545adcfc7":{"kind":"REWORK","_number":26,"created":"2025-11-19 08:02:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/26"}}},"commit":{"parents":[{"commit":"87268f4582912e55854a6864633046120ab90fca","subject":"[OVN] Wait for the \"Chassis\" creation in the HCG tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/87268f4582912e55854a6864633046120ab90fca"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-19 08:01:39.000000000","tz":60},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe88393b3daee05edbd26c7662546d0545adcfc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe88393b3daee05edbd26c7662546d0545adcfc7"}]},"branch":"refs/heads/master"},"e94e7447aa621792b7c31d71d1c581d473128ec4":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-11-20 14:21:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/962155/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962155/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962155/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962155/27"}}},"commit":{"parents":[{"commit":"78593ce9b853ef51184592c13101a409024defb7","subject":"[OVN] Wait for the \"Chassis\" creation in the HCG tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78593ce9b853ef51184592c13101a409024defb7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 07:02:01.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-11-20 14:21:31.000000000","tz":0},"subject":"[OVN] The external networks GW chassis must the same as the GW LRP","message":"[OVN] The external networks GW chassis must the same as the GW LRP\n\nThe external networks use ``HA_Chassis_Group`` to bind the port to a\nchassis. The gateway ``Logical_Router_Port`` registers use\n``Gateway_Chassis``. In order to route the external ports traffic\nthrough a router gateway port, it is needed that both the external port\nand the gateway port are collocated in the same chassis.\n\nThis patch detects when a network (private interface in a router) is\nconnected to a router and if this network has external ports. In this\ncase, the Neutron API changes the external ports ``HA_Chassis_Group`` to\nmatch the set of ``Gateway_Chassis`` used in the gateway port.\n\nThe mechanism used to detect any port addition or deletion to a router\nis the OVN event ``LogicalRouterPortEvent``, loaded by the OVN L3\nplugin.\n\nNOTE: along with this patch it is needed a mechanism to permanently sync\nany change in the ``Logical_Router`` ``Gateway_Chassis`` set and the\nassociated ``HA_Chassis_Group`` of the networks connected to this\nrouter. This will be done in a follow up patch in order to submit\nsmaller and better reviewable patches.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I125d87ffbb33541ff3a4bb2a90915d20d0a171ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e94e7447aa621792b7c31d71d1c581d473128ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e94e7447aa621792b7c31d71d1c581d473128ec4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
