)]}'
{"id":"openstack%2Fneutron~962188","triplet_id":"openstack%2Fneutron~master~Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5","project":"openstack/neutron","branch":"master","topic":"bug/2111276","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-04-14 20:50:14.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-04-16 13:28:43.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-04-16 11:29:45.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"31291":{"account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"last_update":"2026-04-16 13:28:43.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2026-04-16 13:28:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5","subject":"new job tempest-multinode-with-bgp","status":"MERGED","created":"2025-09-24 13:47:34.000000000","updated":"2026-04-16 13:30:58.000000000","submitted":"2026-04-16 13:28:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962188-bug/2111276","meta_rev_id":"f63112e3a7c2a77268a434282ab67e3e91b708fb","_number":962188,"virtual_id_number":962188,"owner":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-16 13:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-04-14 20:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-04-16 11:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-04-16 11:29:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-04-16 05:29:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-24 13:47:47.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-09-24 13:47:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-24 13:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-24 14:54:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-10-08 11:11:13.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2025-11-04 15:16:38.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2025-12-04 13:53:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-03-16 10:11:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-04-14 06:52:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-04-14 15:16:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"b536be1bc682e7d5fe36553d0b184ee472902b9b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-24 13:47:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02d92a8ee07f25336da65e5999d63b92da35473d","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-24 13:47:47.000000000","message":"Patch Set 1: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 18, column 19","accounts_in_message":[],"_revision_number":1},{"id":"56d900ccebffec1632063d5a47d41e0195e61e46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:47:55.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 18, column 19","accounts_in_message":[],"_revision_number":1},{"id":"a246be515d42f6fbfb220e1c963a7c2ff6e724ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:48:34.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  mapping values are not allowed in this context\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 18, column 19","accounts_in_message":[],"_revision_number":1},{"id":"429523e8738650efb8c8257a3a4b66203ddf576b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-24 13:51:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fa9819843b7888d047be04b0eb9cdd099bebf3ed","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-24 13:51:33.000000000","message":"Patch Set 2: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 27, column 31","accounts_in_message":[],"_revision_number":2},{"id":"087c9a0737085aa9d9dad73f3ffbde784816ec4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:51:43.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 27, column 31","accounts_in_message":[],"_revision_number":2},{"id":"cc532b270bca06d6463378107543b562a23652ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 13:51:51.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 27, column 31","accounts_in_message":[],"_revision_number":2},{"id":"8ebcb4bdfeb5c76b379496e5240a5eecadab10cf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-24 13:53:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"93fbf626fe858448d4653892b807f4f239ea6379","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-24 13:54:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8413480ef365cdb7726e5e81d5509018caa9b94d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-24 16:04:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"957a69cf5d8e40df8fa774196071640d02119fba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-24 16:06:28.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"4a940e6eeea62d08c5c14e0e3136ac0b50c6849a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 16:26:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/039440780ff0442ab71f3520dfd7f209\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/a96176d37da249f08ae89ed97dbc94bd : RETRY_LIMIT in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"734a0ef2ee97c3e45be1349198368d7341dae531","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-25 10:10:47.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c9a2cdd7bb77a5e05f1dc263dfcfdb2d32522985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 10:23:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98800cedb551498a81f08ac58c26a026\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/5f05f48837ad4c3bb43596352a31da67 : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9b664eb246208433f8e92bccf43c1934b5794314","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-25 11:21:58.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"6f31d369887558e29c69cbdb89bb7a2a6e66e0aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 13:47:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e66c9b7c5c58485289041916008a74f4\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/2f5d94ed8e074cc898b2c5d41d96abb8 : FAILURE in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dec1ca382e0d9758d8242ae2f9a09b26ffe3770c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-25 14:21:19.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"b52dd8bb1308d82fa544b45316bfee7752ba1df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 16:55:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8914df3aa75144c7a5df547776269ba5\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/4a0f9d218d194a338c49d8c7f68330c7 : FAILURE in 2h 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"473f345457aa595b99e358f7a6354ed10521f5fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-26 06:16:31.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"dd081371ecfb1707094037d2ca02d606cf0eab04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 12:26:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4235b7c8a2a345a5aee88aa7076acd30\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/75f68c411aad4b6a913e2dae92d9af57 : TIMED_OUT in 3h 03m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"03ad095e170cda6d12e4b0edc95684d11857d3f0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-26 12:34:38.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"d97b5053acd9cd693bcd9bf2a4f3b449ba4f5ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 13:03:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fad48bbda0c469f86aa45593ef4383a\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/f287eda8cb1c4f1b9092f8194d4027b9 : FAILURE in 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9860dd6bfa2a9dfe81396774c0f10ffe3246ddc6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-29 14:13:49.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"a02dee9a82463cbd678ce254423235ea6e983ccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 14:47:33.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a9daea3c03d46fdbc2c05cadbad12e8\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/60e3dc8218f646c58584f4d415b82c0e : FAILURE in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e6862336866335a463332f51a245223586034bbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-29 15:05:41.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"42775eb01df278b3283a84cc6ce63224aaf08778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:28:52.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d47a130e939423d928d412ec689178e\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/36b70b60136a48abb2aef99a0f36db1c : FAILURE in 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b4c076c06a4fcdef4f4e19469f378aff2a4c50e0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-29 15:31:14.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"b5a141ee86eafeeeb5d80c5ee323c77b04536b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:51:31.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8faf88b5fcf47f2b0c8ebc5a1d0344b\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d40bcbda907d4fa2aad41aee3483d98f : FAILURE in 18m 44s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ad09542216b0ddf0b5bb4acfd4014cd16cfce466","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-29 15:56:09.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"53e5f3a625e440192a2275d1c0ba9bc4e745bec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 16:43:50.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b10b58cb8494e53b27b8b8fed2bf07a\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/968b35de5b774a968cc238e2be9d03b7 : FAILURE in 45m 32s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a94ffb30a111df8fb4f3f6991575cac81c294386","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-30 08:36:20.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"9ad26959cb39ef4d2bacdbd78e944d78e9ff3739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 09:28:21.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b74949512a49919cfdf6e0227632af\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/75ef0ae70029429bac145fb4d43bde98 : FAILURE in 44m 33s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0f1636b95d7b1248f1c795391d265dacef498d76","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-09-30 15:29:38.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"93d46e35affb037cfc529a25657e43dc0a14e7b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 16:36:12.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f25a6f2dfa2421099657b5316f435ce\n\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/1b8bcec7d88e4442971c3f329f080806 : FAILURE in 1h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c105f6b1bfd3c9b350460464a7853893b9848d58","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-10-01 15:36:15.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"2292bc227e4297a60a0c8cd9e511ab7cf981feec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 17:52:29.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac2f03f4ef8041a48b5ad903d899fe27\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/890e50d3076241c191986b4996e1ffde : SUCCESS in 1h 41m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/27f211b7503d447c96e433bff8871a20 : SUCCESS in 2h 12m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af4d1256798a4d1082393f2b46ff0e17 : SUCCESS in 34m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42988ca6f6944e96a16687386fb787ad : SUCCESS in 51m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e03e3a3cb84045f5bbede38da336c8d6 : SUCCESS in 14m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af23f37982604736baf64c33e48120ac : SUCCESS in 22m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dda38a8667fe4912a8e16096762d0f85 : SUCCESS in 27m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b82c3c74ed7b46a59b01cd5aafaf1aae : SUCCESS in 30m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f07357f53a4381b84482adec8521ff : SUCCESS in 13m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6b9a358509334e40a1f5933fbec070e7 : FAILURE in 47m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8d95e820129545c3978c1929046ee32f : SUCCESS in 1h 10m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/26f815a52f544cf58769f16b536770dc : SUCCESS in 2h 06m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1fcc92fa48124ab4a6cedfdf92265f23 : SUCCESS in 57m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d40c35aa4a44df885b88ebcdf166673 : SUCCESS in 1h 14m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/89a6b6a7736b4459bf034be1a0f31afd : SUCCESS in 1h 30m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0c4880418eab49dfb784a9362111a72b : SUCCESS in 1h 46m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/65fd2d48bd334b34972cde6fc6e05ac1 : SUCCESS in 59m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/371376eb15b246aaabd57a0d623d7e83 : SUCCESS in 51m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/707962cc8f3e4f25bbc1e7a97e83c743 : SUCCESS in 39m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5ecbbb727be94a7ea107d148222a4621 : SUCCESS in 1h 36m 18s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d6a9f1438d434c2b944423de50ef0199 : FAILURE in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"7c37a1be8e3c51bd0d1624b1996faeeec1ec1ae2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 19:25:09.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/902c93bc8f1e431a8c1ce6397ced2647\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b60b0caf5418451a999d2bd3548cbd89 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/426285a13aad4f349af75ebcc4be20d9 : SUCCESS in 23m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2c9d3d413d464564acbe85f5ec788d5b : SUCCESS in 48m 57s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"814f27b8371bdb5156e4c6bb86e4c6655597731b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-10-06 15:26:07.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"095221583cd23041ba40d46ac534c4604e3307d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 17:32:41.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b2b133a3534b08bedb700625c9c861\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3baa51ea9444e03828a4681ca141597 : SUCCESS in 25m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/73b5d0a3f13a4ae18dec0a973f9d4600 : SUCCESS in 33m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/675c651ce13e4e1d90d541ba580d8f2e : SUCCESS in 31m 01s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"8c7d5f30a2b894700d407600ae829646052718ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 17:37:50.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc82a1ff4a5c46608c6d4c01be82681b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/927026edf29c44fd81c1a08986825483 : SUCCESS in 1h 34m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/580ca62e3e7145efb84bf99d4c315292 : SUCCESS in 2h 10m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/716fb1682b174da894dff0485f319479 : SUCCESS in 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb55e98e9e0b4ae18a62f9409bbabb0e : SUCCESS in 44m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10306d124e4e49e08d0bc4a2917bad7f : SUCCESS in 12m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c985237cb61b46b3ab1014edd6f4a2b0 : SUCCESS in 18m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/871a044ee1a546c09bcaecf846ff773f : SUCCESS in 20m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad347d52bc10430eb80254b4f10e105b : SUCCESS in 24m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247513785acb4dea9e56799f96c506ee : SUCCESS in 11m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f97586869d8d46eaa80d10cc141cc8ab : SUCCESS in 1h 02m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/91647d9a84b54248a5992ca92db9f5d3 : SUCCESS in 1h 11m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f07984a956e942cb993f45d20ab50032 : FAILURE in 27m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/988881a0e08d4743a17e44419ff3f195 : SUCCESS in 58m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e73412092f74ffbab760e303bde8055 : SUCCESS in 1h 11m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f9c5e14682ac41d3b36a6fa82f64c9f8 : SUCCESS in 1h 32m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fb30b60d7a1c44c4a49e81984371cdfc : SUCCESS in 1h 30m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6ea111c3052941259f797536b084a8a4 : SUCCESS in 1h 06m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5e57a07be110401ebf64e186521f879f : SUCCESS in 59m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6fffc5b12d5b43f5bf2d623a93782bbe : SUCCESS in 42m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e04e299097b24f1f95f0f2657e2192cb : SUCCESS in 1h 31m 01s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/3a0aefd4b776418d85ca7c418dda0202 : FAILURE in 1h 04m 01s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d2c2e517643d238990b93e8610b9ec5189cda85f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-10-07 08:03:23.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"233e4c865cafbb00a44c45e7c8f68ad828d235d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 08:42:04.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dc1f95545664b59a885ddd604f2ae3e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/958e38cdce814e6d82e5b5c7d46a74e8 : SUCCESS in 35m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50e188ba85d749ee933c0c71b70e0bd2 : SUCCESS in 33m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c1fde2445ed4a4cb3bd9e67aaa00920 : SUCCESS in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f555caa3dd6074fcb645601a1fe77524a31753f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 09:56:53.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e22e586e842435bb05b3c9b69a260af\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7d49f06080084c229b5191a8f3a6efe5 : SUCCESS in 1h 23m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6d1f5814d8cf489ea7224500afaac037 : SUCCESS in 1h 38m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e4162908aca41719946d1636e9982db : SUCCESS in 33m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ce57119032d49c5844577a721fca1ea : SUCCESS in 32m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/122850c4e1ce4fd4986811ce95624976 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a91b0f31fa041b4b902c8dbf8c51c2e : SUCCESS in 14m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e11bd78e74b44b7db0686d22fb4e56b0 : SUCCESS in 21m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/112818673cc34edb9ec964eb14c50211 : SUCCESS in 29m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6e9094b77543c197f71f77566906cf : SUCCESS in 11m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aeef317c93474782b37e85c5972f4366 : SUCCESS in 1h 02m 49s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/74c4683aa69845b8a5ac5a5b1eca6ef6 : SUCCESS in 35m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a11a2386aaf244ec99591cb94f343cb3 : SUCCESS in 1h 47m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e410d0834e274f4b8710cb0b7d19ac23 : SUCCESS in 59m 05s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1630eea64e9b4cc99b45070f9c624774 : SUCCESS in 1h 09m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b8cfd1536454c61abac7b06f0a1c830 : SUCCESS in 1h 26m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/216dc826e7bd428d9d2fed8c8f06e8b7 : SUCCESS in 1h 18m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/26a1051833564f72bc1f10d0d1fa315e : SUCCESS in 1h 02m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1368fed6eca1419eb85fb2516f36b5e9 : SUCCESS in 1h 08m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/33e6dac1ab8146a095209999a1ec489e : SUCCESS in 1h 23m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/016c6a21c7174ef6b84be5362b064426 : SUCCESS in 1h 05m 09s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/c97865d88fb14f0a9201a85ce245a7b3 : FAILURE in 1h 08m 42s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"27c2e6a58235b018660c8aa007cee51bdec73be2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-07 17:40:54.000000000","message":"Patch Set 21: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"a68408c87e9f0891d77d5ea6354462ee7254eda1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-07 17:43:29.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"e08963e3b51efe0d5fa388e4a521df947d46d877","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-07 17:43:53.000000000","message":"Patch Set 22: Code-Review-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 28, column 31","accounts_in_message":[],"_revision_number":22},{"id":"819e0c69f96a4158f4bb70f14452d70a215f6c83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 17:44:06.000000000","message":"Patch Set 22:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 28, column 31","accounts_in_message":[],"_revision_number":22},{"id":"6d2116bd35280bbdcd5e668c237d3ab74769acf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 17:44:08.000000000","message":"Patch Set 22: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 28, column 31","accounts_in_message":[],"_revision_number":22},{"id":"21ce0eed144070464c82631b72298c67b0f723cd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-07 18:48:59.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"2339a4cc432b8f00bb187e67c94b63f0f6369483","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-07 19:04:15.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"6e79defb8288137c0de23a851cbb7e8e36f7ac19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 19:44:56.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e2d1d1520564330984bcb3449c6d9db\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a69f282b64fc47d3860b4e42f2ddc98e : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9a288a87bfd4c89aa2f45af7108fa7a : SUCCESS in 34m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4fe1e3a4bb9c45cb89d25ff9ec3ccd7b : SUCCESS in 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"69e7147a602b7784bc0d291d68ac1bfe88dd9991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 20:30:56.000000000","message":"Patch Set 24: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/963199 is needed.","accounts_in_message":[],"_revision_number":24},{"id":"5cf228f8bb5a26fa69f71129f7999075df92aa5b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-07 20:35:24.000000000","message":"Patch Set 25: Published edit on patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"00beb4eb92535cc80c502ec12d0428e742afc88d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 21:21:01.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc8c48e029c043ae97d6a5f280484189\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e4bb52fa3c24f5bb1053d7126be0830 : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8ca658ce98f4cb3bb7ee49637fc68b9 : SUCCESS in 38m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5636431f4b494d07a43c66614b82e0b7 : SUCCESS in 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"86eb4c8f6cd0e47aeee63c319ed938cb06acb929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 22:41:43.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75b14b15dcaa44abb42c51c0e2fe82fa\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/427f5ca71c67415f9cca844752fa11ac : SUCCESS in 1h 32m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52900fc7c9044e6f9ee7e0ce4303a69e : SUCCESS in 2h 04m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2c027e36ca984a1cb597afa6bcb89604 : SUCCESS in 33m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a98f4ac68a74757bcdcc11afe867dd2 : SUCCESS in 26m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50dee883e24c43ec8f95b0e1511d856d : FAILURE in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec60e90a09b44a1bbc1f79dbfa8df7e5 : SUCCESS in 22m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5633a8de633411090ec29fd3ac5f84a : SUCCESS in 21m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c19a46acd1e848878dadba601e8fe75a : SUCCESS in 26m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9414f6922eb04915b29f0148a3f4a69f : SUCCESS in 12m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9875e14b99104a6e9a5fb33f95dac0cf : SUCCESS in 30m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a9953d6520f946468fc78111f0b9d261 : SUCCESS in 1h 10m 45s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/a6831d5d972d48899ab768519618249d : FAILURE in 29m 28s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/commands.py","accounts_in_message":[],"_revision_number":25},{"id":"7f12a4d3f5c60f80dd63b9c1ff9d5f855ce0d5c9","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-10-08 11:11:13.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"4a4413613b6cab0db61e16f57cdee56cf036136c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 12:25:19.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a337251a29c7456db9df41b1a07baeab\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7a4e736356442f78a4c58c81c26db4d : SUCCESS in 49m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62726acbf04f44768822c7b6ee2489bd : SUCCESS in 45m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/57b3449861524d4e8d31a7fb8eaa9b89 : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"ca6fa5095019f08ec04336952bad4e09e0cac05f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 13:10:29.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd8c8becfdd744ca94033a9978f9e1b2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/205253a5cdd344cab3a0f58828b7072f : SUCCESS in 1h 20m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/702bf6001f1c4b26be334de2bc70a50f : SUCCESS in 1h 57m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a4b5b783af6c4145803d2002bd888836 : SUCCESS in 32m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/913e7e54a1a04f95866e2f6ba978eb7b : SUCCESS in 47m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf4215529e1b4b68b1e5ecb649db151d : FAILURE in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5f089a35c0b43ffbac4df448cf2a489 : SUCCESS in 23m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fefc713e323343878fe472ca3ac57a1f : SUCCESS in 23m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/831917d6b8b24e2485297dbe152bc54a : SUCCESS in 20m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc0e3d222a147e1b763fe7008300c00 : SUCCESS in 13m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7ada72ac5b3040ebb3cb0e4d46d3b6c1 : SUCCESS in 56m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ef8553ea5fbc45c0b53d7df77373282f : SUCCESS in 1h 10m 21s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/7eafb5867f424d87a0deb528e80a5704 : FAILURE in 13m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/commands.py","accounts_in_message":[],"_revision_number":25},{"id":"a43229efcb2ceea4b63f63d6c1571ea7d8ff038a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-08 16:31:27.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"452b4ffd23b1320d611474c60baa9d3dd5cd63de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 18:11:57.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aa475af40c24ef09d6e2fb2f635ee2c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c64332e39ca1477bb3b6b50c6159b425 : SUCCESS in 47m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3817e19d85a4433940626fc5b38fabd : SUCCESS in 45m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e501201dccff412da410c0e321625559 : SUCCESS in 57m 25s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"f65bd39b8d40e95d80e29601c9636276bb1dfcb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 18:16:12.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77c69c29bba2437cb2b0272304a4e48a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5cb999291f4e4af2aa79abdc20733eae : SUCCESS in 1h 20m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a0f427d87dfb47d3822de0cfc438bf66 : SUCCESS in 1h 42m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d2aea438189422b8dc2ea8d73db556c : SUCCESS in 32m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afb0027fdaf344969af70e5832ea8924 : SUCCESS in 32m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2c310dc86c495989e347fd2e066087 : FAILURE in 10m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f0d9edcfe724d55b42ea28a2cf69640 : SUCCESS in 18m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/033850399b944b11b8930bd0f0bfd6a1 : SUCCESS in 16m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bb9f61021614392b4d7e56f99a7a370 : SUCCESS in 24m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff07aad084a449ea7b04e573ecec6b7 : SUCCESS in 7m 44s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e97b4cb874c54ab2ace8dc8bfcacae1f : SUCCESS in 35m 05s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/68425d69e8c74bebbc2ada2eab92e6e6 : SUCCESS in 1h 07m 38s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/a829dc10656e4a29969c375d34510339 : FAILURE in 26m 05s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/commands.py","accounts_in_message":[],"_revision_number":26},{"id":"af6e36015a8935c21a9c854f40dc6ff6fd1896cf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-08 20:58:58.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"f89ec63ad1ea7ddba8f49b1ccee7ba2b427a15fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 22:49:00.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/765ba565e74349e6bf74555af395a8f3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/148971a4809c4eb98260d3c94915ab18 : SUCCESS in 1h 37m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae8648a9a9da4f088fd91c50bc7ffc8c : SUCCESS in 1h 48m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11b548658fd14fc8a028612eb7f64126 : SUCCESS in 34m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f68fde1451bc4dae909eefb88925ecd7 : SUCCESS in 50m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b874e833d5064ab4b021677f87c1d119 : FAILURE in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd471e5ae7ff45989b84805dbdf3763d : SUCCESS in 21m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4311daca1ce54f57aeacb3c1dd7729c5 : SUCCESS in 15m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81a986a6a2d8488ea228a571f45a36ca : SUCCESS in 29m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4389e2b75a9841fcb6166a059201fd15 : SUCCESS in 7m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d5414f1c2a3b4265a98926ec761ba8c5 : SUCCESS in 53m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c5058749d3fa44daa17051264d0144dc : SUCCESS in 1h 10m 10s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/a68ca23040da438aa3c61db2fd11dd47 : FAILURE in 57m 50s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/commands.py","accounts_in_message":[],"_revision_number":27},{"id":"bd0d8359351de3917e0eac0b02cd4b9df639a20d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 23:34:40.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f21b74e6472e4a8bb58ff1ed495dee2e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ec4f91650b946c298513be0dc1279b0 : SUCCESS in 32m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9117a50ac44b4d13b5e21663a81e1adc : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/10ee9c9f918743108a0721e7b9eeeb64 : SUCCESS in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"1584b04801754908b1ea5410e05960f971b2f234","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-09 11:57:23.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"517dfcab06475612e97dc4c52fff2485f8f6f0d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 13:03:49.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24329414dd7544eca733a838421b0030\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5201cd63ddc84eb79a2044a0dd4099c2 : SUCCESS in 42m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10726af7275240929cd1e8f4cb4be297 : SUCCESS in 36m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb9247b1950040bba84122f10880d7bc : SUCCESS in 32m 23s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"ebe8ed525f62ac362f7ad2c03abaa3c2ffb397a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 14:23:13.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d2072c13ed74d6baab66c7f11073aca\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b0aef5ed4ec04501b70906e1b9399803 : SUCCESS in 1h 22m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c457f75974c14fd9932e28ce76b0e604 : SUCCESS in 2h 24m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03522c38dc8f4174b9d6728bd96c3c09 : SUCCESS in 36m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/483f3ee6422342198870ac1d4279705d : SUCCESS in 35m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a138ecb7392b45d78156bd1100bbb2cc : FAILURE in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6a3dcaa98d64ec89693211774ec3bee : SUCCESS in 14m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c644a9e6308a4b93befe2d059d12b7d3 : SUCCESS in 20m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1653dcd8a08d47b69a2b01076b5d555f : SUCCESS in 24m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7791136817654846ac7eff78cd5a7c69 : SUCCESS in 15m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cb0a7f19524d4287afdeb948960a0cab : SUCCESS in 1h 06m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/219e8c03934a4a17ad185a200f8abf06 : SUCCESS in 1h 03m 12s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/9463965e10824e77807e7b76b4601daf : FAILURE in 1h 05m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/commands.py","accounts_in_message":[],"_revision_number":28},{"id":"c89d5a5ae6e8394dcbc697900e82124b83c7746e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-09 15:30:25.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"c52dc859bb900cc86a67e9a0f5e4c4b89305061f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:31:29.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b796dcd556ae4284bc97b6a0f114ef70\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0f4b543ba7984fed9b82d172bfe357fe : SUCCESS in 1h 31m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e316bc8727ce4b719e5b21c4a0c490c7 : SUCCESS in 1h 59m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/148f897630184733b5917042808f5a50 : SUCCESS in 33m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d143a0472064763af4cadc98a008780 : SUCCESS in 28m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c36f053e0c3420994e093414c55bbb4 : FAILURE in 9m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74cc0d440ec54556920a790b2906b530 : SUCCESS in 21m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59485e5b927a4a15be6df727f0937063 : SUCCESS in 22m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03e739ccce1941ac8c98ad64663f2dcb : SUCCESS in 30m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e5d78445deb4bfea39f714c246793b4 : SUCCESS in 12m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f10e0d63931b4b72b2e18ac7043d160f : SUCCESS in 33m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5b0eea665f0340219decd645c6af0879 : SUCCESS in 54m 22s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/96e2aa03f096400e927a6a1a63b21424 : FAILURE in 1h 08m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/commands.py","accounts_in_message":[],"_revision_number":29},{"id":"caadda241fb56ee7e080dcdcc1c8e9bb8229b20f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 18:27:47.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d06c30d2fc364c1dbb35bfb5bf8a2ed3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e3f798fcc5842c1a399699b60969324 : SUCCESS in 52m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/880781b127df4f6f861da06a7bf9ae24 : SUCCESS in 47m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f1dba8f9c21449f9d985c49f30ba008 : SUCCESS in 59m 16s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"76d3ed26689e6f4f10b2d60678cf40ac2eb0c4a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-09 18:56:53.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"7fcfaad2311e94e98e6447506227029282a0b26d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 20:39:44.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf83df1d31654e10bd0cd8c8717bbe6e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec765b26f0ec437bbe364f73c7b769ff : SUCCESS in 44m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/efec7db33fde4bbcbd76a381ca5e3193 : SUCCESS in 20m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fa7eb02d8b364dc4b03f34e8acef1aab : SUCCESS in 21m 44s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"557d85f06dc871bf9f2556d0640ecb6073e57207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 20:58:29.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec32f035852b49979f8fd33452d90971\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b82f14eaa25746f7bafb9d41ae21b64b : SUCCESS in 1h 41m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/808dfed395f34fabad2212650cd01df4 : SUCCESS in 1h 59m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7fb6c921d22f4144a5dca0c4fff99e59 : SUCCESS in 37m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb890343d3fd4489a8e72a8076a40c75 : SUCCESS in 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44442fa5d469432a9c1f448da235f451 : FAILURE in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc663cbb21834d7b9ab44ba108786adf : SUCCESS in 24m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0205acbe7da4f2baabd307e9924ba78 : SUCCESS in 26m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb45002a5db541b692c76d1ce974f64f : SUCCESS in 26m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/179c6a519bdb4dd8a091a6a8246d14c6 : SUCCESS in 12m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/11b41e84ff344bc9aa6acd248d8915a1 : SUCCESS in 53m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2888dc3420ed4463b9d8109ac53d80b3 : SUCCESS in 1h 06m 48s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d235d262a344479d93d311c07670a25e : FAILURE in 1h 05m 29s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/commands.py","accounts_in_message":[],"_revision_number":30},{"id":"4a1e6983ed7e529c84a1462c7983e49baa5e1086","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-10 21:22:04.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"8ea8777c150a6b904ab81ca1dc385e8f500edca7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 21:56:21.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e55cf1791141be8cb844c867953aef\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11441ca2c5b2457f96cf22ea40fcd5ac : SUCCESS in 31m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc0d01c36d494329bab01a1da438aa2c : SUCCESS in 23m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e1d10ad0837d4d1fa10184a2bfe6d619 : SUCCESS in 26m 32s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"1adc5cee462dbf3e512ffef20bb3def7fe739789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 23:29:13.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35ddc74cf85f4e1bba9ace18abe1f40c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e9b4077c70004795951802cf986a0cdc : SUCCESS in 1h 24m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f66c52a1904f43e3b68e1485328f461f : SUCCESS in 2h 04m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a41c03a009bc453da8740466b5986c43 : SUCCESS in 35m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1268b58118d6410cb758e49722504bdb : SUCCESS in 28m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdea6871f26d4949b39ade30f784a6b8 : FAILURE in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65bd7ac4cc5240569c2269356b63dbc4 : SUCCESS in 17m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cf498f400a44ba6b2176cdc64fd3d69 : SUCCESS in 21m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e7324fa92224a25a3263f2374dc13fc : SUCCESS in 19m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dba53e27a2f4e6c8dce4ade9a0f71f5 : SUCCESS in 12m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7f9f8f33678e49a99caf6fd93ba72544 : SUCCESS in 59m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aa6c262333864607aa586a204b1a4ea8 : SUCCESS in 1h 12m 37s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/babd54f002384a79b15ce5173cc07e5a : FAILURE in 33m 11s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/commands.py","accounts_in_message":[],"_revision_number":31},{"id":"c4f524e9c1e0b4f6b239f6f216d30c6758dede47","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-14 19:35:50.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"52435c53aade330ca4fa3b041918b62f229491fb","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-10-14 19:36:11.000000000","message":"Patch Set 32: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 961293,6","accounts_in_message":[],"_revision_number":32},{"id":"1e8253e1b3cd2e67a115cb4a279b99439a5ee77f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-27 18:52:43.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":33},{"id":"16a6ebfab152b0bfcab53d678e512395a0bc83b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 20:49:45.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e995d5f145843b187d2197b0bb887a0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f517aeeb693420e87e7dfb60356fd9e : SUCCESS in 27m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f413a2ba4a724bfe9f17bd3689e6b463 : SUCCESS in 53m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/43f74338a22d4244800dd05ea0739bab : SUCCESS in 53m 10s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"8a2fa2e8526643cd155dc1c929179fe45ee56ea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 21:28:46.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdc37ccfa198458eb03366602b7acaea\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4736781edab74534b40bf7a9ac041cab : SUCCESS in 1h 22m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53df113547ba478a9f3fa6007483018d : SUCCESS in 2h 33m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/309aaaf0cfbc4bc295f4c06df9840934 : SUCCESS in 50m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0072e623cc774a8bbe04d97d8529ed6d : SUCCESS in 36m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0fd75b24ad54c7c9bde56adee0c3bbc : FAILURE in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d32ad996f995419fbede554956ba314c : SUCCESS in 33m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52866f6a0af54679ad7501c7310f0874 : SUCCESS in 27m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e19dbed5d42246e3abc3325d74adb727 : SUCCESS in 34m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b4991cdb0c4879aaa15066cb520097 : SUCCESS in 12m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/884f9de498664bec90d12bfa2a7d27b1 : SUCCESS in 44m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/105abac255d14f3ebd79578727b85ca2 : SUCCESS in 1h 15m 37s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/72aaf510dcf54844b44577ee1fc489fb : FAILURE in 1h 06m 14s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/commands.py\n  Comments left for invalid file neutron/services/bgp/events.py","accounts_in_message":[],"_revision_number":33},{"id":"475f0e9370c4faae34db282a6c747979a4a88bcc","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-28 19:34:24.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"f826aece4b0530d12ab94d2f9b44a9bdb3954d12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:45:02.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/655e1a4c73e34431ac2fa1e79ecda71d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5935a5ce8a7a45d2be313ca36e0ca2de : SUCCESS in 1h 07m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb8b1b3d5e1c459ba78ff5aed12ec130 : SUCCESS in 54m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eaf20034da194881aab256ff3e10e23d : SUCCESS in 31m 30s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"24cfbf601fe0d56a102e2eaa87321b9cf648c63a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 21:32:43.000000000","message":"Patch Set 33:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09e83c72fd7a49d399ce773038361d0c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/484ea2172faf4a2fa083dedb9ef64068 : SUCCESS in 1h 33m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/63efd2b7c59349508116c7f2a697c9a5 : SUCCESS in 1h 55m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee64c97275744aeca656f33a7ad5e5ae : SUCCESS in 34m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff90c2ec3a034f39ac95b1f91ab55e43 : SUCCESS in 1h 02m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cabe3eb4cef42729d9b4baa46cab7f4 : FAILURE in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74f263cef8d64d469757383c13050730 : SUCCESS in 34m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/679848a79d9f4cbdbee49f96e164575c : SUCCESS in 26m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/388882ed1bf74df89667bb06225bafec : TIMED_OUT in 1h 01m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/516af729d549459a864cecaae00a8798 : SUCCESS in 14m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ed45ece2f4214977b328141c680c38dc : SUCCESS in 1h 03m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ac055cb08ff244cc99053e5a0b418d8f : SUCCESS in 58m 40s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/dd31cc174eab477a9cdf021d363af6ba : FAILURE in 1h 10m 37s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":33},{"id":"01452f997a1b1a44f73eacabe9823e8cd1b97996","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-29 17:50:59.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"b8e60e8137d149fce2211fda7fee3e146fcba57c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 18:09:49.000000000","message":"Patch Set 33:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/963223 is needed.","accounts_in_message":[],"_revision_number":33},{"id":"0b1da580076b99dab0fe587a3f496551b8d67d1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 18:09:50.000000000","message":"Patch Set 33:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/963223 is needed.","accounts_in_message":[],"_revision_number":33},{"id":"ddf1b6ae2bd67557728f5d38d2d27db37822148f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-29 18:10:16.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"caceeee868e5ca760b3825d02a6f67b65350f11e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 19:39:46.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a01651b8b1a2422fb906cf21533056ed\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/06d9b539c0e74f7d871cee9a90dcee82 : SUCCESS in 34m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/83dc7d89d76440e6a66dea1ff8251f86 : SUCCESS in 21m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/790f59972e4e4c02b309b958b0c546f4 : SUCCESS in 47m 02s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"69ac1e2f5aed4e86fca65917ad18c38fae1e9163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 20:38:52.000000000","message":"Patch Set 33:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc53f5566c6143c780aae104340faf8c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/72fbe53b50f840ae939f8e3992fe9562 : FAILURE in 39m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/35ccda2f3bd4453b9538a9ec5d4300d4 : SUCCESS in 2h 25m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e11f51fe81604528a12811df90c85c1f : SUCCESS in 25m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97f85c095f8e4fa3a349aedaa2c804b2 : SUCCESS in 58m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1445d33143f4bf0a6b1d86da4ad67e4 : FAILURE in 8m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09040c5e275b4a31b84ef5670334d476 : SUCCESS in 33m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7915b6580c894fa7b4a03d96992dc250 : SUCCESS in 29m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54ba461733d54d169461aa07f2d71d4e : SUCCESS in 42m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f470d9c9fb4143a8842bee9c57c13071 : SUCCESS in 13m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/324963cd6ecb4e42ba5f8f03d6b38d6c : SUCCESS in 31m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/437fe47c06274fbd9bb72c4bdcacde24 : SUCCESS in 1h 06m 45s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/5d2349a74bed4f4fa640fc53b485e75f : FAILURE in 34m 31s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":33},{"id":"5ca303ac22ceaa06fe688570a1eb67597339d003","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-30 20:10:36.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"97bcbef716e0c7f76b6b4c826f57d17cd381cb2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 22:00:06.000000000","message":"Patch Set 33:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ede5f2d5df8c4e8f80be32c7523f893d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/42d23cc59da34cc6baa1ed3206b4aadb : SUCCESS in 1h 30m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ab80560f5e8d4706acd6be2b3ddf0684 : SUCCESS in 1h 47m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3686bfa7fc142b6a692ca6ed138dcdb : SUCCESS in 36m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e799a76c49d54570997b316e50481511 : SUCCESS in 38m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d5897a2791421c8b6e447443fca92f : FAILURE in 9m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7cf1b9c35ab4df0aeb39bfbf372fbe6 : SUCCESS in 26m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/338c57819f7d41d79f57087213674a82 : SUCCESS in 18m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d09abeda0ac54093b39a9d4f7c56f745 : SUCCESS in 35m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0bf1ffcc8149c993e6a524de96cd4a : SUCCESS in 13m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/be6b8f6088b64b51a07f623d5569595a : SUCCESS in 46m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2c8caddbfa8b4ea2bf41fb16ad38b9e4 : SUCCESS in 1h 12m 57s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/c5c76c5f8f9d4368b61788e786e250a4 : FAILURE in 47m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":33},{"id":"404e53b94f55a6b83ecd864eb3da285c00b76fb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 22:14:34.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/839e70d59c564e7aaabb86925946927d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/821edac51e624efc9201e2a692bce7e0 : SUCCESS in 1h 09m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fd354a5d15f640b490682d7b2e775ec2 : SUCCESS in 31m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb7de64edcd5484a89d83aa82cb513c9 : SUCCESS in 47m 50s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"0d247c36a41202e60220512d61e954a088bcbbc2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-31 02:03:52.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"99b8a5dcc6e0339b1829a169154aa19d3e04891f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 04:11:56.000000000","message":"Patch Set 33:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5889082175ab48578069e002222caa85\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4588c03658a845d7adb00ceff6b2d946 : SUCCESS in 1h 17m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d22d5ad817a847b1a972fa5911192c5f : SUCCESS in 1h 58m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4c2d5fece5994ad9ae463bc3024447c3 : SUCCESS in 35m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76d7dcd75f154895bfd9eb1b0721c82c : SUCCESS in 1h 04m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a442b82ff1e4ee288f4fef5e2f81c07 : FAILURE in 9m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1e6dfba007b4cf98b5831ea5b953760 : SUCCESS in 42m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd0039989ddd4ab8b8e0f315ed07b76e : SUCCESS in 31m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6429a3d43574403db78874d35863e939 : SUCCESS in 30m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68790f0abbae469da89791e15a58e09c : SUCCESS in 11m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/17726e750a50484c8edd60d2473c5dd8 : SUCCESS in 56m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b34614716d7d46aea5dd309fce6c3bb6 : SUCCESS in 1h 25m 21s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/cc21188484e04048acbda80f19cb2adf : FAILURE in 54m 26s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":33},{"id":"29d5393c00eef20dad6bcd7b0a10a59b76048702","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 04:18:24.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f10674310d1849c5a7890dde6ef2b6f9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ce5734139764a3588b84681f54dff50 : SUCCESS in 1h 00m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c84d2995cab428fb5e81b5e46d53a07 : SUCCESS in 51m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af66629823df4989b21447b60b8a0fc6 : TIMED_OUT in 1h 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"4e0bfb0adcea305f5411adcb9be0c8a51f44d195","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-31 11:53:36.000000000","message":"Patch Set 34: Published edit on patch set 33.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"baea69fbd96a9f677bdd831c9fc244911f7ddf65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 14:34:04.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1bb1700b457463b83287f07707e52b3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fbbde9efd98d4b16931307d11c00bb51 : SUCCESS in 1h 39m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/57245c86a21e48ec8e0beb7f8f972244 : SUCCESS in 2h 21m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09afd9abbd394066a2121266ed870ea8 : SUCCESS in 44m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8b34d78c98245ab9e17367de8a7d0ac : SUCCESS in 44m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e104eab53b049a1a9f0613165d0f079 : FAILURE in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f5a2d10838246cb8482b3c24ade02fe : SUCCESS in 36m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d631926e2afa44e3b134de3ec8e57aca : SUCCESS in 31m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94bfb5ac59884414a093db1e57f0b989 : SUCCESS in 36m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f737a759404325b370bff1886eae3e : SUCCESS in 6m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/71920ea9ea2d4851a3b6f24d5aee326d : SUCCESS in 1h 00m 49s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7ee8f340f9e04724bada5ea271d49e4e : SUCCESS in 59m 06s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/35470f317a1e46f39aaf62af914304c8 : FAILURE in 1h 08m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":34},{"id":"ba1ac5047e42219579cd120cbf2c40990cdc13b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 14:48:29.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4975ce0b2e2f48f6b1c2d602be980c4c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa56f44a24f84db9b7bcab66d05f650f : SUCCESS in 1h 02m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/328dd93e172f478ea9baba9879b2ca24 : SUCCESS in 42m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/081529930b9b4e018d8716cd090fad08 : SUCCESS in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"66dcb670154d9204365e46aa7b01de173cb9c187","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-10-31 19:47:51.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"fcde3b3ea0eb50825e7d6aa4bcc635891d14a927","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:16:02.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b655ff914d234d45bb35956328fd803d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/26e1886d14d84a069fc81d76659fc3a6 : SUCCESS in 30m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/12dff6967dd74f00ab4529e0e79ecca8 : NODE_FAILURE Node(set) request 3a53ee2488164557b6ddf59f4da7fead failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f015cf1113a44b6889e8489155402ca9 : SUCCESS in 1h 02m 49s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"9760135e6c849980f9e2d5d978f5aa1fd532f838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:43:08.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6746490afc2a400db50aeafe0c017557\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/13a1b1afcebb4325a433e4e2e383a2f8 : SUCCESS in 1h 36m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8384349dcbf44da8bd84047a31e699f4 : SUCCESS in 1h 53m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4321761fcbe8449e82f8e44c78109629 : SUCCESS in 31m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e6909d5aa3242dc92b0d4b56ea98cfd : SUCCESS in 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab8f732a745461c94e2f112ebf05f1b : FAILURE in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1dc0e2be2bb642e2a21af57cd59f34bf : TIMED_OUT in 1h 01m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86e5f11138c94c7f8db8124aeb551fa8 : SUCCESS in 18m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3f63008106f4eabb4b5e53cdd9d2bc4 : SUCCESS in 21m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab9559e673a448d99f07cccfb0abb40 : SUCCESS in 6m 13s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a881d6c3ae3040388c14331ccf80dac7 : SUCCESS in 58m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/eb4700cba7f64a39b7a6b1579d6df1f0 : SUCCESS in 58m 46s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/14e5ab249c424fd1852aaa67e9489055 : FAILURE in 27m 42s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":34},{"id":"b12cbae92046b59080f52e7f7bde2507427b4fa8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-03 16:34:03.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"2748c2f5eac3a4fb8758ae5c79f64fd5561a58d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 17:28:48.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75ae6aec469d491a8c3cffa02000eaee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6719acdc25024997a50f10e89958f2d2 : SUCCESS in 50m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49ed116523d746098590e6011adb61b7 : SUCCESS in 26m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/980c05b304944eb29f99b515358e164a : SUCCESS in 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"bf7cf9cb204e8bff720a3d538837e822f502b8b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-03 17:37:38.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"202c350f3526cd0b390eaf3228a33d7772be5948","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-03 18:39:35.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"67990e6f8c4c8ffeffe62a04abeec2caf63fa1aa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-03 19:23:50.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"527e63db84bf047f21162d4b43d3dcc3712c7272","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-03 20:06:03.000000000","message":"Patch Set 39: Commit message was updated.","accounts_in_message":[],"_revision_number":39},{"id":"f1aeda2ae3e17703c1165c5fcc5d0bc0557bc0b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 20:49:52.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e11918843a9495c9cf852b96ae69e51\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/805f2588f0d54901a0aee7f2e70f1544 : SUCCESS in 28m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b013a3a6f90b41618f65672b33160249 : SUCCESS in 40m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/72668bb82cdc47a187f5ff8fd4ca0670 : SUCCESS in 30m 52s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"603c3cd8b41f4d64359fd798bf2ae15c92b8aaee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 22:51:48.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44b019e2e43d4dd5840f48c803c14f15\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8aa2067765c2426eb3540854175eb9c2 : SUCCESS in 1h 26m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aaa769ab4ffb4c3cac8b36213b42e5c4 : SUCCESS in 2h 41m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/053cb39187e148de99d8687981c2c3df : SUCCESS in 34m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d37889e5e614a7db04523b9b0811985 : SUCCESS in 58m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10374fba9b6c4d0382c07729099eb472 : FAILURE in 10m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ad1f6356611479393c2bc8afa902902 : SUCCESS in 40m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96ff2668a98d487185912ca41af9b314 : SUCCESS in 19m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd35e5f964b14ceebf8c8c52cf9842c5 : SUCCESS in 23m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5af4f40117044eea18aebb402b108fd : SUCCESS in 15m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/186370fc022947f38eb9ababcca90523 : SUCCESS in 44m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3ec982f067704b08b3afc16318d2d174 : SUCCESS in 1h 08m 19s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/db3234bcdd974218831a9b4bc84d3b77 : SUCCESS in 1h 08m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":39},{"id":"2dcd1c08439edb400dfbbb6ac84a12d43d154671","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-11-04 08:52:45.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"fb92789249363c22afdf4df65140c03b12611644","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 09:44:38.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3d944deee64424e8fc93dcb833c939e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8620d7d8ddcf4ed89b8f2d96083518b0 : SUCCESS in 27m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/310a6482776144a09d53e3e8ca88ce03 : SUCCESS in 48m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e0e053ab3c65455bb7b85a0b2a01eae3 : SUCCESS in 44m 42s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"1928fa4c15ac7fc6396e83af53104cc3ddb6b99d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 12:11:48.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/181bd6b597dd48a59535c1679d6e9327\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d420f967fbbd4f518746a450d3a79a23 : SUCCESS in 1h 41m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2cc5552ac164493191cc0a290a229d53 : SUCCESS in 2h 20m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b246be442fb64aaaaca9594cab22b53f : SUCCESS in 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38a79ccb9ce24c28a9d2a264ed9ee8c2 : SUCCESS in 1h 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645cefbd33ae4cbbb2bdd4fe3fa9a29a : FAILURE in 8m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/345479c4f1764f28aa4876ba2296c173 : SUCCESS in 28m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2efce2662b54e5e89132f43fc89037a : SUCCESS in 28m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7497e12df7ff4c20a5a0133180421d07 : SUCCESS in 27m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c6da18d32840fbba34d83ac2262897 : SUCCESS in 13m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a48d0e94037f446b8610f93009870166 : SUCCESS in 1h 03m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2e40fa71609f4d8d8b0b40e562b522de : SUCCESS in 1h 33m 07s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/aa399f9c674841dcb362512d0744f8ee : FAILURE in 1h 35m 45s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":40},{"id":"7bdf34adee83813bf09fb6aefab64cc172bc2736","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-04 15:17:14.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"2da3d8bc5b365987fa900e2348d1d3c0b5554473","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 16:54:52.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bc9f058ccca4ff7af693524873f245f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef35172cd4384634a2e43d2664a7e53f : SUCCESS in 1h 02m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d12aaa5517e94aca9b8c94bd2acc88d9 : SUCCESS in 21m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9c8a221c74bd48b993e717d5f55f0173 : SUCCESS in 1h 00m 05s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"42a994880e6a2d5cebf833b6323bd13d3fdef129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 18:54:00.000000000","message":"Patch Set 40:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9374ae3538294d3cae94ab0cce97b86a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1f4a80e4723e4645be07aa6bcaa60268 : SUCCESS in 1h 22m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3dab45e34a084e1abdb5e28a20f7aa1d : SUCCESS in 1h 55m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/39076454746d4950b7b2059d901068db : SUCCESS in 37m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6e895d59fcb4fb1889aca549efd4dbc : SUCCESS in 1h 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c00bf73ff094c9eb7f78500b76701f0 : FAILURE in 8m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f795b0986124ffa841291d5034ddb11 : SUCCESS in 37m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/098c73a3ea074c0a8668b3cb275acca4 : SUCCESS in 27m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9822d9ac439d4d38aa95e9a30eb464b7 : SUCCESS in 46m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08fc7953f426444db2457add92abcd32 : SUCCESS in 12m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/47b49a29df2240449dc2e3b69360e9cf : SUCCESS in 1h 08m 20s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1c8405f895f24e46b14af39819e54866 : SUCCESS in 1h 11m 43s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/98d8afe3dc9f400b8b43ea86bd8ebda2 : RETRY_LIMIT Host unreachable in 57m 47s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":40},{"id":"1b9391e10efee7b19331c17493944cf98cc3ba03","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-11-04 18:57:36.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"8543e930f3d64bd5104f4d19caca43e15c1813c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 20:00:10.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ab2a286bd041d3aa77405d200340cc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b54fb95b63e24819967c1175f38f66a6 : SUCCESS in 59m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/556605b1ec8f407f9bd12dfd459f4f67 : SUCCESS in 45m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f6e8ff6e1912492cabe60ac1688e32b6 : SUCCESS in 58m 22s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"d80c1028bcb0ee4b88f37023efa65c3aaedf37f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 02:48:25.000000000","message":"Patch Set 40:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fd4b6dc81f749d5aebccdf5c8f5819e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0fb68eff10464255bab57104fef37285 : SUCCESS in 1h 15m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f4e579bd3954c168acf6d52f9f2bd59 : SUCCESS in 2h 03m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb47a4169a464fe9917a74877e52e171 : SUCCESS in 33m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6eec827c23014ed19ba097370a8d790d : SUCCESS in 37m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb97acfa70ab4825992340a404e6ac81 : FAILURE in 8m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85edbeac9ab94fb1a9f1cab58c238a5d : SUCCESS in 29m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f270130835e41e493339d854015fe06 : SUCCESS in 30m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/090fabba2c054b6d96808391da17789f : SUCCESS in 36m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f42108044ef473f89e7afe2e712daf2 : SUCCESS in 10m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8e0e436c125a4830acbd735a587c6c10 : SUCCESS in 45m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d5cc314cbae74a61b21e6339e1de2133 : SUCCESS in 1h 13m 26s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/b1fcf0049d1847e3ac982e781f11c059 : TIMED_OUT in 3h 07m 33s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":40},{"id":"2b7c7d5daa8902d75113cfd116711402a1b569ca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-04 13:53:53.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":40},{"id":"b51c32901f775c91bb2c1b52467321da1c09d9d3","tag":"autogenerated:gerrit:restore","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-12-04 14:03:07.000000000","message":"Restored","accounts_in_message":[],"_revision_number":40},{"id":"4c2566d958878f3a193545dc398980f765907dec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-12-04 14:03:30.000000000","message":"Patch Set 41: Patch Set 40 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":41},{"id":"40fcce9834812552f8beb782d1ec9342568ae851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 17:12:07.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62ebad0fa0c04dce9caf0d0934ac3fa7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9e98b7ac14ba4d1499e95a8073938949 : SUCCESS in 1h 16m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c8e46f64c04e4dbcac20c7abe4a78015 : SUCCESS in 1h 49m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be91e2b9e6274056a88c95688f2397fc : SUCCESS in 31m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0be5d8a902045a2b6114365f18160cf : SUCCESS in 42m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f13bc548d2493b8bff85828ce22461 : FAILURE in 8m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8b3de668e5d4396835a2ba9b2ad486e : SUCCESS in 20m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afc80d364b344677a02ee4f50596a464 : SUCCESS in 21m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc8c25464944210af555153940a3d0d : SUCCESS in 12m 14s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bc7c7ca3266a471db759ed25c3bd30d9 : SUCCESS in 57m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b47d072e5c2b453a87c8efc92491aa20 : SUCCESS in 1h 18m 15s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/169c64bbc5ba4642a8ab46a33654dc8c : TIMED_OUT in 3h 07m 19s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":41},{"id":"5944d7c63f03dfad875c8101623ff7b67fcfde6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 20:25:19.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98fa9727bfe74b8ba560b5b06a90583f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/37eb23e315ea4c2ba2f4355b1dbc3d01 : SUCCESS in 1h 10m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6aa42b03d2a8447d8e51a9b2a4dcdb34 : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/06aefad11eb74a1694e28df4ede05204 : SUCCESS in 1h 03m 36s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"3afe5e11622f7a14300517e99ce28c5ac18e871e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-12-05 10:53:34.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"872654a67d5ee13691127528fc367584e6dadd1e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-01-16 12:40:38.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"b01cb053e64f2a888a6858e3e04f42c4303d3c4e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-16 12:41:04.000000000","message":"Patch Set 43: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 961293,13","accounts_in_message":[],"_revision_number":43},{"id":"b7a95cb17a5aa81ebc213c069e6bb163d1fac72f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-01-16 12:44:21.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":44},{"id":"9d8c1d2f338246c56c658f608e9a33b523c9c931","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-16 12:44:45.000000000","message":"Patch Set 44: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 961293,13","accounts_in_message":[],"_revision_number":44},{"id":"fd39fd3a7c5ea2e5dc1cd44a256613768723a93c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-01-16 12:47:23.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":45},{"id":"4c25236c88c83b1d7b581ab58283a1ac376c2011","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-16 12:47:48.000000000","message":"Patch Set 45: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 961293,13","accounts_in_message":[],"_revision_number":45},{"id":"5c29206617c8e84f8cc9ae330c5a3d1519d86e78","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-01-23 09:00:39.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":46},{"id":"a7d41eba0b2fa9bb69705faa4bf2624fbdc604bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 10:36:11.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ff7c733a4b64b25a047ed4a05790f39\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6e6a44b548764b69bdf1cb428b7933bf : SUCCESS in 54m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c6a5ed2da0084f2eb9a74376f08d6968 : SUCCESS in 43m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad9bb091d075463faf4c992d1290db7d : RETRY_LIMIT in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"f882033e35617074cfacc9b2f4225a76e0771a0c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-01-23 10:47:23.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":47},{"id":"848337fb14fccbbedc50373d1e7882b4650170fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 12:48:40.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b766fa81274c9f8f4e7c7c7ffbba55\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce15c2ebf84245e9a6aebf336fdc9ddf : SUCCESS in 54m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f69ae62c1f044d2db8c671ca6086cbd9 : SUCCESS in 45m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/82a310b7e2f24341ba089346e7163fba : RETRY_LIMIT in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"f12d91629ba3ba12a5dd93a8593de963bb5254a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 14:14:21.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25e1678856cc4318a052257dbc5f993a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e5781eac07ce4ccb9cc4d8535d46ec42 : SUCCESS in 1h 28m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/37dd307d73b64f5e95a9e3eea1728919 : SUCCESS in 2h 19m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/013edbe7690846399be06a5e6aedee81 : SUCCESS in 31m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0f0c13531734fe297cf604b15314dea : SUCCESS in 58m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fab0dd10a354896b98e99b49d5c609d : FAILURE in 25m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d47542827344137ae22bfbdae13bf7a : SUCCESS in 34m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43f2c47e6dc3452dae40f897c74c583a : SUCCESS in 22m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bce065e2bf14a1c94aa1bdf45969b5e : SUCCESS in 9m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/20c15d9edbf84d9e99b1bd2241e7d6be : SUCCESS in 1h 01m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f046679fa8d5474a809ce35c190b3210 : SUCCESS in 1h 09m 17s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/eea1fa1b8f73498b9bbcc44be49d2a0e : SUCCESS in 3h 19m 29s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":47},{"id":"6c861f83c4b0e7c8ee9c8c3c36d9c03ab7bcc463","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-01-23 14:15:51.000000000","message":"Uploaded patch set 48.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":48},{"id":"2f178b55b66aa77a43a98991f73ea0fb2d3c52a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 15:13:15.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63c6c29e064d4d5995c7c72e9c41573d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d193a896ec4a40efa498d80178cce3ec : SUCCESS in 54m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3625b55833f94212b77a44756d6944e9 : SUCCESS in 42m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0e01a514503e431ca26c04e380010b88 : RETRY_LIMIT in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"a2ac52374b51e7a2fce6556a7ec70e857a298f60","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-01-23 15:28:13.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"aabdf850adba7cf12b18ee730488237ad37f81e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 19:00:52.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21d5b30dd0524e87981c1cd3f0337433\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ecc38a37cae4e4498afdb1888ba87bf : SUCCESS in 54m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/724b65bf172643a6b9acca71e8f04d39 : SUCCESS in 46m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b6f006d2f90f451bb2b8dd55fd288365 : RETRY_LIMIT in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"241ba0b3c96dccc1200896633826e0d34f31bb68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 19:09:40.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e866a459b70f47ceb756c9bcd9b05071\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eca6f9b32b1f4b9eabdf169aa4052495 : SUCCESS in 1h 18m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/96d6d0798c92495ba5c560dc3bba45ff : SUCCESS in 1h 39m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/152bf9c391824d028be396ecaab1bbc1 : SUCCESS in 33m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/703b162265ba409cbca6861de62626e1 : SUCCESS in 1h 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3933de729bb848bfaaab446ff5bd0d09 : FAILURE in 26m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f008cdbcbcd41c99f0ff0e970131d2e : SUCCESS in 29m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef4192674de845b0bfd1f1f25840f855 : SUCCESS in 23m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad20c074f1974d78bf736a2f09c0677b : SUCCESS in 12m 31s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f347d6a01e6e444ba5b3ae3f2f3d7869 : SUCCESS in 41m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c28868bf4ea04a96bb04dc649a7c3085 : SUCCESS in 58m 50s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d495e75bcab6457e86daf662f8939977 : FAILURE in 3h 33m 11s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/services/bgp/helpers.py","accounts_in_message":[],"_revision_number":49},{"id":"bcea9a5bd74ab470345cd2f8f5ecf6252851f059","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-01-26 08:54:45.000000000","message":"Uploaded patch set 50.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":50},{"id":"66137b95b80cbcb8dfb8b9e0061ee2e54313d5bc","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-01-26 08:55:04.000000000","message":"Patch Set 50: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 962727,14","accounts_in_message":[],"_revision_number":50},{"id":"45ba987a75807aece949ce945350c22dcc75785d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-03 07:35:07.000000000","message":"Uploaded patch set 51.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":51},{"id":"d82a9d61ab62e7fb8b62be8c12b41c6e296c1865","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 08:28:44.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/508a4c3d491d442ea90006af9c2d8257\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e8fa61daf2d41ee8786e646edfbeb92 : SUCCESS in 51m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3df91a807fd4423094d4988e79f22a45 : SUCCESS in 48m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7c19c0b35ca94ee49394ef6428780a6c : RETRY_LIMIT in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"ce704f6517fb50e05352e470d2be54792c7c4d58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 09:37:18.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f95de4fe8bd49a09a088d6573c9267e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cb8ef06896c6435498109bdece082d6f : SUCCESS in 1h 11m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8f467e46ed134b2d999a22cd4a81ca00 : SUCCESS in 1h 54m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/334116f0a44c4a1d86eed807d0edd977 : SUCCESS in 31m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ba22e18a5614ccb93b95f50b98b45cc : SUCCESS in 45m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042fb3cf3f844985b103c9652434b751 : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ef1766eba0a47d9ba73986f2e700c39 : SUCCESS in 20m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/427707d4a0a44d4f8da293bda15ce532 : SUCCESS in 24m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7512eeb9d78a492591ccc4ea5e7e482d : SUCCESS in 11m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2e98495a66304fd68478cbe483692c7a : SUCCESS in 1h 07m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f8e49363ae3b4e74ba696bc4a49adac6 : SUCCESS in 35m 18s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/03f09977d686408ebe3b62649a0c688c : SUCCESS in 2h 00m 28s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"332ed35414135c582ff9ed9b376939acab1586d5","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-03 14:13:00.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"d46a9d4e98e0d4727913663b6ed481c355d39892","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 15:04:07.000000000","message":"Patch Set 51:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/devstack/+/963199 is needed.","accounts_in_message":[],"_revision_number":51},{"id":"d0b7a2de5c06d8e68487dcb9f0058af7b890e75c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 15:04:22.000000000","message":"Patch Set 51: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/devstack/+/963199 is needed.","accounts_in_message":[],"_revision_number":51},{"id":"25a0d9fdb577260ffae298ca2dd553d3d30ecc18","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-03 15:04:53.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"f6428daf1e0bcb7450eee686ad500c9aa12fa86e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 16:03:00.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48b708e4d0424a58b7944e91caf5ce69\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a954fe5e90b042a9a1570c4c6f1fc095 : SUCCESS in 52m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4aeb4daaf1ba4af2a39abfc87ce4fa58 : SUCCESS in 43m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9d290f14aac49a7a5968f70afb5fd07 : RETRY_LIMIT in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"a6d0f8fde9d2ddf54c5be6802995c7b4b24b18e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 17:38:54.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a897af425d04453bc642de9f43752e6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/be6047bd0d014cdb9f6497c94648b584 : SUCCESS in 1h 15m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/74c65dcef7e042e0870745c0a90acd93 : SUCCESS in 2h 30m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8528da0a5e4047e8b45a1f1c65060758 : SUCCESS in 32m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/050cfa5c26f04f4992c948af37e0b057 : SUCCESS in 1h 00m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f2350b263e4a7ab999bb2d45471414 : SUCCESS in 26m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5316846843b443b6a6340d5a473940e0 : SUCCESS in 21m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d753360b560492b94a569580058eefb : SUCCESS in 24m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6973346dff6e4cfdaa3c18e695bc3386 : SUCCESS in 12m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1a3d251d6cd84b96bc6074015db4b90c : SUCCESS in 49m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5206fb10dfca4da294127addb9f1e25c : SUCCESS in 35m 56s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/474bdbdb358944f1ad64ebb5e35bf45e : SUCCESS in 2h 17m 00s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"18ec53765a1388ae0eb625f9503d4cb447c08b4a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-03 19:49:19.000000000","message":"Uploaded patch set 52.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":52},{"id":"fa8a9006996dbb44acf7cf0cfa7948466b61bb8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 20:47:09.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee9e5cc1210747aca1609e5ac9747c84\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ba274c1cdc704beeb7ee6e35b690db50 : SUCCESS in 53m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22e51761e5814f07a14aa59daf750833 : SUCCESS in 39m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d856f28b7d13423c8ef14a19b47239d7 : RETRY_LIMIT in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"119a2b1a91d60d3e3686828ee31bef663f54ae0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 00:07:41.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc5b566acb694c7dac7e658d2d5ef0d0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/95c5f1937ee141f28f81acc76120373a : SUCCESS in 1h 30m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/20c5fe501b8140e3a685b0e806c0ea10 : SUCCESS in 1h 51m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7de340637b3404291ad4670820692db : SUCCESS in 30m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1d0c27ae383492aa39d5ef100cfd86e : SUCCESS in 1h 13m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcdaed9bd0fd4bb6b0d8add6c9072cff : SUCCESS in 25m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81537fa403c140029dfaebd3d21eebe3 : SUCCESS in 34m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc1edf29166046fcb6dc25a6b7457224 : SUCCESS in 18m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f22fde03ecda4fd0b631039bd1a23e82 : SUCCESS in 5m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aa955a0eed9548bd8bd547e7c885f59f : SUCCESS in 30m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e1f638f22f8042c3860ad88d54de3ad5 : SUCCESS in 1h 13m 51s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d080b0137e8544a28f4e561a52fb0495 : TIMED_OUT in 4h 09m 55s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"e2f61a8d8e2d1a6aaf2f2b96352fc32b28438ecf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-04 11:48:10.000000000","message":"Uploaded patch set 53.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":53},{"id":"37e971b2bff75ca3d12d31de195297839c6da5ed","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-04 12:01:11.000000000","message":"Uploaded patch set 54: Commit message was updated.","accounts_in_message":[],"_revision_number":54},{"id":"a5e9330de1dc5cc85c14cc51eaf36c7deeac8e78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 13:01:25.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/912b6f8974f24bb0be2a3720fdf1fac6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4924a25da8545348abbdcdd112b3e09 : SUCCESS in 32m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d654424967bd48949f10a5c6b09ff121 : SUCCESS in 35m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce01417013de4a359a6084e38bc3d573 : RETRY_LIMIT in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"f9d1463aee27983e60ba3ea374ef243cbde293e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:11:03.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d10deaaa8994471bb671d145cdbe675e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/21502e1a0a4a4d9491a155c33048744f : SUCCESS in 1h 38m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bcd03074c1c4473f823c883cbdbb8b2f : SUCCESS in 1h 42m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/56f23e0fb26c40d5a6bd1d956ae230dc : SUCCESS in 32m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efc6da9bcc584c41b00d74ea846b76b8 : SUCCESS in 1h 01m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a0496b54e442d3ac6ec36c56c3c832 : FAILURE in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ab1f36f8b0a40c29c90774e1d7d488d : SUCCESS in 25m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40da12b99ec04c7ba8f7e2747a3bfa21 : SUCCESS in 25m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c0acdbe4042438a96197df98f2abc08 : SUCCESS in 13m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b30be74bffc540af84301627d9e0b4ab : SUCCESS in 1h 01m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/63cd09db242c4028bc9266dfb4aa9ebf : SUCCESS in 42m 02s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/7b31cee20537458a951c0a22933b8823 : POST_FAILURE in 3h 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"494e45d67e5ebc0ea1ac29a2fd90e3f8a7a2a032","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-05 13:40:48.000000000","message":"Patch Set 55: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":55},{"id":"8b8cb583837b2556e20bb5ff29ce15e0d43fa01b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 15:35:03.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4791bc6cef8413a9751b6d3b1f0194c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91b00dbca65f4a53bdbf4e385028a343 : SUCCESS in 55m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/447fd42030bc4e1293978c57f74709f4 : SUCCESS in 47m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f24944757954855b6fb4b7094cfc182 : RETRY_LIMIT in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"d057b97c4423661d4e46a6355df5a8ceaa52b1fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 17:30:33.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41c91dc511d84730a16c18a3d65de8d0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ae57e17477144c4a809e2c80a6a61dce : SUCCESS in 1h 31m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/540e886e5ebf495ea78c614d050f879a : SUCCESS in 2h 22m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b8f7c69821b04af9b4d75bddf325cfc4 : SUCCESS in 31m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/635e6151c21b40a3824fea2880fda946 : SUCCESS in 47m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29e1359a420648bdb883437e708cf7d2 : FAILURE in 7m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9e1e8e858114e6db00beb29b215ddad : SUCCESS in 42m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab238e36fb524518b7a584f396d2ebb0 : SUCCESS in 21m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc1a653e04f40dd88e7a2f8243dc48e : SUCCESS in 11m 15s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/36c29b5f2feb4125b8f2df86f7000172 : SUCCESS in 34m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8f4f8a4e71554bc1baa9017e19d6f6f1 : SUCCESS in 57m 18s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/49f730839431496e93e67e8a4488954c : POST_FAILURE in 3h 42m 35s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"f6a2b42ee5505473155c13ecbb86ac2922c13491","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-09 13:24:54.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"3fe2ef6763158310c027112695cc0afe42e98b66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 14:28:36.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f6b28ccde564cbd8d254fe5c06823c2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2189169c94d04eff93c754f170cfb729 : SUCCESS in 1h 00m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad1780c2a22a4a029daa28d9dccfde30 : SUCCESS in 42m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c2ef667292045fb9e0092c7ff192625 : RETRY_LIMIT in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"2c79aafd3d9f4e8edb62fe4446176614e4ef5772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:24:16.000000000","message":"Patch Set 55:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf96103e38284a2d9ac1964aea16db3f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e9ab266a3d0545adb8c2f3c8e5aca0df : SUCCESS in 1h 20m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/262119b0795248628a7e17d26c2e2a6f : SUCCESS in 1h 45m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9141d826969243ef8a7f7afe15689b19 : SUCCESS in 31m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c94662637864448e88f62563d3b0b3f4 : SUCCESS in 43m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4838de7db2384facab5ce38d6ec4d482 : FAILURE in 9m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1414bca7dc2b4821a6785230ce3d9d55 : SUCCESS in 27m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b053096bbfa4e7d91cd7ef4b59f7f34 : SUCCESS in 24m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40129b9a18b7498590367f5c1514b387 : SUCCESS in 12m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3fc38b684517408aad3e5a6da6ba53fa : FAILURE in 17m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5e4a662b82aa4d0a815c6b638b18978b : FAILURE in 19m 36s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/977076a7aab24dab902135c33ef8dd4f : POST_FAILURE in 3h 47m 00s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"ff8821216be80353735e775860191204577d1ab0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-09 18:39:12.000000000","message":"Uploaded patch set 56.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":56},{"id":"4c83fa2d638f58a3a5aaea3906233610c14f3953","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 19:21:26.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c48c77beecb4bbd92aea544e9be3842\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/65ee78e9d56444a5912ab34a11e6416e : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/452b0fb8eb434635ae7d839d91f5e2f7 : SUCCESS in 40m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/86a3ffebe03a4d97abd769f61e7cf980 : RETRY_LIMIT in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"0185460682a856b7ee1bc1ffd2ba53c09d35522a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 20:58:10.000000000","message":"Patch Set 56: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/963223 is needed.","accounts_in_message":[],"_revision_number":56},{"id":"fe78fbed669488365b68ddb31d09a4a25b95c1e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-09 21:06:29.000000000","message":"Topic tempest-multinode-with-bgp removed","accounts_in_message":[],"_revision_number":56},{"id":"ffb5a73c48cd85e1197facb95817fc142389ca39","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-09 21:09:08.000000000","message":"Uploaded patch set 57: Patch Set 56 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":57},{"id":"d672a58f1b49751d61189f6920fab47a9d98802f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 22:17:20.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd45b40586e434387c85f6d1ed4df7f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8af242ac7974d0d82adc5ba9f497897 : SUCCESS in 25m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2cadfbbaaadc42af99fd0abbbedb36c8 : SUCCESS in 40m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94d3ddad16dd47e3b668c099ed73f289 : RETRY_LIMIT in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"e9011477b1ac9af3c92f52bd2fd1a5a34e1ce2ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 07:03:58.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a760d2bec9a48f6923d082de5f199cc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d4186be4e35542f3b9bd9c24dc36c607 : SUCCESS in 1h 22m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5b3b03291e784df7a9f94458cdbe6176 : SUCCESS in 1h 48m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dcd37b9f952a41ceac280a1175c7b8f7 : SUCCESS in 31m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be12b2b17f84479bb9973a31a72decd4 : SUCCESS in 37m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be35c59cb7794ea8b84e2213798e8683 : SUCCESS in 21m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d4ddaadbe88423495f9d37aea53b5c5 : SUCCESS in 26m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e12e56eaeaa142bbb31d16d260d65bcd : SUCCESS in 25m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1758383f2d941c0a5960b7106d89dd7 : SUCCESS in 10m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0328a7bb04d14c419eaf97aea9af493c : FAILURE in 9m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6eb2618f90db4584b8a4e061b24e10a1 : FAILURE in 17m 04s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/f891005f38294189a0c17fb5e277d46b : RETRY_LIMIT Host unreachable in 3h 08m 40s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"db5bcd5fdf996c5d60b5014e23e3b7b03b4b09a8","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-10 13:16:47.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"fff27d2073cabbcc424d08cd8ac28e2191c07549","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 14:27:54.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd24fa9fde72434d9ee6505f407b5f5b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d55b37530abe42fda209dee61d156289 : SUCCESS in 58m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6634a87626c0413d9fa1d04c40217acb : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6f9f9cbf1e54439f8b70b35d4236533b : RETRY_LIMIT in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"38fab0569c5a21d5a2abf5432ce3577a3d8e8beb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-10 15:54:49.000000000","message":"Uploaded patch set 58.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":58},{"id":"e1754283e1fdd03315bd22d0e65b67633909cf81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 16:56:30.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c39b65859d3c4b09b71bcc3bccc43960\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4934303f9044b08947503118d16d443 : SUCCESS in 53m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eaaf84a038114bb08b63b9813dd8e931 : SUCCESS in 41m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/65a1cea894d843c3bfeaab3ac86b9cae : RETRY_LIMIT in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"d9c30aa751d4291de2fc630c2157b3086749de08","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-10 17:12:48.000000000","message":"Uploaded patch set 59.","accounts_in_message":[],"_revision_number":59},{"id":"9c853c8ccd6fed5b8b6de5bd5537657278c5a4cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 17:45:12.000000000","message":"Patch Set 59:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/208890b11729401d8a20ada377d238d8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/395fcb388f60490cb6cafcbe31cff685 : SUCCESS in 29m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a1394900e73e45c49ec094bd325dc4b2 : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62fff48388c04d448da8e1301178f2d7 : RETRY_LIMIT in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"eab860a65763edc4e44284708bcab1a480d904dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 21:08:20.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/812dabf30602486ea5ce521c4de06e09\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cc362675ade846e1b486dd9a781636fa : SUCCESS in 1h 33m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bfea5edca47b439f9d1538adc34f71ea : SUCCESS in 2h 04m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6bc1e1220f654640ac6dcdaa03a4874c : SUCCESS in 33m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a389baf9d74f4169805961424787a4ec : SUCCESS in 1h 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30dbcc1eea9c4c84b03654da3612d78a : SUCCESS in 25m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcf74f06944c4f5e924492fe46c508a8 : SUCCESS in 22m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b78e65097d294593a0c3c07b0be3244a : SUCCESS in 26m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ccc6f3d8ff44dfbd779c8a07bf8c1f : SUCCESS in 9m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/81770935c6ab41e9954a02905d6369ec : FAILURE in 16m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d6032d2f93c640189d143df377998fea : FAILURE in 15m 15s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/4c1e19b7f262415c9a76d3123d17f396 : POST_FAILURE in 3h 49m 27s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"a89f56bd6980a6c370a15a2178cc7c63af3659da","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"real_author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-16 11:12:05.000000000","message":"Patch Set 60: Patch Set 59 was rebased on behalf of \u003cGERRIT_ACCOUNT_8655\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"_revision_number":60},{"id":"abd48dba42e49ffba4681041623887ef85ae2a0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 12:59:42.000000000","message":"Patch Set 60:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df0b95ca3f324f52b2d70c9277a27f87\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b48704c4ebd441d3a534daa3bf5c818f : SUCCESS in 28m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5aa4abc621e4697969a5b71d58f5560 : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1b60414588404ada8bfff3ff89b82586 : RETRY_LIMIT in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"9390dd7da2204c519666898bf353c45fcae65f50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-16 13:57:47.000000000","message":"Patch Set 61: Commit message was updated.","accounts_in_message":[],"_revision_number":61},{"id":"e95e295aff3483a043b7cc78baf96e2e163d7992","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-16 15:46:59.000000000","message":"Uploaded patch set 62.","accounts_in_message":[],"_revision_number":62},{"id":"af7eee3dbd6465296596672ebd06cff4b6acc070","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 20:08:58.000000000","message":"Patch Set 62:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4842c77dcfb46b6ab1789dbf54f7aff\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3233775235c14aada95a75678ab986df : SUCCESS in 27m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b66cee812e243b7b3f71fe1040e5c96 : SUCCESS in 26m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a4f05ea1602c42548ef20c17a813899b : RETRY_LIMIT in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"28e33b56d5b8ca67ba7e4e9782c6031297edc02f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 21:49:05.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0ce42672ce54172bc044fad01b3aa4f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/41307cf6b69945d9add4ac227e13dce2 : FAILURE in 45m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3576f59f1f454bf7ba062cc9d4e36685 : SUCCESS in 1h 37m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad0e213d3e0a4369bd8f35899d9a2d82 : SUCCESS in 32m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e75db5f81ba94c8ab74d910da1a17a1a : SUCCESS in 47m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b404afac2164ed39321c0d60622dfb1 : SUCCESS in 27m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a8ebf21e6f0457f8c655fd1d330607b : SUCCESS in 39m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20c542dea84d43ea9b63e62205b02e28 : SUCCESS in 27m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ddbe66bb96a4fcc943de1d1393e9089 : SUCCESS in 13m 56s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0f28b967cb5f4cf8a4bd053807ca6137 : SUCCESS in 1h 05m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d09511153d9040e8b7bb7f3f789ecb2c : SUCCESS in 54m 10s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/4e4758406f2a4d14ad372a4fc3b6efb7 : TIMED_OUT in 5h 53m 00s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"0a17892ee9c9d4fa76371d227421127fc354e476","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-17 23:44:59.000000000","message":"Patch Set 63: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":63},{"id":"97a65d5c53165b16e04a30b96a43576c1ae8ea44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 00:43:45.000000000","message":"Patch Set 63:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea62f3359b9745c0bf4b2477be0e922c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f19049d8f0eb4e78a22b6b5fc8fcbe87 : SUCCESS in 26m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60465efb85a34c34bfb358fa76e7f667 : SUCCESS in 21m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/92508d63f2f84f29bd56b9f50dc41d21 : RETRY_LIMIT in 3m 36s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"688f8a693fba98a6f6aa4cf600b256b3df00e1a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 05:06:29.000000000","message":"Patch Set 63: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15e07a16bab24a7abb6d818f9501b375\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/32ceb63bd0174b678968a03ed0dd8253 : SUCCESS in 1h 28m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d9568a22e629433b88e24ebcc4898f5c : SUCCESS in 1h 47m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ef46069dd1e14087bda17a6ed5d6029f : SUCCESS in 34m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48de3ed43da44646ba159ca40a9ffffe : SUCCESS in 38m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d16e51684e94419a6df686e796386aa : SUCCESS in 18m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/027394a05a03456787c939d84981686f : SUCCESS in 21m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43c21312e3e14db1b1ea6564424abb2e : SUCCESS in 25m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e69a7b7b5f545f7858ae27bcb6c06f8 : SUCCESS in 6m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c387a9ab34df4973a30f2aca8577c8d6 : SUCCESS in 1h 03m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/19e683f29da4408097cf0e0cbb8ae06a : SUCCESS in 1h 16m 29s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/a0979cec89df4521b342c0e84210585f : POST_FAILURE in 5h 08m 15s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"4156aafc749410f8f5635dc6e44576b4e202a565","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-19 00:14:35.000000000","message":"Patch Set 63:\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"1306120653635980407534edc54e84e0444eef63","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-19 12:59:39.000000000","message":"Patch Set 63:\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"c3f31c7a53f079e8fb0c263b3ae900f66936d0fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"real_author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-19 13:59:51.000000000","message":"Patch Set 64: Patch Set 63 was rebased on behalf of \u003cGERRIT_ACCOUNT_8655\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"_revision_number":64},{"id":"aadd655e73325db7e469c46b843507bfc7bf0a8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 14:39:22.000000000","message":"Patch Set 64:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d92660764e46f59afdba1ca1539bc5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d0a56481a4d493cbefce7884d76c9dc : POST_FAILURE in 24m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c805f5d1671542d48446519f85fd2185 : SUCCESS in 36m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ab840fbca5564430b452b0d8ad340342 : RETRY_LIMIT in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"66ba9ba505ad21b8309a73350e2e283d21657f7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 20:01:18.000000000","message":"Patch Set 64: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af28780591e8400989a0a24f13d3bab8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d59b4251086242b597018ded10017d63 : SUCCESS in 1h 17m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2c7bf0b923a24818a18528d09964840d : SUCCESS in 1h 41m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/afca206b3eff47fb9b972de40ef305e3 : SUCCESS in 30m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43ea9c00127a4801a7988ebf9b6dbb91 : SUCCESS in 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f232760e4f0c4bf39fd1d0f07f895026 : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7906df57d94a476cb90bd43a9338f283 : POST_FAILURE in 27m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53c3818b86e14a60a9c8bf618f9f91f5 : SUCCESS in 22m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c23fcb978564d3fb84173b9a4bd6bb3 : SUCCESS in 10m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ad74c807c328400a91b876fb9ae56417 : POST_FAILURE in 53m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/70a47c87b3d84d61bcbf06390cd132ee : POST_FAILURE in 1h 00m 27s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/d47619b6b2414e5c8874620f65e2b66d : TIMED_OUT in 5h 49m 57s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"9cbf6d03f426af07eb7c084a863015e861c0d527","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-19 21:27:41.000000000","message":"Uploaded patch set 65.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":65},{"id":"dab65a488d8f13afc91858444dc35039c54f65a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 21:55:08.000000000","message":"Patch Set 65:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d92ca4820b8646b9a2b5ef264909abee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7fc076035d6d48a985d8e3fca6bfad7f : POST_FAILURE in 25m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d786525b9894ccda99b1fa48525ab2b : SUCCESS in 21m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/27c21dadb4544a50843495cc726b5392 : RETRY_LIMIT in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"c64f7464a7b4526741cd7297bd1c177a99cf5c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 22:03:52.000000000","message":"Patch Set 65: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/neutron/+/977173 is needed.","accounts_in_message":[],"_revision_number":65},{"id":"def79ac4e843a5e7b9691bc1e0817f7243ac6a98","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"real_author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-24 15:38:34.000000000","message":"Patch Set 66: Patch Set 65 was rebased on behalf of \u003cGERRIT_ACCOUNT_8655\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"_revision_number":66},{"id":"6ba8bb5835b19335eae1db47e9287d77867dd39e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-24 15:39:24.000000000","message":"Patch Set 67: Commit message was updated.","accounts_in_message":[],"_revision_number":67},{"id":"9151338d2ca910f104ab055fec866c65ce03088b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 16:30:00.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37c6516f1902485991ae7b585d564d7d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ec35ea1677f4e0e8723ab83b9681678 : SUCCESS in 28m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/035bc8aeafd642e1bd110fe9aea3953a : SUCCESS in 36m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a8d48c82bab64425ab414b828fdf448c : RETRY_LIMIT in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"384676551d9e73c687454a942c70ca93aeb64ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 21:37:50.000000000","message":"Patch Set 67: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2c1d138e989413cac32d23b58997198\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d39fd6d05cbb4a989f92e197b620f065 : SUCCESS in 1h 18m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7a71d03a9bfa434d8237774677c2c3ec : SUCCESS in 1h 43m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c483b5fa71a24a52bd7ba07849127eae : SUCCESS in 31m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05c68a20b0b34853a2ba8bdf48ad80b5 : SUCCESS in 57m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a71674c7083d45f7a4652ba2d6cc3cc4 : FAILURE in 7m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd20e79615124b5b8fa9a55af4d30c97 : SUCCESS in 31m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be8a6ae9e08a443ba1e62700b86c6d37 : SUCCESS in 23m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fac3a3a246c0433c9c9d72e0f750b71e : SUCCESS in 24m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0de16a7b304d6694514ba843134f06 : SUCCESS in 11m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0c7eebb7d7bc44d9a99bb7a1a3f2baf8 : SUCCESS in 48m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9a4eeba1576b425782183de853285c99 : SUCCESS in 53m 50s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/63f07c6885214b39a5c7271b1aed1d0f : TIMED_OUT in 5h 50m 43s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"fe7f0e88ba419baa7f3fa0adfb5410ad79bfe8a0","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-25 22:03:55.000000000","message":"Patch Set 67:\n\n(1 comment)","accounts_in_message":[],"_revision_number":67},{"id":"adf013a5d301f31f93748108b2abe18874dbe439","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 22:47:01.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2414dd0bd28e4029b6d32debe7cc09fa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6fd6d1fda2b4485b18e9dc6cf4e42ee : SUCCESS in 24m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2d6ac815782b4bb387233ce4eb444c51 : RETRY_LIMIT in 3m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2bd1c17c4eef4d3cbc1e8307cf420b25 : SUCCESS in 40m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/38f4cebad95e462aaee898ceba20942f : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":67},{"id":"b846fe5f88a35d7d068ae6a163718c8dc06b9fb2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-25 23:48:35.000000000","message":"Uploaded patch set 68.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":68},{"id":"0848ba668138908b3dd22e35b03cfc0fe32b24bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 00:21:26.000000000","message":"Patch Set 68:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2e2634ae0a045ecaa0d5e89828227ec\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97212ccf21a640a7b1d44fe84a77e807 : SUCCESS in 23m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d7f2d5314393491185f858d59ad16c70 : RETRY_LIMIT in 3m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dff4bfcb3991410fa337b1a109e0e3a1 : SUCCESS in 30m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32272bed7fe944c1935a8143aa69bf8c : SUCCESS in 19m 20s","accounts_in_message":[],"_revision_number":68},{"id":"4b51a08dc7bb9c2e227fbc3f8b8025a31111d777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 09:25:46.000000000","message":"Patch Set 68: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f81a1ad70874a438363394a94f51ead\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7af19585075b4414be6af48e208a13e9 : SUCCESS in 1h 23m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/baa4a7e958dd430ba703c101605c767f : SUCCESS in 1h 52m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6212a0dabc9a4a34b2afe8c77f95c763 : SUCCESS in 32m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ab5107056d94c70aefdaeba5b57e776 : SUCCESS in 37m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1274b50188a64769b4d617a26ca655c1 : FAILURE in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38b582b6fa904f159a00cb08dd619f21 : SUCCESS in 20m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/271868cbfeb545a19d271a18175eda4a : SUCCESS in 17m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7cd23e7b81fc4561b888abcbd55c5baa : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d624eaf7137a469ab6ed7bed0381c91e : SUCCESS in 9m 55s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e7fffee85e1b493faddc21b319c8a2d1 : SUCCESS in 35m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5e36ea937bf048d3bbce1d6cf273713f : SUCCESS in 1h 11m 45s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/ba75a153cb00458b845c6f472e550086 : TIMED_OUT in 5h 51m 48s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"de1e90010ba7ee7999658f3df37aac3edba07b16","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-26 23:28:48.000000000","message":"Uploaded patch set 69: Patch Set 68 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":69},{"id":"4d1c12cea2918a42d89c28419d87871ea89b6e7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:28:01.000000000","message":"Patch Set 69:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce03f76503254c37a7c8d0db67569f9a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a80c73b20e641f8a868355d46bb2d36 : SUCCESS in 57m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/19cfeb4c62f048c39acbb5d91098dfa4 : RETRY_LIMIT in 5m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fe53e1ce952247f1bc05c934f85e9834 : SUCCESS in 55m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a8eb563ba7042058a4bd9973812e2c9 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":69},{"id":"dba1646a8e6767f0eda1f9d70df647c41d74ff07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 04:37:21.000000000","message":"Patch Set 69: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/650fe734e3624233b2371b19fc4b7b16\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/45d8a444f67e46b583cc762a9283489c : SUCCESS in 1h 31m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fdb1f48e73448ec8ef2ff8e6cba55ee : SUCCESS in 1h 00m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/16079aca9a5e45128e82436660b9d255 : SUCCESS in 1h 24m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2da71006b23b4418bb78ba78227619dc : SUCCESS in 1h 40m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02686946c46340a39c0856f07bda9519 : SUCCESS in 33m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81753695edeb46e89b61417d166752a1 : SUCCESS in 52m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fc0fd3a177d4a278c66e611e04ce408 : FAILURE in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41acdd17ef654b1694f3be93e9a5a64d : SUCCESS in 28m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/894ecae86b114fbd8f2b0b1d3cf96627 : SUCCESS in 18m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bf88b432aa5e4730b8be9b3a927b06b6 : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72cd36d92d524ae1a754df7ba1cd4282 : SUCCESS in 11m 22s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f1ebb811c675494ea7b3cbd5d412b578 : SUCCESS in 58m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8d1a11253f6e435c89299a6d2a6cd2d5 : SUCCESS in 1h 16m 10s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/bf6d2d882f4b406f857036f9bbb5a351 : POST_FAILURE in 4h 56m 46s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"f6fe45a4da1701adf19d47b3a059eaf07a93f914","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-02-27 11:32:21.000000000","message":"Patch Set 69:\n\n(1 comment)","accounts_in_message":[],"_revision_number":69},{"id":"f26557b4970683c7c13762eb52b3071f4c257233","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 12:41:00.000000000","message":"Patch Set 69:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ada0d8ae65434ba89958240397b012\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a19f7c56b72749f7876480b868647105 : SUCCESS in 55m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ab3f1e5a0be843e8be9790c39806cd35 : RETRY_LIMIT in 4m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/871f8d4fd1864f95ab87503fb182212b : SUCCESS in 54m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/08459607ddf34bc594a06ed7f66a5887 : SUCCESS in 23m 46s","accounts_in_message":[],"_revision_number":69},{"id":"adbf1c7767034ff6038807ccd4e7d0148d54baeb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-02-27 15:33:25.000000000","message":"Uploaded patch set 70: Patch Set 69 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":70},{"id":"8d8b873ccde5a5b5632fd5b63187cfde6935ed3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:22:18.000000000","message":"Patch Set 70:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd186e1d74554cc5898e2bcba6045bf0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/174eaebce1764b7882856f75e0b231f1 : SUCCESS in 25m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/29c9f073c75446efb0044ce9799460a2 : RETRY_LIMIT in 3m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/059ddee3d9e04b3dacfcc7828961f761 : SUCCESS in 27m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/462ce1f8de1044a688628781648804c9 : SUCCESS in 37m 40s","accounts_in_message":[],"_revision_number":70},{"id":"a5972616c0cee062a5bb3e690e3503370c7b5899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 20:35:17.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f21cb46a842d470aa3717f9d7568612e\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/238ede8accb74d9585fd6660880ddab2 : SUCCESS in 1h 18m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a590766aa92b4763a4ab656ef1289621 : SUCCESS in 1h 05m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4737273a3d7f4b22b9699b323943d343 : SUCCESS in 1h 20m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/94a8eacceb4a461ab086e3d2ac1226e1 : SUCCESS in 2h 28m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f4c1873821de4c5a9ee0b520c411f80e : SUCCESS in 38m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab46a78cb5a94d58b02f01eda9c682cf : SUCCESS in 1h 00m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c205fc937f3a40378d71e783e8efb55c : FAILURE in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55534a4f869f40c9a86d3cede1177371 : SUCCESS in 31m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e624b17a629d4e1fb65bd20507da42a5 : SUCCESS in 21m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/90954d5e9da9448a9e6199d13e46142c : SUCCESS in 22m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b403a5ed67634a6bb054830b6e6580d4 : SUCCESS in 11m 13s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d05a7cf8a779460d8079fb8a239581b1 : SUCCESS in 51m 40s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5bfe47a5a320464ab4b5dbd00fb176df : SUCCESS in 1h 11m 45s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/2f882c80b55f4752aa10341dc1c076e1 : POST_FAILURE in 4h 54m 12s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"f6b5b91b9cefa26aa03e934e70548dbde820f199","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"real_author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-10 15:07:07.000000000","message":"Patch Set 71: Patch Set 70 was rebased on behalf of \u003cGERRIT_ACCOUNT_8655\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"_revision_number":71},{"id":"74fe8395f95e468f944822e6f45c8f32c1d41b3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-10 15:07:33.000000000","message":"Patch Set 72: Commit message was updated.","accounts_in_message":[],"_revision_number":72},{"id":"3f8f13e71ca51d3884e33fb4f3472c3ae9b4c66f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 15:44:15.000000000","message":"Patch Set 72:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/612c7c3644bb4bd2993f18557169532c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4906b42a5f264618844dee9735c4ee0a : SUCCESS in 32m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/431261967b0b45a19dfee94371911a90 : SUCCESS in 19m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/231e56b1a3464b62b979cafc48762bab : SUCCESS in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"ad8f10eb54286d5c6c0cd9fc6a4946fd785264f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-10 17:31:52.000000000","message":"Patch Set 73: Patch Set 72 was rebased","accounts_in_message":[],"_revision_number":73},{"id":"047ecde2267edd3cfb246415cdb683cdfbc4a600","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 19:08:16.000000000","message":"Patch Set 73:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f21d909b1a9a4e7689795b93d3a1254b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f29138f6ff684510b398900dbaf3f0d7 : SUCCESS in 25m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0bfa4b32c3b24f4381269c722cbc06cc : SUCCESS in 24m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d29e8cbe14aa4277b656edb29522df2f : SUCCESS in 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"cbbe1dbe90fc8537b24cb01608737d8ef48ffdb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 22:56:16.000000000","message":"Patch Set 73: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41308325e53446b486d5a2c25b5f8f8c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/962c30090f954c649d4595b9638d670b : SUCCESS in 1h 33m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a2ba933dd4c41afbc042c31b06c2523 : SUCCESS in 1h 09m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e0c9543150514c08a5b3cc6a7e105738 : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a245772e1949489a8bd63c0a04e440f1 : SUCCESS in 2h 07m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/23259777fffc4ac7981809e3ec64736a : SUCCESS in 37m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/075bf35b303c4e1e963ddb638896b842 : SUCCESS in 38m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ca64d7866b4faa88255b85a331b34d : FAILURE in 7m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57ab9608a934424aa6a96735f4b7e929 : SUCCESS in 36m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/49691f8c4a8c4cda8543b9472ab71a7d : SUCCESS in 18m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/05d85b71bf0f4b75aa9456e70d71c6bf : SUCCESS in 33m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c261316849469eb8ad795b5b6a62ba : SUCCESS in 14m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1113ad6d49a644b3ab8ec354fbe82657 : SUCCESS in 53m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c0774266d653488dbf9d924e82be0a6c : SUCCESS in 1h 19m 20s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/1d8be50b02b748c09251b05e0fd50b62 : POST_FAILURE in 5h 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"acd5270003ea19a69b06cfb97e8cabcccae9d1bd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-11 09:36:15.000000000","message":"Patch Set 74: Patch Set 73 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":74},{"id":"628dd2363dab8b0771f13b8c8c259defe2eb02b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 10:34:01.000000000","message":"Patch Set 74:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c62d33c06da245dfba424134a55e6b59\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dbc94005ed3840a4a37196cbad21d349 : SUCCESS in 24m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/380723f07a5747c18f13dabbce9075ec : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/995630a831bc48f89a5e0915dc70aee2 : SUCCESS in 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"271305828a3d1a411561b0134659f6c47c621c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 14:49:06.000000000","message":"Patch Set 74: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d597ffcc33c54c74b0b8a5f50bd61aa2\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/29452a2b9dc34e908b6838896009a6ad : SUCCESS in 1h 31m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7d34b548bf9419c9b3eab6b9244c91c : SUCCESS in 1h 00m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/05f958baf76e4dce82f428b36379ce38 : SUCCESS in 1h 21m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/26fd768369db45459e514c9148b6eb38 : SUCCESS in 1h 59m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a901c5ecb94c437095d19e2ceedd8087 : SUCCESS in 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce6f13b39bf5432cadbd44f416ccc4f2 : SUCCESS in 1h 02m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c96a06074f4b2991bd83d7045269e8 : FAILURE in 7m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/776be14055e44789b10a810c23af15db : SUCCESS in 43m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/286f3442b187443d972d72ce5cd40953 : SUCCESS in 26m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c102add686c40c980b3ba9923ecebdf : SUCCESS in 34m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e800777325994729a0acdacd5bccbe18 : SUCCESS in 13m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c59fd88c6d7641e99a926b0a11eb854c : SUCCESS in 1h 07m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/50d53308715346288914c0d759cf67a8 : SUCCESS in 1h 12m 02s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/1ee92373ca7c454f84c0c36d9ac6038e : POST_FAILURE in 5h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"d98edb77074f700aec337d7d3cbf2111ca7c9a89","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-11 14:51:16.000000000","message":"Uploaded patch set 75.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":75},{"id":"3764bc0b51eb4b6521135505b44c4e27d8046ce6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-11 14:51:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":75},{"id":"e658421155f9b8d33856ea690bc3513012d5d2c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 15:31:08.000000000","message":"Patch Set 75:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86ab93da7499439fbe6d04df8a14529a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff8fe5a64f1844cbb257005855a8dead : SUCCESS in 31m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c86a1896a9ad404894ab1aaf267453c7 : SUCCESS in 21m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1635812d355640fda92f8d7a8b888808 : SUCCESS in 37m 36s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"d940698aa5209d42898c36114e3b340644ab8563","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 17:32:23.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da81442f6d84b05883d88d857084e72\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0f86f99b01b743d1979740f28c16c7a2 : SUCCESS in 1h 25m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66227ff27a3b404d88d332dd51eab2fe : SUCCESS in 33m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a5410888fc1a4be995c5c7d1c573c698 : SUCCESS in 1h 28m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b7d69626261d49f994bd77200a98267e : SUCCESS in 1h 55m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e54374f6433c418e9bffae2a4c7a9cc4 : SUCCESS in 35m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65852f48bb224e68820e84370f4fc8ea : SUCCESS in 46m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e118973eb84cada6e0bbfbc44b9128 : SUCCESS in 23m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b641ef1da59d457880b4e9117c9deaa3 : SUCCESS in 37m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/110274eb536f434f8735ac44497f99f2 : SUCCESS in 21m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/10953df6d36143b4b5f0381b43ce2ce8 : SUCCESS in 38m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b7e2cd4e914f818317618917396733 : SUCCESS in 7m 32s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1c3a6731872449bb8ac2b244c73cfde2 : SUCCESS in 1h 01m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4895f98e48034cef9c9c6c702152f7c7 : SUCCESS in 1h 04m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ae6052a934944232848d383197010f0a : SUCCESS in 1h 34m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/043dedc694d34453b383fc641caeaed2 : SUCCESS in 40m 42s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ceb94ab031154ceeb09fe6c5e9565867 : SUCCESS in 1h 08m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e852a2ce5b864856bffef43ec3d37f7f : SUCCESS in 1h 50m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a3073b30e8774b86aece2b939b281897 : SUCCESS in 1h 41m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/10acf8371b8d41b39bfb378a0ec601b9 : SUCCESS in 1h 14m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/11088bee924e4748a4809837e7f3ef54 : SUCCESS in 31m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7467a6d5020346e08ffab5d4de4229e5 : SUCCESS in 1h 24m 00s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/896388a2f5e14c3dba95f9b277191d8f : FAILURE in 2h 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"cb0ff61d4d8d6b9df65306662e925738ae3f9b52","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-12 21:54:05.000000000","message":"Patch Set 75: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"72ca90355746a9e2090b170931535a8ba88cdf78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-03-12 22:02:23.000000000","message":"Uploaded patch set 76.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":76},{"id":"28a2c07f244cab377ccfb1f97abcbbfe90f1c230","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 22:33:39.000000000","message":"Patch Set 76:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/701b90818e6f42a1ac01e1914fae1e72\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd5f710bcc664180ac6550204da6b766 : SUCCESS in 26m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a49bdda0e74c4fe28b441e73209900d6 : SUCCESS in 18m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7fa6c2f7df00431c90db020a7d984bda : SUCCESS in 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"b631d6667dd09b1bf3331e44084a390f38cfc1d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 00:16:37.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/255736b4b01e43699dda27864f0f03f2\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fb5a93e82d624df4a2489aa660ebe073 : SUCCESS in 1h 15m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f19016c927c64a59b075c9788e1a290a : SUCCESS in 1h 08m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/348712f192ce466e8a2d54f96c775b0c : SUCCESS in 1h 24m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f44a984767104e05aa26737d30bd9c20 : SUCCESS in 2h 07m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc838cbf18b24af4945ee22a4b3c36b5 : SUCCESS in 32m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb1c26a341cf4df5a1f9ed91f49d030b : SUCCESS in 47m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272cd60657824193936f4b6d08eeec18 : SUCCESS in 30m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/641155487e0d420fa26559a117ee67e0 : SUCCESS in 36m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/558dd66724e643d4b527983c387caefb : SUCCESS in 15m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/858df5c2c4de4e75ab6d042d6b1fd937 : SUCCESS in 31m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3732e2b7e5f44de0bec80226f4578c44 : SUCCESS in 13m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3da898100e4741458bdc54c01bb867b6 : SUCCESS in 57m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fcd84fce79a44b61b39ad3d141d5b01e : SUCCESS in 1h 11m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8e23491d3d3148b79d581038dd2742ed : SUCCESS in 1h 19m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c05aadb6f8b4cc28a43c3b33a4cf875 : SUCCESS in 43m 24s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/684dd9f670f9449e8b6e0b0cd38fea0a : SUCCESS in 50m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8b5beb7f79c4477aa3c95ba2325399ae : SUCCESS in 1h 48m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b89b7348e950404581a247173489d4e4 : SUCCESS in 1h 03m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d24b900ecff94f98a59f9ac10757beb6 : SUCCESS in 1h 06m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a9786f6b223949cfbffe3b5ee7e3f94c : SUCCESS in 31m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/32e5505753094919a2bb0788235a9191 : SUCCESS in 1h 28m 15s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/1343c1b4f6b14b35b9453fdf844381eb : SUCCESS in 1h 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"fa75c92d4d0ac14c9ff8bd3a1b754374941f3705","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-13 07:26:59.000000000","message":"Patch Set 76: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":76},{"id":"752500d76f241be08abeb32816cf02d0759c2a65","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-16 10:11:09.000000000","message":"Patch Set 76: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":76},{"id":"ba0407d2e29446bc960cc58d743e02b10d8c70ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-16 10:40:50.000000000","message":"Uploaded patch set 77.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":77},{"id":"fcdf542b14c89b4298e491ba81fd6411cb16df8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 10:42:47.000000000","message":"Patch Set 77:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/devstack/+/963199 is needed.","accounts_in_message":[],"_revision_number":77},{"id":"45c98490ec4328db7ad01867c151a876ef434757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 10:43:06.000000000","message":"Patch Set 77: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/devstack/+/963199 is needed.","accounts_in_message":[],"_revision_number":77},{"id":"0e7a8d0405e1f2df97bd0a9efff198a8ea10ac29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-16 10:43:16.000000000","message":"Uploaded patch set 78.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":78},{"id":"34de43c1c9b9af188321c7e10f7838ec41476f57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 11:56:37.000000000","message":"Patch Set 78:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc13a82bba74713b024101484379cbd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea0c82dd047c4185b61c6a5356da48ab : SUCCESS in 31m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8991b0957e1c48bbbd6e9dba489fcebf : SUCCESS in 39m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7db7bc1ec2634be78b8f270f87af5c7d : SUCCESS in 1h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"28457081df49e8c265287c1a97102f3c2749aa3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 14:21:19.000000000","message":"Patch Set 78: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60ff6f93341c447a93063d619db81ef1\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cfd12621ef32432c9caa5b17303dd4be : SUCCESS in 1h 26m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ff728e6a5a74bd58c50cfacb6ea3260 : SUCCESS in 1h 03m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d85f36c88ee148b6bb8a7cacc6a48d8d : TIMED_OUT in 2h 50m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/343b1d4ae30049c7b0d25a8ba3a71a21 : SUCCESS in 2h 17m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0f0133b8507c46d994658f6d32c24c38 : SUCCESS in 36m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b1ef5aa9e0d4f86a3465e09e94c58d3 : SUCCESS in 1h 00m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3eab392919b4f20a2372aa99e354761 : SUCCESS in 22m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5dc9da654ba6415cbcb531361dfd649a : SUCCESS in 36m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0215aa1b0774cc6889e1a9825b91fa8 : SUCCESS in 28m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8a5aebe0c4ee40ef965e2f8e93fdf2eb : SUCCESS in 30m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f0709bcdf954def83d3d891692b8d19 : SUCCESS in 11m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/23fae56e756e439ba5b5370d9c37f967 : SUCCESS in 1h 03m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/06c739f199d8490380e3f999898be339 : SUCCESS in 55m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a52b350b1bc64565a889a3ffee7bf04c : SUCCESS in 1h 23m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2f6a00a8d1e4d4984e34f7261a213c9 : SUCCESS in 54m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2297b00ce1d49a8bf3018d288b7b467 : SUCCESS in 1h 12m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4adbc1efc12544e2a983e323c9ea02e2 : SUCCESS in 1h 30m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/273de158ecc349d895866908fae8fddb : SUCCESS in 57m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d8483726a18a45d9a11616ee73a5222f : SUCCESS in 54m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5b6c3f13c2d94bf594a197c9d8ab7ff1 : SUCCESS in 1h 09m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ccd3e9d3c3394c979e32a6e0b7a3f4f5 : SUCCESS in 1h 23m 24s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/fafd0c0c5fa2456492ad9fb5cd3d37fe : SUCCESS in 2h 06m 39s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"101d3b88b8c3c40e982b26f7014ce13077d430a9","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-03-16 14:33:28.000000000","message":"Patch Set 78:\n\n(1 comment)","accounts_in_message":[],"_revision_number":78},{"id":"995ae11e0fb5e0dde2e5b610dbac1d102bcb83a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 17:02:54.000000000","message":"Patch Set 78:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da48b766ac14e87ae5be2c66fa8e1c3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea70855715bb4e31a9124492d9c9d21a : SUCCESS in 59m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7e3143714226442a8c1564965e5c8614 : SUCCESS in 41m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e26f8855650c49f486243c9019a5fb46 : SUCCESS in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"e3b7619458564364f9cc1383e5260cb7e0be40f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 17:34:52.000000000","message":"Patch Set 78: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a39094a636fe4d228bd4e2c68338dba7\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5bed540cc2814de59d81be9bb3c0b220 : SUCCESS in 1h 20m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f595477e39148428ff169d5db01b142 : SUCCESS in 1h 04m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c7ff10fdad114de78fb476889cf82a3b : SUCCESS in 1h 18m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/431bd5d3647d4f8daf42069574590ae8 : SUCCESS in 2h 45m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/467b084dfea7472da302b8e56311b94e : SUCCESS in 37m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3bcf2add4354988b18106f9fffad2b1 : SUCCESS in 37m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a07b549c214bb7a8e7309b708fad78 : SUCCESS in 25m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc5868ba3c174b4fa714c52ce3c56d56 : SUCCESS in 30m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87cdf2c1a0954d91af26d88891aadc25 : SUCCESS in 24m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24390616f80249d6877ea5c7c24045cb : SUCCESS in 30m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cbce7cb597d47389a96ef1b0c6984eb : SUCCESS in 14m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9e3e5fc3ff30418eb5242ca9d58ea50f : SUCCESS in 59m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b4078642b4ab4ef7b1882f235bc6e5e4 : SUCCESS in 1h 13m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/306eff7481f847d0883541b63ac8db4c : SUCCESS in 57m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1549ec557ab14cce831c84b6acb706db : SUCCESS in 1h 11m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b46518be3c374d4dae23bde8f305e88f : SUCCESS in 58m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a12f462a217d403ea051da0b912cf94a : SUCCESS in 1h 53m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/85f7c7853ddf4240a95a709a303949fc : SUCCESS in 1h 35m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/75c39698b46740eba63eeb71d28321b1 : SUCCESS in 1h 09m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b5ff80c6fffd483b851d155077137e8b : SUCCESS in 1h 08m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd1c219df9a74b4293b34681829d0c59 : SUCCESS in 44m 56s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/a213336895224dfaa62d4ecff4bda01c : SUCCESS in 1h 59m 51s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"35654ee75abdcaf4ce3aa5693633bb6c41f4694d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-17 08:43:05.000000000","message":"Patch Set 78: Code-Review+2","accounts_in_message":[],"_revision_number":78},{"id":"74373161f3a7e29fcc93cae3a91f5f306160eaa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-04-10 12:53:06.000000000","message":"Patch Set 79: Patch Set 78 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":79},{"id":"ccc092c6bbb35c2d3802c102f8c62c5697805fdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 13:55:00.000000000","message":"Patch Set 79:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/152424c29c2f4434bcf3658ae416043a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c9e770df2454483b658f227b5cc1e36 : SUCCESS in 26m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ba58d84a088447568cd357437cda22c6 : SUCCESS in 18m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/842142c3e0c54e02920373130410a582 : SUCCESS in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"7b8c9ec35022f8b42a2103c24e71e0724eade7d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:11:06.000000000","message":"Patch Set 79: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2f30979cf9f4cbb8e49c26781a70e45\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fb79c16e8de24aa8ad08fa5d6f86226f : SUCCESS in 1h 04m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59b66b155d434ae5a7c24708c2ae68c9 : SUCCESS in 48m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d54c9e73207844e5aedb17a7d6b2c49b : SUCCESS in 1h 27m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ea573dd7520c4fde8a218bcec50403d3 : SUCCESS in 2h 14m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5f84f2efd4d403e89973c2e7c8eaffd : SUCCESS in 36m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75ceff8472a443238af51246ca95d016 : SUCCESS in 1h 14m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f1a58a7bc994f238e6f3d25b793718c : SUCCESS in 19m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db5397084b00467a863170f4fe686fa8 : SUCCESS in 29m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b21119948a9499999cb7df2e1788a81 : SUCCESS in 25m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/73338701e05c4ee4b7d0f9241c6de89c : SUCCESS in 32m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da75bac48924f4a91acb65ae853fa03 : SUCCESS in 12m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/50f66d17807346bfb94cd7c6f1f534b6 : SUCCESS in 1h 08m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dfc723939f2747e4b25504930606dd83 : SUCCESS in 1h 21m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/137826263d99495dad66e7baca3133b9 : SUCCESS in 1h 03m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/48ab6ff863794d61a40f3ffd1823c129 : FAILURE in 36m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2297dc05a31143a8be7244011546bbe3 : SUCCESS in 55m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/94d5172714c64412b8229415bc5e33fd : SUCCESS in 52m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b2ecf345314e41c19201709088ef06e5 : SUCCESS in 1h 24m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/687237936ed64cd4a8c1a8ac5c4e82a7 : SUCCESS in 1h 40m 31s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/f0cd12bfe5764e8f917cbcf842dc2103 : SUCCESS in 2h 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"bd5a50b402cf12264c0fc482f38173ef4b56df4d","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-04-11 06:05:39.000000000","message":"Patch Set 79:\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"03f9ee4919e5c3c95169e1c9cc3d2b57856556b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 06:39:36.000000000","message":"Patch Set 79:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/032e949cd10f4cd180e7f42ff88b37ec\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/320520cab1844e35945232c7c49e7b12 : SUCCESS in 23m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5096d8c41e4b4f2b82eed93ea951a739 : SUCCESS in 31m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5d8d8931c4ba4949885f429feb6ee893 : SUCCESS in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"363f9211b23c6be86975d5d418af741c327580c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 09:05:02.000000000","message":"Patch Set 79: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ebe92414f3740cbbd2e4b1690194358\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d4533eec738243599cee39b6e16b35ff : SUCCESS in 1h 25m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/571395d259914d0e88137fef4ec7d748 : SUCCESS in 1h 03m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f15eaa88da4b44208e5030818654fb9d : SUCCESS in 1h 16m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f1ed478ccb7145a78c543d0ec9155f90 : SUCCESS in 2h 56m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a1c8d2727463497f848fc9554613bb0c : SUCCESS in 30m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1db7245a7f8245ef8475d5a22f0a8b7c : SUCCESS in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29defba4450344b180a5127f3aec13c1 : SUCCESS in 19m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0f6db665508448188d111ec0b5653c1 : SUCCESS in 35m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/effc2a97d25c4602b165e44cadcf8e15 : SUCCESS in 21m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bdd2a995eb93460eb02b1b25d229f9b5 : SUCCESS in 22m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4846859b1ed140759abdcb59c750ee8b : SUCCESS in 9m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd74b02dcfc54e8a9fd8d82aa9324ea2 : SUCCESS in 1h 19m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ecf615342634fb3be74f29c111ba8b4 : SUCCESS in 39m 03s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6e5bbb7c7b74d90ad397424a87a0232 : SUCCESS in 1h 13m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6a0b73d72a34493a9e882c2c012fd68d : SUCCESS in 1h 08m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/540fd8fd8e1e4b96accf387af69aaa01 : SUCCESS in 1h 37m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/00b90d43fcb24c3fb0dbcd6d7bc5ddda : SUCCESS in 46m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c72f56a9f494b64aaea63169082de75 : SUCCESS in 1h 21m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f6daa67e79cf40ce9189c2829bf6219b : SUCCESS in 1h 02m 04s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/7530cbb725f44230ab09519eedd0f622 : SUCCESS in 2h 06m 54s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"5f3567664ba5c48209cdd243914575989ee0c5fb","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-13 15:04:25.000000000","message":"Patch Set 79: Code-Review+1","accounts_in_message":[],"_revision_number":79},{"id":"f2424d8503297359c8040f4f0ec19f99f7fb4ac1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-14 06:52:41.000000000","message":"Patch Set 79: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":79},{"id":"27f18251ba063eb707a1925b5ef6ec9a8ab29097","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-04-14 08:21:32.000000000","message":"Uploaded patch set 80.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":80},{"id":"646d8347d56b0d4c5a412bd8074e695d0b6292ad","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-04-14 08:21:48.000000000","message":"Patch Set 80:\n\n(3 comments)","accounts_in_message":[],"_revision_number":80},{"id":"cc1cbde7c936fe6a201cb4b332b1b247696ab394","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 08:57:17.000000000","message":"Patch Set 80:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb65d8b74ee45e2ad7333ac09b08e5d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e1ac5e00e36041b3902b50924662298b : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f672941984d24649a7157fefea59217a : SUCCESS in 31m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a7131f1805494696a29584c186e8e0a6 : SUCCESS in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"558e2b1a0b562ec3a2116ab277b902be8b4be9a6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-14 09:40:53.000000000","message":"Patch Set 80:\n\n(1 comment)","accounts_in_message":[],"_revision_number":80},{"id":"1788ce44fa75ec3a435ed9b75e15c51727653654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 10:49:35.000000000","message":"Patch Set 80: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95e71b5d4e5148189c1d181e58f112e5\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4f5fae886b654449b6524894e1259b66 : SUCCESS in 1h 21m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01bfbc9eeacc4f6a9792b5e1e3538519 : SUCCESS in 35m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/91253386dd3946bbb04ace67cf3f68a5 : SUCCESS in 1h 26m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/031673475aa149bea77fb648f0a94273 : SUCCESS in 1h 58m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c971ff381304da5bd9eed2ee87c8367 : SUCCESS in 33m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ffb144e88d044a3994422fa1a9b97e0 : SUCCESS in 53m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6b70895ab3456a983b8c702abbff44 : SUCCESS in 27m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3303a10ad3f41ae9fde4fee89a4ae09 : SUCCESS in 34m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d18a513fde564e7c98f4ae9437c48161 : SUCCESS in 18m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2938d53fe2cd40f3972c0413fa476951 : SUCCESS in 36m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcee3b50f5ce47d997c21de71bb83d7d : SUCCESS in 13m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c9f17ae31e5b4e56abd3ffe51f410ac6 : SUCCESS in 1h 17m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/757290c5e4f34886a02962e11899849d : SUCCESS in 1h 06m 32s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/945aa7afcbbd45d3886e1ee012e59785 : SUCCESS in 1h 13m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4e104da02ea4472f9b6cac7585927873 : SUCCESS in 1h 20m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d6002b971423424386264eafe5079406 : SUCCESS in 57m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fed3159118724bdc899dab523e2eaa4a : SUCCESS in 1h 08m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a8660aff9b9041d4b40aafdb1510e077 : SUCCESS in 1h 19m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7829fa76ce7349f7b4564f9e3fd86319 : SUCCESS in 51m 34s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/900a3812671b4ed38b2f1d4d0dde3928 : SUCCESS in 2h 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"175406ae6f0271e15290dc47077985ef47efef64","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-14 15:10:20.000000000","message":"Patch Set 80:\n\n(1 comment)","accounts_in_message":[],"_revision_number":80},{"id":"1ed5bd7dca970f498386535109947fb96565cfac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-04-14 15:37:08.000000000","message":"Uploaded patch set 81.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":81},{"id":"28f9beb1ccab0190ad9bff00562cb74e028f4c8a","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2026-04-14 15:38:53.000000000","message":"Patch Set 80:\n\n(2 comments)","accounts_in_message":[],"_revision_number":80},{"id":"f614f0b773a74e489a043acb7d867d6d396abb53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 17:45:00.000000000","message":"Patch Set 81: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec424dad957043acb4db993fba943f90\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ece0c27d9d4d4b15ad1c7dd88202ec30 : SUCCESS in 1h 38m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b9381f5924f41c3a193a496f2e91fb6 : SUCCESS in 1h 04m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8622f0d4122843158fba6e06c8572324 : SUCCESS in 1h 30m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c5bb845332d944b193fd11dc0304354f : SUCCESS in 1h 59m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e54a6fb1fd1b4091a72e22f62d33bc0e : SUCCESS in 33m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14b3215f9ead4cdb9e03a178b505fa3a : SUCCESS in 39m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e428539793f487db6fd9a07675882a6 : SUCCESS in 28m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e04147fc2a8d404ba5dd8ddb50529d59 : SUCCESS in 42m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3932fc2c38b4ce5a8ebcff5aee515d0 : SUCCESS in 16m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0918e8cafcc8419e9d8b10457f8c93ac : SUCCESS in 29m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/393e821777b447b8bda05778a5de4d45 : SUCCESS in 10m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cbfd0cb07d594b949917ca4379288e3d : SUCCESS in 1h 25m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7881e9492f124d5ca54b9501a42125b7 : SUCCESS in 1h 09m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/14c2d03e01ca4366874d83fd2b12319a : SUCCESS in 1h 34m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b2141c35f10348cd9de258842deb7871 : SUCCESS in 2h 03m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aa7d683d05314fb4bd4a7ac0a99b99d9 : SUCCESS in 1h 46m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/882246ac06bd401fb5c54a1d57ee8408 : SUCCESS in 56m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4e4fbe4dd51f42b3811ff3f4bde5eebc : SUCCESS in 59m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37433422b8af4cd2b969764e922dc387 : SUCCESS in 1h 33m 32s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/7d8f07e4ff4a4c7cbf0978e45a682a92 : FAILURE in 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"4f0fbfcad74542b2dd3eaf1e61efea287910670c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2026-04-14 18:06:01.000000000","message":"Patch Set 82: Published edit on patch set 81.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":82},{"id":"6c155750dbfad4406123369df034bd8a9611095a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 20:20:06.000000000","message":"Patch Set 82: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae2bcfc8d21f4e079342eea73a6f1493\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4634ed1c2aa140ca9fcd39d29ae117b1 : SUCCESS in 1h 16m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7de588b9f964e4fac9081a4f9ac5809 : SUCCESS in 1h 10m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/78fb52d16df64b74864a605fdec1e76c : SUCCESS in 1h 21m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d624677e1fd0410e91d2fe09188f85cf : SUCCESS in 2h 08m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/acbd624d5ed74ecfa1d1330d82210b1d : SUCCESS in 34m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e83cd5d1aea648be8490493d364a40f4 : SUCCESS in 52m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d04eb10e47a438a8559c73d3e79c45c : SUCCESS in 14m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62ed0e12daff4547861e244b18462101 : SUCCESS in 36m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/186af150951242ec85cf59a4081fb184 : SUCCESS in 20m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/af74474ff0774ce3940cc8be1bdaf0a6 : SUCCESS in 27m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d19a33dcf3465e8ff2f67ef1bbc8cb : SUCCESS in 8m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/790a319047d14143ab2b887e6359f1d0 : SUCCESS in 48m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9a2debc46d845729d807134d4ffc8bb : SUCCESS in 1h 02m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c7c74b27f064cd4a0d571b06677eb9a : SUCCESS in 1h 17m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/738bc7be72e249eba338f2f5bffd48b9 : SUCCESS in 1h 20m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6494d84ea5e54033b756d9303677224d : SUCCESS in 1h 02m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d76ccd7223204b76a35c45709e2b1515 : SUCCESS in 1h 08m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bcb7b8f176db461d801b8fbec6f4c338 : SUCCESS in 1h 25m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c2c54bb6e344e49991f9492eadb9fb0 : SUCCESS in 1h 12m 52s\n- neutron-ovn-bgp-tempest-multinode https://zuul.opendev.org/t/openstack/build/285e90a1fd9248899970501177b7d45e : SUCCESS in 1h 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"c7ba1a1711d16847772d8fcd752b5484841be806","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-14 20:50:14.000000000","message":"Patch Set 82: Code-Review+2","accounts_in_message":[],"_revision_number":82},{"id":"68ca6eaceab513bc1e40bd66558a49a5914ad1f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 01:11:30.000000000","message":"Patch Set 82:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15aad67189f54a83b679b1fbef8c10ca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e64f5f719d8840c5ad256c7c360a2543 : SUCCESS in 44m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d9f676724281441888f33ad1c3021d64 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/507606f8150a480e96580467c44b3225 : SUCCESS in 27m 02s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"6bb899397e3ed7fd8d51e0e1b7daf0eb9d8ce41b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-04-16 05:29:59.000000000","message":"Patch Set 82: Review-Priority+1","accounts_in_message":[],"_revision_number":82},{"id":"def42a5a06d94493adf56ab8748c8c258f5a3e23","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-16 11:29:45.000000000","message":"Patch Set 82: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":82},{"id":"c6111609885769c71f18a465212ddd263b62294f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 11:30:32.000000000","message":"Patch Set 82: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":82},{"id":"2adbec54fef3a42bec91ae7638b2a981daf6a928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 13:28:41.000000000","message":"Patch Set 82: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c48321dd59c74c6d9d7573a09472f13f\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf67929818a4462ea0313540f5eeb31d : SUCCESS in 1h 15m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbf0d4a415ef48419292aae59ddaaddd : SUCCESS in 37m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bc4b5f25a5a64aa0a8bada18b254849c : SUCCESS in 1h 30m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f108e81ecab945dd9faae2f6c360f196 : SUCCESS in 1h 57m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0fe5c7b8c994213bf98683fe1dc4a94 : SUCCESS in 20m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb79e82e52df48358944ed495e6f5816 : SUCCESS in 29m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a875b031c8d45048f338a452280c471 : SUCCESS in 18m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07000e60894a47f080c84d55920a317e : SUCCESS in 25m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cdfd36dfdc749cba9ce127611538e98 : SUCCESS in 6m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/417081ae50b6440084c6b2400f65df56 : SUCCESS in 1h 15m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/109cc3bb6bb44c238dddad5f0b5af205 : SUCCESS in 1h 51m 10s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1bbd555fcf4e4cfaaedbd8dbd0d5f6d7 : SUCCESS in 1h 11m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e66053fdf3e493d987c56097027c0a1 : SUCCESS in 1h 14m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3fdd19589b3a4d27b11c856ced6869e5 : SUCCESS in 1h 08m 05s","accounts_in_message":[],"_revision_number":82},{"id":"0a9f558d63aeba4656237766ffae2263c420c2d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 13:28:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":82},{"id":"f63112e3a7c2a77268a434282ab67e3e91b708fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 13:30:58.000000000","message":"Patch Set 82:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddabc9bdf59d4de383f5562a53ac4a3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a306fd3ea4f94948b451fd6910abd60c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":82}],"current_revision_number":82,"current_revision":"b437cc9f5165cbeb69a7dd3cf688dea52b43be53","revisions":{"2ebff51cd867f3e757f8b7633dd3900c673a3819":{"kind":"REWORK","_number":1,"created":"2025-09-24 13:47:34.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/1"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:21.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ebff51cd867f3e757f8b7633dd3900c673a3819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ebff51cd867f3e757f8b7633dd3900c673a3819"}]},"branch":"refs/heads/master"},"a793bbf855ddef530615a62f60c324b97c7a7fbe":{"kind":"REWORK","_number":2,"created":"2025-09-24 13:51:19.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/2"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:49:55.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a793bbf855ddef530615a62f60c324b97c7a7fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a793bbf855ddef530615a62f60c324b97c7a7fbe"}]},"branch":"refs/heads/master"},"b2d833f671f4e404aa1f38d353cdc4f388f36a9e":{"kind":"REWORK","_number":3,"created":"2025-09-24 13:53:35.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/3"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:53:28.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2d833f671f4e404aa1f38d353cdc4f388f36a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2d833f671f4e404aa1f38d353cdc4f388f36a9e"}]},"branch":"refs/heads/master"},"8e85dffb9e9abbb6f9446668eb5acc52cb7ed5d7":{"kind":"REWORK","_number":4,"created":"2025-09-24 13:54:52.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/4"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:54:45.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e85dffb9e9abbb6f9446668eb5acc52cb7ed5d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e85dffb9e9abbb6f9446668eb5acc52cb7ed5d7"}]},"branch":"refs/heads/master"},"d88554263424dd7a0e24cd999471b3c437d61ebd":{"kind":"REWORK","_number":5,"created":"2025-09-24 16:04:27.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/5"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 16:04:20.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d88554263424dd7a0e24cd999471b3c437d61ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d88554263424dd7a0e24cd999471b3c437d61ebd"}]},"branch":"refs/heads/master"},"e064b1b67c47a5c378cf659ad1d674991d0340e1":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-09-24 16:06:28.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/6"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 16:06:16.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e064b1b67c47a5c378cf659ad1d674991d0340e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e064b1b67c47a5c378cf659ad1d674991d0340e1"}]},"branch":"refs/heads/master"},"80cf5ddbc251678606b3e9377bf715c112548e4f":{"kind":"REWORK","_number":7,"created":"2025-09-25 10:10:47.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/7"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-25 10:10:38.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80cf5ddbc251678606b3e9377bf715c112548e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80cf5ddbc251678606b3e9377bf715c112548e4f"}]},"branch":"refs/heads/master"},"ab3a294470a87f3831fc5ab0878ff9b71516074f":{"kind":"REWORK","_number":8,"created":"2025-09-25 11:21:58.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/8"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-25 11:21:51.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab3a294470a87f3831fc5ab0878ff9b71516074f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab3a294470a87f3831fc5ab0878ff9b71516074f"}]},"branch":"refs/heads/master"},"279aa10388e0cf76723d8d2ccf10e3d45c1c7717":{"kind":"REWORK","_number":9,"created":"2025-09-25 14:21:19.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/9"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-25 14:21:12.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/279aa10388e0cf76723d8d2ccf10e3d45c1c7717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/279aa10388e0cf76723d8d2ccf10e3d45c1c7717"}]},"branch":"refs/heads/master"},"a8d2828f2acc84250949619703f87dd738d4eaca":{"kind":"REWORK","_number":10,"created":"2025-09-26 06:16:31.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/10"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-26 06:16:21.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8d2828f2acc84250949619703f87dd738d4eaca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8d2828f2acc84250949619703f87dd738d4eaca"}]},"branch":"refs/heads/master"},"f7cec320fd923ed588d1afed049b431c5c0660e6":{"kind":"REWORK","_number":11,"created":"2025-09-26 12:34:38.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/11"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-26 12:34:32.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7cec320fd923ed588d1afed049b431c5c0660e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7cec320fd923ed588d1afed049b431c5c0660e6"}]},"branch":"refs/heads/master"},"8623a535df591e2f41ddc3a2fcfc355d9879b979":{"kind":"REWORK","_number":12,"created":"2025-09-29 14:13:49.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/12"}}},"commit":{"parents":[{"commit":"783ce1cd4d99ca20fb705d6540fbce5865fe07d7","subject":"Merge \"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/783ce1cd4d99ca20fb705d6540fbce5865fe07d7"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-29 14:13:40.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8623a535df591e2f41ddc3a2fcfc355d9879b979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8623a535df591e2f41ddc3a2fcfc355d9879b979"}]},"branch":"refs/heads/master"},"1908ba1ed3792e780e80d9628f858a4af3c0455a":{"kind":"REWORK","_number":13,"created":"2025-09-29 15:05:41.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/13"}}},"commit":{"parents":[{"commit":"783ce1cd4d99ca20fb705d6540fbce5865fe07d7","subject":"Merge \"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/783ce1cd4d99ca20fb705d6540fbce5865fe07d7"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-29 15:05:33.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1908ba1ed3792e780e80d9628f858a4af3c0455a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1908ba1ed3792e780e80d9628f858a4af3c0455a"}]},"branch":"refs/heads/master"},"9c7814398f1f5317032c9bf56a1fd1959d282c23":{"kind":"REWORK","_number":14,"created":"2025-09-29 15:31:14.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/14"}}},"commit":{"parents":[{"commit":"783ce1cd4d99ca20fb705d6540fbce5865fe07d7","subject":"Merge \"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/783ce1cd4d99ca20fb705d6540fbce5865fe07d7"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-29 15:31:07.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c7814398f1f5317032c9bf56a1fd1959d282c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c7814398f1f5317032c9bf56a1fd1959d282c23"}]},"branch":"refs/heads/master"},"7562a6faea8d8e8e222a1e6f9527538ef40e0db6":{"kind":"REWORK","_number":15,"created":"2025-09-29 15:56:09.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/15"}}},"commit":{"parents":[{"commit":"783ce1cd4d99ca20fb705d6540fbce5865fe07d7","subject":"Merge \"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/783ce1cd4d99ca20fb705d6540fbce5865fe07d7"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-29 15:55:58.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7562a6faea8d8e8e222a1e6f9527538ef40e0db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7562a6faea8d8e8e222a1e6f9527538ef40e0db6"}]},"branch":"refs/heads/master"},"6265bf0f5a6bc5dbcfdeae7e9190fa415971b8d3":{"kind":"REWORK","_number":16,"created":"2025-09-30 08:36:20.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/16"}}},"commit":{"parents":[{"commit":"1d921fdf7bf0431bb082204d23b90dbf9b2fe094","subject":"Merge \"Add a new option to skip loading Trunk extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d921fdf7bf0431bb082204d23b90dbf9b2fe094"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-30 08:36:10.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6265bf0f5a6bc5dbcfdeae7e9190fa415971b8d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6265bf0f5a6bc5dbcfdeae7e9190fa415971b8d3"}]},"branch":"refs/heads/master"},"3b97eb3e162173b7ad7fb74bca12de51ff619b88":{"kind":"REWORK","_number":17,"created":"2025-09-30 15:29:38.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/17"}}},"commit":{"parents":[{"commit":"1d921fdf7bf0431bb082204d23b90dbf9b2fe094","subject":"Merge \"Add a new option to skip loading Trunk extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d921fdf7bf0431bb082204d23b90dbf9b2fe094"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-30 15:29:09.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b97eb3e162173b7ad7fb74bca12de51ff619b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3b97eb3e162173b7ad7fb74bca12de51ff619b88"}]},"branch":"refs/heads/master"},"75fd185fc0891c8859412b045e2639b29b003e3a":{"kind":"REWORK","_number":18,"created":"2025-10-01 15:36:15.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/18"}}},"commit":{"parents":[{"commit":"df4b70b71f60379bdd8b9fbc0d685942787c6bf7","subject":"Merge \"Zuul: do not use USE_PYTHON3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/df4b70b71f60379bdd8b9fbc0d685942787c6bf7"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-10-01 15:36:03.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75fd185fc0891c8859412b045e2639b29b003e3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75fd185fc0891c8859412b045e2639b29b003e3a"}]},"branch":"refs/heads/master"},"adf8e5d984e210b88a05ad43661253a6d16519e0":{"kind":"REWORK","_number":19,"created":"2025-10-06 15:26:07.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/19"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-10-06 15:25:58.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/962727\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adf8e5d984e210b88a05ad43661253a6d16519e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adf8e5d984e210b88a05ad43661253a6d16519e0"}]},"branch":"refs/heads/master"},"125916759cf2f8c201b8a9c8e71e72f355eef0ef":{"kind":"REWORK","_number":20,"created":"2025-10-07 08:03:23.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/20"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-10-07 08:03:15.000000000","tz":120},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/962727\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/125916759cf2f8c201b8a9c8e71e72f355eef0ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/125916759cf2f8c201b8a9c8e71e72f355eef0ef"}]},"branch":"refs/heads/master"},"524865ce51ef71fe0eb633238c1972d02acb1d36":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2025-10-07 17:40:54.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/21"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-07 17:40:54.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/524865ce51ef71fe0eb633238c1972d02acb1d36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/524865ce51ef71fe0eb633238c1972d02acb1d36"}]},"branch":"refs/heads/master","description":"Edit commit message"},"dd25992c47416fdf2018e8614ff27251604e6c08":{"kind":"REWORK","_number":22,"created":"2025-10-07 17:43:29.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/22"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-07 17:43:19.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd25992c47416fdf2018e8614ff27251604e6c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd25992c47416fdf2018e8614ff27251604e6c08"}]},"branch":"refs/heads/master"},"c09e88bbddc0db3a02e4c8b590b22f1835050b07":{"kind":"REWORK","_number":23,"created":"2025-10-07 18:48:59.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/23"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-07 18:48:52.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c09e88bbddc0db3a02e4c8b590b22f1835050b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c09e88bbddc0db3a02e4c8b590b22f1835050b07"}]},"branch":"refs/heads/master"},"5619380ce54cb20544dc4bfb6e00fc6d45dc0a13":{"kind":"REWORK","_number":24,"created":"2025-10-07 19:04:15.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/24"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-07 19:04:09.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5619380ce54cb20544dc4bfb6e00fc6d45dc0a13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5619380ce54cb20544dc4bfb6e00fc6d45dc0a13"}]},"branch":"refs/heads/master"},"6add7bd6b5b7d11c7c4000cf5baa070fc1056b47":{"kind":"REWORK","_number":25,"created":"2025-10-07 20:35:24.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/25"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-07 20:35:24.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6add7bd6b5b7d11c7c4000cf5baa070fc1056b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6add7bd6b5b7d11c7c4000cf5baa070fc1056b47"}]},"branch":"refs/heads/master"},"25504175255582414af7c69c21021d91697f25ab":{"kind":"REWORK","_number":26,"created":"2025-10-08 16:31:27.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/26"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-08 16:31:21.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25504175255582414af7c69c21021d91697f25ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25504175255582414af7c69c21021d91697f25ab"}]},"branch":"refs/heads/master"},"f12e5ae403d246d6b226452869cfefa338d765d0":{"kind":"REWORK","_number":27,"created":"2025-10-08 20:58:58.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/27"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-08 20:58:52.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f12e5ae403d246d6b226452869cfefa338d765d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f12e5ae403d246d6b226452869cfefa338d765d0"}]},"branch":"refs/heads/master"},"16bfaefe1c4dcc6ffe3b91873dff0518320c8764":{"kind":"REWORK","_number":28,"created":"2025-10-09 11:57:23.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/28"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-09 11:57:16.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16bfaefe1c4dcc6ffe3b91873dff0518320c8764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16bfaefe1c4dcc6ffe3b91873dff0518320c8764"}]},"branch":"refs/heads/master"},"c6c511cd9619aadc160be72b81ac5728e03a2f61":{"kind":"REWORK","_number":29,"created":"2025-10-09 15:30:25.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/29"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-09 15:30:19.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6c511cd9619aadc160be72b81ac5728e03a2f61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6c511cd9619aadc160be72b81ac5728e03a2f61"}]},"branch":"refs/heads/master"},"15228e7b078ec70857d65b6e38a1cf5b350af592":{"kind":"REWORK","_number":30,"created":"2025-10-09 18:56:53.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/30"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-09 18:56:43.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15228e7b078ec70857d65b6e38a1cf5b350af592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15228e7b078ec70857d65b6e38a1cf5b350af592"}]},"branch":"refs/heads/master"},"e3b06ab2bf8fc409326870f5c3a3c4785c7743e8":{"kind":"REWORK","_number":31,"created":"2025-10-10 21:22:04.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/31"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-10 21:21:59.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3b06ab2bf8fc409326870f5c3a3c4785c7743e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3b06ab2bf8fc409326870f5c3a3c4785c7743e8"}]},"branch":"refs/heads/master"},"e808865605a07c93105ab92b9e4ad2d2dd965556":{"kind":"REWORK","_number":32,"created":"2025-10-14 19:35:50.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/32"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-14 19:35:42.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e808865605a07c93105ab92b9e4ad2d2dd965556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e808865605a07c93105ab92b9e4ad2d2dd965556"}]},"branch":"refs/heads/master"},"7eebe17a3d70bc8d094afd5e62a6b3f1ef64a89d":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2025-10-27 18:52:43.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/33"}}},"commit":{"parents":[{"commit":"820cce52553a293e52a0b859c5cbabfcd8679b70","subject":"Merge \"[eventlet-removal] Don\u0027t use eventlet in the unit/functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/820cce52553a293e52a0b859c5cbabfcd8679b70"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-27 18:52:32.000000000","tz":-240},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7eebe17a3d70bc8d094afd5e62a6b3f1ef64a89d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7eebe17a3d70bc8d094afd5e62a6b3f1ef64a89d"}]},"branch":"refs/heads/master"},"07e7d17fc660477167651b5e246b1b86e34bd616":{"kind":"REWORK","_number":34,"created":"2025-10-31 11:53:36.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/34"}}},"commit":{"parents":[{"commit":"820cce52553a293e52a0b859c5cbabfcd8679b70","subject":"Merge \"[eventlet-removal] Don\u0027t use eventlet in the unit/functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/820cce52553a293e52a0b859c5cbabfcd8679b70"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-10-31 11:53:36.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07e7d17fc660477167651b5e246b1b86e34bd616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/07e7d17fc660477167651b5e246b1b86e34bd616"}]},"branch":"refs/heads/master"},"bf3cbc75df7cb509a089f3822130436db1f06d8e":{"kind":"REWORK","_number":35,"created":"2025-11-03 16:34:03.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/35"}}},"commit":{"parents":[{"commit":"820cce52553a293e52a0b859c5cbabfcd8679b70","subject":"Merge \"[eventlet-removal] Don\u0027t use eventlet in the unit/functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/820cce52553a293e52a0b859c5cbabfcd8679b70"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-11-03 16:33:56.000000000","tz":-300},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/965947\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf3cbc75df7cb509a089f3822130436db1f06d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf3cbc75df7cb509a089f3822130436db1f06d8e"}]},"branch":"refs/heads/master"},"380d1d25f2aa0bea6226c165bf91cf179c20be5c":{"kind":"REWORK","_number":36,"created":"2025-11-03 17:37:38.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/36"}}},"commit":{"parents":[{"commit":"820cce52553a293e52a0b859c5cbabfcd8679b70","subject":"Merge \"[eventlet-removal] Don\u0027t use eventlet in the unit/functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/820cce52553a293e52a0b859c5cbabfcd8679b70"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-11-03 17:37:32.000000000","tz":-300},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/965947\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/380d1d25f2aa0bea6226c165bf91cf179c20be5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/380d1d25f2aa0bea6226c165bf91cf179c20be5c"}]},"branch":"refs/heads/master"},"262f7293c484bcc893d3dc2e9230400eb04aee46":{"kind":"REWORK","_number":37,"created":"2025-11-03 18:39:35.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/37"}}},"commit":{"parents":[{"commit":"820cce52553a293e52a0b859c5cbabfcd8679b70","subject":"Merge \"[eventlet-removal] Don\u0027t use eventlet in the unit/functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/820cce52553a293e52a0b859c5cbabfcd8679b70"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-11-03 18:39:29.000000000","tz":-300},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/965947\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/262f7293c484bcc893d3dc2e9230400eb04aee46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/262f7293c484bcc893d3dc2e9230400eb04aee46"}]},"branch":"refs/heads/master"},"b4414e14de59b692c021e7ecacfeb321074e3f48":{"kind":"REWORK","_number":38,"created":"2025-11-03 19:23:50.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/38"}}},"commit":{"parents":[{"commit":"820cce52553a293e52a0b859c5cbabfcd8679b70","subject":"Merge \"[eventlet-removal] Don\u0027t use eventlet in the unit/functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/820cce52553a293e52a0b859c5cbabfcd8679b70"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-11-03 19:23:44.000000000","tz":-300},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/965947\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4414e14de59b692c021e7ecacfeb321074e3f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b4414e14de59b692c021e7ecacfeb321074e3f48"}]},"branch":"refs/heads/master"},"3093feb3119c3da7a5aff6603ac3dd40b5e8635b":{"kind":"NO_CODE_CHANGE","_number":39,"created":"2025-11-03 20:06:03.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/39"}}},"commit":{"parents":[{"commit":"820cce52553a293e52a0b859c5cbabfcd8679b70","subject":"Merge \"[eventlet-removal] Don\u0027t use eventlet in the unit/functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/820cce52553a293e52a0b859c5cbabfcd8679b70"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2025-11-03 20:06:03.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3093feb3119c3da7a5aff6603ac3dd40b5e8635b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3093feb3119c3da7a5aff6603ac3dd40b5e8635b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"498c9967b6f15c47486210496e1f6650f79a4469":{"kind":"REWORK","_number":40,"created":"2025-11-04 08:52:45.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/40"}}},"commit":{"parents":[{"commit":"10c6a610ea4301212e12d1ad434d8b22efcd85c6","subject":"Merge \"reno: Update master for unmaintained/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10c6a610ea4301212e12d1ad434d8b22efcd85c6"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-11-04 08:52:32.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498c9967b6f15c47486210496e1f6650f79a4469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/498c9967b6f15c47486210496e1f6650f79a4469"}]},"branch":"refs/heads/master"},"255550ad26705a2a7509202377bc68d06b373866":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2025-12-04 14:03:30.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/41"}}},"commit":{"parents":[{"commit":"a18fa5d3b0c62a125504d9b87658704cb4195eb3","subject":"Merge \"Remove a missing Windows configuration file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a18fa5d3b0c62a125504d9b87658704cb4195eb3"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-12-04 14:03:30.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/255550ad26705a2a7509202377bc68d06b373866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/255550ad26705a2a7509202377bc68d06b373866"}]},"branch":"refs/heads/master","description":"Rebase"},"c8dfaa62d80d5d9f58dd6ebd4efe2f0dec6c55ff":{"kind":"REWORK","_number":42,"created":"2025-12-05 10:53:34.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/42"}}},"commit":{"parents":[{"commit":"4d2c562d2250a02b88524fcddae2066b1b17abe2","subject":"Merge \"Fix ignored metadata options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d2c562d2250a02b88524fcddae2066b1b17abe2"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-12-05 10:53:24.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8dfaa62d80d5d9f58dd6ebd4efe2f0dec6c55ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8dfaa62d80d5d9f58dd6ebd4efe2f0dec6c55ff"}]},"branch":"refs/heads/master"},"eb8cd899170015e633203c3167ba782a5f77e115":{"kind":"REWORK","_number":43,"created":"2026-01-16 12:40:38.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/43"}}},"commit":{"parents":[{"commit":"4bbd427d9f58ac6bc35a2dcdc07584cc5ab415c6","subject":"Merge \"Make sure OVN PG is present before adding a SG rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bbd427d9f58ac6bc35a2dcdc07584cc5ab415c6"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-01-16 12:40:12.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb8cd899170015e633203c3167ba782a5f77e115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb8cd899170015e633203c3167ba782a5f77e115"}]},"branch":"refs/heads/master"},"16fcf8acbf88820c7002caa632f9f2aa37e7e5a0":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2026-01-16 12:44:21.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/44"}}},"commit":{"parents":[{"commit":"4d2c562d2250a02b88524fcddae2066b1b17abe2","subject":"Merge \"Fix ignored metadata options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d2c562d2250a02b88524fcddae2066b1b17abe2"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-01-16 12:43:56.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16fcf8acbf88820c7002caa632f9f2aa37e7e5a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16fcf8acbf88820c7002caa632f9f2aa37e7e5a0"}]},"branch":"refs/heads/master"},"607e389134b59c889235a676c27781dd6356b20e":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2026-01-16 12:47:23.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/45"}}},"commit":{"parents":[{"commit":"4bbd427d9f58ac6bc35a2dcdc07584cc5ab415c6","subject":"Merge \"Make sure OVN PG is present before adding a SG rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bbd427d9f58ac6bc35a2dcdc07584cc5ab415c6"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-01-16 12:47:13.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/607e389134b59c889235a676c27781dd6356b20e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/607e389134b59c889235a676c27781dd6356b20e"}]},"branch":"refs/heads/master"},"eed1df4e17eade82424140eef20843d0db83c811":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2026-01-23 09:00:39.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/46"}}},"commit":{"parents":[{"commit":"264477f8a5ca22179badadd772509207ca667485","subject":"Merge \"Add tap-mirror-both-direction as supported extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/264477f8a5ca22179badadd772509207ca667485"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-01-23 09:00:24.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eed1df4e17eade82424140eef20843d0db83c811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eed1df4e17eade82424140eef20843d0db83c811"}]},"branch":"refs/heads/master"},"65fefd0fa7609bf4baab04da4d8ae8c9faf2402c":{"kind":"REWORK","_number":47,"created":"2026-01-23 10:47:23.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/47"}}},"commit":{"parents":[{"commit":"264477f8a5ca22179badadd772509207ca667485","subject":"Merge \"Add tap-mirror-both-direction as supported extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/264477f8a5ca22179badadd772509207ca667485"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-01-23 10:44:03.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65fefd0fa7609bf4baab04da4d8ae8c9faf2402c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65fefd0fa7609bf4baab04da4d8ae8c9faf2402c"}]},"branch":"refs/heads/master"},"61d76960a0e6f7dcb390bbb6adc239e8cf1e0bc3":{"kind":"REWORK","_number":48,"created":"2026-01-23 14:15:51.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/48"}}},"commit":{"parents":[{"commit":"1ded5f7b79422b3999e4595d64d6aa32977269f7","subject":"Merge \"Don\u0027t delete from OVN DB ACLs not managed by Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ded5f7b79422b3999e4595d64d6aa32977269f7"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-01-23 14:15:32.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61d76960a0e6f7dcb390bbb6adc239e8cf1e0bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61d76960a0e6f7dcb390bbb6adc239e8cf1e0bc3"}]},"branch":"refs/heads/master"},"5fd284b99d473738d2a0bcd11c2c4df471658edf":{"kind":"REWORK","_number":49,"created":"2026-01-23 15:28:13.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/49"}}},"commit":{"parents":[{"commit":"1ded5f7b79422b3999e4595d64d6aa32977269f7","subject":"Merge \"Don\u0027t delete from OVN DB ACLs not managed by Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ded5f7b79422b3999e4595d64d6aa32977269f7"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-01-23 15:28:03.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fd284b99d473738d2a0bcd11c2c4df471658edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fd284b99d473738d2a0bcd11c2c4df471658edf"}]},"branch":"refs/heads/master"},"b978406cd887394d0029f98985e5771c19861b62":{"kind":"REWORK","_number":50,"created":"2026-01-26 08:54:45.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/50"}}},"commit":{"parents":[{"commit":"1ded5f7b79422b3999e4595d64d6aa32977269f7","subject":"Merge \"Don\u0027t delete from OVN DB ACLs not managed by Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ded5f7b79422b3999e4595d64d6aa32977269f7"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-01-26 08:54:31.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b978406cd887394d0029f98985e5771c19861b62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b978406cd887394d0029f98985e5771c19861b62"}]},"branch":"refs/heads/master"},"3c63003ac566423435e56326025204fac38880a3":{"kind":"REWORK","_number":51,"created":"2026-02-03 07:35:07.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/51"}}},"commit":{"parents":[{"commit":"6e32b2cd36452ac28c55fd18ec5bbc4ca4587861","subject":"Merge \"Optimize ``get_subnet[s]_dhcp_options`` methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e32b2cd36452ac28c55fd18ec5bbc4ca4587861"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-02-03 07:34:54.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c63003ac566423435e56326025204fac38880a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c63003ac566423435e56326025204fac38880a3"}]},"branch":"refs/heads/master"},"5a81a8c637f1458f9037299e3a2c6968fe9da556":{"kind":"REWORK","_number":52,"created":"2026-02-03 19:49:19.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/52"}}},"commit":{"parents":[{"commit":"6e32b2cd36452ac28c55fd18ec5bbc4ca4587861","subject":"Merge \"Optimize ``get_subnet[s]_dhcp_options`` methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e32b2cd36452ac28c55fd18ec5bbc4ca4587861"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-03 19:49:13.000000000","tz":-300},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a81a8c637f1458f9037299e3a2c6968fe9da556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a81a8c637f1458f9037299e3a2c6968fe9da556"}]},"branch":"refs/heads/master"},"b3e16cbaf366327825854a9d5185a3c7c218ef4e":{"kind":"REWORK","_number":53,"created":"2026-02-04 11:48:10.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/53"}}},"commit":{"parents":[{"commit":"6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe","subject":"Merge \"Add information to network ip availabilities\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-02-04 11:47:57.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3e16cbaf366327825854a9d5185a3c7c218ef4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b3e16cbaf366327825854a9d5185a3c7c218ef4e"}]},"branch":"refs/heads/master"},"478e938f49ea852843b2a745ec83427b4270eb85":{"kind":"NO_CODE_CHANGE","_number":54,"created":"2026-02-04 12:01:11.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/54"}}},"commit":{"parents":[{"commit":"6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe","subject":"Merge \"Add information to network ip availabilities\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-02-04 12:00:53.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975570\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/478e938f49ea852843b2a745ec83427b4270eb85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/478e938f49ea852843b2a745ec83427b4270eb85"}]},"branch":"refs/heads/master"},"9b53e4c8ffc0c1c249df1dbd9923acd065077449":{"kind":"NO_CODE_CHANGE","_number":55,"created":"2026-02-05 13:40:48.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/55"}}},"commit":{"parents":[{"commit":"6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe","subject":"Merge \"Add information to network ip availabilities\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-02-05 13:40:48.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975570\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975674\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b53e4c8ffc0c1c249df1dbd9923acd065077449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b53e4c8ffc0c1c249df1dbd9923acd065077449"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d66e883921da069b2015ad6c4a1b1d688f120038":{"kind":"REWORK","_number":56,"created":"2026-02-09 18:39:12.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/56"}}},"commit":{"parents":[{"commit":"6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe","subject":"Merge \"Add information to network ip availabilities\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0e6a5bc91ed3c91c2cbec75d4ea2e06ca89dfe"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-09 18:39:05.000000000","tz":-300},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975570\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975674\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d66e883921da069b2015ad6c4a1b1d688f120038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d66e883921da069b2015ad6c4a1b1d688f120038"}]},"branch":"refs/heads/master"},"97b2d76ea1d8861eee369ca57a85381c9ec7b3d8":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":57,"created":"2026-02-09 21:09:08.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/57"}}},"commit":{"parents":[{"commit":"8ae94f620ddd722571bacd59ef8508e0290cd195","subject":"bgp: Implement connecting BGP and Neutron world","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ae94f620ddd722571bacd59ef8508e0290cd195"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-09 21:08:50.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97b2d76ea1d8861eee369ca57a85381c9ec7b3d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/97b2d76ea1d8861eee369ca57a85381c9ec7b3d8"}]},"branch":"refs/heads/master"},"25ad18f3f987e69ba1f6f1a53621077e3d5468e9":{"kind":"REWORK","_number":58,"created":"2026-02-10 15:54:49.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/58"}}},"commit":{"parents":[{"commit":"e1cad6c8d21409ae34f6f21dcc7b1ae118d2ce98","subject":"bgp: Implement connecting BGP and Neutron world","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1cad6c8d21409ae34f6f21dcc7b1ae118d2ce98"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 15:54:24.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25ad18f3f987e69ba1f6f1a53621077e3d5468e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/25ad18f3f987e69ba1f6f1a53621077e3d5468e9"}]},"branch":"refs/heads/master"},"a517af5f04af05a92a6dab8c23db0fa571e2e005":{"kind":"REWORK","_number":59,"created":"2026-02-10 17:12:48.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/59"}}},"commit":{"parents":[{"commit":"e1cad6c8d21409ae34f6f21dcc7b1ae118d2ce98","subject":"bgp: Implement connecting BGP and Neutron world","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1cad6c8d21409ae34f6f21dcc7b1ae118d2ce98"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-10 17:10:14.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a517af5f04af05a92a6dab8c23db0fa571e2e005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a517af5f04af05a92a6dab8c23db0fa571e2e005"}]},"branch":"refs/heads/master"},"fab67395101958fb55f789cba8c9372bfe7e590c":{"kind":"TRIVIAL_REBASE","_number":60,"created":"2026-02-16 11:12:05.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"real_uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/60"}}},"commit":{"parents":[{"commit":"3301651bf05f1711a5f2029a4b9a791313adf53c","subject":"Add neutron-id to DHCP_Options entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3301651bf05f1711a5f2029a4b9a791313adf53c"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-16 11:12:05.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fab67395101958fb55f789cba8c9372bfe7e590c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fab67395101958fb55f789cba8c9372bfe7e590c"}]},"branch":"refs/heads/master","description":"Rebase"},"5d3ef8826f43b28cc74f80d155fafcf92c0def61":{"kind":"NO_CODE_CHANGE","_number":61,"created":"2026-02-16 13:57:47.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/61"}}},"commit":{"parents":[{"commit":"3301651bf05f1711a5f2029a4b9a791313adf53c","subject":"Add neutron-id to DHCP_Options entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3301651bf05f1711a5f2029a4b9a791313adf53c"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-02-16 13:57:47.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d3ef8826f43b28cc74f80d155fafcf92c0def61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d3ef8826f43b28cc74f80d155fafcf92c0def61"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a3035b5c13f29799abab6dfb03c8fbf417c90231":{"kind":"REWORK","_number":62,"created":"2026-02-16 15:46:59.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/62"}}},"commit":{"parents":[{"commit":"3301651bf05f1711a5f2029a4b9a791313adf53c","subject":"Add neutron-id to DHCP_Options entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3301651bf05f1711a5f2029a4b9a791313adf53c"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-02-16 15:45:58.000000000","tz":60},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/963223\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3035b5c13f29799abab6dfb03c8fbf417c90231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a3035b5c13f29799abab6dfb03c8fbf417c90231"}]},"branch":"refs/heads/master"},"3d15c6a836d3acc142c8cf3dcf7e5b4c8e0fcd5e":{"kind":"NO_CODE_CHANGE","_number":63,"created":"2026-02-17 23:44:59.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/63"}}},"commit":{"parents":[{"commit":"3301651bf05f1711a5f2029a4b9a791313adf53c","subject":"Add neutron-id to DHCP_Options entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3301651bf05f1711a5f2029a4b9a791313adf53c"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-17 23:44:59.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/977173\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d15c6a836d3acc142c8cf3dcf7e5b4c8e0fcd5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3d15c6a836d3acc142c8cf3dcf7e5b4c8e0fcd5e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"be07a5305dd449ce60cc6800d841cd323d1e4174":{"kind":"TRIVIAL_REBASE","_number":64,"created":"2026-02-19 13:59:51.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"real_uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/64"}}},"commit":{"parents":[{"commit":"7a666213cc06df5735d11b2c534ff1068072f7b4","subject":"Merge \"bgp: Small fixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a666213cc06df5735d11b2c534ff1068072f7b4"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-19 13:59:51.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/977173\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be07a5305dd449ce60cc6800d841cd323d1e4174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be07a5305dd449ce60cc6800d841cd323d1e4174"}]},"branch":"refs/heads/master","description":"Rebase"},"80e84ea6ac89eb68fc8d2aa17cfc9c93b6c0c37a":{"kind":"REWORK","_number":65,"created":"2026-02-19 21:27:41.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/65"}}},"commit":{"parents":[{"commit":"5c9d33b39ea71f217918262c3306ebd121cb2541","subject":"bgp: Connect BGP and Neutron worlds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c9d33b39ea71f217918262c3306ebd121cb2541"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-19 21:27:15.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/977173\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80e84ea6ac89eb68fc8d2aa17cfc9c93b6c0c37a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80e84ea6ac89eb68fc8d2aa17cfc9c93b6c0c37a"}]},"branch":"refs/heads/master"},"28988c4e41058a693d162cf78b74783cb4af68ea":{"kind":"TRIVIAL_REBASE","_number":66,"created":"2026-02-24 15:38:34.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"real_uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/66"}}},"commit":{"parents":[{"commit":"422c76f6d871d7d20c68a299bac3015ebaafec77","subject":"bgp: Connect BGP and Neutron worlds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/422c76f6d871d7d20c68a299bac3015ebaafec77"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-24 15:38:34.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/975567\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/977173\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28988c4e41058a693d162cf78b74783cb4af68ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28988c4e41058a693d162cf78b74783cb4af68ea"}]},"branch":"refs/heads/master","description":"Rebase"},"8d8818b2f8f353925a0b3aa1f757e906a884643d":{"kind":"NO_CODE_CHANGE","_number":67,"created":"2026-02-24 15:39:24.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/67"}}},"commit":{"parents":[{"commit":"422c76f6d871d7d20c68a299bac3015ebaafec77","subject":"bgp: Connect BGP and Neutron worlds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/422c76f6d871d7d20c68a299bac3015ebaafec77"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-02-24 15:39:24.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/977173\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d8818b2f8f353925a0b3aa1f757e906a884643d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d8818b2f8f353925a0b3aa1f757e906a884643d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"30a807a9e6b8296f0294df53d5b15afb44737128":{"kind":"REWORK","_number":68,"created":"2026-02-25 23:48:35.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/68"}}},"commit":{"parents":[{"commit":"422c76f6d871d7d20c68a299bac3015ebaafec77","subject":"bgp: Connect BGP and Neutron worlds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/422c76f6d871d7d20c68a299bac3015ebaafec77"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-25 23:48:25.000000000","tz":-300},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/977173\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30a807a9e6b8296f0294df53d5b15afb44737128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/30a807a9e6b8296f0294df53d5b15afb44737128"}]},"branch":"refs/heads/master"},"7e7773cb898ad8c9dccd003e88417331dbc855ba":{"kind":"TRIVIAL_REBASE","_number":69,"created":"2026-02-26 23:28:48.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/69"}}},"commit":{"parents":[{"commit":"4683dcec5375cb702dd0c584adede86bf5478aae","subject":"bgp: Connect provider switch when created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4683dcec5375cb702dd0c584adede86bf5478aae"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-26 23:28:33.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/977173\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e7773cb898ad8c9dccd003e88417331dbc855ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e7773cb898ad8c9dccd003e88417331dbc855ba"}]},"branch":"refs/heads/master"},"0d69d88c15d29c2859bceed09757dbca1adc4f55":{"kind":"TRIVIAL_REBASE","_number":70,"created":"2026-02-27 15:33:25.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/70"}}},"commit":{"parents":[{"commit":"56c0c46213c1893f3629ecac80e9cf85fe7b3721","subject":"bgp: Connect provider switch when created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56c0c46213c1893f3629ecac80e9cf85fe7b3721"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-02-27 15:33:05.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/977173\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d69d88c15d29c2859bceed09757dbca1adc4f55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d69d88c15d29c2859bceed09757dbca1adc4f55"}]},"branch":"refs/heads/master"},"fc27bb131a1de22e4b0bb2ba7d28eea47225d47e":{"kind":"TRIVIAL_REBASE","_number":71,"created":"2026-03-10 15:07:07.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"real_uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/71"}}},"commit":{"parents":[{"commit":"b312e1b61d447c4fd71e509348d8fadd4d311f00","subject":"bgp: Implement BGP resources deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b312e1b61d447c4fd71e509348d8fadd4d311f00"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-10 15:07:07.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/977173\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc27bb131a1de22e4b0bb2ba7d28eea47225d47e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc27bb131a1de22e4b0bb2ba7d28eea47225d47e"}]},"branch":"refs/heads/master","description":"Rebase"},"aaadfb69c96bd1f547a1361944e327eead4d3a07":{"kind":"NO_CODE_CHANGE","_number":72,"created":"2026-03-10 15:07:33.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/72"}}},"commit":{"parents":[{"commit":"b312e1b61d447c4fd71e509348d8fadd4d311f00","subject":"bgp: Implement BGP resources deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b312e1b61d447c4fd71e509348d8fadd4d311f00"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-10 15:07:33.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aaadfb69c96bd1f547a1361944e327eead4d3a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aaadfb69c96bd1f547a1361944e327eead4d3a07"}]},"branch":"refs/heads/master","description":"Edit commit message"},"949820e0c18de66604524eeea02ef12f2a76f34f":{"kind":"TRIVIAL_REBASE","_number":73,"created":"2026-03-10 17:31:52.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/73"}}},"commit":{"parents":[{"commit":"80445514fcb4a8fa51f7fa07060b423b89fbb6b6","subject":"bgp: Implement BGP resources deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80445514fcb4a8fa51f7fa07060b423b89fbb6b6"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-10 17:31:52.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/949820e0c18de66604524eeea02ef12f2a76f34f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/949820e0c18de66604524eeea02ef12f2a76f34f"}]},"branch":"refs/heads/master","description":"Rebase"},"48132f1a591eb143c35524b363ddf845aae0cea5":{"kind":"TRIVIAL_REBASE","_number":74,"created":"2026-03-11 09:36:15.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/74"}}},"commit":{"parents":[{"commit":"4cc323bfbf15300d22bf728c9d1a22f073c110ca","subject":"bgp: Implement BGP resources deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cc323bfbf15300d22bf728c9d1a22f073c110ca"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-11 09:36:15.000000000","tz":0},"subject":"WIP: new job tempest-multinode-with-bgp","message":"WIP: new job tempest-multinode-with-bgp\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48132f1a591eb143c35524b363ddf845aae0cea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48132f1a591eb143c35524b363ddf845aae0cea5"}]},"branch":"refs/heads/master","description":"Rebase"},"9c7dcab2bed6f6d5aeb660fc4c6086a91fe87897":{"kind":"REWORK","_number":75,"created":"2026-03-11 14:51:16.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/75"}}},"commit":{"parents":[{"commit":"51e7110c7d1072082a9e7bd9b509dc2388401763","subject":"bgp: Connect provider switch when created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51e7110c7d1072082a9e7bd9b509dc2388401763"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-11 14:47:11.000000000","tz":60},"subject":"new job tempest-multinode-with-bgp","message":"new job tempest-multinode-with-bgp\n\nNew tempest job that deploys:\n- a controller where openstack runs - bgp is configured and frr is\n  enabled to advertise routes to openstack VM instances to the peer\n  leafs - ECMP is used to distribute traffic towards the two leafs\n- two leafs connected to the controller, with frr running and learning\n  BGP routes to VM instances\n- a spine connected to the two leafs, running frr as well to learn BGP\n  routes - tempest runs from the spine, so receiving BGP routes is\n  necessary for the tests to pass\n\nOnly neutron_tempest_plugin.scenario tests executed for the moment.\nVoting set to false until job is stable.\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c7dcab2bed6f6d5aeb660fc4c6086a91fe87897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c7dcab2bed6f6d5aeb660fc4c6086a91fe87897"}]},"branch":"refs/heads/master"},"231f9fbbae4dda44198332237951775fce1065fc":{"kind":"REWORK","_number":76,"created":"2026-03-12 22:02:23.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/76"}}},"commit":{"parents":[{"commit":"51e7110c7d1072082a9e7bd9b509dc2388401763","subject":"bgp: Connect provider switch when created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51e7110c7d1072082a9e7bd9b509dc2388401763"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-03-12 22:02:16.000000000","tz":-240},"subject":"new job tempest-multinode-with-bgp","message":"new job tempest-multinode-with-bgp\n\nNew tempest job that deploys:\n- a controller where openstack runs - bgp is configured and frr is\n  enabled to advertise routes to openstack VM instances to the peer\n  leafs - ECMP is used to distribute traffic towards the two leafs\n- two leafs connected to the controller, with frr running and learning\n  BGP routes to VM instances\n- a spine connected to the two leafs, running frr as well to learn BGP\n  routes - tempest runs from the spine, so receiving BGP routes is\n  necessary for the tests to pass\n\nOnly neutron_tempest_plugin.scenario tests executed for the moment.\nVoting set to false until job is stable.\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/231f9fbbae4dda44198332237951775fce1065fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/231f9fbbae4dda44198332237951775fce1065fc"}]},"branch":"refs/heads/master"},"24c34a071924f4c5b5060423c938638d411099b5":{"kind":"REWORK","_number":77,"created":"2026-03-16 10:40:50.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/77"}}},"commit":{"parents":[{"commit":"0e467760de055355524dee97ecd9af5850262308","subject":"Merge \"[OVN] Specify required mechanism drivers for the NB and SB db sync plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e467760de055355524dee97ecd9af5850262308"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-16 10:40:09.000000000","tz":60},"subject":"new job tempest-multinode-with-bgp","message":"new job tempest-multinode-with-bgp\n\nNew tempest job that deploys:\n- a controller where openstack runs - bgp is configured and frr is\n  enabled to advertise routes to openstack VM instances to the peer\n  leafs - ECMP is used to distribute traffic towards the two leafs\n- two leafs connected to the controller, with frr running and learning\n  BGP routes to VM instances\n- a spine connected to the two leafs, running frr as well to learn BGP\n  routes - tempest runs from the spine, so receiving BGP routes is\n  necessary for the tests to pass\n\nOnly neutron_tempest_plugin.scenario tests executed for the moment.\nVoting set to false until job is stable.\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c34a071924f4c5b5060423c938638d411099b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c34a071924f4c5b5060423c938638d411099b5"}]},"branch":"refs/heads/master"},"c9546439eae28065e015b459fae2ce784906a52b":{"kind":"REWORK","_number":78,"created":"2026-03-16 10:43:16.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/78"}}},"commit":{"parents":[{"commit":"0e467760de055355524dee97ecd9af5850262308","subject":"Merge \"[OVN] Specify required mechanism drivers for the NB and SB db sync plugins\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e467760de055355524dee97ecd9af5850262308"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-03-16 10:43:09.000000000","tz":60},"subject":"new job tempest-multinode-with-bgp","message":"new job tempest-multinode-with-bgp\n\nNew tempest job that deploys:\n- a controller where openstack runs - bgp is configured and frr is\n  enabled to advertise routes to openstack VM instances to the peer\n  leafs - ECMP is used to distribute traffic towards the two leafs\n- two leafs connected to the controller, with frr running and learning\n  BGP routes to VM instances\n- a spine connected to the two leafs, running frr as well to learn BGP\n  routes - tempest runs from the spine, so receiving BGP routes is\n  necessary for the tests to pass\n\nOnly neutron_tempest_plugin.scenario tests executed for the moment.\nVoting set to false until job is stable.\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9546439eae28065e015b459fae2ce784906a52b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9546439eae28065e015b459fae2ce784906a52b"}]},"branch":"refs/heads/master"},"915a573256a3120c603b0ed6b36c007451f257ee":{"kind":"TRIVIAL_REBASE","_number":79,"created":"2026-04-10 12:53:06.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/79"}}},"commit":{"parents":[{"commit":"2f6f25a5bb9d87052679cebefb53e5d028d4742c","subject":"Merge \"Use constant for ha_failover_strategy value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2f6f25a5bb9d87052679cebefb53e5d028d4742c"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-04-10 12:53:06.000000000","tz":0},"subject":"new job tempest-multinode-with-bgp","message":"new job tempest-multinode-with-bgp\n\nNew tempest job that deploys:\n- a controller where openstack runs - bgp is configured and frr is\n  enabled to advertise routes to openstack VM instances to the peer\n  leafs - ECMP is used to distribute traffic towards the two leafs\n- two leafs connected to the controller, with frr running and learning\n  BGP routes to VM instances\n- a spine connected to the two leafs, running frr as well to learn BGP\n  routes - tempest runs from the spine, so receiving BGP routes is\n  necessary for the tests to pass\n\nOnly neutron_tempest_plugin.scenario tests executed for the moment.\nVoting set to false until job is stable.\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/915a573256a3120c603b0ed6b36c007451f257ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/915a573256a3120c603b0ed6b36c007451f257ee"}]},"branch":"refs/heads/master","description":"Rebase"},"569a1cdfe40c142ba89509307a89ebb8659744c3":{"kind":"REWORK","_number":80,"created":"2026-04-14 08:21:32.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/80"}}},"commit":{"parents":[{"commit":"6c6136eb270263bb8b7251edbbde45e4e1e65119","subject":"Merge \"Ignore designate NotFound errors during PTR delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c6136eb270263bb8b7251edbbde45e4e1e65119"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-04-14 07:46:57.000000000","tz":120},"subject":"new job tempest-multinode-with-bgp","message":"new job tempest-multinode-with-bgp\n\nNew tempest job that deploys:\n- a controller where openstack runs - bgp is configured and frr is\n  enabled to advertise routes to openstack VM instances to the peer\n  leafs - ECMP is used to distribute traffic towards the two leafs\n- two leafs connected to the controller, with frr running and learning\n  BGP routes to VM instances\n- a spine connected to the two leafs, running frr as well to learn BGP\n  routes - tempest runs from the spine, so receiving BGP routes is\n  necessary for the tests to pass\n\nIt includes the following tests:\n- tempest.scenario.test_server_basic_ops.TestServerBasicOps\n- neutron_tempest_plugin.scenario (some test are excluded because they\n  are not supported yet with bgp).\n\nVoting set to false until job is stable.\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/569a1cdfe40c142ba89509307a89ebb8659744c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/569a1cdfe40c142ba89509307a89ebb8659744c3"}]},"branch":"refs/heads/master"},"f2eafb1ada311cc774d170c99f4cd50ae4b51bcc":{"kind":"REWORK","_number":81,"created":"2026-04-14 15:37:08.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/88/962188/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/81"}}},"commit":{"parents":[{"commit":"6c6136eb270263bb8b7251edbbde45e4e1e65119","subject":"Merge \"Ignore designate NotFound errors during PTR delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c6136eb270263bb8b7251edbbde45e4e1e65119"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2026-04-14 15:36:48.000000000","tz":120},"subject":"new job tempest-multinode-with-bgp","message":"new job tempest-multinode-with-bgp\n\nNew tempest job that deploys:\n- a controller where openstack runs - bgp is configured and frr is\n  enabled to advertise routes to openstack VM instances to the peer\n  leafs - ECMP is used to distribute traffic towards the two leafs\n- two leafs connected to the controller, with frr running and learning\n  BGP routes to VM instances\n- a spine connected to the two leafs, running frr as well to learn BGP\n  routes - tempest runs from the spine, so receiving BGP routes is\n  necessary for the tests to pass\n\nIt includes the following tests:\n- tempest.scenario.test_server_basic_ops.TestServerBasicOps\n- tempest.scenario.test_network_v6\n- neutron_tempest_plugin.scenario (some test are excluded because they\n  are not supported yet with bgp).\n\nVoting set to false until job is stable.\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2eafb1ada311cc774d170c99f4cd50ae4b51bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2eafb1ada311cc774d170c99f4cd50ae4b51bcc"}]},"branch":"refs/heads/master"},"b437cc9f5165cbeb69a7dd3cf688dea52b43be53":{"kind":"REWORK","_number":82,"created":"2026-04-14 18:06:01.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/88/962188/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/962188/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/962188/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/962188/82"}}},"commit":{"parents":[{"commit":"6c6136eb270263bb8b7251edbbde45e4e1e65119","subject":"Merge \"Ignore designate NotFound errors during PTR delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c6136eb270263bb8b7251edbbde45e4e1e65119"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-09-24 13:47:01.000000000","tz":120},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2026-04-14 18:06:00.000000000","tz":0},"subject":"new job tempest-multinode-with-bgp","message":"new job tempest-multinode-with-bgp\n\nNew tempest job that deploys:\n- a controller where openstack runs - bgp is configured and frr is\n  enabled to advertise routes to openstack VM instances to the peer\n  leafs - ECMP is used to distribute traffic towards the two leafs\n- two leafs connected to the controller, with frr running and learning\n  BGP routes to VM instances\n- a spine connected to the two leafs, running frr as well to learn BGP\n  routes - tempest runs from the spine, so receiving BGP routes is\n  necessary for the tests to pass\n\nIt includes the following tests:\n- tempest.scenario.test_server_basic_ops.TestServerBasicOps\n- tempest.scenario.test_network_v6\n- neutron_tempest_plugin.scenario (some test are excluded because they\n  are not supported yet with bgp).\n\nVoting set to false until job is stable.\n\nOSPRH-19783\n\nRelated-Bug: #2111276\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/963199\n\nCo-Authored-By: Jakub Libosvar \u003cjlibosva@redhat.com\u003e\n\nChange-Id: Iacbd0d59122e820ff9eb6e1c3c48cd17b3fe57c5\nSigned-off-by: Eduardo Olivares \u003ceolivare@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b437cc9f5165cbeb69a7dd3cf688dea52b43be53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b437cc9f5165cbeb69a7dd3cf688dea52b43be53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
