)]}'
{"id":"openstack%2Fneutron~962191","triplet_id":"openstack%2Fneutron~master~I3346c2a16f69c3b16d1693db42335b133a721bf9","project":"openstack/neutron","branch":"master","topic":"bug/2125553","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-06 10:56:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3346c2a16f69c3b16d1693db42335b133a721bf9","subject":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``","status":"MERGED","created":"2025-09-24 14:33:56.000000000","updated":"2025-10-08 13:27:33.000000000","submitted":"2025-10-06 10:56:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962191-bug/2125553","meta_rev_id":"5b53cf7b199fd0e7993cb3d0b28c42cb62d98c03","_number":962191,"virtual_id_number":962191,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-10-02 16:57:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-06 10:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-06 08:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-10-02 16:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-06 08:14:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-24 15:12:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-24 15:21:27.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-24 16:56:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 16:59:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-10-06 08:14:37.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"6f1e4c58d55fa8c82871bca1efa29fd1dd997de3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-24 14:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f668a7dc59ba5bbf8859040a6a8613f72e318fbf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-24 15:21:27.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/91/962191/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"00507c5d2f4b16e7521f7fed45633a0c618931a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 16:56:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e90e0e89edb4c749429fd8097f59acc\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0ca405275825437485881dee03fccfd3 : SUCCESS in 2h 22m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a2844573588c47eb86b4182041103a2f : SUCCESS in 34m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30a2555877534ce4a0f4d57dd030e30a : SUCCESS in 47m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31e8b9de313c45fe80b2429679407d0f : SUCCESS in 7m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0591c3b57b3a49d3a3edb9409be0a72a : SUCCESS in 23m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79bf8bdfcbf24db0b9b98d6207a3ada2 : SUCCESS in 20m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d65283d01d874aad87f1597d070b2d93 : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9bb89605dc441e9a3d9dc154fa7dc3e : SUCCESS in 7m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c9bb5f057939497e89fda814bac160dd : SUCCESS in 1h 51m 01s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1aec45f78ea344af84e6c1dec20717e9 : SUCCESS in 42m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dbf5db7c41ea481cb8c5554af47e2d73 : SUCCESS in 56m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad0b249b1e8646849a132c4b3f5a900d : FAILURE in 39m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf87175bb23241b7a1c1e4330d923f96 : SUCCESS in 1h 07m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ee852f6b74deafada9b90b774b5af7508d8fe3d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 17:41:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a93567129f5d4d3e96e4659ba6fe6409\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e1ebe16d5364e1c9444406aa7c57d5d : SUCCESS in 40m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d856b24573a542e3b03b02846efecf11 : SUCCESS in 23m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3c2efb3621a43729657749056adecb0 : SUCCESS in 51m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a32070bdf2cb648f42fbd6edd3189c7b9aa3676","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-25 10:17:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"788a9af198db1b1d7e2d7fa11f3609c806715f18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 11:58:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56d92491ba2145e882d6bc3f871fe2dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9f164558ef3740fcb1b9c6faab1a0cc7 : SUCCESS in 1h 04m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69ca7fba091f4f6c9fb672ac6ce0ee66 : SUCCESS in 53m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f2d46aac80dc44539a81cf21f0daf368 : SUCCESS in 1h 01m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a3883ee650abd2797b1ff08517399fc822c2a88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 12:16:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a82d76d9dac4cf4a3897ed5fd42facf\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cf570df26ab94715aa96bac66de022f4 : SUCCESS in 1h 58m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d73221c3a7e04189be3c7fe954944f6f : SUCCESS in 33m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52f0cdb386ce45fcb566c0e320344dea : SUCCESS in 28m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9db7d79f3354d508c1b7d9f34af1bf4 : SUCCESS in 14m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3147d10e540247758ed4ee004ca4caf9 : SUCCESS in 17m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee21333d8a2b43f48a0bee6c1310de30 : SUCCESS in 14m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/442d3314a28944a38b74bc00ac8e6316 : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4692f084a69a403d806f358399f3ebc7 : SUCCESS in 12m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cf781aec83f544b59de75dcc584c4f10 : SUCCESS in 1h 39m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4847d06cafa1472a96c68c99e973eab1 : SUCCESS in 35m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8080bbb6133349d99ae4d1ecf95a78ab : SUCCESS in 58m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c8840f6cdd74ecca145b2640ed3eecc : SUCCESS in 1h 23m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/05f68ab840424411a508da225dc7c880 : SUCCESS in 1h 23m 46s","accounts_in_message":[],"_revision_number":1},{"id":"47f24d6571f80bcbd30b5c6fb4fbf0f777c30b96","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-25 13:22:14.000000000","message":"Patch Set 1:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/91/962191/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7fa940c4377fc8f42e84f589b6cb2a94757d3737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-26 20:43:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"93fc2f8a72b25ea8cc7ea627dd6c3ab4a586bf8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 21:58:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb90b997fda418fb367e3a7271c4ec3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9570d48a9ea74b42ade10d0b3942e430 : SUCCESS in 1h 12m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/75251fda133b4f62a60fa18cefcf52d2 : SUCCESS in 26m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/700b23dd1c814d0a991bde9fd3a7a29b : SUCCESS in 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d67f332c1130c434e6afd770aeccfeed2b8cba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 23:14:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/297cf01d1d5a4fab84ae12659202710c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e3becd8c1b7848319740d1d47ab660df : SUCCESS in 1h 58m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da507a6ed42d423fb63e653d67f0cb54 : SUCCESS in 32m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02bddbdd9bd946fd83958fe4d8701d79 : SUCCESS in 31m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a4a09b0ba74849a389fceae321ab53 : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41f15f1d59154bcbaaef670f513b43c4 : SUCCESS in 16m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/941bb6524dae4c6cb864c612143d568d : SUCCESS in 20m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17c2a8ae375e4da1a27773d960f0181d : SUCCESS in 23m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cbd071dc1a34a42b69dd17a333e2f4f : SUCCESS in 12m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b179899c990548b0b08776a53e3db2e6 : SUCCESS in 39m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d96eec63e41546ed8a9d7b4ce1fec0ce : SUCCESS in 2h 10m 20s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6e551eae33c46b38c5737cb89438d4f : SUCCESS in 32m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4a05f28036124fe4a0242a6c07e3794b : SUCCESS in 58m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/25bcf77deb4b433f802ba6bc199ff116 : SUCCESS in 1h 11m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ec8fcdd95bec436b99977a0ea042702b : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":2},{"id":"2bd19c3fed423ddee9229a424f9b2c20e93c4088","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-27 01:23:40.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/91/962191/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"767f9fcdce71589043bdab0c409c93ca06869143","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-02 16:02:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3529985f35172c20976b4960b7b81b6eebab8a96","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-02 16:57:53.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/91/962191/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"66d2d5c31d3a204a2700d6e7b26019d1e3bc69eb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-02 16:59:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ef1d468a8291e3ffce97bfe31343c53935a86097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 18:07:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcccd5bf43f8405697188d9e15427322\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4cd1974823d544a68ebfd2c5daddc1ff : SUCCESS in 2h 02m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6fb75e628344365844a8b9a2827bcfa : SUCCESS in 35m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b8320d0467e498c98dfaf456fe6d564 : SUCCESS in 45m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1945c8059c4d48ab9643f4856913f9e5 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d459f71f0a98410c8fba6ed95cfc1894 : SUCCESS in 17m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5a1f6cf75684918bcd0f2d81d6f72a8 : SUCCESS in 14m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4c47d3fb7fb41c096a7cdbd8685bfc5 : SUCCESS in 27m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd623cf18d3048da924e081d29c7c018 : SUCCESS in 13m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a796f3b32f7b4d0b8744a1a7b526a2a3 : SUCCESS in 1h 10m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5b03e185b9c345759514311a408ef90b : SUCCESS in 1h 39m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef3a4cf432b842e3beb051b87c2f8b44 : SUCCESS in 56m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/de35bfb57a5946ff84fda9bf34b56b43 : SUCCESS in 1h 04m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f72d9151f7184eeb80d9b938f83e9962 : SUCCESS in 1h 18m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b321ab74d5cd4254a5ec75fb86e28ee1 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e2dfb8d0428248064ea32754c65682fe92b09f48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 18:20:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b34fbdd45ce45afbc797ee2817da698\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b36278853c844bbf9f4afa284296cb34 : SUCCESS in 48m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8e68b766be8453b807cb3eed59988c2 : SUCCESS in 46m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03b8999161db4033b93986e1508e0800 : SUCCESS in 1h 01m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8a958998fe837f4c2ae9fa389a23c825e3863f7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-10-06 08:14:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f82e4ff9c303d92b3641f83baddc4170b4531cc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 08:14:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"57234d55114f1d8ea0dae9dcae528e12e98e3a06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 10:56:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75cfdd62e9324243ac4f9c8d597e2ecf\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e4b2d90db9a74715a3cddf5530a7abfd : SUCCESS in 2h 40m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f146f31241d4356b268be4867254b8e : SUCCESS in 15m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bc84547b2e54d89802a500e48496f14 : SUCCESS in 17m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25077b3c578c40eab218fa9c60bb4805 : SUCCESS in 21m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/517d3b9c33854922b08f3e343f4e33a3 : SUCCESS in 7m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/21bc1acdb35e4ee490fdfb100763e720 : SUCCESS in 1h 48m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0fa79d460ea4e5eba1347ea4092a2d7 : SUCCESS in 1h 07m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/72c14948bbff4870859306e480282944 : SUCCESS in 42m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e23cf7347bbf4596992cfd8bbaacadd7 : SUCCESS in 1h 19m 52s","accounts_in_message":[],"_revision_number":3},{"id":"b06f337c798e4567adf1efb3a57c45276b5b3eb7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 10:56:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9c6c9a466c50a3d41addaceaf4ed832f5e36263d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 10:57:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3379f7b8ce17447eafdf0626eb0a8514\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c16b416fe94bd69b8002bdee17e48b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3},{"id":"5b53cf7b199fd0e7993cb3d0b28c42cb62d98c03","tag":"autogenerated:gerrit:revert","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-08 13:27:33.000000000","message":"Created a revert of this change as I201d7681c1a1b0c6e3f54c46ba5d4be586c709ff","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4238ccb45069eb75c1af960d367080e731414a8d","revisions":{"686076fba452b5bcdbde5a66aca4e5c5470d43c7":{"kind":"REWORK","_number":1,"created":"2025-09-24 14:33:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/962191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/962191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/962191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/962191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/962191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/962191/1"}}},"commit":{"parents":[{"commit":"e0ca9a0d68fbbcb98820f488accf2f84fb8c9639","subject":"Merge \"Fix typo in warning log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0ca9a0d68fbbcb98820f488accf2f84fb8c9639"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 11:49:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 14:33:43.000000000","tz":0},"subject":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``","message":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``\n\nWhen a ``HA_Chassis_Group`` for an external port is created, the method\nalways discards the chassis bound to the port. A logical port can have\nseveral chassis stored in the ``Port_Binding`` register, in the ``name``\nfield (the main one) and in the ``additional_chassis`` list (see method\n``get_chassis_host_for_port``).\n\nCloses-Bug: #2125569\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3346c2a16f69c3b16d1693db42335b133a721bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/686076fba452b5bcdbde5a66aca4e5c5470d43c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/686076fba452b5bcdbde5a66aca4e5c5470d43c7"}]},"branch":"refs/heads/master"},"063c70c70b131c277e16de42d5e78837784c5f84":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-26 20:43:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/962191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/962191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/962191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/962191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/962191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/962191/2"}}},"commit":{"parents":[{"commit":"783ce1cd4d99ca20fb705d6540fbce5865fe07d7","subject":"Merge \"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/783ce1cd4d99ca20fb705d6540fbce5865fe07d7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 11:49:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-26 20:39:28.000000000","tz":0},"subject":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``","message":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``\n\nWhen a ``HA_Chassis_Group`` for an external port is created, the method\nalways discards the chassis bound to the port. A logical port can have\nseveral chassis stored in the ``Port_Binding`` register, in the ``name``\nfield (the main one) and in the ``additional_chassis`` list (see method\n``get_chassis_host_for_port``).\n\nCloses-Bug: #2125569\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3346c2a16f69c3b16d1693db42335b133a721bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/063c70c70b131c277e16de42d5e78837784c5f84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/063c70c70b131c277e16de42d5e78837784c5f84"}]},"branch":"refs/heads/master"},"4238ccb45069eb75c1af960d367080e731414a8d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-02 16:02:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/962191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/962191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/962191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/962191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/962191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/962191/3"}}},"commit":{"parents":[{"commit":"9c5736b68c47e00fa131c71fbf31c0dec86e78a6","subject":"Merge \"[OVN] Remove the check for external ports support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c5736b68c47e00fa131c71fbf31c0dec86e78a6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-24 11:49:21.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-02 15:58:05.000000000","tz":0},"subject":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``","message":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``\n\nWhen a ``HA_Chassis_Group`` for an external port is created, the method\nalways discards the chassis bound to the port. A logical port can have\nseveral chassis stored in the ``Port_Binding`` register, in the ``name``\nfield (the main one) and in the ``additional_chassis`` list (see method\n``get_chassis_host_for_port``).\n\nCloses-Bug: #2125569\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I3346c2a16f69c3b16d1693db42335b133a721bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4238ccb45069eb75c1af960d367080e731414a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4238ccb45069eb75c1af960d367080e731414a8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
