)]}'
{"id":"openstack%2Fneutron~962402","triplet_id":"openstack%2Fneutron~master~Ibf3b4f108058387d4a9689d6411139e7727b1315","project":"openstack/neutron","branch":"master","topic":"bug/2125553","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-06 10:56:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf3b4f108058387d4a9689d6411139e7727b1315","subject":"[OVN] Simplify the logic to create a \"HA_Chassis_Group\"","status":"MERGED","created":"2025-09-26 20:26:20.000000000","updated":"2025-10-06 10:56:27.000000000","submitted":"2025-10-06 10:56:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962402-bug/2125553","meta_rev_id":"4085b165662ca435e114d139fa9e27eceede4d52","_number":962402,"virtual_id_number":962402,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-10-02 18:31:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-06 10:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-06 08:17:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-10-02 21:49:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-06 08:17:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-26 20:33:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-26 21:34:59.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-26 22:12:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-26 22:37:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-26 23:50:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 11:15:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-02 21:49:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-10-06 08:17:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"93d6077c841084e5fa43810876b22360450e6d1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-26 20:26:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37ac1532c6e4d35de76da26f30c96eded4a207f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-26 20:43:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a2314847bd8e3032da71bd72f46214b2cc107d2a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-26 21:34:59.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/02/962402/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"948058a2d78ac0981f2c1a7ed958f191f5cd8463","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 22:12:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f290d5cb42a4ce3b8f02e4bc45e5ab8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28168cb717674f0da7ae0de7841c12a2 : SUCCESS in 1h 03m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69bc7b6cdfef427eacd895ed91b911ce : SUCCESS in 1h 04m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/22516ed9a7af44e3a9922568f0cd5c56 : SUCCESS in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3e2c0c9346adf27f0fa23dbbbbf32ad708a7ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 22:37:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/963c1954c250470a866ea99f859dbebe\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/670df01d1f064411b2c8241a8700039d : SUCCESS in 1h 45m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d0c2da9fcce546c384ba756699c95d25 : SUCCESS in 30m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78bd9e6788c0483e933d6d417396c615 : SUCCESS in 33m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7941c744923f4b9581d447819dd5a8cd : SUCCESS in 9m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca684894b0e849c394487e01cee2b395 : SUCCESS in 23m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1efb67907ddc40aebd0dce3b039c882b : SUCCESS in 15m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec0c0d24e462454a91dca6720a1354aa : SUCCESS in 27m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e077c17dc184d9cb2c753bfea159a73 : SUCCESS in 12m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/571d5a2ce50f47219ca16f3a5957a3be : SUCCESS in 1h 01m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9979ab1659ff4f6aa15faec67f726f93 : SUCCESS in 1h 48m 29s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8244b10fc9414c54aad85cd512301005 : SUCCESS in 1h 10m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3914d7ca195f4d2e8b253cf8465a51bb : SUCCESS in 1h 02m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e621d28a783e42779c0a09893d4c6c6d : SUCCESS in 1h 16m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/964c7bf499eb4c66a834d8dcc59733e9 : SUCCESS in 1h 33m 57s","accounts_in_message":[],"_revision_number":2},{"id":"cde5c82ee1ba4f45e7df78efc59e21cad04f17fd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-26 23:50:53.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/02/962402/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"815bf23bf0ee78ae2be62bea4bd0cce07b9a7ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-02 16:02:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"341398a8189cc9bee2a5f1ab032a8c550c960229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 18:03:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4170a840de4b4f76830a616fb93a7e2e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f7843c95adf1456bab39866091ed3567 : SUCCESS in 1h 57m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4954316db28041a698962658228ac692 : SUCCESS in 33m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7621e1eeaaa4849b5e6d692df8d0808 : FAILURE in 42m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5609b974bc1c4785987b9e94d905c8ee : SUCCESS in 7m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8866cae469db47549ef44d2246763567 : SUCCESS in 21m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c18c4002f88e48b480ec44fda5be7b6f : SUCCESS in 19m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ce9ef0099d84f5e8623759a87fb68a9 : SUCCESS in 16m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/583dc711bdc54bd48bb7e73246d83eab : SUCCESS in 6m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bc8f716832534eec98d851b714b31475 : SUCCESS in 1h 06m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fb740e24c2f64b73be4726af6aaf5334 : SUCCESS in 1h 29m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6cb218835a74cfba9e22c1c521f7de4 : SUCCESS in 41m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ad5a9fff4b574234ab82307820e7d4e3 : SUCCESS in 1h 08m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5d13f9852cda45e8aea0fe5faf4c58a2 : SUCCESS in 1h 11m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/822616a5c0264c30977a5a12ea71ea71 : SUCCESS in 1h 15m 54s","accounts_in_message":[],"_revision_number":3},{"id":"866e8a1ab9fde2352d419bc44aa8218ebcb08649","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-02 18:31:06.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/02/962402/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"514a7e083bde034ab8826ecbd0b1cb14b497e899","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 18:41:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74bd478e2e3e40c6a92a181ec325574c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce76a95fd12943e28994a6d532cd684d : SUCCESS in 23m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/962c3a8e75a6485880d371ee5a8a2d61 : SUCCESS in 21m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c47b99d8f33842d58fd48e4b6e9601b6 : SUCCESS in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eb5ee79ab30416960e314aaef5f0628eef8dd162","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-02 21:49:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a7b492aeed222d1b4141f552860df3ace3b84fe7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-05 15:54:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ba74605922119a93f13974b6cf21933b59756fa","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-05 16:39:29.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/02/962402/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"f7d722c0e4b63326598a2ea3072d0dd77254715d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-05 16:50:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e939f07ea254bf799d514df04740bab\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f49c376791a5440884b1fe5425530d77 : SUCCESS in 22m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f341a561dc564cc5820d6fcfeeac6f16 : SUCCESS in 45m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b8071593ab3d43768a28eb9fbf23c680 : SUCCESS in 54m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"03eef2ee52c6c949a1a3d98f24c15a5237fd2487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-05 17:40:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe1dd71be0148d48c707b47fd9f08f2\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5b8f42890527452f9ff5b63c3cd0bc6c : SUCCESS in 1h 44m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4daa73cdcd6445381d00357da3d2046 : SUCCESS in 32m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7f48ec764f349baa0ef29829aae4969 : SUCCESS in 27m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b88df5220045bc99c93d842281fbec : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47a2171fbf924a92ab025fed69d93b54 : SUCCESS in 16m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/907ff81382cb44598041fcb5d0b95d9c : SUCCESS in 17m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/770198fc632242e5a63d0534f698f669 : SUCCESS in 22m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aabeac11fa32428585d6447e50628ca7 : SUCCESS in 7m 35s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4a6ad055fcaf4363b074d33fd6e83103 : SUCCESS in 33m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/05c85e27cf224c42b4ae2d1efd99104f : SUCCESS in 1h 31m 44s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/57b5a62c074e4e6aab74a33e355b8822 : SUCCESS in 43m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7e607038f0d44f09a41b19dcf7a7c462 : SUCCESS in 57m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d78b12bd388b4de986932d112479c11e : SUCCESS in 1h 06m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e5a4011aeb6d458da22dc71b0d1acf3d : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":3},{"id":"cccd09259f1d06052abe2b4d43248ff3783ca650","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-10-06 08:17:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"550a55a433eece4ad86536d7e1a48fa9c60ef2cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 08:17:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"35e43c7ae49cfbae3595ba19fd079a11928a55af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 10:56:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd93f230cef4d3b9e6e7fd7c87102da\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a20299a6fe67471c81b44996543cfa48 : SUCCESS in 2h 19m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7604b58b11714a4099c1ccf84991de3e : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0a3e3a531204db1a1b67a45d85a6c49 : SUCCESS in 17m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19c745e1d7b740049f8313ff42370078 : SUCCESS in 22m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d76a0e59dd460580b1c358065681cd : SUCCESS in 9m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8a709256aa404c8fbec942e52f04b53b : SUCCESS in 1h 49m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd46865dc4e64aa3aa94cc84adbaea68 : SUCCESS in 1h 11m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f2faf41355ad4b80adc8d61a4d0268aa : SUCCESS in 59m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1c51a92ccabc4f4b9158d9e1a1704122 : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":3},{"id":"4085b165662ca435e114d139fa9e27eceede4d52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 10:56:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5cfdb6dc78d74de2a1d85e8c48f775694f3ccfce","revisions":{"c13ffa7b0a82661fffa64069407773191a32f95e":{"kind":"REWORK","_number":1,"created":"2025-09-26 20:26:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/962402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/962402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/962402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/962402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/962402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/962402/1"}}},"commit":{"parents":[{"commit":"686076fba452b5bcdbde5a66aca4e5c5470d43c7","subject":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/686076fba452b5bcdbde5a66aca4e5c5470d43c7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-25 15:05:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-26 20:25:14.000000000","tz":0},"subject":"Simplify the logic to create a \"HA_Chassis_Group\"","message":"Simplify the logic to create a \"HA_Chassis_Group\"\n\nThis patch simplifies the logic for the \"HA_Chassis_Group\" creation.\nThe method creates 3 list/sets:\n* A list/set of chassis to be deleted.\n* A list/set of chassis to be added.\n* A list/set of chassis to be kept.\n\nIf the \"HA_Chassis_Group\" is already present and the highest priority\nchassis is not deleted, it must keep the highest priority to avoid\nrebinding the external ports into another chassis.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibf3b4f108058387d4a9689d6411139e7727b1315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c13ffa7b0a82661fffa64069407773191a32f95e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c13ffa7b0a82661fffa64069407773191a32f95e"}]},"branch":"refs/heads/master"},"a27fe4bf7db4b85ee4e9a6114d6f42ca60093d6a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-09-26 20:43:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/962402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/962402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/962402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/962402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/962402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/962402/2"}}},"commit":{"parents":[{"commit":"063c70c70b131c277e16de42d5e78837784c5f84","subject":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/063c70c70b131c277e16de42d5e78837784c5f84"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-25 15:05:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-26 20:39:28.000000000","tz":0},"subject":"[OVN] Simplify the logic to create a \"HA_Chassis_Group\"","message":"[OVN] Simplify the logic to create a \"HA_Chassis_Group\"\n\nThis patch simplifies the logic for the \"HA_Chassis_Group\" creation.\nThe method creates 3 list/sets:\n* A list/set of chassis to be deleted.\n* A list/set of chassis to be added.\n* A list/set of chassis to be kept.\n\nIf the \"HA_Chassis_Group\" is already present and the highest priority\nchassis is not deleted, it must keep the highest priority to avoid\nrebinding the external ports into another chassis.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibf3b4f108058387d4a9689d6411139e7727b1315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a27fe4bf7db4b85ee4e9a6114d6f42ca60093d6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a27fe4bf7db4b85ee4e9a6114d6f42ca60093d6a"}]},"branch":"refs/heads/master"},"5cfdb6dc78d74de2a1d85e8c48f775694f3ccfce":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-02 16:02:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/02/962402/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/962402/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/962402/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/962402/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/962402/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/962402/3"}}},"commit":{"parents":[{"commit":"4238ccb45069eb75c1af960d367080e731414a8d","subject":"[OVN] Always exclude the local chassis creating ``HA_Chassis_Group``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4238ccb45069eb75c1af960d367080e731414a8d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-25 15:05:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-02 15:58:05.000000000","tz":0},"subject":"[OVN] Simplify the logic to create a \"HA_Chassis_Group\"","message":"[OVN] Simplify the logic to create a \"HA_Chassis_Group\"\n\nThis patch simplifies the logic for the \"HA_Chassis_Group\" creation.\nThe method creates 3 list/sets:\n* A list/set of chassis to be deleted.\n* A list/set of chassis to be added.\n* A list/set of chassis to be kept.\n\nIf the \"HA_Chassis_Group\" is already present and the highest priority\nchassis is not deleted, it must keep the highest priority to avoid\nrebinding the external ports into another chassis.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibf3b4f108058387d4a9689d6411139e7727b1315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cfdb6dc78d74de2a1d85e8c48f775694f3ccfce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cfdb6dc78d74de2a1d85e8c48f775694f3ccfce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
