)]}'
{"id":"openstack%2Fneutron~962406","triplet_id":"openstack%2Fneutron~master~Iff2ebe33d5ca454e06b914313ca124c105c4dedb","project":"openstack/neutron","branch":"master","topic":"bug/2125553","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-06 10:56:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iff2ebe33d5ca454e06b914313ca124c105c4dedb","subject":"[OVN] Remove check for external ports support","status":"MERGED","created":"2025-09-26 21:17:51.000000000","updated":"2025-10-06 10:56:42.000000000","submitted":"2025-10-06 10:56:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962406-bug/2125553","meta_rev_id":"fb10f8bc1ebf52115698792c70e539919453a460","_number":962406,"virtual_id_number":962406,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-10-02 21:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-06 10:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-06 08:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-10-02 16:58:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-06 08:39:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-26 22:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-26 23:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-27 02:08:54.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-01 00:21:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-10-02 16:58:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-10-06 08:39:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"db0863103c311fab4f701bbaf44ab757aec16525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-26 21:17:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3348506041db5b469928dd879305cd97e181f718","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 22:22:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75575fcc0894472f95891e2cfba6da43\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b1d2ddf69c842c2a7117206f33bc8c8 : FAILURE in 1h 00m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe85e4418ca5400db75956c19151b3b6 : FAILURE in 24m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/145e509bf6da42809b0adc371f5e72f1 : FAILURE in 24m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d57da3bc342d722eb01e553f3df49496f21d4a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 23:19:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2845d31b99b458585d6cae4d90c56da\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1f8dca44515c4047a4c89845757d7881 : SUCCESS in 2h 01m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa4e067de4144b24be398a36a2345191 : FAILURE in 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd9246173ea44f188098a5cf98ebca50 : SUCCESS in 14m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00065fa46174428ab22fbe82954cf169 : FAILURE in 17m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8071371badc4124a14661a407361113 : FAILURE in 13m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a6dc4810d5b949a08576241cfeda5ee9 : FAILURE in 16m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e90c01670d549379869d23ac389aaa3 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48211d3cbc3b407c9e36ea4eb392aa38 : SUCCESS in 17m 29s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b12c6272ec734d2e94390c4b16346b9a : SUCCESS in 1h 03m 26s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/039b72879451481fbb9be3ec15838ef1 : SUCCESS in 1h 36m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c64e9fded95f43969bf5a0f2b13a5947 : SUCCESS in 1h 11m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e5d7bdb5c13c4529acdf3ff98971aff5 : SUCCESS in 43m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e3db6e58427457d8fef3fa10cc6b2e2 : SUCCESS in 31m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b314568334a546368fa2e5cc4bbc50a5 : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f6e6bf7b270ad82e419a401fd67b981ca1636158","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-27 02:08:54.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/06/962406/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"711c280c700d4a2c6cb2940d3f61fa5a957f2ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-29 08:28:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"40ca31d74cea630c6b23d698081bec04e7a94fd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 09:24:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5401be42e633491bbf1987e5be48ffa0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6e9cf86df8cd4cb08c304333511c83cb : SUCCESS in 53m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aace44f175644b06985279f6e154028e : SUCCESS in 31m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d8d891e9736d484c8e738dc73815216e : SUCCESS in 39m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff265dc1c9872307e54788a08c81a40d49e6f597","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-29 09:41:06.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/06/962406/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"54382994e10bfc7325f1d2492b3cf066852efe41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:20:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8b152fd7634412a9b9552b51aad779a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7123aca780354c42b1034e33120abd08 : SUCCESS in 1h 51m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53a7cf55b99f4877b8d54ac00a4c30e1 : SUCCESS in 43m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aafb6174c33b4f1781df26be1a4f28fe : SUCCESS in 11m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03544e796495458c8d0c30fe9e8d4a3a : SUCCESS in 18m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74a96539942845f2a5a54a9b50cf9d11 : SUCCESS in 18m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c6ec40e075746bba79889dbf906c998 : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df7ef6acb0241ecbf7f86d1e085d371 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dda032499f64812a152830c35b2188a : SUCCESS in 13m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bea313f8da6f42f08b3455709024dd29 : SUCCESS in 34m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/064099b4fd2d43d0b4c8c70f46da460a : SUCCESS in 1h 38m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/c733f3f2e48042b4bad468e4c7ad4119 : SUCCESS in 1h 03m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b6d4a49809fa4720a39e10843ba76b7a : SUCCESS in 59m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/53a11018c1f0420a83a9854add6dc1bf : SUCCESS in 1h 20m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/db088eee841f4efb88b457f55a792bcd : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":2},{"id":"02c8de7bcd26bdfe89d51d99b7c3df90731d4fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-02 16:02:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0eb9c3c92784216eae760fb7f20e631b825495f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 16:38:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5da587a51e5b4c6eb1cfc6f40e13bf64\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff6e6cf40c714b8386d34230a8a212f1 : SUCCESS in 32m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49b6aa4810d44e26991d036698a3c646 : SUCCESS in 13m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0944c36a8f143c6b90603c63e95ade8 : SUCCESS in 21m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b06af4e4d304fb08844c390e2191e82 : SUCCESS in 19m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8b4ed384ec545409ab634ed3d7e6960 : SUCCESS in 29m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa26e6a4c0046079bfa05c5a1aef3a5 : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/681cc34e08d8405791e4aaf075a29a9a : SUCCESS in 15m 27s","accounts_in_message":[],"_revision_number":3},{"id":"2181dfef379624c016f94f8602a248bb09160f52","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-02 16:58:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5b320110dfadf55d10b7067ea67623fad32e053c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 19:35:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e1a8d34935b49b88a6e33a79afb155c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e92e6ad6908f4fa2b7bf39e5e08e056e : SUCCESS in 44m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6bdf828c7f764c00bcd389b3fa937d9a : SUCCESS in 41m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e60d431546b44288a4bc2b0b3a92cf99 : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"08297986848e7bbab47df6403ea5b5be8bdfb3c3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-02 21:32:16.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/06/962406/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"d6c53b7f0e218c5ed70caf1709c92a1357185c19","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-10-06 08:39:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e8a2205f98aa540253108c18c5b8f1205d1dbfec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 08:39:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b2b98ad28897e3a832d21632e76584be90e9100","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 10:56:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78a7dcb8d1c74512868c954fbaea5306\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61198c619f0c40c9a55e35dbe538e534 : SUCCESS in 12m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61795daddaab46cba69532410f0bb01f : SUCCESS in 22m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec7585c2b98042c686f6ee0735194dc3 : SUCCESS in 14m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3655071828044455a7af4aa2b76fb57e : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57d1f822e26c465aaddea0c1094147d4 : SUCCESS in 19m 06s","accounts_in_message":[],"_revision_number":3},{"id":"fb10f8bc1ebf52115698792c70e539919453a460","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 10:56:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bc4ecf2290cbf808704d126e132b3c2d5043a02","revisions":{"17df2a3ef404f3e96c33acf4e0ceccfe68c6b7ed":{"kind":"REWORK","_number":1,"created":"2025-09-26 21:17:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/962406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/962406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/962406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/962406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/962406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/962406/1"}}},"commit":{"parents":[{"commit":"bfa2fbebd37043aa282bd0a54bb6715c519586bd","subject":"[OVN] Remove maintenance method \"check_for_ha_chassis_group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bfa2fbebd37043aa282bd0a54bb6715c519586bd"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-26 21:09:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-26 21:17:40.000000000","tz":0},"subject":"[OVN] Remove check fot external ports support","message":"[OVN] Remove check fot external ports support\n\nThe field \"ha_chassis_group\" was added to the \"Logical_Switch_Port\"\ntable in [1], that was released in OVN v20.03.0. Neutron enforces\nnewer versions of OVN to work properly.\n\n[1]https://github.com/ovn-org/ovn/commit/b31c76000bef314b68e776d318d1ce4cf152450b\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Iff2ebe33d5ca454e06b914313ca124c105c4dedb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17df2a3ef404f3e96c33acf4e0ceccfe68c6b7ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/17df2a3ef404f3e96c33acf4e0ceccfe68c6b7ed"}]},"branch":"refs/heads/master"},"a19ec75f369498ea167a90c211f8f5295d14722a":{"kind":"REWORK","_number":2,"created":"2025-09-29 08:28:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/962406/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/962406/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/962406/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/962406/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/962406/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/962406/2"}}},"commit":{"parents":[{"commit":"bfa2fbebd37043aa282bd0a54bb6715c519586bd","subject":"[OVN] Remove maintenance method \"check_for_ha_chassis_group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bfa2fbebd37043aa282bd0a54bb6715c519586bd"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-26 21:09:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-29 08:27:54.000000000","tz":0},"subject":"[OVN] Remove check for external ports support","message":"[OVN] Remove check for external ports support\n\nThe field \"ha_chassis_group\" was added to the \"Logical_Switch_Port\"\ntable in [1], that was released in OVN v20.03.0. Neutron enforces\nnewer versions of OVN to work properly.\n\n[1]https://github.com/ovn-org/ovn/commit/b31c76000bef314b68e776d318d1ce4cf152450b\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Iff2ebe33d5ca454e06b914313ca124c105c4dedb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a19ec75f369498ea167a90c211f8f5295d14722a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a19ec75f369498ea167a90c211f8f5295d14722a"}]},"branch":"refs/heads/master"},"7bc4ecf2290cbf808704d126e132b3c2d5043a02":{"kind":"REWORK","_number":3,"created":"2025-10-02 16:02:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/962406/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/06/962406/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/962406/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/962406/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/06/962406/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/06/962406/3"}}},"commit":{"parents":[{"commit":"817a443a20c1978e3cd12fe9ced1515e5e135dd2","subject":"[OVN] Remove maintenance method \"check_for_ha_chassis_group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/817a443a20c1978e3cd12fe9ced1515e5e135dd2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-26 21:09:07.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-02 16:01:35.000000000","tz":0},"subject":"[OVN] Remove check for external ports support","message":"[OVN] Remove check for external ports support\n\nThe field \"ha_chassis_group\" was added to the \"Logical_Switch_Port\"\ntable in [1], that was released in OVN v20.03.0. Neutron enforces\nnewer versions of OVN to work properly.\n\n[1]https://github.com/ovn-org/ovn/commit/b31c76000bef314b68e776d318d1ce4cf152450b\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Iff2ebe33d5ca454e06b914313ca124c105c4dedb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bc4ecf2290cbf808704d126e132b3c2d5043a02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bc4ecf2290cbf808704d126e132b3c2d5043a02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
