)]}'
{"id":"openstack%2Fneutron~962485","triplet_id":"openstack%2Fneutron~master~I16d9fbf5aabc44fc1dd5bd7512de7c317798e2fa","project":"openstack/neutron","branch":"master","topic":"extraroute","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-01 18:39:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I16d9fbf5aabc44fc1dd5bd7512de7c317798e2fa","subject":"Add \"extraroute-atomic\" extension to OVN L3 plugin","status":"MERGED","created":"2025-09-29 12:57:24.000000000","updated":"2025-10-01 18:40:44.000000000","submitted":"2025-10-01 18:39:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962485-extraroute","meta_rev_id":"72640aec1e2f1dff025deb0e0bba18c4c7956939","_number":962485,"virtual_id_number":962485,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-09-30 11:56:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-01 18:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-01 12:21:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-10-01 16:12:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-10-01 16:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 13:45:54.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 13:47:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-29 15:09:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-30 08:39:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-09-30 08:39:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-10-01 12:21:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-10-01 16:12:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"006757251cff09a7c1e68498995b1a1ea7164d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-29 12:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb3277610474f38da42b3ea5fdd1df6e22fd1ba5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-29 12:59:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"321c8db6c30308ad1d28da99357408391c6569c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-29 13:45:54.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/85/962485/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"f91836381a3189a2a1cb0823b73ca08b06360221","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 13:47:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e78c18e1f133401fb5e0d1539939a53a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94a58b2d93bb4f9491ba1c7df1232114 : SUCCESS in 46m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6f1c26c6edff48a4ad0123ff4539ebd2 : SUCCESS in 19m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/809868ed4c774ac58fa1549e70d9c025 : SUCCESS in 37m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca4ab02e9692f2d78c95fa689b92f2b1a5dda3eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:09:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eaa9dee3a684ca5b8cd24c37529f5f1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/faef5872f3a847c88542a4b3324e25ca : SUCCESS in 2h 11m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07c125633ae940779d1dad458cc70bf8 : SUCCESS in 32m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b602d97c552b45f3b6d3438f20aa432d : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eedf77dd8e4144c9a78eb64aa7e5042e : SUCCESS in 12m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c24c717a71d547f18403c0f482d529eb : SUCCESS in 22m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d934ceac955c4623a8d6bb39e52331fd : SUCCESS in 16m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14311a00bce047319576de810ebed40e : SUCCESS in 22m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17aa3682620d4099963625a5c7a170d6 : SUCCESS in 7m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b3f7780a826341df8d685ca29ccaf06b : SUCCESS in 1h 05m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f58a898398ed40d4832a7ed54b502da1 : SUCCESS in 1h 26m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7447c23fad404e5088a427d5ab97a721 : SUCCESS in 1h 02m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b2b3ad20924c42f788f6b1397671895a : SUCCESS in 54m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/153f86f062bd4016814af86db071b502 : SUCCESS in 31m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/084bbdccc7fe424485a8e91264868829 : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3440ace4b4c19e4c6de0f8d9e8f696f604ba545b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-30 08:31:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2af0f0b776f329cab16fc5098c3cd7b21dce78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-30 08:36:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d95278c71163d29043b9721daf9a73f9d9f91952","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-30 08:39:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bbba0b3f9beadcb17fd591f6a9d5566b6b7dd03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 10:14:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b754a3803b7494f8dc19cec406e1131\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b72de909def34bd3b2936d4e3a01766c : SUCCESS in 1h 05m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ae9c82698be74b3f9b286d527f96b1bb : SUCCESS in 1h 01m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f06c138a0e64f2d81e30bb766f751f3 : SUCCESS in 1h 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"75ce6b0a6710d603d2b918864fdfe64f388b5d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 10:56:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/931b7ee7d07f421c84d13cce75ef7338\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/afe6c03541d94d3881a5d1eafa2abb10 : SUCCESS in 2h 18m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e4a149e653ee4527a6856f22e18a95ac : SUCCESS in 31m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8112e7ed53e14662bf399d844c1af5cc : SUCCESS in 27m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87783f7fed64b9cb9e09b449efe334e : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e80928e3f5a4b6c8ca8335b20fde8d8 : SUCCESS in 24m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a77057ae7b094a61b4226fd00df72fae : SUCCESS in 20m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a75de60eff0400dbb2287d1360c4ef7 : SUCCESS in 30m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2414452ed1d2455b9cb3dba5e5e0da45 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/774bf184109d42aba7aeee42e7af9215 : SUCCESS in 23m 02s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/69db0bd7fa5b40f29a6a59ea582ea7b9 : SUCCESS in 34m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/64c5575901a2447995b64ba5811ac6a5 : SUCCESS in 1h 39m 16s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/32e0e5fc4fa24801b6cf60b61c619c96 : SUCCESS in 56m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f23dd2bbdbb341acb85f5113c1e44939 : SUCCESS in 1h 03m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df5579a4530d4ce98d33ebed43bd0bba : SUCCESS in 52m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0b6dbfc5b88e4032b54e37ed9d5a7bb8 : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":2},{"id":"477ae4e06b85d70d8618906a6e6d6a12fe9f75b5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-30 11:56:23.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/85/962485/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"736caa8277f6367ab9e8aee6793d6d6bb474dce5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-01 12:21:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6dfdd706a12c16e704c889e8a65938dff1d43477","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-01 16:12:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"72ffc5165b277eb4a65ed510c145acd477dc0d1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 16:13:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eead4fa50ed9f00ca3f1789c83456917a1169552","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 18:39:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49783e89433a44fab14074d3ab4c5bac\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/643553037f704934885010e57bcb8169 : SUCCESS in 2h 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a715943d1044bdeac4785da3bd8e5a4 : SUCCESS in 13m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94da02765b9f48c38cdc2151f94ff931 : SUCCESS in 25m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c54b79f501424c619faaffdd14c7c25b : SUCCESS in 18m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf5b33c8ea54e0fb612c53e2c2dcda0 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b5b30a28b7b4848a6a054f82bd2a8b7 : SUCCESS in 15m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5be46e3b7ed84d5aa7f84f5af8baf262 : SUCCESS in 1h 52m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f66807b9d1946d58f0d896121f8b8d0 : SUCCESS in 33m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cde5ab3065454edaadf662dd52b53f41 : SUCCESS in 41m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e95f53d71c8c498f9207e849b008c7d3 : SUCCESS in 1h 03m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ca31baa6741253764ecd7394e7b77ebf556f4095","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 18:39:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"72640aec1e2f1dff025deb0e0bba18c4c7956939","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 18:40:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b79a35960821477da91b31e757b050b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd73bfecfac4451187e8738d6d01747f : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d92a3ebc4a34484a67220c0cb4719a1 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c872d5f6529154eb209fe91a374b88ba8705e779","revisions":{"46493047b0aaad8daef0bfddff4ad21bec53fca9":{"kind":"REWORK","_number":1,"created":"2025-09-29 12:57:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/962485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/962485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/962485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/962485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/962485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/962485/1"}}},"commit":{"parents":[{"commit":"783ce1cd4d99ca20fb705d6540fbce5865fe07d7","subject":"Merge \"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/783ce1cd4d99ca20fb705d6540fbce5865fe07d7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-29 12:56:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-29 12:56:51.000000000","tz":0},"subject":"WIP \u003d\u003d DNM \u003d\u003d Add \"extraroute-atomic\" extension to OVN L3 plugin","message":"WIP \u003d\u003d DNM \u003d\u003d Add \"extraroute-atomic\" extension to OVN L3 plugin\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I16d9fbf5aabc44fc1dd5bd7512de7c317798e2fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46493047b0aaad8daef0bfddff4ad21bec53fca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/46493047b0aaad8daef0bfddff4ad21bec53fca9"}]},"branch":"refs/heads/master"},"c872d5f6529154eb209fe91a374b88ba8705e779":{"kind":"REWORK","_number":2,"created":"2025-09-30 08:36:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/962485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/962485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/962485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/962485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/962485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/962485/2"}}},"commit":{"parents":[{"commit":"1d921fdf7bf0431bb082204d23b90dbf9b2fe094","subject":"Merge \"Add a new option to skip loading Trunk extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d921fdf7bf0431bb082204d23b90dbf9b2fe094"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-29 12:56:51.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-30 08:36:13.000000000","tz":0},"subject":"Add \"extraroute-atomic\" extension to OVN L3 plugin","message":"Add \"extraroute-atomic\" extension to OVN L3 plugin\n\nRelated-Bug: #1826396\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I16d9fbf5aabc44fc1dd5bd7512de7c317798e2fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c872d5f6529154eb209fe91a374b88ba8705e779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c872d5f6529154eb209fe91a374b88ba8705e779"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
