)]}'
{"id":"openstack%2Fneutron~962569","triplet_id":"openstack%2Fneutron~master~Ia4f685077a8d72bf28f66daf21225d96f57ddef6","project":"openstack/neutron","branch":"master","topic":"bug/2125553","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-07 01:17:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia4f685077a8d72bf28f66daf21225d96f57ddef6","subject":"[OVN] Add method ``sync_ha_chassis_group_network_unified``","status":"MERGED","created":"2025-09-30 08:32:36.000000000","updated":"2025-10-07 01:18:05.000000000","submitted":"2025-10-07 01:17:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962569-bug/2125553","meta_rev_id":"b50d89318106bd04fe2df78e2130b82afef6c345","_number":962569,"virtual_id_number":962569,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-10-02 19:17:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-07 01:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-06 08:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-10-06 23:33:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-10-06 23:33:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-30 10:02:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-30 10:13:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-30 11:10:55.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-01 00:21:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-10-06 08:49:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-10-06 23:33:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"69f78eb94a7a228cdc47ab7d9948159a00bba5d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-30 08:32:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"799ec95f3dcbda597d7b74bd5c0e28ffdf4b28e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 10:02:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac5457c8e3c4f14b44dab91f00c09a9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d18e422f1e84108b1e853f4661c9a5b : FAILURE in 1h 02m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cff8912c364e4384b9591bf945471056 : FAILURE in 1h 00m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b3152b5b9bb24d48975e478a242012e3 : FAILURE in 1h 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e1df78df56bc781118a28bc3a2abb548503447d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 10:13:46.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edaef89700be4b5da4aca1e31e819249\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/048a66d662cc44b698800d09d637a1d4 : SUCCESS in 1h 40m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3e169fef8c04082b4bc18d0cf41c7f9 : SUCCESS in 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05a7f0873c3c4faba6e44394bd9626e5 : FAILURE in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa8b7a74e93b415e9def11f067c7cafc : FAILURE in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a4da401776c4a62ac916ee82bff0b8f : FAILURE in 22m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9d6ac915a7442bdbb8ecf555b0d1cec : FAILURE in 18m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b7e47bf9d7740348ca833d2f586abbe : FAILURE in 27m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d27ba0d5544c4f99bfbeed110d8bb6b9 : SUCCESS in 11m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e37b1d3f9119446789f6f966347c3204 : SUCCESS in 1h 01m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/876a376e9f824c37807f111840c281ef : FAILURE in 37m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/11be3a148525474a813eff4f62c29692 : SUCCESS in 32m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0a2d26b8081743329ec638d10b7518a6 : SUCCESS in 1h 09m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad0e6f3906e344a290cce8c8704c7e8f : SUCCESS in 55m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/92df1ed2b7964feb8e3bfa4ade421695 : SUCCESS in 1h 24m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b5485c6fa4eaf4cba21403809bc2359d027b0dbc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-30 11:10:55.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/69/962569/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"834d00106d50509d0a676856e579d85a1fcee573","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-30 11:45:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"052c5b5a670538f74fae887cd2f1b16e1bbc2bad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 12:32:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cebeaa434f14e9882d604345075b2a0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5aa4e81d03ad48adb7025583bedf7521 : FAILURE in 25m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a2b9edf1d43432d86b26ee5869e9416 : FAILURE in 35m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/86e94e1eaee24bafa58cf68c48f750e8 : FAILURE in 44m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f955c66459e4514581eec2c1f88e791b8712dcef","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-09-30 12:42:43.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2025/Arista_Tempest_Testrefs/changes/69/962569/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3b07073f97fcbd208bf3001ae801d5dd8bf6573e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 13:43:16.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c44fff3a2714677a0019440b9b6e9c9\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f884a62396774dfdb9bc7d2cb1301c1a : SUCCESS in 1h 56m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/98c33cc817694cfc8ded90e4af822b44 : SUCCESS in 31m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c45985aab114a8b8fc618e593f0495d : FAILURE in 45m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bae123f99b745489f38d5cd6db51a78 : FAILURE in 9m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf9587be922b4b298ae5989b28610412 : FAILURE in 14m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/388c76a0c1314aeab3ed08aff4468d64 : FAILURE in 13m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23cf2a0077054a9cbb609421c2b0e48b : FAILURE in 23m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b9cd519e77458c9fd508e26e7eeadb : SUCCESS in 11m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c36158a4f33d4f40b6d0eadbc1cf02fc : SUCCESS in 50m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/31e41ed9c79a4d29af9d2cbcd747345d : SUCCESS in 1h 23m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/883c93221ffb4caeb9155338f5cee6db : SUCCESS in 34m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/529ab2bb6c924b1ba2791d6bfdb0bdce : SUCCESS in 1h 05m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a20b3c033b594ccc89f4afe957ee6519 : SUCCESS in 1h 13m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/56ea6711c73b41bda2bd72c339911ff2 : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":2},{"id":"1b594dfd705489954d13bf5f70ee3947b76c3699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-01 09:18:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"37fcbdf755e120ef3aaa4f20d26a54cdba9acd27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-01 09:18:53.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e3c4d14984488e932a3d3882adafa27f1ea3518a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-01 10:04:29.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/69/962569/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"dccc7fdbb037f3c89af730d470403f1bf863b3ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 10:23:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7770f7634d64990bedca6d1bbd5ee70\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/698dad42180e4cfd9969cba8b2a29f1d : SUCCESS in 22m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d8dbe41ba27c4ad78263c9962eaa2f46 : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e490ff30cf5146d0815075810bae3013 : SUCCESS in 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f35913a243a82df043ba2c355a8e7cc83a66205b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-01 11:37:42.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/69/962569/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"7fdcee4a2c4c8c7c09e3d26c05f01946a8660ef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 12:05:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c1310da70c4447aaedec482e8a20ccb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/658d6b2012c843ccaa242a0ba0fc4bfd : SUCCESS in 2h 44m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2179242ec3934443a6d297be386970cb : SUCCESS in 34m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9919cd9c635e494aa6d56c5842586e46 : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/141f7d50a0414a88b6a5590022ed4865 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d2743ec37314f7dade628c55ff3b12c : SUCCESS in 21m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95285cfc0889491199febcb81da1aa66 : SUCCESS in 15m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e885ebeab1d5491d96a4df69ff567ae7 : SUCCESS in 18m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7a44aca8a84ec9b9424cd75f389149 : SUCCESS in 6m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/224cf69d95374826892e78a28b49952d : SUCCESS in 1h 12m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4d17d255d52b43f9a23e7bc3b3c7fd49 : SUCCESS in 1h 44m 53s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/dbc7c3cd494f4f33b5713866e42c61e1 : SUCCESS in 35m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b454566cbdf24954aa088a4ea3c391a8 : SUCCESS in 57m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1f72fc9c27a04729ab14cc283130af89 : SUCCESS in 51m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/627efa1c7a6940b38a27fa3739b45d89 : SUCCESS in 1h 20m 01s","accounts_in_message":[],"_revision_number":4},{"id":"d8c504bb1f3b6428625c2c2a66c1811108355f75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-02 16:02:03.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"27fa838439b0dc94aad0a3ebe7b3d55ff88ce7d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 17:58:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fed9f77429f422895d8987dd5b35ef3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/86543727ad144e03a9839666e14c96bc : SUCCESS in 1h 54m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c3a93f7dcc89442098c0a53811442f41 : SUCCESS in 34m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71a5c6b1a0c6411d9ed1b4d9e7aa6407 : SUCCESS in 45m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c78c30178f243bfad5875662dc4677a : SUCCESS in 18m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abf20c9e6ddd48768b584b0661aaccde : SUCCESS in 24m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d30d36c53454083887b1690a48a7976 : SUCCESS in 20m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9327353227ce4006bd3a2deefd50acd0 : SUCCESS in 40m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c823489b3d4fb89f1ca8de484aedbf : SUCCESS in 13m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/af79d871974b42ef99cccfd0f5f0cef8 : SUCCESS in 1h 03m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/05a130d7d70841f99bb9b6169066f38f : FAILURE in 33m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5c3cf0c3fe24fe5b514f8feca23f1f5 : SUCCESS in 1h 13m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a8b486ea73a849a1bc9471620151554a : SUCCESS in 1h 03m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1769abf38f3f4f25848a8026fcb521b7 : SUCCESS in 1h 13m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3479b6161998458f88fce0777f83eecc : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":5},{"id":"6c0d194b0748291f59c6cafc543ee3935ffc1177","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-02 19:17:04.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/69/962569/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"f05d6ae90c4f99a9835ec00d87d1521c5a20a30d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 20:09:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8304fd069f364d8b8ace9e0eb36762eb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/105a9f4e6bcd4b12aa90ae979abddd31 : SUCCESS in 18m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e6db290279954611a118f75fe017c991 : SUCCESS in 22m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c9e343bdac2462b97b070fd211749a6 : SUCCESS in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7491278633aef187c341646d245c63da013233ab","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-06 07:21:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9fbab2b984896688fdb0241a2238ec572519cd7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 07:57:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a46b87846c544fbabfc3d30958211b6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42930252787d48868d3f6efd06cb696b : SUCCESS in 33m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2470432dc99442c4985984b36bf4ba29 : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fa2844e60dd9409bb351d646290a2f86 : SUCCESS in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"88a7455b961382c5c22547c4f5762cb0da0a7ad7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-06 08:08:33.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/69/962569/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"581af20aa52fe92ec11efb796cf66689b6bc81c4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-10-06 08:49:17.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8f3043c1c050fd46dd52c47ba94ad2a3773d0427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 09:40:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/816c134abde44bb7b193ce04470f8c98\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ebba4c2db6d24b2c9aa1b007a2576567 : SUCCESS in 2h 17m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e05200d4adc54b23b112b3b88674e45b : SUCCESS in 22m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39305a8dcfeb406b975f35fe2a25809b : SUCCESS in 26m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/584160ccc025404184c4ee4f45707394 : SUCCESS in 15m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/859fb0bdc4d44f598a10d2cb9e4c5b8f : SUCCESS in 15m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6b6b26fcdbe48eaa213e90c28094394 : SUCCESS in 23m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e6faaaf41b8441ea8353116042f77a3 : SUCCESS in 16m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1661988d83694cc8b81926a5e141c4aa : SUCCESS in 7m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/49e7f8603ca44ce88eec168346baca69 : SUCCESS in 1h 14m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6de76be6f18446f9b795ec5a60eb962b : SUCCESS in 1h 45m 51s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a16d17365b344659a1cd9b6b0b242af4 : SUCCESS in 33m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d8495357714c4430b713f828c8484fbd : SUCCESS in 55m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7859e581d0f34dd085895e331a5eaff4 : SUCCESS in 57m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3006d7c1b81c4ae9a20a63fc594fe1e3 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":5},{"id":"360fbe410e22ccc2c5144f39121da658629cf2a8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-06 23:33:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"11863d568f2fd3f28f5838f83f105bdab635e48a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 23:33:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b090c37a082a5866589f391b8eb441576bf9d34e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 01:17:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9203fc05c17a4e61a94eca08d672ee25\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8b528d001a504e25b323cd838ddec6e6 : SUCCESS in 1h 43m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0013ed8e85314820913f8ddb7c5e9085 : SUCCESS in 13m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2b01f3f0bc54265ab21b84719b7a2b7 : SUCCESS in 20m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68d5f29880be467ebde21e6f5a602254 : SUCCESS in 18m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e769f9ee344983a455307315d61ebe : SUCCESS in 12m 03s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/829196d502534fd79d3279756f4661c6 : SUCCESS in 1h 33m 15s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/474c9fd39fda40478fefe4f0ced7af00 : SUCCESS in 51m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6806100b3f0a4e6887e28e7995cfd950 : SUCCESS in 55m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c13c9dd2400646608d0ce42620b87968 : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":5},{"id":"6dad785503f40ca14d4e635d53bdfeb8f6f54d27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 01:17:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b50d89318106bd04fe2df78e2130b82afef6c345","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 01:18:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ac49de356df4bf58e0941fc14458777\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed338a2b4ca44706b7c33425d8cc5df0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a5ca6d2bb8a2d6df92ecb09ad19898e7e2532d91","revisions":{"26d3d16c72db80f9ea79fe5dcd4d6b3d04988292":{"kind":"REWORK","_number":1,"created":"2025-09-30 08:32:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/69/962569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/962569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/962569/1"}}},"commit":{"parents":[{"commit":"00c21725f6abf33e9b4f176e779d0c19b7e83420","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00c21725f6abf33e9b4f176e779d0c19b7e83420"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-29 10:34:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-30 08:32:17.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] sync_unify_ha_chassis_group_network","message":"WIP \u003d\u003d [OVN] sync_unify_ha_chassis_group_network\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia4f685077a8d72bf28f66daf21225d96f57ddef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26d3d16c72db80f9ea79fe5dcd4d6b3d04988292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26d3d16c72db80f9ea79fe5dcd4d6b3d04988292"}]},"branch":"refs/heads/master"},"dd6e94527feeab41cb86b2eed17272e52a649d2c":{"kind":"REWORK","_number":2,"created":"2025-09-30 11:45:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/69/962569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/962569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/962569/2"}}},"commit":{"parents":[{"commit":"00c21725f6abf33e9b4f176e779d0c19b7e83420","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00c21725f6abf33e9b4f176e779d0c19b7e83420"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-29 10:34:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-30 11:41:18.000000000","tz":0},"subject":"[OVN] Add method ``sync_unify_ha_chassis_group_network``","message":"[OVN] Add method ``sync_unify_ha_chassis_group_network``\n\nThis new method is meant to be used when a network with external ports\nis connected to a router. In this case, it is needed to create a single\nscheduler for all the ports connected to the router and matching the\nsame chassis as the gateway router port.\n\nThis method uses the network ID to assign the group name. It also uses\na defined chassis priority (that will match the gateway router port\nassignation). The created ``HA_Chassis_Group`` must have a set of\n``HA_Chassis`` registers that should have the same chassis name and\npriority as the gateway router port ``Gateway_Chassis``.\n\nNOTE: in future developments, the router port ``Gateway_Chassis`` will\nbe replaced with a ``HA_Chassis_Group`` register.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia4f685077a8d72bf28f66daf21225d96f57ddef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd6e94527feeab41cb86b2eed17272e52a649d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd6e94527feeab41cb86b2eed17272e52a649d2c"}]},"branch":"refs/heads/master"},"3641ae6f6a1c815373f1c2b9c5e1aa3be043e298":{"kind":"REWORK","_number":3,"created":"2025-10-01 09:18:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/69/962569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/962569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/962569/3"}}},"commit":{"parents":[{"commit":"00c21725f6abf33e9b4f176e779d0c19b7e83420","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00c21725f6abf33e9b4f176e779d0c19b7e83420"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-29 10:34:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-01 08:52:29.000000000","tz":0},"subject":"[OVN] Add method ``sync_unify_ha_chassis_group_network``","message":"[OVN] Add method ``sync_unify_ha_chassis_group_network``\n\nThis new method is meant to be used when a network with external ports\nis connected to a router. In this case, it is needed to create a single\nscheduler for all the ports connected to the router and matching the\nsame chassis as the gateway router port.\n\nThis method uses the network ID to assign the group name. It also uses\na defined chassis priority (that will match the gateway router port\nassignation). The created ``HA_Chassis_Group`` must have a set of\n``HA_Chassis`` registers that should have the same chassis name and\npriority as the gateway router port ``Gateway_Chassis``.\n\nNOTE: in future developments, the router port ``Gateway_Chassis`` will\nbe replaced with a ``HA_Chassis_Group`` register.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia4f685077a8d72bf28f66daf21225d96f57ddef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3641ae6f6a1c815373f1c2b9c5e1aa3be043e298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3641ae6f6a1c815373f1c2b9c5e1aa3be043e298"}]},"branch":"refs/heads/master"},"9e14fa625f525f592f36f1d21526d599c1eac351":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-10-01 09:18:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/69/962569/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/962569/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/962569/4"}}},"commit":{"parents":[{"commit":"00c21725f6abf33e9b4f176e779d0c19b7e83420","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00c21725f6abf33e9b4f176e779d0c19b7e83420"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-29 10:34:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-01 09:18:39.000000000","tz":0},"subject":"[OVN] Add method ``sync_ha_chassis_group_network_unified``","message":"[OVN] Add method ``sync_ha_chassis_group_network_unified``\n\nThis new method is meant to be used when a network with external ports\nis connected to a router. In this case, it is needed to create a single\nscheduler for all the ports connected to the router and matching the\nsame chassis as the gateway router port.\n\nThis method uses the network ID to assign the group name. It also uses\na defined chassis priority (that will match the gateway router port\nassignation). The created ``HA_Chassis_Group`` must have a set of\n``HA_Chassis`` registers that should have the same chassis name and\npriority as the gateway router port ``Gateway_Chassis``.\n\nNOTE: in future developments, the router port ``Gateway_Chassis`` will\nbe replaced with a ``HA_Chassis_Group`` register.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia4f685077a8d72bf28f66daf21225d96f57ddef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e14fa625f525f592f36f1d21526d599c1eac351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9e14fa625f525f592f36f1d21526d599c1eac351"}]},"branch":"refs/heads/master"},"a5ca6d2bb8a2d6df92ecb09ad19898e7e2532d91":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-10-02 16:02:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/69/962569/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/962569/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/962569/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/962569/5"}}},"commit":{"parents":[{"commit":"633ff6093bf7de8c724b310cac18b2b1a709effc","subject":"[OVN] Add the physical network to the ``Logical_Switch`` register","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/633ff6093bf7de8c724b310cac18b2b1a709effc"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-09-29 10:34:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-02 16:01:36.000000000","tz":0},"subject":"[OVN] Add method ``sync_ha_chassis_group_network_unified``","message":"[OVN] Add method ``sync_ha_chassis_group_network_unified``\n\nThis new method is meant to be used when a network with external ports\nis connected to a router. In this case, it is needed to create a single\nscheduler for all the ports connected to the router and matching the\nsame chassis as the gateway router port.\n\nThis method uses the network ID to assign the group name. It also uses\na defined chassis priority (that will match the gateway router port\nassignation). The created ``HA_Chassis_Group`` must have a set of\n``HA_Chassis`` registers that should have the same chassis name and\npriority as the gateway router port ``Gateway_Chassis``.\n\nNOTE: in future developments, the router port ``Gateway_Chassis`` will\nbe replaced with a ``HA_Chassis_Group`` register.\n\nRelated-Bug: #2125553\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ia4f685077a8d72bf28f66daf21225d96f57ddef6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5ca6d2bb8a2d6df92ecb09ad19898e7e2532d91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5ca6d2bb8a2d6df92ecb09ad19898e7e2532d91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
