)]}'
{"id":"openstack%2Fneutron~962589","triplet_id":"openstack%2Fneutron~stable%2F2025.2~I0304c8a9dc0b0bcb18a0e5e4360b5b1756cc6f29","project":"openstack/neutron","branch":"stable/2025.2","topic":"2025.2_release","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-09-30 19:32:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0304c8a9dc0b0bcb18a0e5e4360b5b1756cc6f29","subject":"[2025.2 only] Switch to 2025.2 neutron-tempest-plugin jobs","status":"MERGED","created":"2025-09-30 14:13:43.000000000","updated":"2025-09-30 19:33:37.000000000","submitted":"2025-09-30 19:32:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962589-2025.2_release","meta_rev_id":"05edc5c6282afa689847557d56c13a46e81c1bbf","_number":962589,"virtual_id_number":962589,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-30 19:32:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-09-30 17:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-09-30 17:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-09-30 17:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-09-30 17:03:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-30 15:00:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-09-30 16:12:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-30 17:03:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-09-30 17:27:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"cdefbc1cd9817e60d58f48ec9f13b6e9b98ae209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-09-30 14:13:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"623dedd887c95bda1923d5b273cabcb7a9db1ba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 16:12:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b60a464348430cb00781fcad695335\n\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/cf07dd2136d84c9998714170db967cba : SUCCESS in 1h 25m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/69d3dbd8ca3344d3a0dd0d02f0e32298 : SUCCESS in 1h 52m 00s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/7273ea1b805c4f49880b8fd6f9dccbbb : SUCCESS in 1h 40m 26s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/68563a3e91614a3e971283905f333d2c : SUCCESS in 38m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f09b167eb45e412c93e75af0cbb6c6a9 : SUCCESS in 14m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aba9544c48e473b872a0c55c5772405 : SUCCESS in 13m 44s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f15626cd01044dc68760fba2ac0efbdc : SUCCESS in 1h 53m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9169733c27ff4d7399eec5b740fa6cb2 : SUCCESS in 1h 00m 49s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/571654fa26384fcebfe2595ef87c917e : SUCCESS in 1h 10m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/90ef666144d54f3aa16b6e209a620e82 : SUCCESS in 1h 29m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9ef438d243dd494abccda1f846267bad : SUCCESS in 1h 23m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a8a25c6e9a324ed18189c2b45443f47e : SUCCESS in 59m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b223abb286c548ed918de5d999c5f1bb : SUCCESS in 53m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/14139bf3202141cfb18e9cd321ffe4ef : SUCCESS in 1h 01m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/13190f28fe9c48fca23f549fab69ccac : SUCCESS in 1h 18m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b03c7b46d42134b83966f8f7a0deed0864815481","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-09-30 17:03:18.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"6fe02f8d23f50481a810c951b55879f966ffae9e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-30 17:27:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4149f322b3c5ffc07283bf15ac1e244f490bb509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 17:27:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a42a52dccd084762d41f57149774d888870ee33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 19:32:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6427248b68f4265ab32a4963fb5f68f\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/a06952b887de4bedbb828acc9e7f1ddd : SUCCESS in 2h 04m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ee571c6cfec40b083b9909e6f38c228 : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e672ad65f164f06ae3600bd6bd4cb67 : SUCCESS in 12m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c5ba20e5e57d4c7b981b2a6352b1f050 : SUCCESS in 1h 46m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f8eb2eba8b6746838283357d8d034eff : SUCCESS in 1h 44m 19s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/318a62e237dd427987a5dd92cc8b848a : SUCCESS in 1h 01m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d595df6a9243434dae18a55dde78e804 : SUCCESS in 55m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/38de0e8dfdc8468ab18418979db8b798 : SUCCESS in 45m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/911c61a9ffb348d499ba8934acfe66da : SUCCESS in 1h 20m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3718ea5044c3b03ab0c37878898970e77ed59afc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 19:32:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05edc5c6282afa689847557d56c13a46e81c1bbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 19:33:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d67ad554eac422886c94a8ec855a500\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b29781e24b84da2a63ce0609f5a504a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e03b0a376f71255fb875fed8be48741a9edb0224","revisions":{"e03b0a376f71255fb875fed8be48741a9edb0224":{"kind":"REWORK","_number":1,"created":"2025-09-30 14:13:43.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/89/962589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/962589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/962589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/962589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/962589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/962589/1"}}},"commit":{"parents":[{"commit":"253ede2b1b1de76e2452c4e2021040945469d4d6","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/253ede2b1b1de76e2452c4e2021040945469d4d6"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2025-09-30 14:13:08.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2025-09-30 14:13:08.000000000","tz":330},"subject":"[2025.2 only] Switch to 2025.2 neutron-tempest-plugin jobs","message":"[2025.2 only] Switch to 2025.2 neutron-tempest-plugin jobs\n\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\nChange-Id: I0304c8a9dc0b0bcb18a0e5e4360b5b1756cc6f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e03b0a376f71255fb875fed8be48741a9edb0224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e03b0a376f71255fb875fed8be48741a9edb0224"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
