)]}'
{"id":"openstack%2Fneutron~962755","triplet_id":"openstack%2Fneutron~stable%2F2025.2~I1dc6eabbb666e5923d9c18465d10cdf95e472915","project":"openstack/neutron","branch":"stable/2025.2","topic":"fix-rbac-unit-tests-stable/2025.2","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-10-02 23:29:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1dc6eabbb666e5923d9c18465d10cdf95e472915","subject":"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs","status":"MERGED","created":"2025-10-02 08:07:55.000000000","updated":"2025-10-02 23:30:45.000000000","submitted":"2025-10-02 23:29:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962755-fix-rbac-unit-tests-stable/2025.2","meta_rev_id":"85caabf76cd798a3f96b0a0391da55a7f829e1a0","_number":962755,"virtual_id_number":962755,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-02 23:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-10-02 12:31:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-10-02 15:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-10-02 15:46:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 10:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 12:31:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-10-02 12:31:40.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-10-02 15:46:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"7d87c0a7f7907caf75a5e51d9403f64c183c3ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-10-02 08:07:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f30145a8c41e9e7bd1b978c79fbc33dd046b862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 10:32:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89bf1e496a64424eac6e457d5810c72f\n\n- neutron-tempest-plugin-openvswitch-2025-2 https://zuul.opendev.org/t/openstack/build/60096f4167574c93b50612db9c248697 : SUCCESS in 1h 49m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-2 https://zuul.opendev.org/t/openstack/build/1c3acf30ad4a462b96220c1bbe656204 : SUCCESS in 1h 45m 37s\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/cd54ae365eee431a95e61122656b6662 : SUCCESS in 2h 23m 09s\n- neutron-tempest-plugin-designate-scenario-2025-2 https://zuul.opendev.org/t/openstack/build/24a1d5f332e742e88a3a287cec60bb16 : SUCCESS in 34m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5a9977b57044610ad426a5b0e377347 : SUCCESS in 48m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/100f690f32f24f0ab4494ab89d2bce5c : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c82879cb7f064d0da6d2ecea4a906918 : SUCCESS in 23m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/980b22a479814a2e9b6d864e4cbb5596 : SUCCESS in 20m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65417fded2f84b9c81c09d56c05b7057 : SUCCESS in 28m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7d4250a3bbc40eaa8265ca66bcb8699 : SUCCESS in 12m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fbc2f51ddd4f4924b6595d7728fc45eb : SUCCESS in 1h 47m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f22ef66efb3f4a878d63365c37e16e7b : SUCCESS in 59m 27s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2498b3ba0354197bc716a42aa4a19a8 : SUCCESS in 1h 10m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b84c67d2f9474bca984fb3d2b5645278 : SUCCESS in 1h 41m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f525c6f273a24b94b9f51094a09bb94b : SUCCESS in 1h 32m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/52904324c5644d1aaebcb946581a5ad3 : SUCCESS in 1h 04m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/df917bc2c93946c9a475ce03085e7aad : SUCCESS in 26m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fdd29512b452453a8167e72616fc041b : SUCCESS in 1h 12m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fc7afb5157b14867ae13ee2f310bba49 : SUCCESS in 1h 28m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8802a0312cce6e2d8010938ec424a939cae83814","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 10:52:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf549beab85b44c2a949dd3522aec5f8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5fe960ea394f4feebdbf2d1722783ae4 : SUCCESS in 55m 20s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab272f17202043d780457b34aece514b : SUCCESS in 18m 28s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6b916a9b288a4fe5a4eed12bd429d18f : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5e84cbc9d24b2debe5c14919a7f4ad38e847d6a3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-02 12:31:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f80b0202cb2b9e4fe8f07f44e41c8587cf1128d4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-10-02 15:46:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae76ddf023ed8af1a64b9ebaebaa0ff749e46281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 21:35:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc451d9723f5c695abe0e2af00ce726be244d2e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:29:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52b71f95f40c4b5a8871f68f0199f650\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/87a24e30b1314646bb4cb8e17e7cc91a : SUCCESS in 1h 41m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a140118d26b24fe8a990ef9d950fd78e : SUCCESS in 12m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d960d3e048345dc8bba5b1a0ba60494 : SUCCESS in 26m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21a5f17b1087451e90c0643a23cd75c0 : SUCCESS in 20m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfbcde34fa134570b9ec24a50059de5d : SUCCESS in 11m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8163dc02fc2549049efc4f41bcc84c58 : SUCCESS in 1h 49m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/871e40f45a97491c9f92c7bfd790ff31 : SUCCESS in 1h 38m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/cafa823e4e5d4c58b15d5e9280fc07da : SUCCESS in 1h 06m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd95395409914125b984ba1806d803db : SUCCESS in 59m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/77e89d4ed536403a92c935239847ec12 : SUCCESS in 1h 10m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/22505e704dbb4b1190cedec10208bf00 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"28e04b82080d37e120765ce71243b815d708b254","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:29:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"85caabf76cd798a3f96b0a0391da55a7f829e1a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:30:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f539ffa4af814609a9f86054a5bf27eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dadb7c3970c48b8a39189ba4a0d93cb : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"346dfed03d2a472147247655b1bef4c48565d8d5","revisions":{"346dfed03d2a472147247655b1bef4c48565d8d5":{"kind":"REWORK","_number":1,"created":"2025-10-02 08:07:55.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/55/962755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/962755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/962755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/962755/1"}}},"commit":{"parents":[{"commit":"e9aa546d30021d517646ecfc36545b1fcdbfc168","subject":"[S-RBAC] Fix policies for the local_ip association APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9aa546d30021d517646ecfc36545b1fcdbfc168"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-25 09:53:40.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-10-02 08:07:24.000000000","tz":120},"subject":"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs","message":"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs\n\nThis patch updates l3_conntrack_helpers API policies so that POST, PUT and\nDELETE actions are allowed for the PARENT_OWNER_MEMBER role and GET is\nallowed for the PARENT_OWNER_READER.\n\nAdditionally this patch fixes unit tests for the api policies for that\nAPIs so that owner check is done during unit tests and issues like the\none mentioned above can be catched by unit tests.\n\nCloses-bug: #2125660\n\nChange-Id: I1dc6eabbb666e5923d9c18465d10cdf95e472915\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n(cherry picked from commit cb3331e52580a67fd6d65b3e44eca9b14fc9cefb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/346dfed03d2a472147247655b1bef4c48565d8d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/346dfed03d2a472147247655b1bef4c48565d8d5"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
