)]}'
{"id":"openstack%2Fneutron~962756","triplet_id":"openstack%2Fneutron~stable%2F2025.1~I4c26403e237afdb8934dda65fef3b12c99a5f689","project":"openstack/neutron","branch":"stable/2025.1","topic":"fix-rbac-unit-tests-stable/2025.1","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-10-02 23:29:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c26403e237afdb8934dda65fef3b12c99a5f689","subject":"[S-RBAC] Fix tests for port RBAC policies","status":"MERGED","created":"2025-10-02 08:22:39.000000000","updated":"2025-10-02 23:30:41.000000000","submitted":"2025-10-02 23:29:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962756-fix-rbac-unit-tests-stable/2025.1","meta_rev_id":"3a698a267970bff45b3938d20765190146b2b630","_number":962756,"virtual_id_number":962756,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-02 23:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-10-02 09:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-10-02 21:38:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-10-02 21:38:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 09:00:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 09:44:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-10-02 12:31:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2025-10-02 21:38:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"274c7ac9faa1ea9a383769f2bf7c4bd26f712bc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-10-02 08:22:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"608023557c47e01bb86499f311032ddc32bb099c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 09:00:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20440fd2d414cbe9448579e556487bd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/504f05cb2f58423eabcf7579f88db97f : SUCCESS in 37m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88db4aafb101431a9f92c545d2459a6b : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2806c8db524a4b798410ac7523974b18 : SUCCESS in 21m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec0c94b01bf546df8eefa933bcbb2f2c : SUCCESS in 21m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d116c4643b1246b9b3817b73931497c9 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d71566775839e017632eb779748324ee76f6fd48","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-02 09:44:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b11af03abb839114af7ceac10b02fe99c08ea3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 10:33:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed8d8bdddfd64e449b061d6481e1dab4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a08e00bf9c684ad1aabf541e105a581a : SUCCESS in 37m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a17d5c5ae2849a1acb3c94521823512 : SUCCESS in 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea6a21a143e4f0d10beae03dda2854f53d4df234","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-10-02 21:38:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78dc06e981b5443717d4262e520fa2fc2c8475c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 21:38:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"900628611ade78b0d0a06bd5dabd33ee0510056a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:29:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7943b337117945edb3f797c36ea79188\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d152eda8e064c91826e6cfbfb5626e7 : SUCCESS in 11m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e34bb4e7b2c54baea950406421a2af6e : SUCCESS in 19m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95a5061048b14b43bda904d53f9dfd81 : SUCCESS in 25m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3601a06381454820b7521bcda4a9f3bf : SUCCESS in 10m 12s","accounts_in_message":[],"_revision_number":1},{"id":"411b81977c864ef22fc44f23d59c89f143e55f92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:29:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a698a267970bff45b3938d20765190146b2b630","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:30:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e13d19bee28149b4b2db1ee41c97c642\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2446bc0ea58644779451397a061c5f52 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1555e14677a85b76291bcf79a2af7135a1fbfa30","revisions":{"1555e14677a85b76291bcf79a2af7135a1fbfa30":{"kind":"REWORK","_number":1,"created":"2025-10-02 08:22:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/56/962756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/962756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/962756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/962756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/962756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/962756/1"}}},"commit":{"parents":[{"commit":"0eca108015d080e8031a88e7498a0303f776f37f","subject":"Merge \"Decode the metadata response before sending it to the VM\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0eca108015d080e8031a88e7498a0303f776f37f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-09-24 14:46:32.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-10-02 08:21:56.000000000","tz":120},"subject":"[S-RBAC] Fix tests for port RBAC policies","message":"[S-RBAC] Fix tests for port RBAC policies\n\nExisitng tests didn\u0027t mock \"get_network\" function properly thus\nthere was no \"network:tenant_id\" field in the target object set and\nOwnerCheck check was failing for the \"net_owner\".\nUnit tests were green because they badly expected e.g. for the\nProjectMember that e.g. port with given fixed_ip or mac_address will not\nbe allowed (exception was raised because net_owner check failed) even\nthoug it should be possible through the net_owner rule in policy.\n\nThis patch fixes mocking get_network function in unit tests module for\nthe port RBAC policies and it also updates unit tests accordingly to\nmake them pass.\n\nChange-Id: I4c26403e237afdb8934dda65fef3b12c99a5f689\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n(cherry picked from commit a48b715e8b1ea0ef1e316637df29dc040007b20c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1555e14677a85b76291bcf79a2af7135a1fbfa30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1555e14677a85b76291bcf79a2af7135a1fbfa30"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
