)]}'
{"id":"openstack%2Fneutron~962862","triplet_id":"openstack%2Fneutron~stable%2F2024.1~I9cd6e77e47ce16dcf2fc46fc076f1d7ee7a4f9d3","project":"openstack/neutron","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2025-10-07 12:37:59.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-10-07 14:16:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9cd6e77e47ce16dcf2fc46fc076f1d7ee7a4f9d3","subject":"[OVN] Set always the GW LRP \"gateway_mtu\" option","status":"MERGED","created":"2025-10-02 17:56:36.000000000","updated":"2025-10-07 14:17:22.000000000","submitted":"2025-10-07 14:16:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"962862","cherry_pick_of_change":962456,"cherry_pick_of_patch_set":3,"meta_rev_id":"1b5d1f3b1ced7d39a901470809978f53ebb2a1cd","_number":962862,"virtual_id_number":962862,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-07 14:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2025-10-07 12:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2025-10-06 08:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2025-10-07 12:37:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 17:56:36.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-10-02 18:27:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-03 09:04:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-10-06 08:18:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-10-06 23:18:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"56659d07d622db1bccde7109b54de3d3a5eb82a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-02 17:56:36.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.\n\nThe following files contain Git conflicts:\n* neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n* neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py\n* neutron/tests/unit/services/ovn_l3/test_plugin.py","accounts_in_message":[],"_revision_number":1},{"id":"9b59af8f7bac70db179ccff811621351054408b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 18:27:12.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c10c0c5a49714713949fcc5ce6db46a8\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/712c157c319449e492e4c8798eb8754b : FAILURE in 14m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2553e9d2dd134cc0b71b4c467202cadb : FAILURE in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc90add1f81d4fc5b898f8aae8c6bede : FAILURE in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03f346d28f9b45ecb2bb0550cb835cf7 : FAILURE in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e591d3a5a9f41d1bc1e6dac23505fef : FAILURE in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a87fa6c03308426ea6baf88c9a5cff11 : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f886cb820b9e44fb9cdc43f5c3907e5f : SUCCESS in 23m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/79064deeb1cd4bb4b3fe99cd21722442 : FAILURE in 10m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bfa8ce261f0647c8b675dc93bbc65c6a : FAILURE in 20m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d31b5f40c344032b47f71794f390730 : FAILURE in 25m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ff8141f21a484e379ca63afd76ebadb0 : FAILURE in 21m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f473cb52caeafd1be51eace1c91319615a9cb4fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-02 21:42:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d6badf6fe8582dcc2a3367b94d69b36bf1c30144","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-02 21:42:47.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"0651eba1344e52437e93fad30cd7462505d6db10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:29:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8ca6b8a2ffa4c2aa44e087f2af35f79\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/5fe3ec0618574d4baacddac742862b31 : SUCCESS in 1h 41m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/035223118d874e398e03c6002d2d2c8b : FAILURE in 38m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76778e0b05984c90b2892dc47cb1dab0 : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/315c1010d39a4d4392ff33a3e0007f0b : FAILURE in 22m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1120c638ffe4cb0b638745719620e93 : FAILURE in 16m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f028e8bbfcac452a862df25523405222 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a90c75aeeffd43aeb72eac41ee3380b6 : SUCCESS in 19m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a696c0971ec742759b784adb1850d7d0 : SUCCESS in 1h 41m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e2c3f518a9e74bb59bf22bb4a41f7518 : SUCCESS in 42m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/336f058c3fa44573bcf63d9b7c506c0a : SUCCESS in 1h 34m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9244c5c3d2834ed09b0fe04d5a821a00 : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":2},{"id":"06965f6a976f96ee2f280e86460785b71ac7610c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 00:23:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0775da7efd274c9ca336f28ebe511e44\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d727f261251b4de78a828f190c7f7d58 : FAILURE in 1h 00m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cbd165fe3e1f4465b099e64bc7d0ebd6 : FAILURE in 46m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"070dea910be0b55a9826780a58a46509cf625fff","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-03 09:04:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f17d1172b802bd1dabd6feb41855c21b1442c805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-03 20:38:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fd10656946f26c0c52602d6fa9375c3ae296025c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 21:25:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57aba3e9364b442c8e714cbca5855bc7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc26e1fb2e1e46e9830b5f9e914a59c1 : SUCCESS in 45m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ed7c932afc6b46f1a45d611849e66394 : SUCCESS in 41m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55f33fb55c8f9f1db30841b5c28a28b328bc4029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 22:21:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b53ca460ee854fbfbbf181508ef2fcb1\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/d1321a41d602494b8d9697fd7bd61955 : SUCCESS in 1h 23m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cad045a5f3bb4ea78f72c366eb72ea49 : SUCCESS in 39m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cbbc54a2eb34763b66ecadad224bb21 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96360349d9cb4cdf86e45581e166a62f : SUCCESS in 27m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83fd6e30a25c4e62b437768868581492 : SUCCESS in 24m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07b655a662c4603acf06b1650d96b23 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0efa9c5087434c07ac2ecdafff7454c8 : SUCCESS in 12m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/94719463c48c4358928f98c7d52c344d : SUCCESS in 1h 38m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5e452c3e1e4b41169b2c39433bf40736 : SUCCESS in 45m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4523aca5766b4dc2aeaf9db482edc506 : SUCCESS in 1h 27m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c5335697172b4001b1e22dee8377101b : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":3},{"id":"dfb80d30795ad9c96aa35cdc9994adcc8b72f537","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-06 08:18:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"775d57528c5e5ef6a0304487ca68e975df76633b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-10-07 12:37:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9511b7402b0efade4937a3a961235d0abb039870","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 12:38:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"004e694674ea1a4d968e326377f46c9bd1c01ecf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 14:16:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5a0235a3db44fe88c3a78c3b19e8f5f\n\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/83ad76f8987242999648ea6e2997e33b : SUCCESS in 1h 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d689c56b8c4482a4689f24a121395c : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bc5573b09a74eda82e289397063d688 : SUCCESS in 27m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f6abe20ca0743288ad15e0168dd68c8 : SUCCESS in 16m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43965a06e2a4eb2814d7056f38fdf64 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac12a0efaddf4c4c93ff3add700443fc : SUCCESS in 12m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/16d7d9c53fbd4fa08285c7835a568832 : SUCCESS in 1h 16m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5fb02d6f5c22431cae4d83b2b7435ca3 : SUCCESS in 1h 01m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20661b7c60374ce0ba9f534b4b275eef : SUCCESS in 41m 50s","accounts_in_message":[],"_revision_number":3},{"id":"e03a82e6a97e34593d559f7afe4b815b555aabeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 14:16:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1b5d1f3b1ced7d39a901470809978f53ebb2a1cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 14:17:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efa026a790c24b18b3d5ed11c4862ffe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f98ffc6e15b43328ce0eafef26b2b28 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1984cc77b5d54a298a9d0051c9fff43c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f9ee4663fe08d8f3936b6c4f10e51b6363605cc","revisions":{"e745a63e88dd2e654bb536a21b74f6a5af345d47":{"kind":"REWORK","_number":1,"created":"2025-10-02 17:56:36.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/962862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/962862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/962862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/962862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/962862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/962862/1"}}},"commit":{"parents":[{"commit":"3f42ac5cafabfb810f89b545a7f4cdada18ac715","subject":"[stable] Pin rally\u003d4.1.0, rally-openstack\u003d3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f42ac5cafabfb810f89b545a7f4cdada18ac715"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-04 14:02:02.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-10-02 17:56:36.000000000","tz":0},"subject":"[OVN] Set always the GW LRP \"gateway_mtu\" option","message":"[OVN] Set always the GW LRP \"gateway_mtu\" option\n\nNOTE: stable/2024.2 only, as this is a combination of\ntwo commits in one to not introduce a regression by\ndoing them separately. There were small changes required\nto unit tests as the code had changed in 2025.1 where\nthis was originally merged.\n\nFirst commit:\n\nIf the configuration option \"ovn.ovn_emit_need_to_frag\" is set, write\nalways the \"LSP.options.gateway_mtu\" value, using the minimum MTU\nvalue of all networks connected to the router. That will limit the\nN/S traffic MTU regardless if the gateway network MTU is greater or\nless than the internal network MTU.\n\nConflicts:\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py\n  neutron/tests/unit/services/ovn_l3/test_plugin.py\n\n(cherry picked from commit d080b35782c705b105da61af3949d67832efd2c8)\n\nSecond commit:\n\nRevert the network IDs filter in ``_gen_router_port_options``\n\nIn [1], a change in the ``network_ids`` filter was introduced. This\nfilter is used to retrieve the networks belonging to a router. However\nthis is used in two places:\n* To set the gateway MTU. Here all the networks attached to the router\n  must be present.\n* To calculate the \"redirect-type\" value in the Logical_Router_Port.\n  Only the internal interfaces must be present; the gateway port should\n  be discarded.\n\nNOTE: patch [1] is present since 2025.1 (Epoxy).\n\nConflicts:\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n\n[1]https://review.opendev.org/c/openstack/neutron/+/937026\n\nCloses-Bug: #2105383\nRelated-Bug: #2032817\nChange-Id: I9cd6e77e47ce16dcf2fc46fc076f1d7ee7a4f9d3\n(cherry picked from commit db030749278324b52b6d09f4d55a5d1e288d0abe)\n(cherry picked from commit 2489d8009e5202af0c93c473c6771f0b6eaacb1b)\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n(cherry picked from commit 197c920cdb75418fa4072e1e175789a61eaab804)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e745a63e88dd2e654bb536a21b74f6a5af345d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e745a63e88dd2e654bb536a21b74f6a5af345d47"}]},"branch":"refs/heads/stable/2024.1"},"9571ae0379a019f92ff4fb2559355fdd9feadf11":{"kind":"REWORK","_number":2,"created":"2025-10-02 21:42:31.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/962862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/962862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/962862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/962862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/962862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/962862/2"}}},"commit":{"parents":[{"commit":"3f42ac5cafabfb810f89b545a7f4cdada18ac715","subject":"[stable] Pin rally\u003d4.1.0, rally-openstack\u003d3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f42ac5cafabfb810f89b545a7f4cdada18ac715"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-04 14:02:02.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-10-02 21:40:59.000000000","tz":-240},"subject":"[OVN] Set always the GW LRP \"gateway_mtu\" option","message":"[OVN] Set always the GW LRP \"gateway_mtu\" option\n\nNOTE: stable/2024.1 only, as this is a combination of\ntwo commits in one to not introduce a regression by\ndoing them separately. There were small changes required\nto unit tests as the code had changed in 2025.1 where\nthis was originally merged.\n\nFirst commit:\n\nIf the configuration option \"ovn.ovn_emit_need_to_frag\" is set, write\nalways the \"LSP.options.gateway_mtu\" value, using the minimum MTU\nvalue of all networks connected to the router. That will limit the\nN/S traffic MTU regardless if the gateway network MTU is greater or\nless than the internal network MTU.\n\nConflicts:\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py\n  neutron/tests/unit/services/ovn_l3/test_plugin.py\n\n(cherry picked from commit d080b35782c705b105da61af3949d67832efd2c8)\n\nSecond commit:\n\nRevert the network IDs filter in ``_gen_router_port_options``\n\nIn [1], a change in the ``network_ids`` filter was introduced. This\nfilter is used to retrieve the networks belonging to a router. However\nthis is used in two places:\n* To set the gateway MTU. Here all the networks attached to the router\n  must be present.\n* To calculate the \"redirect-type\" value in the Logical_Router_Port.\n  Only the internal interfaces must be present; the gateway port should\n  be discarded.\n\nNOTE: patch [1] is present since 2025.1 (Epoxy).\n\nConflicts:\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n\n[1]https://review.opendev.org/c/openstack/neutron/+/937026\n\nCloses-Bug: #2105383\nRelated-Bug: #2032817\nChange-Id: I9cd6e77e47ce16dcf2fc46fc076f1d7ee7a4f9d3\n(cherry picked from commit db030749278324b52b6d09f4d55a5d1e288d0abe)\n(cherry picked from commit 2489d8009e5202af0c93c473c6771f0b6eaacb1b)\n(cherry picked from commit 197c920cdb75418fa4072e1e175789a61eaab804)\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9571ae0379a019f92ff4fb2559355fdd9feadf11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9571ae0379a019f92ff4fb2559355fdd9feadf11"}]},"branch":"refs/heads/stable/2024.1"},"9f9ee4663fe08d8f3936b6c4f10e51b6363605cc":{"kind":"REWORK","_number":3,"created":"2025-10-03 20:38:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/62/962862/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/62/962862/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/962862/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/962862/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/62/962862/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/62/962862/3"}}},"commit":{"parents":[{"commit":"3f42ac5cafabfb810f89b545a7f4cdada18ac715","subject":"[stable] Pin rally\u003d4.1.0, rally-openstack\u003d3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3f42ac5cafabfb810f89b545a7f4cdada18ac715"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-04 14:02:02.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-10-03 20:38:23.000000000","tz":-240},"subject":"[OVN] Set always the GW LRP \"gateway_mtu\" option","message":"[OVN] Set always the GW LRP \"gateway_mtu\" option\n\nNOTE: stable/2024.1 only, as this is a combination of\ntwo commits in one to not introduce a regression by\ndoing them separately. There were small changes required\nto unit tests as the code had changed in 2025.1 where\nthis was originally merged.\n\nFirst commit:\n\nIf the configuration option \"ovn.ovn_emit_need_to_frag\" is set, write\nalways the \"LSP.options.gateway_mtu\" value, using the minimum MTU\nvalue of all networks connected to the router. That will limit the\nN/S traffic MTU regardless if the gateway network MTU is greater or\nless than the internal network MTU.\n\nConflicts:\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py\n  neutron/tests/unit/services/ovn_l3/test_plugin.py\n\n(cherry picked from commit d080b35782c705b105da61af3949d67832efd2c8)\n\nSecond commit:\n\nRevert the network IDs filter in ``_gen_router_port_options``\n\nIn [1], a change in the ``network_ids`` filter was introduced. This\nfilter is used to retrieve the networks belonging to a router. However\nthis is used in two places:\n* To set the gateway MTU. Here all the networks attached to the router\n  must be present.\n* To calculate the \"redirect-type\" value in the Logical_Router_Port.\n  Only the internal interfaces must be present; the gateway port should\n  be discarded.\n\nNOTE: patch [1] is present since 2025.1 (Epoxy).\n\nConflicts:\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n\n[1]https://review.opendev.org/c/openstack/neutron/+/937026\n\nCloses-Bug: #2105383\nRelated-Bug: #2032817\nChange-Id: I9cd6e77e47ce16dcf2fc46fc076f1d7ee7a4f9d3\n(cherry picked from commit db030749278324b52b6d09f4d55a5d1e288d0abe)\n(cherry picked from commit 2489d8009e5202af0c93c473c6771f0b6eaacb1b)\n(cherry picked from commit 197c920cdb75418fa4072e1e175789a61eaab804)\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f9ee4663fe08d8f3936b6c4f10e51b6363605cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9f9ee4663fe08d8f3936b6c4f10e51b6363605cc"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
