)]}'
{"id":"openstack%2Fneutron~962871","triplet_id":"openstack%2Fneutron~master~Iac731787ed858fe0c0f752e3bdf400005c5f40be","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"21040":{"account":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"last_update":"2025-10-10 11:47:18.000000000","reason":"Change was submitted"},"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-10 06:15:30.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-10-10 08:37:16.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-10-10 11:47:18.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-10-09 01:31:20.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Iac731787ed858fe0c0f752e3bdf400005c5f40be","subject":"Fix periodic_resync thread in DHCP agent","status":"MERGED","created":"2025-10-02 19:45:23.000000000","updated":"2025-10-10 11:48:38.000000000","submitted":"2025-10-10 11:47:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962871","meta_rev_id":"98fe9143018dba46177b7cbb3adb639415269253","_number":962871,"virtual_id_number":962871,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":1,"date":"2025-10-08 19:08:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-10 11:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-10-10 06:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-09 20:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-10-10 06:15:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-10-10 06:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 19:50:58.000000000","updated_by":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"reviewer":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"state":"REVIEWER"},{"updated":"2025-10-02 21:40:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-03 01:22:58.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-03 14:38:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-10-03 14:38:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-10-08 08:28:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-10-08 13:48:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8b488f6584e7ee22a2c398ad41ea06b48213714d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-02 19:45:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b8e6880c58ad95ef54606091e0959f8b1f01b69","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2025-10-02 19:50:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"04415d2e8e822131fe7689eaaa978813d66e461c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 21:40:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/270fabeeb921412799ed42c142ad4871\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d84978b4c01048f39cface5ebdad67dc : SUCCESS in 1h 27m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c3b04813b9147f89551917bac84939d : FAILURE in 44m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13742279b1de4c8a9f1cf66723671938 : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b835fc7ed1e4bb988738631705de004 : FAILURE in 15m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da22f9618e844316994e94ad19dfb9f8 : FAILURE in 15m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a6596b967cf425a8f40c72a92d0050a : FAILURE in 26m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7103e70320014d1aa433cfed2b3edd4b : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86cbc5149eac4a4ea961cdbd010ed632 : SUCCESS in 22m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b62a4653ccf44d6f958f52dbc4b045fb : SUCCESS in 57m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5ff50080237e4dc583383edfe02e6965 : SUCCESS in 1h 30m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/27d57346dfb64392b0c61f69fe940e11 : SUCCESS in 58m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c8d32f31076241f0aa2f1997e93cdb6b : SUCCESS in 1h 48m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/798e03673c4a4c0d9eeea0f7f4bed70c : SUCCESS in 1h 35m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c74e9790240f45acb09239e2844aecb6 : SUCCESS in 26m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0629fd804ad9390354785aa1a461f0d2cb41af12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 23:55:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a23396c99384aa7a49c90cf58538b2b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a8006c5395d41868c9c5dffb2ce3f10 : FAILURE in 18m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cff901d121c74e51bf504a9855853c12 : FAILURE in 44m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eee2a706c96e4666a1493f5cd61de58f : FAILURE in 35m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3020fa5fd704bbed035cd0db53b4fa3c5f6f2818","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-03 01:22:58.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/71/962871/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"6dc96f39ff977d3302f5f7ae17c39aefa236cc60","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-03 14:38:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6c7c4872e1459a2ee421cddcad59d5bff0611f8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-06 12:19:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f759dc345184b7c5fab06595f3f1c273e1e570e4","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-07 17:45:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e19b193040416c6b77ea87c29f65aa4602296b79","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-07 17:50:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c81206832b9fc2d92ef77cc38bced057dde3279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-08 08:18:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c45937e22b5c95f0eff629ca254b26a832a6e570","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-08 08:28:21.000000000","message":"Patch Set 2: Code-Review-1 Review-Priority+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"533d70264f7101b89437e1bdeec0e21c520caca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-08 08:51:05.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a3c23cdf5a0e53a5bf23359c9e23b967874002db","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-08 08:51:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6252c82213630646f020026caf58e981bdf1d4f1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-08 09:04:28.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/71/962871/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"f106ae255997baa6964d4530c8d47b0ac887f3b4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-08 09:50:55.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/71/962871/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"dc58e3f8712f7ea49bb922159eb02ffe395254e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 10:45:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61654f553c8a41adae05688cad5e0bef\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/522a4d97c254450ca5f472afd02655ed : SUCCESS in 47m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/153f82e1bb8d47b5ba5f18178af577ff : SUCCESS in 28m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fe0698869ef54cbbbb832d0a1c66d3d3 : SUCCESS in 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef7ee52479950bd27a4cdf9715ae30621dbe980b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 10:52:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/408fcac41dee4596aa2f8be0f0d2bfde\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/850fa2ff5f1446d398acd7f144513806 : SUCCESS in 1h 17m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e62e196833aa482bb1415ccc4b159510 : SUCCESS in 46m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c286a7d9d080408e85be35d9d024b8d5 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0efc6f81e974b178fe6a5177ea11d65 : SUCCESS in 22m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c51313f36534b558351afe0ecadd23d : SUCCESS in 20m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/929d6357058b4821b084479eda185df5 : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc67108a37c4074888cd4a1f5c46d4c : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0800a8ce01f94d4bab084c538f7c634e : SUCCESS in 26m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a3ab37b49d3d4a25b7621ddd1b646388 : SUCCESS in 53m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/260aac8a06d84cbdbdfe48f09357425f : SUCCESS in 1h 55m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7f7802b255c4873a212206253f51c03 : SUCCESS in 56m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3eee01cbfaeb4492a26790ff0125f249 : SUCCESS in 1h 49m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d608f6de5e104196bbd6ac586002605d : SUCCESS in 1h 30m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e3922e07a81e4de4b97519f56ebdd5cb : SUCCESS in 26m 14s","accounts_in_message":[],"_revision_number":3},{"id":"22ce1791a3374236f3b03c3504016da176dd2fd3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-08 13:48:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dbe849e48b712fb3b30f9d1a0b6fb76d3b27f4da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-08 13:54:16.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4d80121af89b4708eb5366b70d9ca04aa881e32d","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-08 13:54:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fc836ed2acddcbe28784f3a044810143b5d85316","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-08 15:28:01.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/71/962871/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ecee32ae1eaf367845e08bc4b7728c15ec95f6c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 16:15:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a174f925b445669c1f91eb1f4d4cad\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1985aa247c9f48eea3f6dc46fff561ca : SUCCESS in 21m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e58c46c3fdce464a91dec34d4061fdf5 : SUCCESS in 41m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/618c972085f843b3b17a93a18fc1158f : SUCCESS in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d21222f4a078062ff418b25a5d94358cb99f98d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 16:27:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6958531b472c426e994ff299da9c023f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8bb049de1a4c4efd9ab0b8736a1057e0 : FAILURE in 2h 31m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34a4d586140d4704b91e12f0f0f74562 : SUCCESS in 53m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d9e174bb5814aa990d6b97039fc15e0 : SUCCESS in 10m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/449fa6e28887426c809a01a897d6ee10 : SUCCESS in 23m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e4e299de0bb40f5a65e83de35a6a8d1 : SUCCESS in 28m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f76ed3af66bf465fb67da3805dd501e2 : SUCCESS in 31m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92ac8b820724c8f8aea845f5bdbae27 : SUCCESS in 13m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b4a2e1b7dd09483a90d66bda5e8f80fa : SUCCESS in 1h 02m 18s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b717f3c2c8514c2a8f13102b1d745b7b : SUCCESS in 1h 46m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5137cfffa9b942faadd738f0a15ba144 : SUCCESS in 35m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f83ed614f77843d19c734e0375ed59e3 : SUCCESS in 1h 32m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e7cc9f294b849e3823ed3d1eac1647f : SUCCESS in 1h 09m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e137d1b0c4dc4fc6960dd6c57402567a : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":4},{"id":"ac620f4eb36840e8372c3970d521401b824df099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-08 17:47:33.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"fede7d37478246d56c0dfb7b66e996122f4686f0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-08 19:08:27.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/71/962871/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"3fee33efc3e78fae32556c01608b85b1d5541470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 19:20:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f858d47e77b48f2849e4fc65c847500\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3d39b2f7a18846e898219b22b2dbe44b : SUCCESS in 1h 30m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc295472ff0e4fe89ca6e82eb713e118 : SUCCESS in 32m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca014a7977044947bf66ab484fdb0999 : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe21627761af4f5fa885d3133de557c0 : SUCCESS in 23m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d2a15d1e68a4e619139c6c80da046ed : SUCCESS in 13m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1e8d62cf7bf42a7bcc71a8397d6c5c5 : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba586a11fad4a4bb9747a014ee77dfc : SUCCESS in 6m 09s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d5472d1bac014055bf88597f0515f651 : POST_FAILURE in 1h 01m 09s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cc534501ce664d41b5a8f0e5e420c6bb : SUCCESS in 1h 26m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f472130e25554de6a0b7b87e3fb53fec : SUCCESS in 1h 02m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e72c3c7943844b3090b53cd522249722 : SUCCESS in 1h 07m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e18621d58974f0a906f8d4b10c86b82 : SUCCESS in 1h 13m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c405977882a5441e9c094665a210158a : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":5},{"id":"8ab94a33daa10f31e7469bb39e7ce4992c028cfa","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-08 19:37:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e5cdf5fd1595f9f033b0cea0b18a0eaec4b2a3e3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-08 20:26:05.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/71/962871/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"9af0c9eb02f31e2c3f828d65ad7082deeffc95dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 21:19:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9825658233fd460c8cfc9ead54fec672\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3efc731fbc2e4fe9b897a77e29bf43c5 : SUCCESS in 34m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a27db3a911e84a019afc00c5092e0ac4 : SUCCESS in 39m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/153205837e7f43a0b13b3b66b37f7b0c : SUCCESS in 52m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"004acad73a54fc326777a5286c2c16611efc380f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 21:29:04.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd8b0c090d1c4534bb2660b89b377399\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/496449f5777f4094bf0f5b41c91f1b7a : FAILURE in 1h 32m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6890258b7aa6464e9d65f236e6f7cba9 : SUCCESS in 55m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fd1afcd02064206816ca2d74df9b76e : SUCCESS in 14m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6108df72811a41188c5d452060ca0b8b : SUCCESS in 19m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dba711e47db849df8026e8a628c21ba9 : SUCCESS in 17m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f633df2b059482ea77bc79cf37f601d : POST_FAILURE in 21m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4f62036cd644cf880fc35741cca9c1 : SUCCESS in 11m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0903337b3f0747df86105a7198a513cd : SUCCESS in 56m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/092085585d494ce8981c818a119ada6d : SUCCESS in 1h 44m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/79ac47398d6b47508ec08511433a1dd4 : SUCCESS in 1h 00m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/858f7f8d1ae74a50b9100d83ee043aa9 : FAILURE in 1h 10m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/63d032635db54a6bbd5c219f85f12271 : SUCCESS in 1h 33m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/14aeda581e0b4bef8a4b2058bc53c5bb : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":5},{"id":"4da438bb14f77f0fbb0905814182f4101adcd0e4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-09 01:31:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e1abf6fd391e5e92925dec51a616b2054494be06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 02:08:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e53e55316d5a403ca3ef607621ac1af4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d893a1e04d604e9294575af08a3b863f : SUCCESS in 28m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c8487f0815c46d7a96aedcd29674f73 : SUCCESS in 34m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6d0abdda8e624f5fabf413aa6c082235 : SUCCESS in 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4f41abd5d29c86dffafd0ee264b610cbda622b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 03:42:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5273d0a054214485b3622812a1655ce0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/400bafd60f3f43c981138f822c84db7f : SUCCESS in 1h 35m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b729f57f029b4e6cb0c4a72dad860311 : SUCCESS in 51m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64baf0d72a3e414fae391f9c81ded240 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6e82dfcb16b45b88fd4029e27dc7951 : SUCCESS in 25m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa8cd248685c4c0cada810498d8e9054 : SUCCESS in 14m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b4890f216af4f8f9d9e5531b6dfa039 : SUCCESS in 17m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d89978e5474bb9aada5ff33289044c : SUCCESS in 12m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5ffb23c079a24d3c9076c217370547e0 : SUCCESS in 1h 02m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/20dca87dcd1a49aca36ddd814fad7f11 : SUCCESS in 2h 05m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c09a95383e44f1ba5e248da918af876 : SUCCESS in 28m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4186efee43424c9d91ff94ce10eef32f : SUCCESS in 1h 31m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/629f312114f24ff5876a3e0fa515f770 : SUCCESS in 1h 46m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/13e39cf4435f40a29f3c76f9ce43065b : SUCCESS in 30m 17s","accounts_in_message":[],"_revision_number":5},{"id":"8e5ce37bdf4d6294870f8145d2258594a7b4b1af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-09 04:57:24.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/71/962871/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"1c0c2ecc4d7495cefaa9dcb399e9ab44910321cd","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-09 19:41:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4b74273c6d726e4ff96a9fedc228f34fc878ff4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-09 20:37:25.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"91167ad12d76cf0d62991554377892e8ac938c6a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-10 06:15:30.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bdbd82f0983965ea100afe0fbfc8f16edc9aa1c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 06:15:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b3149d98b21e87e0c8ccba38f2edce0950ee072c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 08:09:08.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03da667c49064e798f3c005e400e1b0d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4731445fc4d1497da45b02a295e4f8c4 : FAILURE in 1h 51m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080dd1c8184340e88fc4f87522be6d17 : SUCCESS in 13m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/733e07f3098a4873b4e3cac3c32e80f0 : SUCCESS in 22m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/611b9489f5304546a3458e9a76519737 : SUCCESS in 21m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1278ddf1601c4868896a3f597ab656d9 : SUCCESS in 12m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c916fd0dda0647b2af792670e452f59d : SUCCESS in 1h 49m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0b2ff6fc01994351945fb1d65fb677bf : SUCCESS in 1h 42m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/edd7032866214515b2b24d6f68389488 : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":5},{"id":"6d23856a4c844811c712e6c334e576ac6baf3059","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-10 08:37:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf80241fe82e0e0449105444c7012d5e7f79f2f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-10-10 09:41:45.000000000","message":"Patch Set 5:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2025/Arista_Tempest_Testrefs/changes/71/962871/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"e94ecba9505c1ed66f1c73090a12e378a69bea86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 09:43:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3732e15dbc0c4b2aad51e299900cc705\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb93a1cf8e564192850f44e9f108b6db : SUCCESS in 51m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c787514fafd4cd3858b67c4f863d997 : SUCCESS in 46m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f203b50ac4e24f8da81787c43f0c92ed : SUCCESS in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ad342c0baa4c84fdd39bbf26668b871b5b1caf46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 10:15:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45a138cc79574acea5e17996b3d11938\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a58b77aa311a4140a2454acf07288dd8 : SUCCESS in 1h 37m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c23428c41d6348228a4013518dc45e03 : SUCCESS in 47m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8051877b8f2b4eefb152e2f13d871d0f : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0baff48372c4f1a9f1d87d97239779a : SUCCESS in 14m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dbc84b939464464384eaa1ede1915df1 : SUCCESS in 20m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8966d0c9b882427a908728217d1a87ef : SUCCESS in 16m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c88783703346e7ad863f1fc7932404 : SUCCESS in 8m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cc7e5343716d4ec184be9e67354d85bc : SUCCESS in 35m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ea6011b9804b40c8a0593dadbea2f3cc : SUCCESS in 1h 23m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8323b9290654dc0a53c9bb6066ff1af : SUCCESS in 42m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2672eb2050c741a294cd6b6a815c681a : SUCCESS in 1h 07m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fd120779f8944845a4a3b7a19d6d74db : SUCCESS in 1h 26m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/859a4409f1ca4ad78d8af1fc18a538f8 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":5},{"id":"937995f11afd0ff7d92bf6e8e26e831414a8b7ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 10:15:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e46772c74f216e012f5bde403d98ad06a64dac6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 11:47:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe99b1e5f529494f9a8fc600933200e9\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1488d122115f4099a6052f476588d2a5 : SUCCESS in 1h 28m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9af4e1b65e7497dba4e77aaa75e62cf : SUCCESS in 14m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddc5630253274874be33c10a706f6e2a : SUCCESS in 16m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/671deaf92f2a42a88db9a4ccaed8f5cb : SUCCESS in 22m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee1f4daf73f45dea72f280224dd1ac4 : SUCCESS in 12m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/10f17e6652a5449f871edb31f01a159d : SUCCESS in 1h 29m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eb2e155f72814ac8bdac53eaa1accf95 : SUCCESS in 1h 29m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8cacb6b08ce4c89971442540b90ac31 : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":5},{"id":"4710c2632734b0ac978340e46fa45036fa24dd6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 11:47:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"98fe9143018dba46177b7cbb3adb639415269253","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 11:48:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8439532a7565464083284d551b62f18c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5e5c8edbec4c33bd4f988bd900d41b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"84b9b62e6a4b35d480d274231b024aa6d69bda82","revisions":{"e84f348f462bbb80ce73cca9821e62f29294fd85":{"kind":"REWORK","_number":1,"created":"2025-10-02 19:45:23.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/962871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/962871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/962871/1"}}},"commit":{"parents":[{"commit":"9c5736b68c47e00fa131c71fbf31c0dec86e78a6","subject":"Merge \"[OVN] Remove the check for external ports support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c5736b68c47e00fa131c71fbf31c0dec86e78a6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-02 19:42:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-02 19:45:12.000000000","tz":-300},"subject":"Fix periodic_resync thread in DHCP agent","message":"Fix periodic_resync thread in DHCP agent\n\nThe thread tries to use threading.Event as a\nthread target. This prevents DHCP agent to start.\n\nChange-Id: Iac731787ed858fe0c0f752e3bdf400005c5f40be\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e84f348f462bbb80ce73cca9821e62f29294fd85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e84f348f462bbb80ce73cca9821e62f29294fd85"}]},"branch":"refs/heads/master"},"c9d38d2a6e753df2b91eb9b2d37b85c666bb5b6d":{"kind":"REWORK","_number":2,"created":"2025-10-08 08:18:16.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/962871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/962871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/962871/2"}}},"commit":{"parents":[{"commit":"9c5736b68c47e00fa131c71fbf31c0dec86e78a6","subject":"Merge \"[OVN] Remove the check for external ports support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c5736b68c47e00fa131c71fbf31c0dec86e78a6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-02 19:42:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-08 08:15:45.000000000","tz":-300},"subject":"Fix periodic_resync thread in DHCP agent","message":"Fix periodic_resync thread in DHCP agent\n\nThe thread tries to use threading.Event as a\nthread target. This prevents DHCP agent to start.\n\nCloses-Bug: #2127029\nChange-Id: Iac731787ed858fe0c0f752e3bdf400005c5f40be\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9d38d2a6e753df2b91eb9b2d37b85c666bb5b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9d38d2a6e753df2b91eb9b2d37b85c666bb5b6d"}]},"branch":"refs/heads/master"},"fb663277bfe9ba6916679ec3daa57dcb300e6c2d":{"kind":"REWORK","_number":3,"created":"2025-10-08 08:51:05.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/962871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/962871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/962871/3"}}},"commit":{"parents":[{"commit":"9c5736b68c47e00fa131c71fbf31c0dec86e78a6","subject":"Merge \"[OVN] Remove the check for external ports support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c5736b68c47e00fa131c71fbf31c0dec86e78a6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-02 19:42:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-08 08:50:49.000000000","tz":-300},"subject":"Fix periodic_resync thread in DHCP agent","message":"Fix periodic_resync thread in DHCP agent\n\nThe thread tries to use threading.Event as a\nthread target. This prevents DHCP agent to start.\n\nCloses-Bug: #2127029\nChange-Id: Iac731787ed858fe0c0f752e3bdf400005c5f40be\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb663277bfe9ba6916679ec3daa57dcb300e6c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb663277bfe9ba6916679ec3daa57dcb300e6c2d"}]},"branch":"refs/heads/master"},"299021463f381bf7d634c271c9580e12a1b46cc0":{"kind":"REWORK","_number":4,"created":"2025-10-08 13:54:16.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/962871/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/962871/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/962871/4"}}},"commit":{"parents":[{"commit":"9c5736b68c47e00fa131c71fbf31c0dec86e78a6","subject":"Merge \"[OVN] Remove the check for external ports support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c5736b68c47e00fa131c71fbf31c0dec86e78a6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-02 19:42:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-08 13:54:01.000000000","tz":-300},"subject":"Fix periodic_resync thread in DHCP agent","message":"Fix periodic_resync thread in DHCP agent\n\nThe thread tries to use threading.Event as a\nthread target. This prevents DHCP agent to start.\n\nCloses-Bug: #2127029\nChange-Id: Iac731787ed858fe0c0f752e3bdf400005c5f40be\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/299021463f381bf7d634c271c9580e12a1b46cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/299021463f381bf7d634c271c9580e12a1b46cc0"}]},"branch":"refs/heads/master"},"84b9b62e6a4b35d480d274231b024aa6d69bda82":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-10-08 17:47:33.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/962871/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/962871/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/962871/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/962871/5"}}},"commit":{"parents":[{"commit":"e1d337e15fd445d29bc50067b139e390c5145303","subject":"Merge \"[S-RBAC] Fix policies for local_ip_association\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1d337e15fd445d29bc50067b139e390c5145303"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-02 19:42:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-08 17:47:33.000000000","tz":0},"subject":"Fix periodic_resync thread in DHCP agent","message":"Fix periodic_resync thread in DHCP agent\n\nThe thread tries to use threading.Event as a\nthread target. This prevents DHCP agent to start.\n\nCloses-Bug: #2127029\nChange-Id: Iac731787ed858fe0c0f752e3bdf400005c5f40be\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84b9b62e6a4b35d480d274231b024aa6d69bda82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84b9b62e6a4b35d480d274231b024aa6d69bda82"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
