)]}'
{"id":"openstack%2Fneutron~963291","triplet_id":"openstack%2Fneutron~stable%2F2025.2~I4631094d7e2bfda9ddff5feb8c96960ea0cd436d","project":"openstack/neutron","branch":"stable/2025.2","topic":"bug/2126588","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-10 09:45:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4631094d7e2bfda9ddff5feb8c96960ea0cd436d","subject":"[OVN] ``HAChassisGroupRouterEvent`` is for router HCG only","status":"MERGED","created":"2025-10-07 14:33:17.000000000","updated":"2025-10-10 09:46:56.000000000","submitted":"2025-10-10 09:45:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963291-bug/2126588","meta_rev_id":"5f46721f288de5ed30d3f28e2c452bc01df44e2c","_number":963291,"virtual_id_number":963291,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-10 09:45:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-10-10 06:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-10-07 17:04:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2025-10-10 06:31:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 16:07:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-07 16:38:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-07 17:04:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-10-10 06:31:50.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"338dafdc3e3c4468e8e4e9a868c703d7c9b42acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-07 14:33:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cac5c75fa3d369ea1aa491df1ef4d77026061e6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 16:07:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f9725b8677941419e0eb601a7cbf334\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9de9fd21b23e4b88b471fdaf0f2de083 : SUCCESS in 53m 06s\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d7abf9fff9143b5b64718f0cff962ef : SUCCESS in 49m 31s\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/26202fc2e9994ed1ba3909dce5229757 : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"92cf49e389f8c1254d43bad7f05e98f9432089d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 16:38:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10e7dfd67117478d93056a60d34f9de1\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/3cb859e4cd0b4127982ebffadb509d59 : SUCCESS in 2h 04m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07fbab67efd14dc19242cc587dcba38f : SUCCESS in 26m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7706f96e28814dc7b612cb0db6266c5e : SUCCESS in 7m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35f1cfd4534c4e3f9371792efc2cfc1d : SUCCESS in 21m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/feddc157cfb2491d85a365029f010cec : SUCCESS in 20m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08950c3cf508403a8dde6a56f3f56fbc : SUCCESS in 16m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d651adbe344fb3bf6e19d3a9c96131 : SUCCESS in 7m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4938e5be89c147b1a0aa747f6965bf86 : SUCCESS in 1h 28m 50s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcebb577a99e414093ae3c32ca35162f : SUCCESS in 1h 07m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/989d95aee62b4c92ac90c5b71d951c37 : SUCCESS in 57m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad005876a92d4acca02fdf6638e3ea08 : SUCCESS in 1h 19m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c1eba43156ae4e5b8abb5f161c77ba9b : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8e19c1912bcebd7af5f7d4a397e5fcfa976a49f1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-07 17:04:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"97ea2cdc3c9abdce91ab874330397ace52de004c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-10 06:31:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b07c92713f48f1ab3659bbc4a1a8226fccfd8e02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 06:32:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9af814b2e7952d13f2fd6401139f51f55072656e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 09:45:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba1009fabb5d48a785fec8f8bcf53733\n\n- neutron-tempest-plugin-ovn-2025-2 https://zuul.opendev.org/t/openstack/build/7248f96c5f7247a8a4cf65a4fb36a202 : SUCCESS in 2h 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb51682b6f845edb41acfc49cf811ab : SUCCESS in 12m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a0f11e6e99a4dbe9d0fc8ab46faf5f8 : SUCCESS in 21m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1788dc09912467791799ff3e552b24a : SUCCESS in 15m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac0da595eb774fea8f31ca63129ed455 : SUCCESS in 12m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8a225dc4ce45462987eb2dbe273c6029 : SUCCESS in 1h 29m 07s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/28a64c7e245c40deb57f7c56d588c23b : SUCCESS in 1h 12m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/debc0489067d4cab8a3fde762272b979 : SUCCESS in 54m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5720fe3318f04c5c92d9ea411e6824b7 : SUCCESS in 1h 26m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cc5b904ccfd04714eeb3bf6c5108b4ef2a1cc89a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 09:45:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5f46721f288de5ed30d3f28e2c452bc01df44e2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 09:46:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34d795572e2842e2a642c82c87ecedb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b199916b16e434ea8f687fa22c7d197 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa1d08d812b64e9351ecd8cf2f5c6c136a26d250","revisions":{"fa1d08d812b64e9351ecd8cf2f5c6c136a26d250":{"kind":"REWORK","_number":1,"created":"2025-10-07 14:33:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/963291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/963291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/963291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/963291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/963291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/963291/1"}}},"commit":{"parents":[{"commit":"62418e1ee7cde076222f9034fd12436e6bff3630","subject":"[stable/2025.2 only] Drop CentOS 9-stream jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62418e1ee7cde076222f9034fd12436e6bff3630"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-02 07:15:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-07 14:33:03.000000000","tz":0},"subject":"[OVN] ``HAChassisGroupRouterEvent`` is for router HCG only","message":"[OVN] ``HAChassisGroupRouterEvent`` is for router HCG only\n\nThe event ``HAChassisGroupRouterEvent`` handles the ``HA_Chassis_Group``\nregisters created for routers, when the external network is tunnelled.\nThis event should only attend ``HA_Chassis_Group`` that have the key\n\"neutron:router_id\" in the external_ids field, that identifies that is\ncreated for routers only.\n\nCloses-Bug: #2126588\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4631094d7e2bfda9ddff5feb8c96960ea0cd436d\n(cherry picked from commit d7e80423fd300f4cd407a86221932146d74d3a04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa1d08d812b64e9351ecd8cf2f5c6c136a26d250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa1d08d812b64e9351ecd8cf2f5c6c136a26d250"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
