)]}'
{"id":"openstack%2Fneutron~963293","triplet_id":"openstack%2Fneutron~stable%2F2024.2~I4631094d7e2bfda9ddff5feb8c96960ea0cd436d","project":"openstack/neutron","branch":"stable/2024.2","topic":"bug/2126588","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-10 09:45:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4631094d7e2bfda9ddff5feb8c96960ea0cd436d","subject":"[OVN] ``HAChassisGroupRouterEvent`` is for router HCG only","status":"MERGED","created":"2025-10-07 14:33:48.000000000","updated":"2025-10-10 09:46:56.000000000","submitted":"2025-10-10 09:45:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963293-bug/2126588","meta_rev_id":"e018368249b3e40ecacd820dac563b422af0dc6a","_number":963293,"virtual_id_number":963293,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-10 09:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-10-10 06:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-10-07 17:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2025-10-10 06:46:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 15:49:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-07 16:33:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-07 17:04:59.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-10-10 06:46:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"f29e2889530d459dc21fea6a589a51b13f590084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-07 14:33:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d2f5d407fccd70c178c324575f5318ea1ce8db1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 15:49:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87f72e58fc8e46d48c97dc8745bb7c3f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40eb3b882f1846f9b67b1509b8b0a7b0 : SUCCESS in 22m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8dedb7455c8e416a98b1a1ac087323a1 : SUCCESS in 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81cbba2a6e6c660c3c38d3a51a9a14fe84cb5f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 16:33:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19ecc6ee2eb94705a7005f3b2c74e575\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/a05f3acad4ea4a0f8a362e687b5bc105 : SUCCESS in 1h 57m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6edf14f8a6a046d78ad4b4ef7acf8759 : SUCCESS in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053498e4784f45d7b55ddb40307ba561 : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a999368b275c4fd5aae8b53546094da0 : SUCCESS in 37m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7221bedb5d934f75a7b3f3c2feaa0435 : SUCCESS in 34m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e1b2b40b5c84946934049760c34a913 : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db126414e6544aba018afede4dee908 : SUCCESS in 17m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6d794114f674e1fa4a2867f8b7f9073 : SUCCESS in 1h 36m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/795fe9a5dd534892a56a3c9e6cb26bb7 : SUCCESS in 1h 10m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8cde7591b2c540ca9a3be9b9cdf87017 : SUCCESS in 47m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9b5131405f3746959c69cb63fa42a62f : SUCCESS in 55m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb63acd58baa4dba968354bc99819c4b : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f8960452d74b5345608a18e298e60ec81f536418","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-07 17:04:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3257184238457c0347036106dbbcaf7144f541ca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-10 06:46:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ea3cd46853a1151b1ca30760dc72a3ae586b550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 06:46:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e3d5f05cbe42d4fc44ae896c7e0e28dc2de1e8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 09:45:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/646e1a84de8a4aac8adddef321121392\n\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/c7b65aed046849fd9b221f13665e9ec0 : SUCCESS in 1h 20m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/011248f6483244c2a24ace875d08470f : SUCCESS in 9m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe08e28a0ed74bc8bd9b07fa9a3fd533 : SUCCESS in 23m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2f27a43674546fbbbeac6230f7cad5e : SUCCESS in 19m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb16d304380d4a4cadfebc836710f98f : SUCCESS in 10m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c930a256ac74d4cb8b85558dd574498 : SUCCESS in 1h 32m 47s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1415a60e07b406e9fad0db193026b0e : SUCCESS in 1h 02m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d008e6ff7a934721930fb4adafbee9de : SUCCESS in 57m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3228c9a3e8fa42628617d41a42c84cd6 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9d5322211de5cf524f018614b423c9642c4f4490","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 09:45:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e018368249b3e40ecacd820dac563b422af0dc6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 09:46:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa110fb1158d447ba1cc1f661b2da5ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/384c209ae6094425bc0dfb23318c8066 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9493185ec586009ce574269e07285e35995649f6","revisions":{"9493185ec586009ce574269e07285e35995649f6":{"kind":"REWORK","_number":1,"created":"2025-10-07 14:33:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/93/963293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/963293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/963293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/963293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/963293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/963293/1"}}},"commit":{"parents":[{"commit":"78ebe3ba943de0577733d6aef6e920c1b807a3d9","subject":"Merge \"[S-RBAC] Fix policies for the l3_conntrack_helpers APIs\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78ebe3ba943de0577733d6aef6e920c1b807a3d9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-02 07:15:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-07 14:33:42.000000000","tz":0},"subject":"[OVN] ``HAChassisGroupRouterEvent`` is for router HCG only","message":"[OVN] ``HAChassisGroupRouterEvent`` is for router HCG only\n\nThe event ``HAChassisGroupRouterEvent`` handles the ``HA_Chassis_Group``\nregisters created for routers, when the external network is tunnelled.\nThis event should only attend ``HA_Chassis_Group`` that have the key\n\"neutron:router_id\" in the external_ids field, that identifies that is\ncreated for routers only.\n\nCloses-Bug: #2126588\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I4631094d7e2bfda9ddff5feb8c96960ea0cd436d\n(cherry picked from commit d7e80423fd300f4cd407a86221932146d74d3a04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9493185ec586009ce574269e07285e35995649f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9493185ec586009ce574269e07285e35995649f6"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
