)]}'
{"id":"openstack%2Fneutron~963571","triplet_id":"openstack%2Fneutron~master~If16353710a6e2c9fc649abcf53edd7b736b19f4c","project":"openstack/neutron","branch":"master","topic":"ironic_periodic_ci_job","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-10-14 15:30:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If16353710a6e2c9fc649abcf53edd7b736b19f4c","subject":"Add the ``ironic-tempest-ovn-uefi-ipmi-pxe`` CI job","status":"MERGED","created":"2025-10-09 14:49:05.000000000","updated":"2025-10-14 15:32:19.000000000","submitted":"2025-10-14 15:30:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963571-ironic_periodic_ci_job","meta_rev_id":"aadfcb9f765ce587d3b444adf841af7f3b88dc89","_number":963571,"virtual_id_number":963571,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 15:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2025-10-14 13:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2025-10-13 08:19:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-10-14 13:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-10-14 13:38:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-09 16:36:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-09 20:20:29.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-10-10 06:17:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-10-10 06:17:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-10-10 06:17:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2025-10-10 07:50:43.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-10-14 13:38:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-10-14 13:41:50.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"3251e4077c45ef3924730c6718431d8a5ff660a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-09 14:49:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10a7535588d5c1db963512c008deb523045870c0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-09 14:49:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eea302ce6f5492839209a166ad2b8e0e233a2762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 16:36:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2c42f7ffb3b4205ab13f2125df49edc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4d8a4334f69442f2940acb62810f9945 : SUCCESS in 1h 37m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9af9364461be4f68aeb70952a8a7534a : SUCCESS in 1h 44m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e62bb004a4f9444485b0cc6e2acb3503 : SUCCESS in 33m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/744dcf872dc0464ab0702e808f90e853 : SUCCESS in 32m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66052913961b4e0291a989ae9eafa8e5 : SUCCESS in 15m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee3b8b233bb843b9b910603e66759870 : SUCCESS in 24m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/559e40b3e76143279e300b0dd71e5aa4 : SUCCESS in 20m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/639fc7c778d549dd9fd8ff5852d9b173 : SUCCESS in 25m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cceb6eafc0014ba5b28212079b7a58b3 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"60d5015a82069ae1b477809cea2b8afb3de4c9f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:18:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/824b519fab5947038b3c0d156cdbe4cd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ba57017ba2b6477285604638ab61317b : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b969eb3ea3245399964b27ee5421e4c : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01f9b6b666624327b2977d0496243eb9 : SUCCESS in 53m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2894b0acc65b0e555704a3935e60c33e4c9bac61","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:47:51.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ad27a31a2b541b68b55f1819f64382c\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/65491662d1d6406796093b1a00b63b6d : FAILURE in 2h 56m 26s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/917ce3f4a6fd43f49ff1e6bf837a7795 : TIMED_OUT in 2h 48m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/a4a186e140994ab3b31a23bd60ddceb0 : FAILURE in 1h 46m 19s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e28da6b2920745edaddb75de7a59fa73 : SUCCESS in 1h 54m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0dcfea8fba3c4017b521f02761248c60 : SUCCESS in 1h 51m 22s\n- openstack-tox-py312-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2b1bdff36cec4e06a828a9beb2129726 : SUCCESS in 22m 00s\n- openstack-tox-py312-with-oslo-master https://zuul.opendev.org/t/openstack/build/c241a1249e1f4a529954ee99a757b067 : SUCCESS in 30m 00s\n- openstack-tox-py312-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/f0a97f919d4e4662b02450db1f264266 : SUCCESS in 17m 28s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/5858f0da06ff4a60bd1168c771599356 : SUCCESS in 1h 37m 09s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/ba55658fee2d465db8f158ac1f089387 : SUCCESS in 2h 36m 02s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/e04cd75d658349aeacad309e305071d3 : SUCCESS in 1h 04m 10s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/335df5f295a34ca19dad4d5607d36c87 : SUCCESS in 1h 02m 41s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/4c41f17a265f4fdf927af7f70e286e7c : SUCCESS in 1h 23m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6023a3c1196849bc8477afd7cb1be869 : SUCCESS in 1h 10m 20s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/24fec46938e64712ab8c982a9d4be948 : SUCCESS in 1h 32m 05s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/b344f685e36046cf9502988c206951dd : FAILURE in 2h 09m 49s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/034c5ee2081d4d71b332b9913bc3545d : SUCCESS in 31m 58s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4820ea6158184ba8ab37d70e650b2d73 : SUCCESS in 30m 08s\n- neutron-tempest-plugin-ovn-with-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/3d51600bc93e44cbb4e80a9b396aa238 : SUCCESS in 2h 49m 49s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/53ee64e9de5a43d289351ab011cc5e2f : FAILURE in 1h 04m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ee606183f052fd39af83495b9d59285fa61b2730","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-10-10 07:50:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae6a5bd179d7ec8b16079caa25e64dabf33bfd6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-13 08:19:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f291d2fe7dc9aadbb7ffbab4ec75295be79fc60b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-10-13 08:19:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"016fd4f57ff8a0b189c0a00a325865e8440a7a27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 09:09:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d19ccb195fc646d6a2b9722178eb6c32\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/daf9161510b74f87a22318a4614a370c : SUCCESS in 48m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c522b71775a4aaebd52da3508886e30 : SUCCESS in 26m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e0c295c8023e4aec9965a41e6e7d6c5e : SUCCESS in 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d464f0df388eb42e668ccc05d890195d88f5035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 10:07:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e81b30da57c4e938d9315fcb4a983c4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2a6d4f9b0ed24b528fdbad6f5b19c9d9 : SUCCESS in 1h 24m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4f7231bb77cb468199157060908be7e9 : SUCCESS in 1h 46m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/476aaae01261437bbf1ee33c7edcc4d1 : SUCCESS in 20m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b633113fd0c41c29373dd43bb77ac1e : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237af52dd0be426495315fc3d006867e : SUCCESS in 9m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31d9fc49fd8a4556b180c8bf244e394e : SUCCESS in 31m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c3ff481a5824b1e8ecd5c4b663c4e54 : SUCCESS in 30m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3211477a22f24e06974186dc6185dc29 : SUCCESS in 24m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af625d01e6df4effb5e5983ccc99409b : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":2},{"id":"b1dbd8d6c88fb26b2a1ba5a8c3f9cedf19aafbd5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 11:27:42.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a90a83dc35c4ff0941a9f6ec8c9b9aa\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/04cba495939445aca20817cbbf073a86 : FAILURE in 3h 00m 54s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/4e67a40e2b5242e2b2d4412ba8c54f8e : SUCCESS in 1h 30m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/8dbf947819fc4d298941ab4e1473c2e5 : FAILURE in 1h 48m 39s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1c077857c8ff462d83b8f5346f538821 : SUCCESS in 1h 15m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fc2080bd3e1242828fe283df0f5c028e : SUCCESS in 1h 37m 34s\n- openstack-tox-py312-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/29fcc9f061e842ee8ff8aca8b68a983e : SUCCESS in 14m 36s\n- openstack-tox-py312-with-oslo-master https://zuul.opendev.org/t/openstack/build/8c3e05b4e32e499d91d84f9d67df6e90 : SUCCESS in 24m 36s\n- openstack-tox-py312-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/6a6950ef0ed74047992bab1edd610bcf : SUCCESS in 15m 18s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/d675e81613104dc9ba5469f985d47e62 : SUCCESS in 1h 46m 26s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/3c16a17f2eae465a94fac044081b2a63 : SUCCESS in 1h 47m 12s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5dd026de724842228ccac4adaf5f8089 : SUCCESS in 52m 04s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9f411338936149c587cec06201ece80c : SUCCESS in 1h 14m 59s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/b0e69c5d30e44f90ae25dd5961ee9b10 : SUCCESS in 1h 06m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8713213939e3413189369e71a71c095c : SUCCESS in 1h 11m 53s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e3ba0469559443b3b0066022ffdb3c3b : SUCCESS in 1h 13m 34s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/8e149e046c184950939ae8e1a30ea5b3 : SUCCESS in 1h 58m 32s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/dc536eea608b427fb28d1b319ef0f3ee : SUCCESS in 24m 20s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/82a0275a1fa94ee299a3163ddd29da5b : SUCCESS in 15m 03s\n- neutron-tempest-plugin-ovn-with-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/8089f8c6a9fe4100b3a838ade5a8ba5c : SUCCESS in 2h 07m 44s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/a30556b88c204d5695079ae18ef648f2 : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":2},{"id":"fcfb65bf58c86c19cc9d56a424f76a075791c331","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-10-14 13:38:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dad9a7939cb857932716f42d0cdbcbdae5e9cb75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 13:38:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc399eba2340af1753cfe8fa05027e5ec8131472","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-10-14 13:41:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ec8ff3a17e807005ada1eb38d495b207b46161db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 15:30:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13eb65537817474e9307929e08f64248\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/66aaf637053845429a3831ff1c51c56c : SUCCESS in 1h 25m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/658f29e88894471f95a94eb0aec151a9 : SUCCESS in 1h 49m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/798ff1ec326f4a02af07ccdc740bf961 : SUCCESS in 13m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff9a122310254d8ebb40fc3303026bf2 : SUCCESS in 26m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/766998de0cf84fe193b6f3dba65b14a9 : SUCCESS in 17m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8857f2bc53af4d5dad1aa4825e3e9c21 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a0d07aa7279ad4ec9e151c7f1465e8ed27ce0af3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 15:30:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"aadfcb9f765ce587d3b444adf841af7f3b88dc89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 15:32:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23ffd9dddce9463a8ac530c188b576c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5769846c94fe4ab8835e088970bf33e3 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f55051a2cf5e66cf22d29aa66bb088e2b30ebc03","revisions":{"8f32b45f71093876d6634ca31fa8249db8c1d310":{"kind":"REWORK","_number":1,"created":"2025-10-09 14:49:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/71/963571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/963571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/963571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/963571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/963571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/963571/1"}}},"commit":{"parents":[{"commit":"e1d337e15fd445d29bc50067b139e390c5145303","subject":"Merge \"[S-RBAC] Fix policies for local_ip_association\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1d337e15fd445d29bc50067b139e390c5145303"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-09 14:48:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-09 14:48:17.000000000","tz":0},"subject":"Add the ``ironic-tempest-ovn-uefi-ipmi-pxe`` CI job","message":"Add the ``ironic-tempest-ovn-uefi-ipmi-pxe`` CI job\n\nThis job ``ironic-tempest-ovn-uefi-ipmi-pxe`` is added to the\nexperimental and periodic queues.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: If16353710a6e2c9fc649abcf53edd7b736b19f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f32b45f71093876d6634ca31fa8249db8c1d310"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8f32b45f71093876d6634ca31fa8249db8c1d310"}]},"branch":"refs/heads/master"},"f55051a2cf5e66cf22d29aa66bb088e2b30ebc03":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-13 08:19:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/71/963571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/71/963571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/963571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/963571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/71/963571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/71/963571/2"}}},"commit":{"parents":[{"commit":"d032eaf11d0c414fd8acd09e4a49f8bf5ea6c08f","subject":"Merge \"Fix periodic_resync thread in DHCP agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d032eaf11d0c414fd8acd09e4a49f8bf5ea6c08f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-10-09 14:48:17.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-10-13 08:19:46.000000000","tz":0},"subject":"Add the ``ironic-tempest-ovn-uefi-ipmi-pxe`` CI job","message":"Add the ``ironic-tempest-ovn-uefi-ipmi-pxe`` CI job\n\nThis job ``ironic-tempest-ovn-uefi-ipmi-pxe`` is added to the\nexperimental and periodic queues.\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: If16353710a6e2c9fc649abcf53edd7b736b19f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f55051a2cf5e66cf22d29aa66bb088e2b30ebc03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f55051a2cf5e66cf22d29aa66bb088e2b30ebc03"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
