)]}'
{"id":"openstack%2Fneutron~966495","triplet_id":"openstack%2Fneutron~master~I447572a9d014459400c82038fd30819b3630bace","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"38298":{"account":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"last_update":"2026-04-28 13:55:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I447572a9d014459400c82038fd30819b3630bace","subject":"Fix bug/2129926 delete recordset without PTR","status":"ABANDONED","created":"2025-11-08 03:57:03.000000000","updated":"2026-04-28 13:55:42.000000000","total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"1361be42d3775bb4035ee4b1b8bb82fd1ff8fc93","_number":966495,"virtual_id_number":966495,"owner":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-21 10:56:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"disliked":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"all":[{"value":-1,"date":"2026-03-03 13:06:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":1,"date":"2025-11-21 08:49:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-08 04:51:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-08 06:03:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-21 08:48:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-11-21 08:49:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-03-03 12:22:59.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"CC"},{"updated":"2026-03-03 13:06:29.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"}],"messages":[{"id":"a0d7a43dda7abf56b27aa31ff7594335c6cf0259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2025-11-08 03:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bb6b4e13f31907b2b5df5c038321a1b245c9ce8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-08 04:51:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6117aaf7b994b8b9f8ca5333a06a724\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eca74975ea284399acfface8f72e4061 : SUCCESS in 52m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e6aaf125ca54eb79f96b67ee685f959 : SUCCESS in 30m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/96e508ffe8de48bea91cecdc4d5db301 : SUCCESS in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c50f952005729e97633caf707299e9bbc7c4ff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-08 06:03:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85fe27663d444f3a9e8e3d0d1940c62b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b5400a414f12496e8ef586adc568d10d : SUCCESS in 1h 17m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/658223b26adf4165b04feadce45ae406 : SUCCESS in 2h 02m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/48d398b8dc5240b2a16edda8377e30f1 : SUCCESS in 33m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d159a51a0c741fca8e200ebf8bed46d : SUCCESS in 1h 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca618b4b563b45c390acdd08da5e8c97 : SUCCESS in 14m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4e3ae60b0be4a6cb0fa3d2ba29ef3a7 : SUCCESS in 34m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5516c85debb24c458ce519ac85fed864 : SUCCESS in 27m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbf664e1112f439da6f92d4342896cbd : SUCCESS in 37m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49964c4abaf343adab5c0e9885a0fa88 : SUCCESS in 13m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e145938ceabf471bad9e9c196b22221d : SUCCESS in 56m 32s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9e3c697df3694a67b43a6fde22d8d054 : SUCCESS in 51m 19s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5168a907198b41eca3d5a4ed6d61a44d : FAILURE in 30m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ee4671745234f8094c0768c2a05dfe8 : SUCCESS in 1h 08m 11s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/9423e30cc06d47ca9b9076e67cf68725 : SUCCESS in 1h 11m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f3545bdc63324ba49bc352caafd28b7a : SUCCESS in 1h 54m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d01ce661bb6b4620974efad5f2c449b6 : SUCCESS in 1h 10m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/efcab8d5904f453c9a637e23fb56b415 : SUCCESS in 1h 03m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/de0d0333e3b0479c90d9ecc927784731 : SUCCESS in 1h 15m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b751b44ba7d4f68934f9bcfa3080971 : SUCCESS in 1h 22m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6f5b62b541494db0acc93510840c1a89 : SUCCESS in 1h 18m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e75d1c55f1cf7911cccccd1fe666b9028e86f6fd","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2025-11-10 15:47:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f14a0d7078e3f2bae6e7bda908f5335eba80562","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2025-11-17 15:38:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7109b73bf85bc01d738164e250ae1b336681a40","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-21 08:48:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f59d6e055694c3d9f78a7ba0a659adc49e2bc3f4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-21 08:49:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"51516129e8f53a714f70ba6a3b3218bee5b2128b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 10:48:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a602a9ea0c5444c6a64dd8d7af050323\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/66d964568fb24840ad22fd5d1740c660 : SUCCESS in 1h 04m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d7b46d0fccd641d886a0d450c7587b93 : SUCCESS in 24m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f747723b072483196533b48a437c705 : NODE_FAILURE Node(set) request bb9d44a7de3843b783e9fa8a175f88c4 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35e147624d3895151d7f4acc82d8c479ac3982ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 10:56:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3b25db94054489a23463ed7b8126a3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e633e9b6ba444f71957520ce2d34ca62 : SUCCESS in 1h 31m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/86b255ed640f42709ed39636f4b37aa9 : SUCCESS in 2h 02m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8be720e0226a440cb6dd748ca122ef06 : SUCCESS in 35m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04e792b52f174415b8f808c1a9c1e400 : SUCCESS in 43m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d29f7159a8c648248f8546b75fa6521d : SUCCESS in 13m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/370985e7c7c94a4bbf75b2c66957d577 : SUCCESS in 35m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb1a37bdc9c946d1aa5cab3617b533fc : SUCCESS in 22m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17d23dd285e4a499cc0c4af573c1959 : SUCCESS in 14m 08s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c0d543868b6746f6ac0bf814494a0782 : SUCCESS in 30m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f18eba70d2fc4740a3a3ef8e65f4de1d : SUCCESS in 59m 50s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0ddc8a5176c94092b00f9e96643ce27b : SUCCESS in 1h 34m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e787f2757c4642a19987f7a2e9d334af : SUCCESS in 1h 05m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/220d3d5d7e914fe78ae0a7a2d0c3d9c8 : SUCCESS in 50m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9ed1fd9f9f4b4fc0a5cc468a9c5dac80 : SUCCESS in 1h 48m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a9a4d8dc8a2b4e4c90d38b00f4c9dd8d : SUCCESS in 1h 43m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/76900d5cc2964b21812c9f9634fd0dfe : SUCCESS in 52m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e69c81df40734431b1e85e763eaea90f : SUCCESS in 54m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c65f71d498f745ab8c82fbed663854f5 : SUCCESS in 1h 22m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e71152ad889145d6a7be079058557288 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":1},{"id":"710c1502af879c3fc3af131f3a887d598f0343b4","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-03-03 12:22:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f11ca381f027b7bd48c86f49d4a8115bce66ba6b","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-03-03 13:06:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4ce4299e66414bce9522f3dc5bc0e173722e0b5d","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2026-04-28 12:17:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1361be42d3775bb4035ee4b1b8bb82fd1ff8fc93","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"date":"2026-04-28 13:55:42.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/neutron/+/978567 was merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f83da1235fd2b3248ad0f1fc0d4b3a686253e9f7","revisions":{"f83da1235fd2b3248ad0f1fc0d4b3a686253e9f7":{"kind":"REWORK","_number":1,"created":"2025-11-08 03:57:03.000000000","uploader":{"_account_id":38298,"name":"Helen Chen","display_name":"Helen Chen","email":"ichen@redhat.com","username":"ingwherchen"},"ref":"refs/changes/95/966495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/966495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/966495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/966495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/966495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/966495/1"}}},"commit":{"parents":[{"commit":"090d6cb0947731c335a465188ebf4d52bf06d607","subject":"Merge \"Remove mocking of eventlet.spawn in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/090d6cb0947731c335a465188ebf4d52bf06d607"}]}],"author":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2025-11-08 03:45:23.000000000","tz":-300},"committer":{"name":"Helen Chen","email":"ichen@redhat.com","date":"2025-11-08 03:55:24.000000000","tz":-300},"subject":"Fix bug/2129926 delete recordset without PTR","message":"Fix bug/2129926 delete recordset without PTR\n\nWhen reverse DNS lookup in neutron is enabled, i.e.,\nallow_reverse_dns_lookup \u003d True,\nmatching PTR records are created when A or AAAA records are created.\nOpenstack also has commands to allow for manual insertion/deletion of\nPTR records [1].  Thus, even when a PTR record is inserted automatically\nwhen reverse DNS lookup is enabled, the PTR record could still be\nmanually removed.\n\nConsequently, when deleting an A or AAAA records, it is necessary to\ncheck that the corresponding PTR record exists first, before deleting\nthe PTR record.  Without checking that the PTR record exists first, the\ncode fails due to designateclient.exception.NotFound [2].\n\nA new unit test is added to test that an A record is deleted gracefully\nwhen reverse DNS lookup is enabled but the corresponding PTR record does\nnot exist.\n\n[1] https://docs.openstack.org/designate/latest/user/manage-ptr-records.html\n[2] https://opendev.org/openstack/neutron/src/branch/master/neutron/services/externaldns/drivers/designate/driver.py#L159\n\nCloses-bug: #2129926\nChange-Id: I447572a9d014459400c82038fd30819b3630bace\nSigned-off-by: Helen Chen \u003cichen@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f83da1235fd2b3248ad0f1fc0d4b3a686253e9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f83da1235fd2b3248ad0f1fc0d4b3a686253e9f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
