)]}'
{"id":"openstack%2Fneutron~967813","triplet_id":"openstack%2Fneutron~master~I0fc4b92a8dbaf5631470a9d74860fe76d8dd6109","project":"openstack/neutron","branch":"master","topic":"bug/2131994","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-11-20 15:32:38.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-11-20 13:31:24.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-11-20 15:32:38.000000000","reason":"Change was submitted"},"37662":{"account":{"_account_id":37662,"name":"Brian Haley","email":"v-brianhaley@microsoft.com"},"last_update":"2025-11-20 15:32:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0fc4b92a8dbaf5631470a9d74860fe76d8dd6109","subject":"Workaround for the OVN/OVS directories in functional job","status":"MERGED","created":"2025-11-20 09:52:00.000000000","updated":"2025-11-20 15:33:52.000000000","submitted":"2025-11-20 15:32:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967813-bug/2131994","meta_rev_id":"813d82182268896b95187cd094cb1f2a420a4427","_number":967813,"virtual_id_number":967813,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-20 15:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2025-11-20 12:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2025-11-20 13:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2025-11-20 13:31:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2025-11-20 12:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":37662,"name":"Brian Haley","email":"v-brianhaley@microsoft.com"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37662,"name":"Brian Haley","email":"v-brianhaley@microsoft.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-20 10:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-20 11:14:46.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2025-11-20 12:19:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-11-20 12:19:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-11-20 12:19:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":37662,"name":"Brian Haley","email":"v-brianhaley@microsoft.com"},"state":"REVIEWER"},{"updated":"2025-11-20 12:19:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"dec0228d923bceed5b66cabc0817990010b76299","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-20 09:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32be2a8ab61fefe67706a60dff4e5acccfcad1cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 10:43:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ada64d4a2a174e32a0655c9901d0d751\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41e504a0090c42a5aed3cc3285507dbb : SUCCESS in 47m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf50da04f4d34690a7041287f011a9ab : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/532936543aca496484c396d13e8b4d2c : SUCCESS in 34m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75b34278158c41d78a16308b36d10b19 : SUCCESS in 29m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a57fe681cef348f2a6da77b9b97f0840 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e08d487a65d84876a9b519da44c7da41 : SUCCESS in 25m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c29ea57a7d69401fbff6f753e522febb : FAILURE in 42m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a034d28235eea23618cfacf3250e3e44cdd619a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 10:54:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b12d6cf5623e44f09bc3640b66628f5f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e7a0e58222e4c269e03365f8072eb9d : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/daead1f4b5804ca6be38b28b65f879bb : SUCCESS in 43m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/11b56fdc9d0b46998f2a95231b401bb0 : SUCCESS in 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8c2f0f75ee2a7a89dd886a122baa39a0d6a849c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-20 10:54:20.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1423692dc634863838d54a7c1fd1cf43cde60ba8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-20 11:14:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5705aafb03d53954d600c2923d9962f057618d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 11:31:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/249262a9226148acb74d6fe63fb0005a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/597db6c018114877a97912f55575aa49 : NODE_FAILURE Node(set) request e9acb218a59b44c3ab28e5f70dbd6ba6 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b36318ecbce14f74bf527694b1f997d3 : SUCCESS in 28m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4753dd7f6032446691c4c4b541a653af : SUCCESS in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f964bbef6fcd149153166b2778f6d488dae3d61","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-20 12:19:24.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"440588351a8b8f818bd404d6728adeb30678ac23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 12:21:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b2f8b369fc4c3eb4cbe7004df7e43e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daf9126db2da4b9da2547dfbc23c1642 : SUCCESS in 1h 01m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f7bc26bbe14df8acbde87a251ba274 : SUCCESS in 8m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e726d5495e54faa868ac3a96bcf849b : SUCCESS in 26m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14d51da5d3014252aa31a00526130758 : SUCCESS in 31m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc9d7b3da434eb9a26e8c7fd5fdad06 : SUCCESS in 12m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdb433e7da7f47f1be583a8bb6a66ac9 : SUCCESS in 25m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/11497bd357344dba9b2a6a91dc34fd2c : SUCCESS in 1h 24m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e91eac65333f2428289c1663572b80e9d0eb7ad6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-11-20 13:31:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3c7b88744bb74efb9008137c8ac57b803fa41c19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 13:31:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"272f00a01ebdc8eb9f9f4934d07c17b0bf873242","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 15:32:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1115a16078084fa19b29d8251513cf3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b405063093fe41c180deb2bef284da08 : SUCCESS in 14m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9d8246a38d748469014a017f4970431 : SUCCESS in 20m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7adfeab14354e55acbe6c60720d933b : SUCCESS in 25m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fbc59dcd436484e915324bb6558eb3a : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8191a098f274aa0b5f82e997c350d78 : SUCCESS in 30m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b14a4b1d0c334cb8acef5d64ea5aa966 : SUCCESS in 1h 53m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9d28aa1dcb5fcc861ee3443a7bfbe45e0bee5994","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 15:32:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"813d82182268896b95187cd094cb1f2a420a4427","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 15:33:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80388a2720b64cd1a1690d15ad73ee77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70955ebe292248dbbc402e24851731f7 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94e1ab5999504a7caf793f9a241f35c7 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2806f566f3740bc65553d343c7dbffd4b3e523c1","revisions":{"f8344ce04aa0a9a78f059e59af117525e5736a4b":{"kind":"REWORK","_number":1,"created":"2025-11-20 09:52:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/967813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/967813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/967813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/967813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/967813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/967813/1"}}},"commit":{"parents":[{"commit":"98af2f898a435928eceba3cb9fae4efa6f242a67","subject":"Merge \"[OVN] Use ``uuid.UUID`` to build a OVN database row\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98af2f898a435928eceba3cb9fae4efa6f242a67"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-20 09:49:57.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-20 09:51:53.000000000","tz":60},"subject":"Workaround for the OVN/OVS directories in functional job","message":"Workaround for the OVN/OVS directories in functional job\n\nThis patch is overriding the OVN_SRCDIR, OVS_SRCDIR and VTEP_SRCDIR\ndirectories until the proper fix lands in ``ovsdbapp``. This patch\nwill unblock the zuul gate temporarily.\n\nRelated-Bug: #2131994\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0fc4b92a8dbaf5631470a9d74860fe76d8dd6109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8344ce04aa0a9a78f059e59af117525e5736a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f8344ce04aa0a9a78f059e59af117525e5736a4b"}]},"branch":"refs/heads/master"},"2806f566f3740bc65553d343c7dbffd4b3e523c1":{"kind":"REWORK","_number":2,"created":"2025-11-20 10:54:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/967813/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/967813/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/967813/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/967813/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/967813/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/967813/2"}}},"commit":{"parents":[{"commit":"98af2f898a435928eceba3cb9fae4efa6f242a67","subject":"Merge \"[OVN] Use ``uuid.UUID`` to build a OVN database row\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98af2f898a435928eceba3cb9fae4efa6f242a67"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-20 09:49:57.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-11-20 10:54:20.000000000","tz":0},"subject":"Workaround for the OVN/OVS directories in functional job","message":"Workaround for the OVN/OVS directories in functional job\n\nThis patch is overriding the OVN_SRCDIR, OVS_SRCDIR and VTEP_SRCDIR\ndirectories until the proper fix lands in ``ovsdbapp``. This patch\nwill unblock the zuul gate temporarily.\n\nRelated-Bug: #2131994\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I0fc4b92a8dbaf5631470a9d74860fe76d8dd6109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2806f566f3740bc65553d343c7dbffd4b3e523c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2806f566f3740bc65553d343c7dbffd4b3e523c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
