)]}'
{"id":"openstack%2Fneutron~967902","triplet_id":"openstack%2Fneutron~master~I6fc382c88cb9e0087b933e080ac6846a1619b1ca","project":"openstack/neutron","branch":"master","topic":"bug/2126791","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-02-17 10:47:51.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-02-17 22:10:57.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-02-17 22:10:57.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-02-17 18:13:49.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I6fc382c88cb9e0087b933e080ac6846a1619b1ca","subject":"[eventlet-removal] Handle systemctl stop for DHCP agent","status":"MERGED","created":"2025-11-20 18:35:55.000000000","updated":"2026-02-17 22:13:13.000000000","submitted":"2026-02-17 22:10:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"967902-bug/2126791","meta_rev_id":"d37f60b74bc5aa01f597687706b19c457ce80c29","_number":967902,"virtual_id_number":967902,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 22:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-02-17 10:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-02-17 18:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-02-17 18:13:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-20 18:36:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-20 18:36:11.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-11-20 18:37:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-20 21:27:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-11-21 19:43:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-02-16 09:30:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"ccc1aa8f741ec00549dbc40a25fb16fead7cccd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-20 18:35:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcf6023e35e64e659c5e6e7051184a9b5b6dc219","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 18:36:07.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 967811,1","accounts_in_message":[],"_revision_number":1},{"id":"073ed0a17604f8b993b83ffae5fdeca61f9cecb5","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-11-20 18:36:11.000000000","message":"Patch Set 1: Code-Review-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 967811,1","accounts_in_message":[],"_revision_number":1},{"id":"847918cbebaab51712bb110071d3852e83823d20","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-20 18:36:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ace66ae1a193c9604b84451638879df9ff04b110","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 18:36:29.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 967811,1","accounts_in_message":[],"_revision_number":1},{"id":"c77839c86d4f5b0ff260355f7f0ea53b12ff1d4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 18:37:17.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 967811,1","accounts_in_message":[],"_revision_number":1},{"id":"f7aa6f9f05b378b952dafa07e5304e106e84ce0e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-21 07:22:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"accc916cfcbf879d7f97a46e5fc32775148c0755","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-21 07:27:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ff54da6000e9a50166019a985b160a2a1c053f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 08:40:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b0e88abc7a34b4a996818268818167f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fffdaff0f39b4f4292e0b57238957c32 : SUCCESS in 26m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ed3d7444320c4053a59141c7896dbc24 : SUCCESS in 56m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dad884bbabc2474a8bf2b596e2b5eedd : SUCCESS in 1h 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58fa7c4357f322eabc3fefbab859a63169e101f0","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 09:58:00.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/099bafe6d3044453b8d26428539e8536\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/c70d7287db064f5eaa9597ab12577241 : FAILURE in 1h 22m 21s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/f34de712b1af4121a3b5b016a8d5c9d4 : SUCCESS in 1h 23m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/bcb6a8154062486e9abe6c918172bd0a : FAILURE in 1h 46m 13s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e8be0324d8f048d093aa180182965470 : SUCCESS in 1h 50m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b40e832fd2e4d9883fadd0b4fa3dc26 : SUCCESS in 1h 46m 45s\n- neutron-functional-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/72e2d472713b4d4b9adeadc4195b2e72 : FAILURE in 25m 28s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aa1d99f821a240e78a893b84b0f44601 : SUCCESS in 44m 46s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0a3f8ddd5d93475691229405847f2906 : SUCCESS in 1h 31m 35s\n- neutron-ovn-tempest-loki https://zuul.opendev.org/t/openstack/build/b0599d9500e74b0caee11ee34aa5cee1 : SUCCESS in 1h 25m 46s\n- openstack-tox-py312-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a2acad71612f49e9bc10bd63c0c01489 : SUCCESS in 25m 24s\n- openstack-tox-py312-with-oslo-master https://zuul.opendev.org/t/openstack/build/4fac88ac6d0f437fa20068a64e0b7e4b : SUCCESS in 33m 17s\n- openstack-tox-py312-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/17e658777b6b48d59b41f8b6b0dc017b : SUCCESS in 30m 43s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0e63ad17b2ef4f8d997f700d98e2fd58 : FAILURE in 42m 31s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/7047eade26a74a02b5821165f0ac69f4 : FAILURE in 26m 48s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/3a9e5097bf4e4d9e93f293c4827e441f : FAILURE in 46m 26s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/73be82fbb18443af8a8d5d1495714018 : FAILURE in 46m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a6cc77cab28a4c749618d63fab06fe49 : SUCCESS in 40m 23s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/dafdd20fa215489a89ce17e580041d42 : SUCCESS in 1h 08m 15s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/55d359eb181a4b96a18d2e78efa595ee : SUCCESS in 2h 10m 44s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/19588a7330b64133b41dcbd50b0f8907 : SUCCESS in 49m 38s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/cd547eba7cda4fd1a34e3cc281c2b604 : SUCCESS in 1h 36m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/36a19904df69438c8e277005867d0e1e : SUCCESS in 1h 19m 25s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/8069b197e9b14f639b0e54677aca4566 : SUCCESS in 45m 41s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d06a0e318b244f918f772d99fdf8ad00 : SUCCESS in 1h 25m 04s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/fb2c7cdc2761476e80c71cad08d9de05 : SUCCESS in 1h 39m 54s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/5f7ac8a307d848aa8132e0f7f018c657 : SUCCESS in 1h 20m 13s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/92986f260532474cbc63abdf60effb18 : SUCCESS in 1h 24m 45s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/49569876b6b545ba8841025ea29c149b : SUCCESS in 2h 24m 53s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/3c6a20752e5d4330996f39990ce982f5 : SUCCESS in 2h 28m 27s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/ee50952a2923421ea6640bb14d3c0496 : SUCCESS in 1h 28m 56s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a968785a80fc486da0097a129ed4aaa7 : SUCCESS in 1h 08m 42s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c579318be6944e22b6c741997c12448d : SUCCESS in 1h 18m 54s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/7b3bce24315d4956bac18c9fe47d0413 : SUCCESS in 57m 03s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/84ceadb6959740a3850498be61a3c8d3 : SUCCESS in 1h 20m 01s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/6b78fd9f6c9c43b5b1a64f8bca995693 : SUCCESS in 1h 03m 45s\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/cf7e53b69dc64fffb799193eb5c98bde : SUCCESS in 1h 14m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/467b71ff12e441418f2f3e0b64ec79bb : SUCCESS in 1h 26m 23s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3fdcf4a615e14ecca202e7ab76cc91b2 : SUCCESS in 1h 15m 19s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/3138927a1b2949799c3a3cf3414f0f12 : SUCCESS in 1h 46m 02s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/8d9984cdf29243c9974f5b67a0a42cf0 : SUCCESS in 14m 51s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9b7c247355634e2f89ebce467298d0b1 : SUCCESS in 29m 07s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/d563bd0fdf384d0d93680280343a9221 : FAILURE in 31m 34s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/8ec0cd6fdc9c4fc29b6fb320d9cbac94 : SUCCESS in 1h 28m 14s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/d98b339b61c4485abc22fab9b1a4fc71 : SUCCESS in 1h 03m 15s\n- neutron-tempest-plugin-ovn-with-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/c47d0d072c4f480fab23cc34a21d35ef : SUCCESS in 2h 00m 47s\n- neutron-ovn-grenade-multinode-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/aaaf674d60654b9ea9eb5ffae646d3dd : SUCCESS in 1h 08m 16s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/b7e5f22653ff48a5888eb58d6e6bbc69 : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b7beeb281d7e96886a358d2a818adfe5728c7f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 10:10:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d819c94e2134f3ba3f907ea7abe3e12\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/74038316a5ff4c45a3d970678faafe7e : SUCCESS in 1h 36m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b9c93b845b63411ebd805b2927f3822b : SUCCESS in 2h 40m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1ddd60ac5de745f98f77d43f1b2688b2 : FAILURE in 25m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f37183e2b914572b1f3055f5a4a2337 : SUCCESS in 53m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c4b02cff4e4807968a9aea3416726b : SUCCESS in 11m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bca0c59c31e454cbe1fdb13bc613d82 : SUCCESS in 30m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/519114963b724215b44417ca40cb61a8 : SUCCESS in 24m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7290883c253548b4b65af751a6486901 : SUCCESS in 12m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3ce7496384484a56a31d3d0878dc037d : SUCCESS in 58m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3de37759967c4f4cb1833936843e40ec : SUCCESS in 34m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8120252d2e1040da8f489caf5af50966 : FAILURE in 43m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4ad5e8d25ee4726947b04e8855badd6 : SUCCESS in 58m 38s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/8256b1e7a1c741888184e243114439af : SUCCESS in 43m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2cc6a117d68f4345b15305b4ea5e6b94 : SUCCESS in 2h 04m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e93d9a8d793b4b4c9e188c6bc28fc8ac : SUCCESS in 56m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/20ce2dbb442649ba825969b583c03af9 : SUCCESS in 1h 11m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f0e171a8c90a4fe3a4e51e984a3e5cbf : SUCCESS in 1h 32m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62113457cffb42959e08096b1d0b47f1 : SUCCESS in 1h 00m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2c6b3ebbb31f481895357bb9a8316ba9 : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ea2feecb4942d8ae7c91740cb924468946eae9e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-21 10:43:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e91663a4c367762aca4788302f8f5af42dded2dc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-21 10:44:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac0cdec04753908e33262572b47db87327f869cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 12:43:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a374b1d0f81488b93d819886306f189\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/63d8fac249b1429a86e56500c91c4412 : SUCCESS in 1h 23m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4e688869948a40d2b295631739951ff3 : SUCCESS in 1h 40m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0b7113d5b7f9487ea5a6f11f8f0be558 : SUCCESS in 32m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98f0b7288ca948fbb6c48b6ffda66aee : SUCCESS in 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78fa0d88a6914a95b13c84e7a5154f00 : SUCCESS in 12m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/690e0fccf66e46f7858296b9b88e90e2 : SUCCESS in 47m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb03facc361a4ae88fd817697b5be56a : SUCCESS in 38m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abced5a5ece4a92aeb7a02159fecf01 : SUCCESS in 7m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6ab7fd7f199748cea048f69cca0704a0 : SUCCESS in 49m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/dffc9bb753e34e19b096c78c4c7bcfff : SUCCESS in 1h 14m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/dfa79aa29e5a4732ae624572c653cb51 : SUCCESS in 1h 33m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0f75b3c5a4e4a84b1d1a492c4620115 : SUCCESS in 1h 02m 25s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2864f824e8e49eb9c395d3fa9bb75fc : SUCCESS in 1h 09m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/698c93da8c3249ef97e150ef83d7463b : SUCCESS in 1h 42m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/128127a21443495ca793094a558a48e4 : SUCCESS in 1h 44m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1c233357987842ae8288fb7d7c16f7f5 : SUCCESS in 51m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/242c4451262e451382c6cfd7867e5b31 : SUCCESS in 1h 14m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1133711ba6a24280801432493691c384 : SUCCESS in 1h 18m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/690c3368e061496a9124b2ddb9cb13de : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":2},{"id":"74992dd6d018f25e1bf1ce506c7b4cc9066c89aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 14:42:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bb0a1c546df4208be30a4447437950f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ae5a7c34e384af986d088f23c18ff7a : SUCCESS in 1h 08m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e15434fdf6d47c5acd67e49b5da8f7f : SUCCESS in 56m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c1fe35904b14c32b091e8e7fb0098ab : SUCCESS in 45m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5fe40e765188f608376498957e595d982c896265","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-21 19:43:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"be2e9bff95d23bb2ed6786ee91c184fb5250436c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 22:48:55.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f109db2545d5452fa50b3dfa9b0d329b\n\n- neutron-tempest-plugin-dvr-multinode-scenario https://zuul.opendev.org/t/openstack/build/6a8e533277084605bfb2014dcd60359d : SUCCESS in 2h 25m 17s (non-voting)\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/f164667040e24411839ae682bdc50c9f : SUCCESS in 1h 15m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/a7179eb51e204906a43dc3aae1718f19 : POST_FAILURE in 1h 38m 54s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ef78a8ab468a47788c0d0b017f4fa944 : SUCCESS in 1h 45m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dab9dfdd0cc8472783d317c30f79b481 : SUCCESS in 1h 34m 35s\n- neutron-functional-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/2dbcb4d913214f259280ee1135253893 : SUCCESS in 1h 50m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/aad1edbb4b7d45189c4a6954c5795a2a : SUCCESS in 1h 18m 13s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/5e8216fc8d1b4c99a9f188f769b14c13 : SUCCESS in 1h 06m 41s\n- neutron-ovn-tempest-loki https://zuul.opendev.org/t/openstack/build/b6c2b598fd1b46a1aef1ce8181ae8126 : SUCCESS in 1h 27m 37s\n- openstack-tox-py312-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/595b1cdc99184ca99465530df0f4757c : SUCCESS in 30m 29s\n- openstack-tox-py312-with-oslo-master https://zuul.opendev.org/t/openstack/build/8b274dba38674907b39219553d16e853 : SUCCESS in 33m 25s\n- openstack-tox-py312-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/53c4535ca32b4a8d80ae9ab0af1afb1e : SUCCESS in 28m 49s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5399f6f321624b7782f2473ec48776bc : SUCCESS in 1h 39m 40s\n- neutron-functional-fips https://zuul.opendev.org/t/openstack/build/d2471d7f902e4bb5af7f677fd49806f0 : SUCCESS in 1h 31m 59s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/da78c1f5263d4a66b996bd06c5d6bb1d : SUCCESS in 1h 07m 40s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/27b3ac78ac524b608da8141bee18de81 : SUCCESS in 1h 07m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8790ed55f68a4c6bbcaa23b1ebfb7d07 : SUCCESS in 1h 21m 26s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/6aac0c26d6a0425faf838297171c16ec : SUCCESS in 48m 45s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/89ee9482d5ef4d1f9146787cae5ac934 : SUCCESS in 1h 06m 26s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/3223793519aa492583ed031e81fc0674 : SUCCESS in 42m 02s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/14744ba6ba3d4fb2b5812f8f2e3639ef : SUCCESS in 56m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/bff6f4bc2f884ef6a2ceb67b15cf409a : SUCCESS in 1h 19m 10s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c2b86914ab049c184c64598f5ac1f5b : SUCCESS in 52m 27s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/36c9b883aa7f4792bbe589fab74608b0 : SUCCESS in 1h 48m 10s\n- neutron-ovn-tempest-full-multinode-ovs-lts https://zuul.opendev.org/t/openstack/build/bbf1a555a9614c01b3596fdcbe6269d8 : SUCCESS in 1h 33m 13s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/ab4823468e024b31975d24befbedbd5e : SUCCESS in 1h 26m 35s\n- neutron-ovn-tempest-ovs-lts https://zuul.opendev.org/t/openstack/build/bf93e41ac8874e86a14a4d45a63bcb3c : SUCCESS in 1h 23m 47s\n- neutron-ovn-tempest-plugin-ovs-master https://zuul.opendev.org/t/openstack/build/1b8b3b2f40064b2393b1076d0de92d8d : FAILURE in 24m 02s\n- neutron-ovn-tempest-plugin-ovs-lts https://zuul.opendev.org/t/openstack/build/08ffdc1f87554a49bbd5e6d34f7cae0d : SUCCESS in 1h 55m 41s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/199aca57b2a94df382476f531e7ed5d5 : SUCCESS in 49m 24s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d64c7c5eb5284c09af0e57491e7b2ca4 : SUCCESS in 49m 56s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/52a679acf2ea4317b0b9d2be5f427687 : SUCCESS in 42m 32s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/380dcd846e2d4c0988f682113598c780 : SUCCESS in 1h 00m 28s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/c04410937fe84de791d38fff5bd91a56 : SUCCESS in 1h 38m 55s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/5a4be92cf6f349b6b0861099e71796c2 : TIMED_OUT in 3h 03m 19s\n- devstack-tobiko-ovs https://zuul.opendev.org/t/openstack/build/3f34446fcbf44ad2b391e7fb74549087 : SUCCESS in 1h 14m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0272fa4ebc684585b59828ef9d0c1c2e : SUCCESS in 34m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9cc8c6d3779946dd8a9e9f55332e3636 : SUCCESS in 1h 28m 49s\n- neutron-ovs-tempest-plugin-iptables_hybrid-nftables https://zuul.opendev.org/t/openstack/build/71629be4b1894814802a37bec79f2ae0 : SUCCESS in 1h 47m 00s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/8ad675502e2341b497eadb8a5187a361 : SUCCESS in 25m 16s\n- openstack-tox-py312-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6cf261478d7f4a3eb79b0279d4b2c41e : SUCCESS in 24m 55s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/78b67a4cbc3c4f0f988d50665e201438 : SUCCESS in 1h 38m 45s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/4adb6f3627554a4f9267e932fd8d486a : SUCCESS in 1h 21m 28s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/9a66498ea49e403bac35d046a6a2e491 : SUCCESS in 1h 19m 13s\n- neutron-tempest-plugin-ovn-with-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/11017f09af79432d929f781c32612b2e : SUCCESS in 1h 53m 50s\n- neutron-ovn-grenade-multinode-ovn-metadata-agent https://zuul.opendev.org/t/openstack/build/51a0e0630b0645abb37cf73331015c71 : SUCCESS in 1h 10m 03s\n- ironic-tempest-ovn-uefi-ipmi-pxe https://zuul.opendev.org/t/openstack/build/467e94ad60ff42fbbb42986ad62e2a3b : SUCCESS in 23m 22s","accounts_in_message":[],"_revision_number":2},{"id":"46891d4a121243d779feb7fff395b7d6b70f704f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-09 18:09:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0ed7f533138a259d9855506473996a4ef9e6cfc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 18:53:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71ffdf0137bd4de49c2097795fe2dc0f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec4213e2acb04a3fb2760d262aaca024 : SUCCESS in 28m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/409d2845cb034d41aa2e7ba39c0213b5 : SUCCESS in 41m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ff2a51e2bd504e88b30bacedd1967fac : SUCCESS in 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1db6a0b0ab23b4055e89a5ae684c311e8fcf1cc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 20:07:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9ecfb3b41244e87acbcc31b70cc5481\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ffff02d67fd149d19e6d539ee3979d0d : SUCCESS in 1h 21m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/63cb05b2d4964a78bfc11c03a1718f65 : SUCCESS in 1h 54m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/27f71082093d49c198a13fe541f48aeb : SUCCESS in 34m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/120d6f0736bf4d9ba113822132ef1d65 : SUCCESS in 1h 00m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4b6dd919ca4ee8919c09c9bec97092 : SUCCESS in 26m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a87e2b9b4b0442faaa259e856843bc8f : SUCCESS in 37m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a2925ba27d546aaa6617f4b7b472d15 : SUCCESS in 22m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd91dc45943d4022964b0ddf52839252 : SUCCESS in 10m 20s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/41e2dba88f714959812e340b3691876a : SUCCESS in 54m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a905d4134dbd468692c4a0ed1c80013e : SUCCESS in 1h 12m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/bfdf6c8472b44dd2b7fc2c7c4d6f911c : SUCCESS in 1h 34m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7744ca2f327b4198915df49a59a83146 : SUCCESS in 1h 06m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/78b987d3b22b4d43a3dfa9185515590a : SUCCESS in 36m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3cf8b9ac6f1b4ff9ad86d6bda05a1059 : FAILURE in 46m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/59eaeb2596404d6dab92fa35bd31c833 : SUCCESS in 56m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6186f02cb5b0491aa98fef9f6adb6724 : SUCCESS in 1h 15m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2642011d002d4d79bf4de36b687a2b7e : SUCCESS in 1h 12m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e64d52787a754b47b599f4eebebe77e9 : SUCCESS in 39m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/30964b720baf47b88866cdf014fa3125 : SUCCESS in 1h 32m 00s","accounts_in_message":[],"_revision_number":3},{"id":"97ec9949b89206701b36c58f0028c29fbe291db8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-07 08:53:37.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1551f41709b8057c8dd626d22ccc21f45fd88c41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 09:50:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc93c13e7d7b4020a85a33b9c3d22d6d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bfc1e1524dfb40d18f254e1428514198 : SUCCESS in 46m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a2abcbb5e0374996b64dd5afddd4a83a : SUCCESS in 41m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9e7c75ad10d44ac3b3c803ffe1b86065 : SUCCESS in 54m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"341f1075e4c8d01b0d0440cc3b232d0e83b017ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 11:26:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4b5689a54084efbb116e3d148df4b95\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7ba6c183bcfe4713be8e2f5c631e5899 : SUCCESS in 1h 20m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/73dccb96f6b34782b0818c69c516be71 : SUCCESS in 2h 31m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/003213975b5b4e74b9ceaf90733e1252 : SUCCESS in 31m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48ed58a58bd0467582fddf85b6f87d3d : SUCCESS in 36m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60edbd6c34244fbb859a2f3983d15fd : SUCCESS in 24m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f283fcb9d8e48de89ba248e523d93e4 : SUCCESS in 20m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26b8a2b3576442a791a7eec371571f77 : SUCCESS in 37m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/219bcdaf1aa74e608e583b7811e563ec : SUCCESS in 5m 38s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/be95400a544042b6aaf2fc3d49681c2b : SUCCESS in 1h 06m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fc8aeedaa2c34bda91db91b8b6b66604 : SUCCESS in 1h 04m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/3d0f037bb0964ee79f2b77d54746b9ba : SUCCESS in 1h 33m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/608822e18e5d4e079e77ee4631937b3a : SUCCESS in 44m 55s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4849985ac38e4d689c31d3f0eb2023ad : SUCCESS in 1h 17m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f8ceb85cfdfd4feba71dd61b0578a9b4 : SUCCESS in 1h 27m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8f9f0af224ef41d2a35aa933afc81d46 : SUCCESS in 53m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c5aced83d319420d9ae40c97cab3d194 : SUCCESS in 49m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/03a59839795d4cf78f608998eebffbda : SUCCESS in 54m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9849a16c62e45a199990ba358e0f827 : SUCCESS in 1h 01m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b1831c513ec84a548da2920e64104da4 : SUCCESS in 1h 20m 49s","accounts_in_message":[],"_revision_number":4},{"id":"513a015b148b0b17f70d5b42357ce8740b69515c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-21 18:53:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7e49ca2dc852e43fcf76bbf9423ecfb2a3faaba0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-21 19:55:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d511b0bb7e1577a30f58c904c7ff5ad746f14acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 21:05:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa448098c4724385bf2239eec6c9d8d8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/991bca5855c9484eba3a9ae634498866 : SUCCESS in 1h 25m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/738913d39b744d3abd94418730a7c361 : SUCCESS in 1h 40m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c16d84969aff40e19519d67831e8c0b8 : SUCCESS in 32m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e6a6e9adcde43c596f8dbea606a6e84 : SUCCESS in 1h 06m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40e93511c33041ba8fca5587631b06c8 : SUCCESS in 26m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77a36fde6ee946929bafa3899b29693e : SUCCESS in 35m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c517c9e827884e25b5897caa141f5165 : SUCCESS in 24m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35915a958bc84c2fa827872733d94f13 : SUCCESS in 12m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/85ed824a2f304123b57207e0ee4f5f19 : SUCCESS in 56m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/20e15e30494d40d49446f4a08cb1b948 : SUCCESS in 45m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/f6782710061f48978c6bf444f86f9517 : SUCCESS in 1h 12m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/76f54ee0f055400f8603dfda5305cb20 : SUCCESS in 49m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1ab1003ed5d45d9931abbe1cb380905 : SUCCESS in 1h 08m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e317e53f65d845e1b58e7131a1a46c54 : SUCCESS in 2h 04m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e7990c029ecf418c8348fd46a0114f76 : SUCCESS in 1h 38m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/411890db23604d68b9ed388dda4c9a5e : SUCCESS in 50m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/74a07b028bcc4045874eb8053c519e81 : SUCCESS in 1h 08m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f0406f94a5004ff3836c46715176dfe5 : SUCCESS in 1h 20m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf882f1b695f48f487db062b452e22b3 : SUCCESS in 1h 29m 31s","accounts_in_message":[],"_revision_number":5},{"id":"375337fe23760462aab3cf796e9ea8d51b3c8cb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 21:14:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24fcc4bd895d4ec7aefa6bcbaff7ef81\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b075b7c3327d4c5c8215193bf8f6c302 : SUCCESS in 38m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/836824c148ee4bf58f2d6280973ca937 : SUCCESS in 24m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd40c4aa67e140888967d60896d2a610 : RETRY_LIMIT in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"23253c4b65d84ecc2f52c1e719d37fdfb83409c6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-23 04:46:39.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9c7a8796d94e9740b4b2320239f36a33e05e194d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-23 15:56:22.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2b24f201cc648d927e6c1d6cc10cc0c48863d88c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-23 15:56:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"57ef5500f2955b005da165d6c4c5107ffba77013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 18:42:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5155e21159e841228cc7512675b20a79\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/626eba0ad83242a3a2152317ecdfa4de : SUCCESS in 1h 22m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a3a127f573c94834bff9792ca59c8a01 : SUCCESS in 2h 12m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0c7602f0fbf945d99eb3fa69fc5a6a79 : SUCCESS in 32m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e53c252454a4588b48ab6c561975d8b : SUCCESS in 48m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d0a52a174c47168942a0b7ab88598d : SUCCESS in 26m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fefd32b157834f998078a6ca73994920 : SUCCESS in 36m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d949041431a94569819ee8cccd893fb3 : SUCCESS in 27m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678695fbc90140e3aed60d00e8a41560 : SUCCESS in 11m 33s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/77ca6da22d2245cc8947341847973437 : SUCCESS in 51m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/13722205f58246bfb316bdd8db0ccf77 : SUCCESS in 1h 13m 14s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fa9607c3bcba4250a0cf8fc938e553e2 : TIMED_OUT in 2h 32m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1587c089e5b74866858454206d6128bc : SUCCESS in 1h 16m 30s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/eac29ab810124434bdfa62d31dcdbff3 : SUCCESS in 57m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a52c569105d14122a3f70b31378e41a2 : FAILURE in 1h 20m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cad2406c7ecc4274bf9e1fcd63dca109 : FAILURE in 1h 21m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/41e2c75b8dcd4cd3a0c20ef7f11e7d2c : SUCCESS in 1h 08m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5a7033317b874b4f9881d1d7c6fac198 : SUCCESS in 1h 18m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d68ef01781834478b41d88ea4525de9d : SUCCESS in 1h 32m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a776e2b873741f2929813795e4f7a09 : TIMED_OUT in 2h 06m 30s","accounts_in_message":[],"_revision_number":6},{"id":"996a5c25f0ac5d63d47c4b3f9f9c672bdb9f55db","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-23 20:26:55.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a5ac439ca6ac21774b47b9ce0c6b63a7e3caaf3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 22:31:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9b369f14922418b8461ee18a4372ea7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/52456b5e07ca4c63948cb56ce2b62ce1 : SUCCESS in 1h 16m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fd1ce437aa3c4b75ad0c23865eec92cc : SUCCESS in 1h 55m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e147489089e54271ab32f46ae9209aa8 : SUCCESS in 30m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f508fcfcb964d149b731e63ea18375d : SUCCESS in 1h 04m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28c9298945d4a1a816ae7b9a0ef829e : SUCCESS in 30m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/617c4a38ad2b4f48a9e0cd629f38207a : SUCCESS in 24m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37624a78f2724356806dedd437f26ec9 : SUCCESS in 16m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf45344fa0a44ef090dd1857ade1533c : SUCCESS in 9m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e93789cac78f4e6dbfcc0d563cd08a31 : SUCCESS in 58m 13s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/757142cb906b4694a501edafd65ed68f : SUCCESS in 1h 09m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/48c7e66e4db5441089a4ce4970a89ccf : SUCCESS in 1h 49m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c31b2236884940c789e410cc5b4ddfa9 : SUCCESS in 1h 12m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/80949db604c34647b3c5be91c5844fd4 : SUCCESS in 35m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/51556488d3e0465c8fc82f92506bfa62 : SUCCESS in 1h 08m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1dfb5d562436447fb404a3e2c354286c : SUCCESS in 1h 38m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f065797a9e1e4632bf8e02d21beec623 : SUCCESS in 58m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ce70af32c7da424e9d389dd273fe1279 : SUCCESS in 35m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5570db4c7ffb42138a4b3ab6733f3b1c : SUCCESS in 1h 28m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4381850da98d4157a091a1ff6be8ce31 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":6},{"id":"7b978eb1a3ed674c60dd277e269dd8ab2e6d25fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 00:11:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40f53fc0a56443f3ab019ddcd80ef67e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1eb0c5c5208b4b128c017eaf91e529a9 : SUCCESS in 56m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b48d7de94fbc4084a2bf552931cbe7df : SUCCESS in 47m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b8f1a1dab59d421eb7f5ce2d6ecc4d1a : RETRY_LIMIT in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"21f6b6bd239f40006fe35a5aaab51abe62cff777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-29 09:47:28.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3736eb92e4111ccb592264a5ad99dbac10244306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 12:12:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b55d30f1007643d98643e4a81613064a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0d544fe228b5403cb41abc5851e6d0fa : SUCCESS in 1h 39m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ac183580d40e4615a995cf35e7bcb0b5 : SUCCESS in 2h 23m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4472b9fd7ec14db2b29a521758beff94 : SUCCESS in 32m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b80b6e2844448689ed572f7f4bd8b33 : SUCCESS in 1h 05m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b2a61b62a94613a67cf279fef329ed : SUCCESS in 10m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0408159e7df44fc3946d194971ea6be8 : SUCCESS in 45m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0671bb9b4c074302893d43159edc9df1 : SUCCESS in 24m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be783e67596e4e7bb5ec4a557106e9ca : SUCCESS in 5m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b10e289d0be04de3b4bbcb4e69db634e : SUCCESS in 30m 49s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fb535946c2234df6a6c060658b20eed4 : SUCCESS in 1h 14m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/503a1b7f23a04587ae659f9172531751 : SUCCESS in 1h 23m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5292dabaa4664b39a760ff6493293013 : SUCCESS in 56m 45s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/757fdeee02764325922135240e337d30 : SUCCESS in 1h 13m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5e931ca5d8a647e8aacbe37323388bb8 : SUCCESS in 1h 57m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8918318bcd5347fea1c69c1f432df6c3 : SUCCESS in 1h 42m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3843eca6721e476aa555e981382c299b : SUCCESS in 1h 07m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/31a621a4a1784fa2aaa2aef4e9c290ee : SUCCESS in 1h 08m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2bfeb1e2c6e844d7a9d070c3a8c08567 : SUCCESS in 1h 25m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6abde7e18727438daf3b105ecdc7bdcc : SUCCESS in 1h 27m 34s","accounts_in_message":[],"_revision_number":7},{"id":"38fab5b74f4bc92d199a46581b8e9916ce418ca8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-29 18:15:09.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"88012181f6ec29ee2a728f7bbb380ba4ad9185ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 18:59:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da1cc7b048b94ba7b01e1809893df9a9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a9f6f47a58a0428c90e4e63cafdd46dd : SUCCESS in 1h 02m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f13d7786267c41138fdc1a957d63d4b5 : SUCCESS in 43m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7643f2a2affa4276814ba10e3508299c : RETRY_LIMIT in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2d176ae477193ff0be1b7087fdb19f19ae84ff8c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-12 09:44:24.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"97dec86fc3764a975d2cf62b8273edf40d37843a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-16 09:30:10.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e4e4606024b51938de25ae00a60c8f992039b60d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-16 09:57:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b48b993e3462ded61067d914793d7c6b137d1ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-17 10:45:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d667389110612a6d8359ba0e69ef9252616e8ba6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-17 10:46:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"51b1683a8c1606e89cefc210e1c369ba96f48865","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-02-17 10:47:51.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"967d55b0b7b107289bbda5f90944e7d266ad7165","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 12:25:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29aa94f9e3744abfb8c3836969e969eb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/26a4259002ed4644b7d5f763f37d62d4 : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f5ac5f7076e42cdaf129ca935441b17 : SUCCESS in 21m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cde39b2412e84ed5aabef5d977c58752 : RETRY_LIMIT in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c2c7b24694c43ca6971e6aa0518f1afe75596b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 13:02:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35d670e8518e42f0a4144b1960b3e76b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1eb4164e109d44b18486e9232540186f : SUCCESS in 1h 31m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/43862050a1d64191b9bb524a2406787c : SUCCESS in 2h 04m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/88bef3a1f66c40579bbc98982d9f7b12 : SUCCESS in 21m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92529239b9854a70b9f4315b0659e14f : SUCCESS in 42m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/616ac08e5d1a4ba79fdcf7be268660ad : SUCCESS in 20m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69cd3cf706c44c5ca1ca94c07b6dbec6 : SUCCESS in 28m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76cae06d4408433bb1cf40e346f175ab : SUCCESS in 18m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f0d606b527457cb17157a47e204540 : SUCCESS in 14m 43s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/16112f652fad49a39a4158e59bc54402 : SUCCESS in 55m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2b96e00bf1f0424d9557b95693961914 : SUCCESS in 41m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9f62b30ed556447da5e64a36b056e1e1 : FAILURE in 31m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc8e0e452baf4feba88686d98ad38ff2 : SUCCESS in 41m 35s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3832004291574cf2abf5f86bbc7ae4fb : SUCCESS in 1h 10m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aff9f9d685cc4c5183eaced8b6611acb : SUCCESS in 1h 06m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c2a3f60b06f44308c14d189d370c3cc : SUCCESS in 1h 40m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9a13352fe1f84739a0ea0f9901c0e851 : SUCCESS in 53m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/49fe35b09626474cb5c478803f4a98e5 : SUCCESS in 1h 29m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/124ff9e7d7704bad818243f31cdf7032 : SUCCESS in 1h 26m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f7d2d0953e774fd78f9616dce89e6526 : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":8},{"id":"c01d60f5bbe1677f35ffcb281d9efd6857c11fb8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-17 15:21:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c2797d89ae5ef12df4490687e9f2a7215856991e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-17 15:22:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1016aa6e8b210c1b5b7ed8791e3aa0097b576e8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:02:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2435f0a2aecd4a95827f3f4b4432fbd8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/722523a380be4f26976563300fe9f624 : SUCCESS in 1h 37m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b13fbec0663241a4aee608ca60716079 : SUCCESS in 2h 17m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/67136fe551f7476286204153933f33f8 : SUCCESS in 36m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f16bdfccceef4a2fb0b18ff0adff9731 : SUCCESS in 43m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39622879fc7543ddb3d646b119c51ec5 : SUCCESS in 28m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/621f6073383d44a8b3e4b27f08d99415 : SUCCESS in 48m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ee2e76d5e9044a7924e17a78f7c2335 : SUCCESS in 22m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d5f31e05654de9b29540023b486361 : SUCCESS in 12m 30s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4d1d4b9243ff4932821086b50b43aa99 : SUCCESS in 1h 01m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1c1c0b11c6a245bf8792dce392a8fd2b : SUCCESS in 1h 26m 55s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/db64fbf80acc43ffb908cb1b593a83b5 : SUCCESS in 1h 29m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e7aa37266ca466bb3771d17fc8ece0d : SUCCESS in 47m 17s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/1639e7085f3e422eba7d8607b76fff52 : SUCCESS in 1h 09m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d82c2ca64e2b40aba7342372ec5d7d78 : SUCCESS in 1h 04m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5bdb084636f94e49928df4cb556edb7d : SUCCESS in 1h 32m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/23098ec1aa1541bab11a1129a9a48ee1 : SUCCESS in 1h 04m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c9a73eb28ba74d28905927be4e24f696 : SUCCESS in 1h 12m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7e6fcc9ef31143a7980dff4fe6e3233e : SUCCESS in 1h 24m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7252c92cf3734a4097c0c426a936bd18 : SUCCESS in 1h 55m 53s","accounts_in_message":[],"_revision_number":8},{"id":"d3d6033f1f911551a682973a25da04d5b7dee455","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-17 18:13:49.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"07f0ed0b4b395f70c643a8d6cabac07925134485","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:14:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9119a08f8df8490787eb3053fc17aa6c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4dea99f2189e471c94e734de6968bb1b : NODE_FAILURE Node(set) request cd1e3f1afc964cbaad18e47f8247f1ed failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/117ac4a5d6e64512b94d0eb730a49dd5 : NODE_FAILURE Node(set) request 15c233a3f1d247f4af22c4b8fa4eef4d failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/85ac22ca7d3b49f9b7685038ed780cfa : NODE_FAILURE Node(set) request 1bb8a51c4bea4509afc366556b6abbfa failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8e86559f033e5971ae476f4974be81d81f99cbe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:14:54.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"067340e4141257cfb38c61058364e1ecca8cf70a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 22:10:57.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4ed9b2a08ab4a65b264a579487f18cf\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e4810522344444c58fe63813628049a7 : SUCCESS in 1h 42m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/05599fbd44914f9c943c70e200db0e2e : SUCCESS in 2h 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2102347fde274e0e9e91ce2e744d5742 : SUCCESS in 11m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/835c111db3774f8e9cab564632767570 : SUCCESS in 33m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6aa192cf661e4514bc9974e5ae8380a2 : SUCCESS in 26m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98adf9078254a51a814df180c1b3776 : SUCCESS in 12m 42s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/81827b109c5046f4bc480ec270e796aa : SUCCESS in 1h 24m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/277539a5642c4056b733feeaee6f2881 : SUCCESS in 1h 53m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e081883b31914f05a5f2100aa76fde2b : SUCCESS in 1h 13m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9256145544934c58af709846a1023c9d : SUCCESS in 1h 09m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7c4258ed061f49a9b245470460c6eaf1 : SUCCESS in 1h 08m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/18662178efd04347a5a623efc415ad39 : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":8},{"id":"68ae8ec0372297171d8d09f3c94c51b9d322db74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 22:10:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"d37f60b74bc5aa01f597687706b19c457ce80c29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 22:13:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/038f656a4a4c423ca410bfbed4ebd29a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8550e3bf2a2e457299ba4d6e8fa41d1e : SUCCESS in 58s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b67a007e43786574e337c21d26e35be38f697d56","revisions":{"66e3864b5041fac3104ac4ca79893ae4d5a42fe4":{"kind":"REWORK","_number":1,"created":"2025-11-20 18:35:55.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/967902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/967902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/967902/1"}}},"commit":{"parents":[{"commit":"b34b2e28214486b282c35368f2fae29f3edad9d1","subject":"Revert \"[eventlet-removal] Handle stop DHCP agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b34b2e28214486b282c35368f2fae29f3edad9d1"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-10-08 08:05:52.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-20 18:33:22.000000000","tz":60},"subject":"[eventlet-removal] Handle systemctl stop for DHCP agent","message":"[eventlet-removal] Handle systemctl stop for DHCP agent\n\nCloses-Bug: #2126790\nRelated-Bug: ##2131801\nDepends-On: https://review.opendev.org/c/x/devstack-plugin-tobiko/+/967812\nChange-Id: I6fc382c88cb9e0087b933e080ac6846a1619b1ca\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66e3864b5041fac3104ac4ca79893ae4d5a42fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66e3864b5041fac3104ac4ca79893ae4d5a42fe4"}]},"branch":"refs/heads/master"},"aca7c3bfcdc6e075a1864107a6ada4fc63454ffe":{"kind":"REWORK","_number":2,"created":"2025-11-21 10:43:30.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/967902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/967902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/967902/2"}}},"commit":{"parents":[{"commit":"20585d865d94741ef1dbe4e030557bf5a6db6d6a","subject":"Revert \"[eventlet-removal] Handle stop DHCP agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20585d865d94741ef1dbe4e030557bf5a6db6d6a"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-10-08 08:05:52.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-21 10:36:51.000000000","tz":60},"subject":"[eventlet-removal] Handle systemctl stop for DHCP agent","message":"[eventlet-removal] Handle systemctl stop for DHCP agent\n\nCloses-Bug: #2126790\nRelated-Bug: ##2131801\nChange-Id: I6fc382c88cb9e0087b933e080ac6846a1619b1ca\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aca7c3bfcdc6e075a1864107a6ada4fc63454ffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aca7c3bfcdc6e075a1864107a6ada4fc63454ffe"}]},"branch":"refs/heads/master"},"52b08b003dedf4fff3927984989c4bd608f14fd0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-09 18:09:36.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/967902/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/967902/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/967902/3"}}},"commit":{"parents":[{"commit":"4d2c562d2250a02b88524fcddae2066b1b17abe2","subject":"Merge \"Fix ignored metadata options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d2c562d2250a02b88524fcddae2066b1b17abe2"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-10-08 08:05:52.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-09 18:09:20.000000000","tz":60},"subject":"[eventlet-removal] Handle systemctl stop for DHCP agent","message":"[eventlet-removal] Handle systemctl stop for DHCP agent\n\nCloses-Bug: #2126790\nRelated-Bug: ##2131801\nChange-Id: I6fc382c88cb9e0087b933e080ac6846a1619b1ca\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52b08b003dedf4fff3927984989c4bd608f14fd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/52b08b003dedf4fff3927984989c4bd608f14fd0"}]},"branch":"refs/heads/master"},"06c746da885e89311759c86fcdc3358c627024b9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-07 08:53:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/967902/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/967902/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/967902/4"}}},"commit":{"parents":[{"commit":"af2aefeb87f6e2cdacf2b164338e440e131692ad","subject":"Merge \"[Docs] Remove note about FWaaS not being supported by OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af2aefeb87f6e2cdacf2b164338e440e131692ad"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-10-08 08:05:52.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-01-07 08:53:37.000000000","tz":0},"subject":"[eventlet-removal] Handle systemctl stop for DHCP agent","message":"[eventlet-removal] Handle systemctl stop for DHCP agent\n\nCloses-Bug: #2126790\nRelated-Bug: ##2131801\nChange-Id: I6fc382c88cb9e0087b933e080ac6846a1619b1ca\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06c746da885e89311759c86fcdc3358c627024b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06c746da885e89311759c86fcdc3358c627024b9"}]},"branch":"refs/heads/master","description":"Rebase"},"d55460cab3204fc341ef5030703d1043c0d5769e":{"kind":"REWORK","_number":5,"created":"2026-01-21 18:53:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/967902/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/967902/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/967902/5"}}},"commit":{"parents":[{"commit":"762694d1bd6092060ce08c29d12575e303ae2973","subject":"Merge \"bgp: Detect BGP bridge creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/762694d1bd6092060ce08c29d12575e303ae2973"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-10-08 08:05:52.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-01-21 18:53:01.000000000","tz":60},"subject":"[eventlet-removal] Handle systemctl stop for DHCP agent","message":"[eventlet-removal] Handle systemctl stop for DHCP agent\n\nCloses-Bug: #2126790\nRelated-Bug: ##2131801\nChange-Id: I6fc382c88cb9e0087b933e080ac6846a1619b1ca\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d55460cab3204fc341ef5030703d1043c0d5769e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d55460cab3204fc341ef5030703d1043c0d5769e"}]},"branch":"refs/heads/master"},"fd40351f3de14e2d576ff6fe87e8dbe439a3d23a":{"kind":"REWORK","_number":6,"created":"2026-01-23 15:56:22.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/967902/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/967902/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/967902/6"}}},"commit":{"parents":[{"commit":"762694d1bd6092060ce08c29d12575e303ae2973","subject":"Merge \"bgp: Detect BGP bridge creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/762694d1bd6092060ce08c29d12575e303ae2973"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-10-08 08:05:52.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-01-23 15:56:21.000000000","tz":0},"subject":"[eventlet-removal] Handle systemctl stop for DHCP agent","message":"[eventlet-removal] Handle systemctl stop for DHCP agent\n\nCloses-Bug: #2126790\nRelated-Bug: ##2131801\nChange-Id: I6fc382c88cb9e0087b933e080ac6846a1619b1ca\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd40351f3de14e2d576ff6fe87e8dbe439a3d23a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd40351f3de14e2d576ff6fe87e8dbe439a3d23a"}]},"branch":"refs/heads/master"},"371ec2b4151980875e5a598caa78fd41a41fdc52":{"kind":"REWORK","_number":7,"created":"2026-01-29 09:47:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/967902/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/967902/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/967902/7"}}},"commit":{"parents":[{"commit":"7f4efcb48a0a78e909b5f71bfdf9e20bfcf06d33","subject":"Merge \"[OVN] The HA_Chassis_Group sync method limits to 5 HA_Chassis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f4efcb48a0a78e909b5f71bfdf9e20bfcf06d33"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-10-08 08:05:52.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-01-29 09:47:12.000000000","tz":60},"subject":"[eventlet-removal] Handle systemctl stop for DHCP agent","message":"[eventlet-removal] Handle systemctl stop for DHCP agent\n\nCloses-Bug: #2126790\nRelated-Bug: ##2131801\nChange-Id: I6fc382c88cb9e0087b933e080ac6846a1619b1ca\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/371ec2b4151980875e5a598caa78fd41a41fdc52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/371ec2b4151980875e5a598caa78fd41a41fdc52"}]},"branch":"refs/heads/master"},"b67a007e43786574e337c21d26e35be38f697d56":{"kind":"REWORK","_number":8,"created":"2026-02-17 10:45:02.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/967902/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/967902/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/967902/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/967902/8"}}},"commit":{"parents":[{"commit":"d209de7ffaa2e1f6ea8b036fe5303ad284ae0257","subject":"Merge \"bgp: Don\u0027t include all bridge IPs in openflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d209de7ffaa2e1f6ea8b036fe5303ad284ae0257"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-10-08 08:05:52.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-17 10:44:44.000000000","tz":60},"subject":"[eventlet-removal] Handle systemctl stop for DHCP agent","message":"[eventlet-removal] Handle systemctl stop for DHCP agent\n\nCloses-Bug: #2126790\nRelated-Bug: ##2131801\nChange-Id: I6fc382c88cb9e0087b933e080ac6846a1619b1ca\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b67a007e43786574e337c21d26e35be38f697d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b67a007e43786574e337c21d26e35be38f697d56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
