)]}'
{"id":"openstack%2Fneutron~968161","triplet_id":"openstack%2Fneutron~master~If7510e4210e7a5fe549163970e501b0cf9766058","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-11-24 16:03:51.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-25 14:02:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7510e4210e7a5fe549163970e501b0cf9766058","subject":"Centralize oslo.service patching","status":"MERGED","created":"2025-11-24 05:41:47.000000000","updated":"2025-11-25 14:03:27.000000000","submitted":"2025-11-25 14:02:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968161","meta_rev_id":"34a4aecf6c2beee387cdee7976bd6e0f93963b4f","_number":968161,"virtual_id_number":968161,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 14:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-11-24 16:03:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-11-25 09:41:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-11-24 21:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-11-25 11:51:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-24 06:44:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-24 08:00:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 08:35:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-11-24 15:15:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-11-24 16:01:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-11-24 21:02:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-11-25 09:41:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"f5c5787c4aed94ae7515f221a4896df6ba3ed079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-24 05:41:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"809922a0e3ee28e6ededdc3df4043c65b3754bd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 06:44:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e5ea248e9974ec6abd7717448e1b419\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35d4c99ed49848689fc15f1ce22a8df9 : SUCCESS in 47m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bcf2d6f6b7a0469780c5181652940513 : SUCCESS in 45m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/37378920ee6c4b8780edd3fcdaa6fe1f : SUCCESS in 1h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"323d072c089eedf26b20c6764766b44d30cb2da1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 08:00:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81d515ae378b4ae7938305cdd3f968f3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bb5169b93bcf416db35ad99521dda54a : FAILURE in 1h 19m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/61635560197a4d878ed5d285f67e40b1 : SUCCESS in 2h 17m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ee39819624944e6b8c9e42f6867b9dd : SUCCESS in 41m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb23b8725254e2cab41c6719f828901 : SUCCESS in 12m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/088e7c9dc514442ab30f7d2cc634cc76 : SUCCESS in 36m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15716c21ed2d4091b695c2bd9dfe6349 : SUCCESS in 25m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/174ed8f805ce4ad9b949ca53136ef194 : SUCCESS in 12m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/399e1c7c88994ea8b824f8ace701087f : SUCCESS in 56m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/69c93d6b4ee44cfca498336510089713 : SUCCESS in 1h 25m 30s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8c1b0423db124006b60b0494fc8599c9 : SUCCESS in 1h 50m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/19f1d041f27a4ed69dbc2ca1afdbf4e2 : SUCCESS in 43m 28s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/2045e5f177384b5b9f2dba04e5977cac : SUCCESS in 37m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f1b412f9f0864bdfb5fbd40ef5721297 : SUCCESS in 1h 46m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e98cc4900fb14e9e84897f7e7243947f : SUCCESS in 56m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f44547021e454667a4be9f28f8855027 : SUCCESS in 1h 09m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/87e55422fe284ea39a145786943dfc0c : SUCCESS in 1h 09m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1c066a87db37407c8c2cd709e7e82baf : SUCCESS in 1h 47m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d68eaa2171a343b6b3790c3a5a26d916 : SUCCESS in 1h 47m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1b5d4d3a137f36032f19ec8a0069d3a72a29c8e4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-24 08:35:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c2ca79801747296d59a8b92778b40e07b73547d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 09:37:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b39e6eb87bec4a30925f7f2a6fb06456\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bca7dd941c804c2bb8a2833cec1cfa5f : SUCCESS in 59m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bcc7d69928bf4619b8774a1406fcc0b9 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/243c12d3d1254733bd77ec0d0ae4f482 : SUCCESS in 45m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57c486bb92bc095ad7b5749163effdd96a98a4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 10:36:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a452871c256a4b8e9ae945c25c51e3ee\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3691d276e3234a44898f2aca5bb5e5fa : SUCCESS in 1h 25m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/46a0e11121734e6893e36d2ae5cfec2e : SUCCESS in 1h 58m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a326320e5745421985b2c0e0d7e996aa : SUCCESS in 1h 03m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/672001270dee41978a99b4daefeedd8f : SUCCESS in 12m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4597fb6c0b204a2ab0362d77e286b927 : SUCCESS in 23m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04b59de9669b45aba0971f768ab20a52 : SUCCESS in 26m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbd612dec0a4f128bcd7be9a6ee884c : SUCCESS in 12m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d43cfc9c2fba490aab8b0e4a0a1d95d4 : SUCCESS in 1h 01m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2e7bcc63eabb41b0bc8b27dad6618848 : SUCCESS in 35m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/61efd7612f944dcda242ec14d1d6b546 : FAILURE in 29m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/afeb918d95c44af183d7120df99d247d : SUCCESS in 1h 04m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bff96d18b224a4f8617c18f23e66ff4 : SUCCESS in 1h 05m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9605f098167745acbcc2fd7b71e3db11 : SUCCESS in 1h 10m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7d6d7ad2eedc44769139dc7b38e32cf1 : SUCCESS in 1h 34m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6abc69f077184b159093a432a5563ba2 : SUCCESS in 1h 03m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0b2fd098ebc0465e917d4cc84e1129df : SUCCESS in 1h 05m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f9e6d8dc16054bb5aa6ea56bcbe9f494 : SUCCESS in 36m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/68389f85303149ca88dcf71a5414821f : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e44b42b7a5048efe6a3d206cb9d2a916dac2aec3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-24 15:15:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47a2b3f79a2b97f148a1bcbedafdb0391365f14f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-24 15:23:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"634e0c4a7458f12cc8ec5226e80957559cfc841e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-24 16:00:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d04b92aefb0e3e27f30796bffba866fdef08933d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-24 16:00:13.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d183e63cb77bcc1faf755efade5e86344de8fe72","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-24 16:03:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"40efc0eb012a6bb12f99f0b8f41ecd02d13f8831","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:18:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc34e0081ce419ba62414b870744d41\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1770697cfe2547b082b5e10f809328e0 : SUCCESS in 25m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/97a0d65217e847328f23292416cad456 : SUCCESS in 55m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e802b8a4938246ea89b7f01f32545b0e : SUCCESS in 1h 06m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74781768a8a09d64dd8d22b65e4590ddc7ce6ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 18:29:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fde19a91918c4df5b24331c88d813735\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/71a395089f8a4b429878eb99eb49aade : SUCCESS in 1h 42m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/658b26abc7d0478d80174ebd113cb6f4 : SUCCESS in 2h 20m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9796663899da436f80dae7adc3f04aef : SUCCESS in 47m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b811183d16b549e1a3f3776bc65531b2 : SUCCESS in 14m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c94693aff7b4d90aa6874d24d889ca9 : SUCCESS in 27m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf3dcd144b4e469690331317067819cf : SUCCESS in 33m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838d2dfa4e0742b9b45c7aa99188a77a : SUCCESS in 13m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/09f55a8df9084ed5b70abd1fee40a666 : SUCCESS in 59m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e567c3f7db6b4e47b751e2bdc42fc290 : SUCCESS in 1h 10m 25s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cbd5952d842c4ba4a5abf2c1b4b85b04 : TIMED_OUT in 2h 12m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b58628dd83c64522ba165dccc0604277 : SUCCESS in 1h 08m 43s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ea06b61091f45dd89868a5cbb13c26f : SUCCESS in 1h 13m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c4d3b649705d426180dc724217fc6512 : SUCCESS in 1h 55m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bf2ae0e99ba94b75a05be82344f202e5 : SUCCESS in 1h 08m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4dcf01e972a24cccafb62b9dc93cb567 : SUCCESS in 58m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6cf861aaff6d4c06aaf177b3248b5dbf : FAILURE in 14m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e0abd5ae8d834b8d98571431d75503bc : SUCCESS in 48m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e320d032497e405eaa6037880dedbed3 : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":2},{"id":"20244c24aec025ab67d1dfcfab694186632e0c01","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-24 21:02:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"82c2c2a29576f993dbaf81b99f6fab7da8653f7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 22:42:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e57e00166044e9b360e73b4c1c2c0a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28d95ac85e65499491f846dabee2e86e : SUCCESS in 27m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1253b91c9f2a4344b1a67a059f8df82b : SUCCESS in 27m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ac39d901c2d941aab5b36f1f3ee6f500 : SUCCESS in 30m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7f4391df080c5b2c16877e9d4e7d89a953c0ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 23:15:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa9a71c97e2145dcae5b869306eefc62\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1125dd0f23ff4ec191199baeb27244f9 : SUCCESS in 1h 18m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9c7f80b9e98e4ec188e948983f6e3106 : SUCCESS in 1h 58m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61b158889bf1468894893589415d4cba : SUCCESS in 1h 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5346bbc95284b67ad3bc0c6f8413687 : SUCCESS in 9m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/912015e447a04aaca94bed7ce3856f63 : SUCCESS in 31m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b4d073636aa4ad3b8c4d511d240099f : SUCCESS in 34m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbb470109e140709a59e6379d11344d : SUCCESS in 11m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/10b0cbeaf15a49f98b4e4176b3fe1782 : SUCCESS in 44m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a0b7aa5591534816aac123877788fabc : SUCCESS in 41m 05s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fb70c9fac30640b7bd2576fc4c143d61 : SUCCESS in 1h 33m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9de43ef5f6bc4fecb65c07f43178e285 : SUCCESS in 1h 01m 33s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/80455dff2b6d4991a1f141ae1fbb493e : SUCCESS in 1h 06m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/92dcd06e6aca4c6094f5a99c797997a1 : SUCCESS in 1h 55m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/350a6be3d1194671a5891f8307c13400 : FAILURE in 1h 32m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8d855217be3d492c93053c4424b7a831 : SUCCESS in 1h 05m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1c6d2258289f4f97bc22776f51c95ab1 : SUCCESS in 45m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d57ab810af6b4ac592f7b8e20ec0211e : SUCCESS in 1h 27m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2b2f9eaa29294b01ba4c9f66f1c936a0 : SUCCESS in 1h 33m 39s","accounts_in_message":[],"_revision_number":2},{"id":"3ca53ddb51852ec46d77a62684caa069b532521b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-11-25 09:41:27.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bdab7f4d25719b6f7ce2b82388887c5db11a948e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 10:17:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff06a71cc244411b5072b2d17263dac\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/083d2ad9f4ec4a17b6befe0f41790075 : SUCCESS in 30m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fea98cbda4034f3988bf1c03c79834a8 : SUCCESS in 28m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8ac9ace4718e4a8689ccc549a7f8ab93 : SUCCESS in 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a08d67ba695e90aabfb57a056f3aa65e20a6e104","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-25 11:51:46.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"276248b84b1a98c4e8f61308593283e3e21e6e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:57:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acc23f76d2ac4514a66c4ec7d58e42de\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b57c1b48fe82475ba153c7ff509bf62e : SUCCESS in 1h 23m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/78d83ac64a3441bda8bed78ec6afa426 : SUCCESS in 2h 15m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40d4098668e04f2484fc94b65bf0b682 : SUCCESS in 56m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d39611f05342b4994bfafa95da1590 : SUCCESS in 13m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6357e1daee64d558f18caf2cf02670b : SUCCESS in 21m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ab2591e597d4000b792d44c7c045266 : SUCCESS in 36m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b95be12950543a9b283b20a51cb0cf3 : SUCCESS in 6m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4b6c86de747a4cb8af1dee72053abc2e : SUCCESS in 55m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bf25853ef11747aa962ec76419471e8f : SUCCESS in 1h 09m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b8786f55199448e6bbec007625f6a7ee : SUCCESS in 1h 14m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0eb8cd880a8b4088b377fec62c9bf3b5 : SUCCESS in 1h 10m 34s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/041e11c0b7cc4e33a7829b7612cb66a5 : SUCCESS in 43m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b3225d0c9cda4b29afb844fd800be87f : SUCCESS in 1h 46m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/86a17fa7b8cb4d03a21506994f22c863 : SUCCESS in 53m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/098067321d564753a8b6519f24e92d65 : SUCCESS in 59m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2dcdd6be34064402a74624b81f772234 : SUCCESS in 1h 03m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aad3207ec2074d7c964aecf61abfe198 : SUCCESS in 37m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bc906d7c2a1846b89f1caaeec27ec04e : SUCCESS in 1h 21m 52s","accounts_in_message":[],"_revision_number":2},{"id":"78ee6df88a605ed24e5a9f2681c1747c2da816de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:58:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8c4b0e2fd795bcb7ba97e595f5a9d57ab89bbf68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:02:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcf82014380a43f7816ee14cc970e3c6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6288507299ff4e308231d99e335e7a0d : SUCCESS in 1h 17m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/abc1e1847eff4395b1daafd76b1e139c : SUCCESS in 2h 01m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa286f712a54651b8f9cee9ebb5006b : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/328b387e7e7f4776b4d7ae5909a06c59 : SUCCESS in 26m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bed76a6fa372435ca181e5254d183076 : SUCCESS in 26m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21a7a34899940a8b62dd972011bbfac : SUCCESS in 12m 32s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/0e88d55124c342949247dc33aaee7681 : SUCCESS in 1h 48m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/374022562bb644c2aa7cccb161f6085f : SUCCESS in 1h 11m 13s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/70a6fa8b329c400d909683bb2bf8e5c0 : SUCCESS in 1h 06m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/258a134a84e04d0bb4fe35b5c1690cb4 : SUCCESS in 1h 05m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/47e2f3e759714c56b30c53329c15f957 : SUCCESS in 55m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6f4a1ff872d946b8ab0c240ac3512301 : SUCCESS in 1h 11m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a8e7f73e3070ade14d2fd0d1ac078e3916c75297","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:02:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"34a4aecf6c2beee387cdee7976bd6e0f93963b4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 14:03:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89b97ce310f2429d9ab7563320e6e014\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/253794d54c8c4891b5f49f88dd044b46 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cdac238f9a2d350ecac39a74f8a089d4072e83b4","revisions":{"5bfe3c53b5ee1b180bba5421a0866bf0f2e4518d":{"kind":"REWORK","_number":1,"created":"2025-11-24 05:41:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/968161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/968161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/968161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/968161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/968161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/968161/1"}}},"commit":{"parents":[{"commit":"cf41465164e834cf5e24033ab24628cecd8d9647","subject":"Use common constant for neutron-ovn-agent process name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf41465164e834cf5e24033ab24628cecd8d9647"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-24 05:39:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-24 05:39:36.000000000","tz":540},"subject":"Centralize oslo.service patching","message":"Centralize oslo.service patching\n\n... to reduce number of noqa: E402 in each entrypoint files.\n\nChange-Id: If7510e4210e7a5fe549163970e501b0cf9766058\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bfe3c53b5ee1b180bba5421a0866bf0f2e4518d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5bfe3c53b5ee1b180bba5421a0866bf0f2e4518d"}]},"branch":"refs/heads/master"},"cdac238f9a2d350ecac39a74f8a089d4072e83b4":{"kind":"REWORK","_number":2,"created":"2025-11-24 16:00:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/968161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/968161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/968161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/968161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/968161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/968161/2"}}},"commit":{"parents":[{"commit":"e46ca37684689dc0e8e291e9b4d43b1992ba9d52","subject":"Merge \"Update IPv6 gateway in OVN after subnet gw changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e46ca37684689dc0e8e291e9b4d43b1992ba9d52"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-24 05:39:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-24 15:59:50.000000000","tz":540},"subject":"Centralize oslo.service patching","message":"Centralize oslo.service patching\n\nMove the patching logic to __init__ which is loaded automatically.\nThis allows us to reduce number of noqa: E402 in each entrypoint\nfiles.\n\nAlso drop override of OSKEN_HUB_TYPE, because os-ken uses native\nhub type by default since 4.0.0 .\n\nChange-Id: If7510e4210e7a5fe549163970e501b0cf9766058\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cdac238f9a2d350ecac39a74f8a089d4072e83b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cdac238f9a2d350ecac39a74f8a089d4072e83b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
