)]}'
{"id":"openstack%2Fneutron~968211","triplet_id":"openstack%2Fneutron~master~Id895c01dd9342a28df99055b0dc2fde9b5e70b1d","project":"openstack/neutron","branch":"master","topic":"bug/2131994","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-11-26 07:12:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id895c01dd9342a28df99055b0dc2fde9b5e70b1d","subject":"Bump ovsdbapp to version 2.14.1","status":"MERGED","created":"2025-11-24 16:02:25.000000000","updated":"2025-11-26 07:13:22.000000000","submitted":"2025-11-26 07:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968211-bug/2131994","meta_rev_id":"f7b86bc6bd48bd5e0a9d184e4ae970df8a1e29af","_number":968211,"virtual_id_number":968211,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-26 07:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-11-26 05:49:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2025-11-25 22:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-11-26 05:49:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-24 17:00:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-24 17:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 20:53:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-11-25 22:49:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-11-26 05:49:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"377a616099726bd1e765122354c183a9c9c91ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-24 16:02:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c53c3f48016052786ff4da9faafe4847b7e77af6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-24 16:02:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d1c846c8eaa69554e04cbad7debc7b224a417dd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:00:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5bc7eb0d2614659886db374dbd4310c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6fe7fa790536410c9b1e5b6ec501e608 : FAILURE in 5m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f377451000448268b8e925eb7e57782 : FAILURE in 4m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9481b1f2299f4685bb649514558c5865 : FAILURE in 8m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9759659518d031bb013063e31227833d2e9e7f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:16:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f97cfd1c839f4f8aa357aaaf5e533eec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a9b404662154d5bae9299ba929d99c2 : FAILURE in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4111435759804938b25890c34d031847 : FAILURE in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abc7be6d1c3544878c9f1821ccbc2160 : FAILURE in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d6b0be6e0214a33a6fc34d774e81851 : FAILURE in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d1b38bab08a4e9fa57019d1d097f3c7 : FAILURE in 5m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8a542fe798134d63b607dfe486131966 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b897d50f0a24b18b6d8f394e8aa2839 : SUCCESS in 25m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8f7317dd5b79465da97d29b393340e99 : FAILURE in 46m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b48861fd51214bf489e97b1abc21440a : FAILURE in 1h 00m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9bb5519fc0034b10b1f265af6ff11710 : FAILURE in 21m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/117f62403cbf4c1e8cc1fa09779e21a5 : FAILURE in 58m 46s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/77bbb6555d4146049faa6b1091aa6971 : FAILURE in 51m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f26b30c8b02845148ce49da3bc2a9877 : FAILURE in 17m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9da23f12a32c481c9526c8b10f83e5fa : FAILURE in 20m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4ec3b140e0e44dbc8c2e9e0c0da24fb4 : FAILURE in 13m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9f0bfa815705476db07b0f2c6c8d3f37 : FAILURE in 17m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d34bf5e8387444ba81a119ba2d7bbe61 : FAILURE in 9m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c6f64c50ec7a414f9a8dd6f47308fe22 : FAILURE in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3f18e3429ef84b4ab8ef93bef2d8850dc6def4c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-25 22:42:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"24dd6c62796e7a8e47be6f86389bc5a9fe98ca31","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-25 22:49:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d2aa438dacae234c7450908795e17af41b450d12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 23:49:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/383b93e3e4c143a19b0b2681a46b07b6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81e6313267794743842f9070481d9215 : SUCCESS in 1h 04m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c372acbeed5947efa7a1f604836f1a42 : SUCCESS in 23m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1664f45a7bfd40ea8c571cfb80464b44 : SUCCESS in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c1cde94b99f73b3f592eb8b49c744ea48f6c3086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 00:43:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0bdcfc7a58649078167a025ec6af72b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/027ba5763a684a7c965bab500e501e4b : SUCCESS in 42m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a986d515f44c4010888b310e28cec69e : SUCCESS in 12m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2aad7b9208374247bc1dca9b9c87ce46 : SUCCESS in 33m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19ffca3908494516b9a76f09e4cba919 : SUCCESS in 37m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2e0c54092434aad9accb3ec0d6245d0 : SUCCESS in 6m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/85fac20096ef419ca2c7acbadc5235aa : SUCCESS in 2m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e566c8eaa8e4f22a933619fedbb8988 : SUCCESS in 18m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/960d6b3ec6d842af808f1f34ef669b54 : SUCCESS in 1h 05m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f76237bcbd8745f88fc1f83e5f5adced : SUCCESS in 41m 04s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1fe50be4911a49228367904b85c7d69a : SUCCESS in 1h 55m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7d85897e02d4142bbdb8ae821a6da3d : SUCCESS in 1h 10m 56s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/da8ec3388dda48ac910bff5f64739407 : SUCCESS in 1h 15m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5d3a70f87bfa44b588e9be272840f918 : SUCCESS in 1h 05m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/feeeadabc54c4524999f3120731c0a69 : SUCCESS in 1h 37m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b2bb6764259d4148bb9ba136b7de096a : SUCCESS in 56m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ad1106dbc9df4fb0add9ef9637386760 : SUCCESS in 35m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/03a88447f11f46a886b809bb50aea3b3 : SUCCESS in 1h 01m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d0c6a04e1ed8493aacc7dd67b2df472f : SUCCESS in 1h 25m 15s","accounts_in_message":[],"_revision_number":2},{"id":"a5627a5a211bd7344f23e92bd2910e69b8248138","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-26 05:49:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b1dd8d4a2a574053533cc42f6d1063ff25e5678","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 05:50:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c4fb773fbb9d59c24258fe675343cdab3ab1d831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 07:12:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00c1fd74afaa4fb3ba082ec1bb37cfe0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72d661cc2db45cd98f2f10f08a24935 : SUCCESS in 11m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52adb0dd341a4e40b38b89a532d97f55 : SUCCESS in 28m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea35318c53974384bcbc9fdeb01987e2 : SUCCESS in 34m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc71aadbe2a4179abc108521c8ad787 : SUCCESS in 9m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b99849dedae54c1fb3ff8c79117ce92c : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/996a3cd840ec4fad85ad532aba466f8e : SUCCESS in 25m 48s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5781fae4b522461a97903a82240d762e : SUCCESS in 1h 10m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4421c435109747c0add446b1c9618dd7 : SUCCESS in 1h 07m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8e513260adb4ddcb9a8f62ef17769fb : SUCCESS in 52m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/957a95e63e0e4c5a8545dd6e52407b31 : SUCCESS in 1h 04m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6ff0b429b65d4bef831a71ca06b84034 : SUCCESS in 1h 12m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9085664fb2814a6e822e9cbab368c007 : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":2},{"id":"462e509ec2ee00f346e7facbb79c7389ea31d54f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 07:12:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f7b86bc6bd48bd5e0a9d184e4ae970df8a1e29af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 07:13:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ede8212967b44b7919e8150ca32fc73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bdf9d9766d3419ca84d4e9c09515daf : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60a433008d284fa78fb9564f323d9642 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2d6eaa61cbf7a5ef54f692ceafe8ecce5932cf5","revisions":{"60914349c3a640757753bba5b75aa2fbcc619ebc":{"kind":"REWORK","_number":1,"created":"2025-11-24 16:02:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/968211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/968211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/968211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/968211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/968211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/968211/1"}}},"commit":{"parents":[{"commit":"e46ca37684689dc0e8e291e9b4d43b1992ba9d52","subject":"Merge \"Update IPv6 gateway in OVN after subnet gw changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e46ca37684689dc0e8e291e9b4d43b1992ba9d52"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-24 16:01:21.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-24 16:01:21.000000000","tz":60},"subject":"Bump ovsdbapp to version 2.14.1","message":"Bump ovsdbapp to version 2.14.1\n\nThis patch also removes the workaround needed for ovsdbapp 2.14.0\nrelease in the functional test job.\n\nRelated-Bug: #2131994\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Id895c01dd9342a28df99055b0dc2fde9b5e70b1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60914349c3a640757753bba5b75aa2fbcc619ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/60914349c3a640757753bba5b75aa2fbcc619ebc"}]},"branch":"refs/heads/master"},"f2d6eaa61cbf7a5ef54f692ceafe8ecce5932cf5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-25 22:42:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/11/968211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/11/968211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/968211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/968211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/11/968211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/11/968211/2"}}},"commit":{"parents":[{"commit":"2ade46584b890c7780460eafda4ae15b17e90d31","subject":"Merge \"Centralize oslo.service patching\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ade46584b890c7780460eafda4ae15b17e90d31"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-11-24 16:01:21.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-11-25 22:42:01.000000000","tz":0},"subject":"Bump ovsdbapp to version 2.14.1","message":"Bump ovsdbapp to version 2.14.1\n\nThis patch also removes the workaround needed for ovsdbapp 2.14.0\nrelease in the functional test job.\n\nRelated-Bug: #2131994\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Id895c01dd9342a28df99055b0dc2fde9b5e70b1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2d6eaa61cbf7a5ef54f692ceafe8ecce5932cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2d6eaa61cbf7a5ef54f692ceafe8ecce5932cf5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
